)]}'
{"id":"openstack%2Foslo.messaging~680629","triplet_id":"openstack%2Foslo.messaging~master~I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac","project":"openstack/oslo.messaging","branch":"master","topic":"nats-python","hashtags":[],"change_id":"I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac","subject":"WIP: Add NATS driver","status":"ABANDONED","created":"2019-09-06 08:57:13.000000000","updated":"2022-10-31 16:01:53.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ccc18d606af545736587b1fae1252f98942fbe1f","_number":680629,"virtual_id_number":680629,"owner":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"date":"2019-10-31 03:55:29.000000000","_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-25 07:24:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":-1,"date":"2019-10-28 16:01:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":-1,"date":"2019-10-28 15:44:10.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-01-26 15:32:27.000000000","_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 07:24:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-28 15:44:10.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-10-28 16:01:34.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-01-26 15:32:27.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2022-08-31 14:06:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-10-31 16:01:53.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"}],"messages":[{"id":"e1c0c62bfea30ecf6bd0932b255e0df2b365208f","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-09-06 08:57:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f840df9b503a78f09ca8d39de7292e80ac42eb6","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-09-06 08:59:02.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e4e6e78c7fa52931028c17ac5db31e1e5dd9a569","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-09-06 09:00:10.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d00843516328f117f65f7776c204bace9840d2de","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-09-06 09:02:07.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f780c1d9e59f206317bfbec6a50397376424e66d","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-09-06 09:04:21.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"a3d53018ca8d37fee3bfbbcabfb710734ebf3c9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 11:11:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/180d6fe096b4459a9147e223890610bc : SUCCESS in 2m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ca46396ada8144d68041dd3f7c3a041c : SUCCESS in 1h 36m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/803e884c8b8249b1b4fcf74342f93fc7 : SUCCESS in 1h 49m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f978d31b3fac4d7d932eef835a71cfb2 : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7785f787b5245c8843066792f40a678 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b921959dfe74f50ae9519bc071d8d01 : FAILURE in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b9ddedaaa9b4f8ca5423cdf2094d327 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1eab290115204dcc855eecff5d86717d : SUCCESS in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70ea3c3820a04676816b1cecae632a00 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c800271de754bd9aea62fb9f1aeefdf : SUCCESS in 3m 59s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/a821141dd5774da591b9dba200434830 : SUCCESS in 6m 39s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/49641ecf838d43509435f23183e35776 : FAILURE in 12m 55s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/cfb56307e4534677b5e0ef33429d8967 : FAILURE in 8m 02s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/79648a55f12d4c848a6cf61b83004982 : SUCCESS in 5m 58s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/43c496d497c749e7a055a88d0c04d13f : FAILURE in 11m 54s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/6d3e78ecf8e946cd84f4314f4b15683c : SUCCESS in 5m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6fd8c7d018a04884863dfa3389561feb : SUCCESS in 1h 46m 03s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e97500997d1d47f1a7045b0593b4d4a7 : SUCCESS in 1h 42m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/0be9f01b8af5424b9b0eda4eba3f1d6f : FAILURE in 1h 56m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/33a04d19d2c141608746cf907a9f4afd : SUCCESS in 1h 43m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/e375ae2c46f648e7a5c8dd5880099ad6 : SUCCESS in 1h 57m 08s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/aa837cfc0fd04e65a30aab7f2a6afc77 : SUCCESS in 1h 04m 36s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/0b7fcc1cff454df39978a95ba8f9b568 : SUCCESS in 1h 28m 26s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/bffa429708e249ba838ba81d0071b689 : FAILURE in 52m 54s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/d7c9317ca05e4e92aa7e3cf39c2d8ce2 : FAILURE in 50m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/3273e47dbe564b189d6ee2cf6eca8f6f : FAILURE in 45m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"359c825e88872306a76310a6c5f36de6c9812c60","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-09-15 12:52:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e5386df628334dec8bc17b626eaadd293d2847d8","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-09-15 12:53:42.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"c45b58d133b8454c20031d77a79f197533de3ffc","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-09-15 12:54:56.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"4dec2cb974837ba43440bd882bc2ea94629242b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 14:59:02.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e12c46221f2d4be28a5bdbd70055faec : SUCCESS in 1m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a9efb7e2c8004061aa20776e302b2aef : SUCCESS in 1h 21m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff873ef7a3ca4bfd8b91b0f138bc1fea : SUCCESS in 1h 28m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b125ae17674149cc8f72165dd4554f1a : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10b37a16edf8479189f1a562cb3dee76 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f363664460d14ee295d1b33e639257f3 : FAILURE in 2m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bd380fc32354996b7838cf4460ccf58 : FAILURE in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2725de09a688480385677a09a4a0ed21 : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1a067fd29054f8abb76bf127c7f5581 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0803a423b87747fb8a469fd4bcedbbfa : FAILURE in 3m 10s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/ca801437e10a4745a417c8ce0f084e90 : SUCCESS in 5m 51s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/0e9ed2432c1249e1846c2f39659d9db7 : FAILURE in 11m 38s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/ef3c9e188a1d433cb692e4fdf1e1ff3a : SUCCESS in 5m 25s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/3956c2ccdf2e456fbf267adccf686f57 : SUCCESS in 4m 38s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/12d8153c7f6f4957a274a22caf54fff3 : FAILURE in 10m 47s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/4f945d328a7c4ab58909beaae8f01d26 : SUCCESS in 5m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/48039abd39ba498c88ba6d4955250701 : FAILURE in 1h 12m 09s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6c8d43d69fdb4d0580632cccc8a6c04f : SUCCESS in 1h 42m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/49e84c2981594e399fa2d0d5491e103e : SUCCESS in 1h 58m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a5d3bac72d614daa9b8dcf77bb7e0fc2 : SUCCESS in 1h 31m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/522c372bb71c4680ba864edf58755c21 : SUCCESS in 1h 39m 12s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f56cb04db8ef4adeb21399f13a674074 : SUCCESS in 1h 05m 06s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/0c06d9084e6c4e89a16b36a978d354ed : SUCCESS in 1h 08m 45s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/ca39fcc2998549e6b1a185d718740ed4 : FAILURE in 49m 12s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/102faa2e28604360a702b66d14de4f96 : FAILURE in 39m 35s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/3f4417452e8245a28ceda7f74f86cfcd : FAILURE in 38m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"abee9a8f5c85994786732339d30170473a64ebf3","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-09-21 14:29:59.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b8026fd88be95d55f3383495182dfb497b7ea4ea","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-09-21 14:35:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e808cdfbd700e9341f500106c6f0ff58f3ef512e","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-09-21 14:39:31.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"bfdf1f95b5df3193ef659d3d9f2e43f4d3656c0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 16:16:47.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c4ce5aab2f5641e1995d00fb606abb4b : SUCCESS in 1m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d3b6bc590d6a4d6d8c794f57ce40dab6 : FAILURE in 4m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7c8524241ac2472cb5c9c4361da4844e : FAILURE in 15m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67a8cebce4204eb88539e54d972a4463 : FAILURE in 16m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93d28f2a4a134a7095f378407ea3973c : FAILURE in 4m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6a0cddff9004290af12e21605ae90c0 : FAILURE in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b461cf86ce34579b4f1d44080002880 : FAILURE in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/906ff8bf127c4770ac39f821603f48a5 : FAILURE in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/461a4e8ba8074658be783b3062085957 : FAILURE in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f9b50d8c8024b71aaf7acd62e019b92 : FAILURE in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f652fef2e76f44b489284ae45a1118dc : FAILURE in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4b70501174349e0bcf13a8624bebb68 : SUCCESS in 3m 54s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/a6159d4d491d400ebd45adaa219501f4 : FAILURE in 3m 44s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/f83e03f80f1649afadf10b21bf0a9dc6 : FAILURE in 4m 33s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/e62766d8016d403d832681af13a7a14f : FAILURE in 3m 48s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/062d63f5466644c7ae6fffefe44a2ab7 : FAILURE in 5m 12s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/b3b7b1063117479fa15576c35c027817 : FAILURE in 3m 40s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/c0aabd149ef3453699cf8b8202663812 : FAILURE in 6m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4867c9a95b7e41cc9f55ded84e2a4e99 : FAILURE in 13m 19s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/b8ad74c719044c2b9878a7541f7175e1 : FAILURE in 16m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/d8f657d71b68488895c4dcfe5667a325 : FAILURE in 14m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2870f690dead436a8557a665238f0569 : FAILURE in 15m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/dbef2a740fbf4936a0ea337cd35eb645 : FAILURE in 20m 09s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/9358d270ace24130983fdf3bafc49253 : SUCCESS in 1h 06m 28s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/71a0e2a35f614ab38470ea5ff2a380ff : SUCCESS in 1h 29m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/502c330bf28446ed9f6a35c7a9a0a9ac : FAILURE in 16m 29s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/9b9621d71bda456d939dacd5ccfdc8a9 : FAILURE in 16m 19s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/9dff7aa611d7427d9703a948b68927d5 : FAILURE in 15m 39s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e8d8403e5c30f21e2de465df275e51bafc27c041","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-09-21 23:38:07.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"7d674423d6a94fd0a39816f8c2bbfc0dc5955e7f","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-09-22 00:25:48.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"95f1a3a35264ca1dcce3277d28dc74ddc36ba5ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-22 01:46:48.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/44496bb8923f4d4395050419134146b2 : SUCCESS in 2m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6812ac30a1fb4304b5653a0efe87f3be : FAILURE in 4m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/aa7baa09dd0940b2a4ea7deeb6f12a3a : FAILURE in 13m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98396f07c59e4cf687dfddf54d4a11f1 : FAILURE in 10m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1d65ec111514a1abb1fa87c56aaa911 : FAILURE in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0e6d1c0d9204e928c0d27971a0bae33 : FAILURE in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/425c9e224928493eaf176460b62e1e6f : FAILURE in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fcec6f7ed51845b698634d348a46ddf3 : FAILURE in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3713a9d45cfe429c8a0cb5e4846d5ff9 : FAILURE in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5222a37587f54e0f8d779e02a5aed7b3 : FAILURE in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cff2759224044f888c7d4549de2f99f6 : FAILURE in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4837d6f0f424c5da4e49e840b3dbc78 : SUCCESS in 3m 46s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/c0b1f6c5276642878512704a57d8a727 : FAILURE in 4m 25s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/73f61effb627446d944676a87f77fc95 : FAILURE in 4m 39s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/b44e807028674fa3b9e893cdc670c7fe : FAILURE in 3m 57s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/6f70e551ed554e488afb71618bab40eb : FAILURE in 5m 41s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/004b6088a2e3443e831fb535d9fb3d1b : FAILURE in 3m 47s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/290c51b97a18421a90b7a66a3ad9565c : FAILURE in 4m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/81266becdfb1460fa6b54155e1eb1bff : FAILURE in 12m 21s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/b85f06f261ba4b0985cb89b51cb4f2b8 : FAILURE in 12m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/02420efe3d1141bfbefdbcb18848a745 : FAILURE in 11m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/999bc0c001914611b59170f36839679a : FAILURE in 11m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/dcd1166fc54d4fa6b282db72c2b3d453 : FAILURE in 16m 01s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/dd60f55a6b0940eca3fcbe2fb1194f8d : SUCCESS in 53m 49s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/edb8dc1998574950bbdc4fe611388594 : SUCCESS in 1h 19m 25s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/e4af81017b9e4d93925f20a03d130d6a : FAILURE in 11m 30s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/cd15a388ca0b4d78a473c5d17f8d7902 : FAILURE in 14m 25s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/eae7245fe8db453f81ebfa808fbca120 : FAILURE in 15m 46s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"26e080193277fca4ab5892e8cc070dc9f95df628","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-08 14:25:01.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"51b570565a92bcf2d7a707b73392e6406557517f","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-08 14:28:00.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"113bfd96cf25898cc770290200b76ef7321dc238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 15:57:01.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fe546526641b40ea95ea6b4add76945b : SUCCESS in 2m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/35bc7efa228a43ad8369c711f47bd39b : FAILURE in 5m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cc056f6ac96b4eb29cd1158d9378ee9e : FAILURE in 15m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cdf97d57710e45f0b52387860cdcf811 : FAILURE in 11m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0a8c27fcd9e438e94d914ccd5845b63 : FAILURE in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/934c02cd5de64868a57b92352d5e90a7 : FAILURE in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13bbea37dafc4ddf80f3ce70d621d230 : FAILURE in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b913cf6c1124c5caa59823022c3afe7 : FAILURE in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc8939cc44bb4698bb3c027dbbc021c0 : FAILURE in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3bc9fdeb916444bbd449a3cc5c1bae4 : FAILURE in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2dff65b0f614f328bf1b269c152ec7e : FAILURE in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60cd99307acc457f8e0732204c57c7c5 : SUCCESS in 4m 44s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/28d6bcb8bc61427dae5f21b9e106fae6 : FAILURE in 4m 12s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/5e5637ced87d4c13906290ca379ca4d8 : FAILURE in 4m 13s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/bd147747c63947c18d7eb2aec8971f23 : FAILURE in 4m 19s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/6bff6bd2dad44597a736510094ffcf80 : FAILURE in 3m 44s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/1af8e9f05c5d459aa708070aafcab949 : FAILURE in 7m 55s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/4bdcc32e74384ea087034bdbbab6fbe9 : FAILURE in 5m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4d304cf76a1f4fe78e667cf190bf6775 : FAILURE in 22m 39s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/61945e158ed64a7e927271bbabe479a6 : FAILURE in 18m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/8ce4842df22e4277a06b222dd4476ca6 : FAILURE in 19m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/61fba37e10964ad692ae96b0d36ef40b : FAILURE in 20m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/458ac2cb099348158b62c4b77c20ec25 : FAILURE in 13m 59s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/ec814e86899a465f824d71fd0382e2c3 : SUCCESS in 55m 46s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/d3f24d05e5dd43419dfb97fe31694b77 : SUCCESS in 1h 12m 37s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/07cc49be64054384bbb0892ecac08dac : FAILURE in 12m 49s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/6e69c40954904eb5b651fb208558da7e : FAILURE in 30m 11s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/e417024f3b104e57b9d299e2f1a6d3f0 : FAILURE in 22m 01s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"997ef1ef53ba0687a19aada4b2d9430a3cc93899","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-11 04:19:29.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"3708274937a952d39b49b63588506c9f42d9f215","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-11 04:38:29.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"d6c2dfcbb23f334bef4b2ba4017d8d164db44ea4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 06:51:36.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6f90c5b3d18f430f98267be63c7ab427 : SUCCESS in 1m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/25ca0e952b2040ce87b7cff44a86a69b : FAILURE in 3m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/753fab02f9ec4c189a881eebc02c5916 : SUCCESS in 1h 40m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/963b81eed58d46a6a6df24b7a2c81de2 : SUCCESS in 1h 22m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3384a2b967d442a9db14d90ed1146df : FAILURE in 4m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c076f0a4b7d34297a47af94a286bfab6 : FAILURE in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c73683b41545e286b4c1f0194e2e51 : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2839f58956344c7d8d60f3185a4bab56 : FAILURE in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7206feb8f3547e7a4c559336da8df53 : FAILURE in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bd01d409b894060af0f237935b672aa : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4207794df0c495f91ad3cd9650e29eb : FAILURE in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2154203e29ba43f6b267ddee66629dd2 : SUCCESS in 3m 51s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/55cc9f2c2f4d44ca9de4185e75a4483a : FAILURE in 4m 32s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/61b1f885c362497c9836844df1bfd97b : FAILURE in 4m 38s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/f149b3bd316d4736a04abe7c0eac6d74 : FAILURE in 5m 22s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/533d7336e2754d41a0b623e5fe89e40d : FAILURE in 4m 32s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/63ea800b5f4f4ab0b86f29b2cbc254ce : FAILURE in 4m 18s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/8a30a76299f84faabb3a910577c9a06c : FAILURE in 4m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0f207941ab4043d88381b6a124350a5e : SUCCESS in 1h 29m 50s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/46baef5660fa4ac6b1dafcfd400d2207 : SUCCESS in 1h 32m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/61750429232d44238fe6f2bd23ec0539 : SUCCESS in 2h 02m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f601fa1a5dcc426eaef2b97fe669b1d6 : SUCCESS in 1h 21m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/6c7d4683e8d141cfad3bd5a973e6af44 : SUCCESS in 2h 11m 34s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/775ebd979576441595b5a533f7779eb3 : SUCCESS in 1h 02m 18s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/d9c3ce28e0164057808e09f5ccdbe15c : SUCCESS in 1h 09m 32s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/7340bb61347940499f07da155075a4cc : FAILURE in 56m 16s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/ea1a6d662ccf4b00b72b395c31b0182e : FAILURE in 42m 14s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/3fa4ca08533140f3a1a8b6006594ba15 : FAILURE in 49m 35s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"88a46e0c327fffd6d3daf18997169820a4a65e1f","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-12 09:38:39.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"98c2f3fc9441378e77df9b44c3b08d1acb09c42c","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-12 10:41:43.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"d1512a1fccab41c296a23c8d9eb5872dfb7bc07d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 13:18:00.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b1f6422b62484695942c4648e7f16fe3 : SUCCESS in 2m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b26f2cc9fb56483494240f167281a0ff : FAILURE in 3m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f80013ad0eb04ee3b424465933f15e7c : FAILURE in 1h 20m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d72d92fcbba490899a97ab96d33951a : SUCCESS in 1h 19m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a452b8ddb6e842d6ab0463169f6341d8 : FAILURE in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f447661e43d342fe98cb739a5cfca394 : FAILURE in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3de2b7465e49c58f1ef5017eb9477a : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ef2df3aa77b4e849a2d7b531da01a19 : FAILURE in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d4f06f862b540f0a65da2d88ad74002 : FAILURE in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee2e410fac574d1bb6f63fdf4da6d828 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c2b445a76644b26a2e8a522764adf77 : FAILURE in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c75f7bc45be04acfaf13249661986423 : SUCCESS in 3m 31s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/578a452a80eb477c8520bce22722d3e8 : FAILURE in 4m 28s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/0a264720aa0c4db7b2971f8058f96e23 : FAILURE in 5m 53s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/473b166340724c6888e2f1adec04bc02 : FAILURE in 7m 18s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/a796cdaef36d4e8c94eb3e397be1cf3e : FAILURE in 5m 50s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/72cd4916e72e48eba0b6c9d1d09cb202 : FAILURE in 6m 35s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/6dcfd25044c04d3c8ec21fccda72372f : FAILURE in 7m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b2b3f271f70e43d0808fd1e5d2a83fe6 : SUCCESS in 2h 10m 05s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a3596c09c8964e519d6fe45d213e2171 : SUCCESS in 1h 59m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/b3ee067deb0349b2afbd3fc293938225 : SUCCESS in 2h 17m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ecd53938579b4575bcbf2df393b4b35f : SUCCESS in 2h 10m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/d43c3d01716d4c4e998c8ebdffea8b2b : SUCCESS in 2h 26m 29s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/a6946d5561314608ad05366cf29fe96e : SUCCESS in 56m 11s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/52435f934c754978bc8aa2fa84efef68 : SUCCESS in 1h 15m 32s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/2332ab07d21d485ea65b31d0f85a64cc : FAILURE in 44m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/f8ac8b775c104810b28a961d7e71a880 : FAILURE in 1h 02m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/42af762de19b444897385f0ba28371b2 : FAILURE in 48m 01s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"8b79330733bb7fe0f3c36a1dce2fb40417190377","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-12 15:52:06.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"2df9fde83c2094036fad217769b79cc337a6c58d","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-13 03:23:22.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"4aa5e52bfa69861e776d1c6c8cbb8c008aba7527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 04:50:59.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ffe6832a9bfa4ee88692ca59148b9a35 : SUCCESS in 3m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/64b209a940854c7e94bc95b4729b2c32 : FAILURE in 3m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/37e71357e3344774aae7e0b1f6935edd : FAILURE in 19m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87d8fc1634a04a1ca1d7b34894a14989 : SUCCESS in 1h 26m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cf80777fae44ff6919fdde693fb6c0a : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c82c7da76864cf7b9d2972dc765475d : FAILURE in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9187e5c6efc14190acb5f55bd327ffeb : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ff21fd97b6848d995bd3548f37b74a6 : FAILURE in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8675915d1a1541bfb8ecc6ff1c15fa6d : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4552fd88434f41cebf23a5e5502e7012 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cd640fba6cd4274a24825a54617f96e : FAILURE in 2m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba4c46f3580143a891ef7aa8b89994b8 : SUCCESS in 4m 01s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/30e0dab6ea92469981477683a2207eae : FAILURE in 6m 00s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/b2865f66d54d427b94bc46d308d16a4b : FAILURE in 3m 51s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/4baceccf981a4bdaa3589b9843b413f2 : FAILURE in 4m 50s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/6e8add33a3e54d1181b6544a4a395e5d : SUCCESS in 6m 57s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/7a8a5a32fdc2498a82cb510a6a42d0fc : FAILURE in 11m 01s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/8f234f4bb55342af9c395b47002a381f : SUCCESS in 5m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c9caee0fd6f2448a9a6b0bc6c9f8087e : FAILURE in 18m 14s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/dcb54559826d47e4aad67d65448d6e89 : FAILURE in 16m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/625dd0c0d15e40f0b2df8f603ec8d892 : FAILURE in 23m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/4ae020170951470896c8f3610e236a45 : FAILURE in 17m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/27a01d581cfc4354b17a013575d2fbd3 : FAILURE in 26m 20s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f38b8a44128b4ea582a680f50d6a4c38 : FAILURE in 40m 44s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/17ba36cd3e3941fa9e55ea3821a13668 : FAILURE in 59m 10s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/5a8f2a8d6ed349688ea692e85e8628c6 : FAILURE in 20m 19s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/a42c676e199941368fac2cf45354fb9a : FAILURE in 21m 17s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/6ac067bdbb91443fa7c96a9fd7b28ea1 : FAILURE in 21m 07s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"da75136b9309545d7f4e9ecfb0db0f680026d743","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-15 06:47:10.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"91cd462977e8598f741b2ce028eeb63cc3584ef8","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-15 07:17:56.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"55a527c7a4081aa576202ede50f8c1ad8dde5a94","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-15 07:32:19.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"cfd1db78eee25835ad41a9fc5ad55c133b39febc","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-15 07:47:23.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"0849bf0705996fde4b8ab201400db02c1f420e33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 09:31:06.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4382b68a2b724f42b74cbe24b6c3fef0 : SUCCESS in 3m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4ae6b1fbb378480097933c59a027337b : FAILURE in 3m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4dad5e2a8a0b476c94b7a3ccf5135b29 : FAILURE in 24m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aea4d2cf6d594011b480114d8ef82c0d : SUCCESS in 1h 37m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/055d356c20714c448d34d3f5a768761c : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e864dbddb7d64c6d83747b48aaa5d556 : FAILURE in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9ef13c60d334a48922509e9dbd400be : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e87e21abf624a369b6436469a742d99 : TIMED_OUT in 41m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4b6d868aa8b4f6096e904a50a807108 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32e8d18429a84c9da4a0518f75ae4736 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6f1bf01984f4d8cab40f19923da4623 : FAILURE in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9c1433b1e0e4c949395e9e072206abf : SUCCESS in 6m 04s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/3dcde173126b469f9ff3122863ce494e : SUCCESS in 5m 23s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/891b8f0ee1b64ac98d77fbb1a355179b : FAILURE in 12m 03s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/e6c7bfa5420b477284d6b43d5f828cab : SUCCESS in 5m 09s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/85c28a063edd40a19b65507260e77c52 : SUCCESS in 5m 17s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/28e7b0df90c84f448de8075e630875cc : FAILURE in 12m 35s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/f2b1d1c57a1d4a56805baeee71480fae : SUCCESS in 5m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e4ebc9702d5e4b3bb767d780a15dc965 : FAILURE in 15m 51s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e7d9689f08644c2c82aa505a6ca2ff5c : FAILURE in 18m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/94173a3de44c43e2af4abe00b250b17b : FAILURE in 33m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8dcd86de321e4e17b8f61ff06a324f7c : FAILURE in 25m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/2707924aaf764ec5808ace3f329b943f : FAILURE in 29m 12s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/3f6258b6dea6459fad5c1f542a4f96f0 : FAILURE in 50m 01s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/740d4b0d04fc4050b740364610f2450e : FAILURE in 55m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/4bb2381d2d4543ee8d72920640f51257 : FAILURE in 24m 29s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/896697617f5d495e832cd302bfffcb63 : FAILURE in 23m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/ddc6238b70b94cb5a7293f442a54704b : FAILURE in 24m 08s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"f813e7f96a332618147f6321c8ff9697c45edcf5","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-16 01:51:06.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"b3aa07599e04a41272bfcee0897a273a6d0e17d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 03:27:11.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dcbd46c005e04a07a2307508745eb1a2 : SUCCESS in 2m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac6bd71cbb314579bac745fc24e6c48a : FAILURE in 3m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bce2248e76bd472ba073daddecb63ff3 : FAILURE in 15m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fcc0a1c5e234194a390886c91a4429d : SUCCESS in 1h 12m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7c4e9698e754c7db2a42402a4903921 : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/085b083b54a4429bbd745b6cf5024f24 : FAILURE in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60b55c7a6b434efaa5db5ae134951f07 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2fa28469528d495abf7c8a0645c4ad10 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff1479d2ce7a45ebacb1823ab853728a : SUCCESS in 11m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9310513defa42ffa74bd6ebcb64bcac : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b32f620e58a1476083e04a7236faac1c : FAILURE in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7699217a16df41a1bf255a881217eec4 : SUCCESS in 3m 58s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/435d327107564c9a8f21c6720377debe : SUCCESS in 10m 29s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/2b96bdf0893e45d1919d3eec57ad574c : FAILURE in 18m 30s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/2f12697baff646268b454a7bc9d96cd8 : FAILURE in 5m 44s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/b9399173781743e9ad3c00c3c8557142 : SUCCESS in 5m 27s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/d1911a194c9d4583a7199da2feecb879 : FAILURE in 14m 16s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/55d346e64bd64a27b86ca4dbf224130a : FAILURE in 6m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/71151637a21b4c89a50ee0cfa8576654 : FAILURE in 21m 14s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3259b53c4d6b41b09babc132d46bde35 : FAILURE in 19m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/d71e3c3f4aba4f20b97b46bc58876626 : FAILURE in 23m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/33a476202d9241acb1fe56d0e6d76d6f : FAILURE in 24m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/9805a6e27cad4820b4f418d5b556b342 : FAILURE in 32m 29s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/28ed6cd038fa42ec9bb45b5896a35a76 : FAILURE in 48m 08s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/9f3209d22595400f95c59c0679332508 : FAILURE in 1h 30m 08s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/1994ae0930f84a94b551af6ed1f86da3 : FAILURE in 22m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/b747e4fbf53348e790d44373d3e22f01 : FAILURE in 20m 35s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/0a85e3d3e81e4d888961fa5b1f0d959f : FAILURE in 24m 30s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"7a26cc2bb91bf64a8c97b61c33d25df2703b02c9","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-21 05:01:25.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"c06749da724e4f58a3c4e3a587634a5d75219c69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 07:26:50.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0159650cb79f495899c43fb621ac4633 : SUCCESS in 1m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6b14c530a7d04efdbff8d1d2e172ed2b : FAILURE in 3m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/04ee8e03f5a84dc792038740d022a88c : FAILURE in 19m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56b658b837fd46b6a56de85ed5c90197 : SUCCESS in 1h 32m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28c63a6373dc49cfb678183a4dff3d30 : SUCCESS in 4m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91174647b42845339037ee06650cca25 : FAILURE in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b24b9cc6f154ff899d0aab674160993 : SUCCESS in 3m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3356331dd2d04d43be35ec630b3e8bc3 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a6849b54f354bb1b7e029d8842a6eba : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/950952188fc542679266cc497549dd0d : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19cd464639ef41d9bcdb17e60338f81e : FAILURE in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2811a6c60b4d4959a3986d13174c119f : SUCCESS in 3m 40s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/1250c5f2f0dc475a818b4c497a1887d4 : SUCCESS in 5m 08s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/f426ec461aa74020af420a694008a282 : FAILURE in 11m 43s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/c5ba910f63fd4716bfbcbdec9ebf958e : SUCCESS in 4m 50s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/8fa23b0221454003b1bc711925fb0712 : SUCCESS in 5m 22s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/ade99a439af3410da80de68abee04909 : FAILURE in 11m 26s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/f5c4e85ac2004d758d65f037b51b2e21 : SUCCESS in 5m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bba8723607ec4f0ab7056ad9e191afed : FAILURE in 18m 41s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ee8e4272474f4c10b901b967acc3d3ae : FAILURE in 17m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/7d15aaa02bf740f98763f53b8fc67264 : FAILURE in 26m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/721362e2367d47b8b612df62aa531ac9 : FAILURE in 19m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/1b48af13078d414a9c9730a78e96f536 : FAILURE in 23m 42s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/b2a5fd4970e44fa187ebe1450d3d862d : FAILURE in 40m 26s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/2377e14ad2f94d1587217fbc02d0ba07 : FAILURE in 54m 30s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/1cf33e2934e44c97badcb5550f096eec : FAILURE in 23m 17s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/779d89930bb945c39d4caf8ae8fe8843 : FAILURE in 21m 47s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/19fa5c4e2329499086c115f4ab436c85 : FAILURE in 21m 15s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"c4ed8511395ec6a57db1ab9175b86e4844fb3a59","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-24 00:13:57.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"50e00f610c81e445c55a1938d68421d05ffb56a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 01:43:20.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0d0f62236df64bffa03a7152dba6f6d4 : SUCCESS in 1m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6f5c5006a037409392f599bc95f3a6e7 : FAILURE in 3m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e95306bf0f7a49778c9c5362ba061fe7 : FAILURE in 12m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ceb7712c718475cb5cc6f142120f79b : FAILURE in 12m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51a8ebad19df4400ba35c24015f0c156 : FAILURE in 3m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a1cbf2667944442a6bcc01078f9593a : FAILURE in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9afb92c58ab4d899b0f03191b227403 : FAILURE in 2m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b5d8bfbb4ae4f96948051d88ca8f870 : FAILURE in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35d1c6a5dbcd490a9200ad7bc049b7f1 : FAILURE in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba75a28fcbb24f7c88f777ca4e4d048d : FAILURE in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4120651b48984baa9693333d6fef4031 : FAILURE in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a2211d26a3545028386dd2b2a0c802a : SUCCESS in 4m 00s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/e937646b7eca4d2781db776a1db8052c : FAILURE in 3m 48s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/03b3eba94fae445fb0617c629d6864ba : FAILURE in 4m 14s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/58b70cabb2f14112acc1f216967ea031 : FAILURE in 3m 18s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/cbcef10f07cf494295ab6906a6236d5c : FAILURE in 3m 33s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/25762f7c6a5d47ddaa79e94839b3f43e : FAILURE in 3m 46s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/0f71797e75454792ad6aa5adb649f540 : FAILURE in 3m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7ac54b33586d4820b97b0d88147194a8 : FAILURE in 13m 36s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/b296bb9b38b146a2ba0a770cff987480 : FAILURE in 13m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/d690b671f42b4ef1aed11863349c3f3e : FAILURE in 15m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/316262692ce14c29ad0a704a3105d404 : FAILURE in 14m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/414495d6b3944f81884e9a6d7603d461 : FAILURE in 13m 12s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/e4af890b77a74b71a8cd088bfb2d7117 : SUCCESS in 1h 01m 07s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/0db2d7f3010843fd99838dfa8495fcdd : SUCCESS in 1h 21m 45s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/92481ad599df490c85ad106be053dba4 : FAILURE in 13m 15s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/fc7165f11c7a49b1b35bc801db2cb5df : FAILURE in 14m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/52d8c4ecbf244e5aacabb1a519a561ad : FAILURE in 15m 08s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"e6d4c7df3ed5fc987ed13ca825875f5e0b81bde8","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-25 04:50:56.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"8f24c48ea1beec700ef9fd680d390080e90b0bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 07:24:11.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/077490376d4b42759fab00045a5277c9 : SUCCESS in 2m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6094bd7eec79436d8833a12349a279a1 : FAILURE in 3m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fdbfaa15d9424aa5b1b9c45847772f4b : SUCCESS in 1h 41m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63f8ee749d3e4e33897bbdc057ea46f9 : SUCCESS in 1h 23m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ac852c027d64af08a76dab4529fed68 : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fe111279f9b4ead928080bc3711cabf : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb3a94ee89974e37a5ef651a1777fa4e : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ed28358d2f447689792ac4f00fd63b8 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2aa0e5dcf2d542cdae84b10b828293a6 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd9257b249354d5280e20f4aed88391b : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a0880bd5254e5b80ea9ee6a6f40f81 : FAILURE in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0949240aa154fe68e2c50ea27826777 : SUCCESS in 4m 05s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/3a45756d5c66454c93c02706714caef5 : SUCCESS in 5m 35s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/5e6b51b1f6f84e53998e949a956493a0 : FAILURE in 11m 54s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/c8c97c2a7a4c48fc971870bba05abebd : FAILURE in 5m 43s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/ab48e27d94ba45209c7fab8bbe9a5aed : SUCCESS in 5m 24s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/40037d66b48648cca491c4825ce5a02b : FAILURE in 11m 40s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/e756a37e31e74f0cbc2fec969121dde1 : FAILURE in 5m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/676d87440e8a4edab3df693b837f72ae : SUCCESS in 2h 04m 12s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/811db179e81b48c0aadcdea0df32ced3 : SUCCESS in 1h 45m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/f2ed069626474bd294606fce6dd4ec89 : SUCCESS in 2h 12m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/28e74ed443204feea19bbeb44b940841 : SUCCESS in 1h 57m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/18ebc5a1fb8d41889b1d1f61e94acf2f : SUCCESS in 2h 25m 44s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f85312d2b47a4718849a2514ca64ffd9 : SUCCESS in 1h 05m 57s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/2eb5b18090bf49ff92e40dd86ad31a04 : SUCCESS in 1h 18m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/23163119314b4408acc34260eb109e77 : FAILURE in 43m 08s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/d6efa6160151458da1894e79f9b6cb1b : FAILURE in 45m 07s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/77388bb3ce854c3a865f9f45ce01f76f : FAILURE in 44m 23s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"7dc52b15b36c3e34f7df6e8ce2d8e63262686a64","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-10-28 15:44:10.000000000","message":"Patch Set 26: Workflow-1\n\n(1 comment)\n\nI\u0027m going to mark this WIP until we have a chance to discuss the addition of a new driver further. We\u0027ll need to make sure that the team can commit to the requirements in https://docs.openstack.org/oslo.messaging/latest/contributor/supported-messaging-drivers.html#policy","accounts_in_message":[],"_revision_number":26},{"id":"92d7b471842b9afe6bfaee2ffe983cdc8c65b0ea","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-10-28 16:01:34.000000000","message":"Patch Set 26: Code-Review-1\n\nHi Akira,\n\nBefore a new driver can be added to the project a \"blueprint\" describing the driver as well as the use cases.  Blueprints are accepted as patches to the oslo-specs project\u0027s git repo.\n\nWhile I don\u0027t have a \"new drivers\" template you can use an existing template as a guide.  There are two driver-related blueprints you can take a look at:\n\nhttps://opendev.org/openstack/oslo-specs/src/branch/master/specs/newton/amqp-dispatch-router.rst\n\nand:\n\nhttps://opendev.org/openstack/oslo-specs/src/branch/master/specs/queens/update-kafka-support.rst\n\nPlease let me know if you have any questions - thanks!","accounts_in_message":[],"_revision_number":26},{"id":"673845705f80b398f575f727dc5b250d6ce01436","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-31 03:50:26.000000000","message":"Patch Set 26:\n\n(1 comment)\n\nBen, thank you for your review. I checked the policy you mentioned and will try to fill the requirement.","accounts_in_message":[],"_revision_number":26},{"id":"c42cc1a2f956ab4a950976d674aeee09b9d9ec8d","author":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"date":"2019-10-31 03:55:29.000000000","message":"Patch Set 26:\n\nKen, thank you for your reply. I\u0027ve started to write a new blueprint based on one for kafka. I\u0027ll post it next week.\n\nBTW... do I need to write another blueprint for the other patchset for msgpack_(de)serialize_msg()?","accounts_in_message":[],"_revision_number":26},{"id":"3f10ee6576e65308aae318ed26e7a9b46cc92568","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-12-04 20:45:04.000000000","message":"Patch Set 26:\n\nHi Akira,  can you rebase this work onto the new feature/nats-driver branch and re-issue the review request from that branch?  thanks","accounts_in_message":[],"_revision_number":26},{"id":"1ba33bfe80aac6fb65d2563d682fab326a6adf84","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 14:13:01.000000000","message":"Abandoned\n\nNo activity since a year and a half. Feel free to reopen it if needed. Thanks for your understanding.","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"ab14a9f2b7b4e44bcbf2cc2c57449018ac878a26","revisions":{"f097c73006b6ac597ed2157b6780d76bf4451080":{"kind":"REWORK","_number":1,"created":"2019-09-06 08:57:13.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/1"}}},"commit":{"parents":[{"commit":"0a9ce0ba5316a8bc3b99a76b708ee26f0e09df8e","subject":"WIP: Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0a9ce0ba5316a8bc3b99a76b708ee26f0e09df8e"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:56:38.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using nats-python *1. At first, I\ntried to use nats.py *2 for NATS python binding but I found that\nasyncio-based binding is not suitable for oslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the\n  main thread because queue.Queue() or asyncio.Queue() didn\u0027t\n  work between inside and outside of asyncio eventloop. Yes,\n  janus module *3 provides a special queue for the environment\n  but it also didn\u0027t work.\n\n    *1:https://github.com/Gr1N/nats-python\n    *2:https://github.com/Gr1N/nats-python\n    *3:https://github.com/Gr1N/nats-python\n\nNATS driver works fine now. I\u0027ve tested it with cinder in my\ndevstack deployment. It\u0027s WIP because:\n\n- No unit test\n- Not tested with other projects like nova\n- Not tested for notification\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f097c73006b6ac597ed2157b6780d76bf4451080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f097c73006b6ac597ed2157b6780d76bf4451080"}]},"branch":"refs/heads/master"},"8949aff45fc87fa706f446f49fbdf758c6d25c86":{"kind":"NO_CHANGE","_number":2,"created":"2019-09-06 08:59:02.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/2"}}},"commit":{"parents":[{"commit":"a488d1dffb176bf11cbdcdebc8b45a38f8992d96","subject":"WIP: Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a488d1dffb176bf11cbdcdebc8b45a38f8992d96"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:58:49.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using nats-python *1. At first, I\ntried to use nats.py *2 for NATS python binding but I found that\nasyncio-based binding is not suitable for oslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the\n  main thread because queue.Queue() or asyncio.Queue() didn\u0027t\n  work between inside and outside of asyncio eventloop. Yes,\n  janus module *3 provides a special queue for the environment\n  but it also didn\u0027t work.\n\n    *1:https://github.com/Gr1N/nats-python\n    *2:https://github.com/Gr1N/nats-python\n    *3:https://github.com/Gr1N/nats-python\n\nNATS driver works fine now. I\u0027ve tested it with cinder in my\ndevstack deployment. It\u0027s WIP because:\n\n- No unit test\n- Not tested with other projects like nova\n- Not tested for notification\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8949aff45fc87fa706f446f49fbdf758c6d25c86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8949aff45fc87fa706f446f49fbdf758c6d25c86"}]},"branch":"refs/heads/master"},"6c4a47a464ba8bc95830cb2af4cf9f91870fc0f6":{"kind":"NO_CHANGE","_number":3,"created":"2019-09-06 09:00:10.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/3"}}},"commit":{"parents":[{"commit":"c988a72f7215be286ffbce115562e7dfa88b4507","subject":"WIP: Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c988a72f7215be286ffbce115562e7dfa88b4507"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:59:59.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using nats-python *1. At first, I\ntried to use nats.py *2 for NATS python binding but I found that\nasyncio-based binding is not suitable for oslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the\n  main thread because queue.Queue() or asyncio.Queue() didn\u0027t\n  work between inside and outside of asyncio eventloop. Yes,\n  janus module *3 provides a special queue for the environment\n  but it also didn\u0027t work.\n\n    *1:https://github.com/Gr1N/nats-python\n    *2:https://github.com/Gr1N/nats-python\n    *3:https://github.com/Gr1N/nats-python\n\nNATS driver works fine now. I\u0027ve tested it with cinder in my\ndevstack deployment. It\u0027s WIP because:\n\n- No unit test\n- Not tested with other projects like nova\n- Not tested for notification\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6c4a47a464ba8bc95830cb2af4cf9f91870fc0f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6c4a47a464ba8bc95830cb2af4cf9f91870fc0f6"}]},"branch":"refs/heads/master"},"e1124f16850e3c757268caa3bd14a0d6dac72d08":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-09-06 09:02:07.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/4"}}},"commit":{"parents":[{"commit":"c988a72f7215be286ffbce115562e7dfa88b4507","subject":"WIP: Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c988a72f7215be286ffbce115562e7dfa88b4507"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 09:01:37.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using nats-python *1. At first, I\ntried to use nats.py *2 for NATS python binding but I found that\nasyncio-based binding was not suitable for oslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the\n  main thread because queue.Queue() or asyncio.Queue() didn\u0027t\n  work between inside and outside of asyncio eventloop. Yes,\n  janus module *3 provides a special queue for the environment\n  but it also didn\u0027t work.\n\n    *1:https://github.com/Gr1N/nats-python\n    *2:https://github.com/Gr1N/nats-python\n    *3:https://github.com/Gr1N/nats-python\n\nNATS driver works fine now. I\u0027ve tested it with cinder in my\ndevstack deployment. It\u0027s WIP because:\n\n- No unit test\n- Not tested with other projects like nova\n- Not tested for notification\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e1124f16850e3c757268caa3bd14a0d6dac72d08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e1124f16850e3c757268caa3bd14a0d6dac72d08"}]},"branch":"refs/heads/master"},"f82d0291bee63fa7e6063c1028737d3af2eb2c23":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-09-06 09:04:21.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/5"}}},"commit":{"parents":[{"commit":"c988a72f7215be286ffbce115562e7dfa88b4507","subject":"WIP: Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c988a72f7215be286ffbce115562e7dfa88b4507"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 09:03:50.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using nats-python *1. At first, I\ntried to use nats.py *2 for NATS python binding but I found that\nasyncio-based binding was not suitable for oslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the\n  main thread because queue.Queue() or asyncio.Queue() didn\u0027t\n  work between inside and outside of asyncio eventloop. Yes,\n  janus module *3 provides a special queue for the environment\n  but it also didn\u0027t work.\n\n    *1:https://github.com/Gr1N/nats-python\n    *2:https://github.com/nats-io/nats.py\n    *3:https://github.com/aio-libs/janus\n\nNATS driver works fine now. I\u0027ve tested it with cinder in my\ndevstack deployment. It\u0027s WIP because:\n\n- No unit test\n- Not tested with other projects like nova\n- Not tested for notification\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f82d0291bee63fa7e6063c1028737d3af2eb2c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f82d0291bee63fa7e6063c1028737d3af2eb2c23"}]},"branch":"refs/heads/master"},"ad56a3cd0a997b44015d1c18e56ee0fde52a6024":{"kind":"REWORK","_number":6,"created":"2019-09-15 12:52:35.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/6"}}},"commit":{"parents":[{"commit":"ebfef8984cf8192c9bf8788bb42526dacfa0d2c3","subject":"WIP: Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ebfef8984cf8192c9bf8788bb42526dacfa0d2c3"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-09-15 12:42:25.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system *1 *2. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version *3 of\nnats-python *4. At first, I tried to use nats.py *5 for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module *6\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.5+ because nats-python uses type\nhints.\n\n    *1:https://nats.io/\n    *2:https://bravenewgeek.com/dissecting-message-queues/\n    *3:https://github.com/yosshy/nats-python\n    *4:https://github.com/Gr1N/nats-python\n    *5:https://github.com/nats-io/nats.py\n    *6:https://github.com/aio-libs/janus\n\nNATS driver works very fine now. I\u0027ve tested it with cinder in my\ndevstack deployment. It\u0027s WIP because:\n\n- No unit test\n- No support for automatic reconnection\n- Not tested with other projects like nova\n- Not tested for notification\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ad56a3cd0a997b44015d1c18e56ee0fde52a6024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ad56a3cd0a997b44015d1c18e56ee0fde52a6024"}]},"branch":"refs/heads/master"},"1f78d71a49ef5e3c34db552b7702a14c4f3bb57f":{"kind":"NO_CHANGE","_number":7,"created":"2019-09-15 12:53:42.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/7"}}},"commit":{"parents":[{"commit":"5ecd6e5158eb5e601798e562cda51b0008e5ba29","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5ecd6e5158eb5e601798e562cda51b0008e5ba29"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-09-15 12:44:01.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system *1 *2. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version *3 of\nnats-python *4. At first, I tried to use nats.py *5 for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module *6\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.5+ because nats-python uses type\nhints.\n\n    *1:https://nats.io/\n    *2:https://bravenewgeek.com/dissecting-message-queues/\n    *3:https://github.com/yosshy/nats-python\n    *4:https://github.com/Gr1N/nats-python\n    *5:https://github.com/nats-io/nats.py\n    *6:https://github.com/aio-libs/janus\n\nNATS driver works very fine now. I\u0027ve tested it with cinder in my\ndevstack deployment. It\u0027s WIP because:\n\n- No unit test\n- No support for automatic reconnection\n- Not tested with other projects like nova\n- Not tested for notification\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1f78d71a49ef5e3c34db552b7702a14c4f3bb57f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1f78d71a49ef5e3c34db552b7702a14c4f3bb57f"}]},"branch":"refs/heads/master"},"4196a4f6f9e8d61ad53f2b0a6124fa933e01f65a":{"kind":"NO_CHANGE","_number":8,"created":"2019-09-15 12:54:56.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/8"}}},"commit":{"parents":[{"commit":"d17743a56b9fd8060eb9a61bdba9a584053d1831","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d17743a56b9fd8060eb9a61bdba9a584053d1831"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-09-15 12:45:10.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system *1 *2. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version *3 of\nnats-python *4. At first, I tried to use nats.py *5 for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module *6\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.5+ because nats-python uses type\nhints.\n\n    *1:https://nats.io/\n    *2:https://bravenewgeek.com/dissecting-message-queues/\n    *3:https://github.com/yosshy/nats-python\n    *4:https://github.com/Gr1N/nats-python\n    *5:https://github.com/nats-io/nats.py\n    *6:https://github.com/aio-libs/janus\n\nNATS driver works very fine now. I\u0027ve tested it with cinder in my\ndevstack deployment. It\u0027s WIP because:\n\n- No unit test\n- No support for automatic reconnection\n- Not tested with other projects like nova\n- Not tested for notification\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4196a4f6f9e8d61ad53f2b0a6124fa933e01f65a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4196a4f6f9e8d61ad53f2b0a6124fa933e01f65a"}]},"branch":"refs/heads/master"},"ece305af49450d4be3ed354dfbc734957142d734":{"kind":"REWORK","_number":9,"created":"2019-09-21 14:29:59.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/9"}}},"commit":{"parents":[{"commit":"ce0e41e9ab5247706db2b0993b913904fcbcd4d7","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ce0e41e9ab5247706db2b0993b913904fcbcd4d7"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-09-21 14:01:44.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system *1 *2. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version *3 of\nnats-python *4. At first, I tried to use nats.py *5 for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module *6\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.5+ because nats-python uses type\nhints.\n\n    *1:https://nats.io/\n    *2:https://bravenewgeek.com/dissecting-message-queues/\n    *3:https://github.com/yosshy/nats-python\n    *4:https://github.com/Gr1N/nats-python\n    *5:https://github.com/nats-io/nats.py\n    *6:https://github.com/aio-libs/janus\n\nNATS driver works very fine now. I\u0027ve tested it with cinder in my\ndevstack deployment. It\u0027s WIP because:\n\n- Not tested with other projects like nova\n- Not tested for notification\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ece305af49450d4be3ed354dfbc734957142d734"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ece305af49450d4be3ed354dfbc734957142d734"}]},"branch":"refs/heads/master"},"0c2ed2136f4f4ce743f6042928418f1ecc765518":{"kind":"REWORK","_number":10,"created":"2019-09-21 14:35:40.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/10"}}},"commit":{"parents":[{"commit":"ce0e41e9ab5247706db2b0993b913904fcbcd4d7","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ce0e41e9ab5247706db2b0993b913904fcbcd4d7"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-09-21 14:26:44.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system *1 *2. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version *3 of\nnats-python *4. At first, I tried to use nats.py *5 for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module *6\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.5+ because nats-python uses type\nhints.\n\n    *1:https://nats.io/\n    *2:https://bravenewgeek.com/dissecting-message-queues/\n    *3:https://github.com/yosshy/nats-python\n    *4:https://github.com/Gr1N/nats-python\n    *5:https://github.com/nats-io/nats.py\n    *6:https://github.com/aio-libs/janus\n\nNATS driver works very fine now. I\u0027ve tested it with cinder in my\ndevstack deployment. It\u0027s WIP because:\n\n- Not tested with other projects like nova\n- Not tested for notification\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0c2ed2136f4f4ce743f6042928418f1ecc765518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0c2ed2136f4f4ce743f6042928418f1ecc765518"}]},"branch":"refs/heads/master"},"48cbe778f5911686673fb614c4b92596db52133d":{"kind":"REWORK","_number":11,"created":"2019-09-21 14:39:31.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/11"}}},"commit":{"parents":[{"commit":"ce0e41e9ab5247706db2b0993b913904fcbcd4d7","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ce0e41e9ab5247706db2b0993b913904fcbcd4d7"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-09-21 14:30:22.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system *1 *2. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version *3 of\nnats-python *4. At first, I tried to use nats.py *5 for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module *6\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.5+ because nats-python uses type\nhints.\n\n    *1:https://nats.io/\n    *2:https://bravenewgeek.com/dissecting-message-queues/\n    *3:https://github.com/yosshy/nats-python\n    *4:https://github.com/Gr1N/nats-python\n    *5:https://github.com/nats-io/nats.py\n    *6:https://github.com/aio-libs/janus\n\nNATS driver works very fine now. I\u0027ve tested it with cinder in my\ndevstack deployment. It\u0027s WIP because:\n\n- Not tested with other projects like nova\n- Not tested for notification\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/48cbe778f5911686673fb614c4b92596db52133d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/48cbe778f5911686673fb614c4b92596db52133d"}]},"branch":"refs/heads/master"},"9c53f47f9aa9db3ce32e60d8ac21835187009b55":{"kind":"REWORK","_number":12,"created":"2019-09-21 23:38:07.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/12"}}},"commit":{"parents":[{"commit":"ce0e41e9ab5247706db2b0993b913904fcbcd4d7","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ce0e41e9ab5247706db2b0993b913904fcbcd4d7"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-09-21 23:28:40.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system *1 *2. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version *3 of\nnats-python *4. At first, I tried to use nats.py *5 for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module *6\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.5+ because nats-python uses type\nhints.\n\n    *1:https://nats.io/\n    *2:https://bravenewgeek.com/dissecting-message-queues/\n    *3:https://github.com/yosshy/nats-python\n    *4:https://github.com/Gr1N/nats-python\n    *5:https://github.com/nats-io/nats.py\n    *6:https://github.com/aio-libs/janus\n\nNATS driver works very fine now. I\u0027ve tested it with cinder in my\ndevstack deployment. It\u0027s WIP because:\n\n- Not tested with other projects like nova\n- Not tested for notification\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9c53f47f9aa9db3ce32e60d8ac21835187009b55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9c53f47f9aa9db3ce32e60d8ac21835187009b55"}]},"branch":"refs/heads/master"},"f5a6ca279013c88dbfc9c0976ea302200d58c325":{"kind":"REWORK","_number":13,"created":"2019-09-22 00:25:48.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/13"}}},"commit":{"parents":[{"commit":"2fb1704831c4ffe0c5ff457ad989876b9a99dd09","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2fb1704831c4ffe0c5ff457ad989876b9a99dd09"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-09-22 00:16:40.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system *1 *2. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version *3 of\nnats-python *4. At first, I tried to use nats.py *5 for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module *6\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.5+ because nats-python uses type\nhints.\n\n    *1:https://nats.io/\n    *2:https://bravenewgeek.com/dissecting-message-queues/\n    *3:https://github.com/yosshy/nats-python\n    *4:https://github.com/Gr1N/nats-python\n    *5:https://github.com/nats-io/nats.py\n    *6:https://github.com/aio-libs/janus\n\nNATS driver works very fine now. I\u0027ve tested it with cinder in my\ndevstack deployment. It\u0027s WIP because:\n\n- Not tested with other projects like nova\n- Not tested for notification\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f5a6ca279013c88dbfc9c0976ea302200d58c325"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f5a6ca279013c88dbfc9c0976ea302200d58c325"}]},"branch":"refs/heads/master"},"a650ccef79a16aae4a1a7e120426d4c203f71497":{"kind":"REWORK","_number":14,"created":"2019-10-08 14:25:01.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/14"}}},"commit":{"parents":[{"commit":"acaad633e83a163ca6ddb2a62f8c14eb4455a9c9","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/acaad633e83a163ca6ddb2a62f8c14eb4455a9c9"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-10-08 13:51:45.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system[1][2]. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version[3] of\nnats-python[4]. At first, I tried to use nats.py[5] for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module[6]\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.6+ because nats-python uses type\nhints.\n\n    [1] https://nats.io/\n    [2] https://bravenewgeek.com/dissecting-message-queues/\n    [3] https://github.com/yosshy/nats-python\n    [4] https://github.com/Gr1N/nats-python\n    [5] https://github.com/nats-io/nats.py\n    [6] https://github.com/aio-libs/janus\n\nIt is WIP because\n* no auto reconnection\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a650ccef79a16aae4a1a7e120426d4c203f71497"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a650ccef79a16aae4a1a7e120426d4c203f71497"}]},"branch":"refs/heads/master"},"c8065d2b9ba6ead3438f70298a24f785650e90d7":{"kind":"NO_CHANGE","_number":15,"created":"2019-10-08 14:28:00.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/15"}}},"commit":{"parents":[{"commit":"82fc6d4767850c22bca77a92a55705b5913f19d3","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82fc6d4767850c22bca77a92a55705b5913f19d3"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-10-08 14:19:42.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system[1][2]. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version[3] of\nnats-python[4]. At first, I tried to use nats.py[5] for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module[6]\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.6+ because nats-python uses type\nhints.\n\n    [1] https://nats.io/\n    [2] https://bravenewgeek.com/dissecting-message-queues/\n    [3] https://github.com/yosshy/nats-python\n    [4] https://github.com/Gr1N/nats-python\n    [5] https://github.com/nats-io/nats.py\n    [6] https://github.com/aio-libs/janus\n\nIt is WIP because\n* no auto reconnection\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c8065d2b9ba6ead3438f70298a24f785650e90d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c8065d2b9ba6ead3438f70298a24f785650e90d7"}]},"branch":"refs/heads/master"},"abdb170a0d65b2bfa0318a5fc450aebdae88e06e":{"kind":"REWORK","_number":16,"created":"2019-10-11 04:19:29.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/16"}}},"commit":{"parents":[{"commit":"82fc6d4767850c22bca77a92a55705b5913f19d3","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82fc6d4767850c22bca77a92a55705b5913f19d3"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-10-11 04:12:40.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system[1][2]. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version[3] of\nnats-python[4]. At first, I tried to use nats.py[5] for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module[6]\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.6+ because nats-python uses type\nhints.\n\n    [1] https://nats.io/\n    [2] https://bravenewgeek.com/dissecting-message-queues/\n    [3] https://github.com/yosshy/nats-python\n    [4] https://github.com/Gr1N/nats-python\n    [5] https://github.com/nats-io/nats.py\n    [6] https://github.com/aio-libs/janus\n\nIt is WIP because\n* no auto reconnection\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/abdb170a0d65b2bfa0318a5fc450aebdae88e06e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/abdb170a0d65b2bfa0318a5fc450aebdae88e06e"}]},"branch":"refs/heads/master"},"7367b77d2ae132051a5a1b6e50bdfa6ac5a26b21":{"kind":"REWORK","_number":17,"created":"2019-10-11 04:38:29.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/17"}}},"commit":{"parents":[{"commit":"82fc6d4767850c22bca77a92a55705b5913f19d3","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82fc6d4767850c22bca77a92a55705b5913f19d3"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-10-11 04:31:37.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system[1][2]. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version[3] of\nnats-python[4]. At first, I tried to use nats.py[5] for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module[6]\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.6+ because nats-python uses type\nhints.\n\n    [1] https://nats.io/\n    [2] https://bravenewgeek.com/dissecting-message-queues/\n    [3] https://github.com/yosshy/nats-python\n    [4] https://github.com/Gr1N/nats-python\n    [5] https://github.com/nats-io/nats.py\n    [6] https://github.com/aio-libs/janus\n\nIt is WIP because\n* no auto reconnection\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7367b77d2ae132051a5a1b6e50bdfa6ac5a26b21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7367b77d2ae132051a5a1b6e50bdfa6ac5a26b21"}]},"branch":"refs/heads/master"},"ad8c5adfb5e2dcc39802caa9188b43ee63d9904e":{"kind":"REWORK","_number":18,"created":"2019-10-12 09:38:39.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/18"}}},"commit":{"parents":[{"commit":"82fc6d4767850c22bca77a92a55705b5913f19d3","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82fc6d4767850c22bca77a92a55705b5913f19d3"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-10-12 09:30:27.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system[1][2]. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version[3] of\nnats-python[4]. At first, I tried to use nats.py[5] for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module[6]\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.6+ because nats-python uses type\nhints.\n\n    [1] https://nats.io/\n    [2] https://bravenewgeek.com/dissecting-message-queues/\n    [3] https://github.com/yosshy/nats-python\n    [4] https://github.com/Gr1N/nats-python\n    [5] https://github.com/nats-io/nats.py\n    [6] https://github.com/aio-libs/janus\n\nIt is WIP because\n* no auto reconnection\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ad8c5adfb5e2dcc39802caa9188b43ee63d9904e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ad8c5adfb5e2dcc39802caa9188b43ee63d9904e"}]},"branch":"refs/heads/master"},"b248e790ec7bf73888d066be2073854ea11ac077":{"kind":"REWORK","_number":19,"created":"2019-10-12 10:41:43.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/19"}}},"commit":{"parents":[{"commit":"82fc6d4767850c22bca77a92a55705b5913f19d3","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82fc6d4767850c22bca77a92a55705b5913f19d3"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-10-12 10:34:51.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system[1][2]. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version[3] of\nnats-python[4]. At first, I tried to use nats.py[5] for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module[6]\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.6+ because nats-python uses type\nhints.\n\n    [1] https://nats.io/\n    [2] https://bravenewgeek.com/dissecting-message-queues/\n    [3] https://github.com/yosshy/nats-python\n    [4] https://github.com/Gr1N/nats-python\n    [5] https://github.com/nats-io/nats.py\n    [6] https://github.com/aio-libs/janus\n\nIt is WIP because\n* no auto reconnection\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b248e790ec7bf73888d066be2073854ea11ac077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b248e790ec7bf73888d066be2073854ea11ac077"}]},"branch":"refs/heads/master"},"7a95aa7069a2177e088bf0e42e8affe8b26ace3e":{"kind":"REWORK","_number":20,"created":"2019-10-15 06:47:10.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/20"}}},"commit":{"parents":[{"commit":"82fc6d4767850c22bca77a92a55705b5913f19d3","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82fc6d4767850c22bca77a92a55705b5913f19d3"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-10-15 05:59:38.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system[1][2]. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version[3] of\nnats-python[4]. At first, I tried to use nats.py[5] for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module[6]\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.6+ because nats-python uses type\nhints.\n\n    [1] https://nats.io/\n    [2] https://bravenewgeek.com/dissecting-message-queues/\n    [3] https://github.com/yosshy/nats-python\n    [4] https://github.com/Gr1N/nats-python\n    [5] https://github.com/nats-io/nats.py\n    [6] https://github.com/aio-libs/janus\n\nIt is WIP because\n* no auto reconnection\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7a95aa7069a2177e088bf0e42e8affe8b26ace3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7a95aa7069a2177e088bf0e42e8affe8b26ace3e"}]},"branch":"refs/heads/master"},"53a0b75f58674e58b089636322bc762b43f55a2b":{"kind":"REWORK","_number":21,"created":"2019-10-15 07:17:56.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/21"}}},"commit":{"parents":[{"commit":"82fc6d4767850c22bca77a92a55705b5913f19d3","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82fc6d4767850c22bca77a92a55705b5913f19d3"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-10-15 07:11:32.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system[1][2]. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version[3] of\nnats-python[4]. At first, I tried to use nats.py[5] for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module[6]\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.6+ because nats-python uses type\nhints.\n\n    [1] https://nats.io/\n    [2] https://bravenewgeek.com/dissecting-message-queues/\n    [3] https://github.com/yosshy/nats-python\n    [4] https://github.com/Gr1N/nats-python\n    [5] https://github.com/nats-io/nats.py\n    [6] https://github.com/aio-libs/janus\n\nIt is WIP because\n* no auto reconnection\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/53a0b75f58674e58b089636322bc762b43f55a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/53a0b75f58674e58b089636322bc762b43f55a2b"}]},"branch":"refs/heads/master"},"4087b6e053988ee25c79705b15850f099dea75f8":{"kind":"REWORK","_number":22,"created":"2019-10-15 07:32:19.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/22"}}},"commit":{"parents":[{"commit":"82fc6d4767850c22bca77a92a55705b5913f19d3","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82fc6d4767850c22bca77a92a55705b5913f19d3"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-10-15 07:24:45.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system[1][2]. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version[3] of\nnats-python[4]. At first, I tried to use nats.py[5] for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module[6]\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.6+ because nats-python uses type\nhints.\n\n    [1] https://nats.io/\n    [2] https://bravenewgeek.com/dissecting-message-queues/\n    [3] https://github.com/yosshy/nats-python\n    [4] https://github.com/Gr1N/nats-python\n    [5] https://github.com/nats-io/nats.py\n    [6] https://github.com/aio-libs/janus\n\nIt is WIP because\n* no auto reconnection\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4087b6e053988ee25c79705b15850f099dea75f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4087b6e053988ee25c79705b15850f099dea75f8"}]},"branch":"refs/heads/master"},"e5eb93adaa720213443a3106bc7cff8fbd1bda81":{"kind":"REWORK","_number":23,"created":"2019-10-15 07:47:23.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/23"}}},"commit":{"parents":[{"commit":"82fc6d4767850c22bca77a92a55705b5913f19d3","subject":"Add (de)serialize_msg_as_bytes()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82fc6d4767850c22bca77a92a55705b5913f19d3"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-10-15 07:41:01.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system[1][2]. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version[3] of\nnats-python[4]. At first, I tried to use nats.py[5] for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module[6]\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.6+ because nats-python uses type\nhints.\n\n    [1] https://nats.io/\n    [2] https://bravenewgeek.com/dissecting-message-queues/\n    [3] https://github.com/yosshy/nats-python\n    [4] https://github.com/Gr1N/nats-python\n    [5] https://github.com/nats-io/nats.py\n    [6] https://github.com/aio-libs/janus\n\nIt is WIP because\n* no auto reconnection\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e5eb93adaa720213443a3106bc7cff8fbd1bda81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e5eb93adaa720213443a3106bc7cff8fbd1bda81"}]},"branch":"refs/heads/master"},"239b254be34c5bbb43ccf6c993858f937da51934":{"kind":"REWORK","_number":24,"created":"2019-10-21 05:01:25.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/24"}}},"commit":{"parents":[{"commit":"5a55f8d13a537bd11316cfcdcffdd7a6aaed6448","subject":"Add msgpack_(de)serialize_msg()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5a55f8d13a537bd11316cfcdcffdd7a6aaed6448"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-10-21 04:55:36.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system[1][2]. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version[3] of\nnats-python[4]. At first, I tried to use nats.py[5] for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module[6]\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.6+ because nats-python uses type\nhints.\n\n    [1] https://nats.io/\n    [2] https://bravenewgeek.com/dissecting-message-queues/\n    [3] https://github.com/yosshy/nats-python\n    [4] https://github.com/Gr1N/nats-python\n    [5] https://github.com/nats-io/nats.py\n    [6] https://github.com/aio-libs/janus\n\nIt is WIP because\n* no auto reconnection\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/239b254be34c5bbb43ccf6c993858f937da51934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/239b254be34c5bbb43ccf6c993858f937da51934"}]},"branch":"refs/heads/master"},"98469d97c81a0f746808af43e096be1e47640b54":{"kind":"REWORK","_number":25,"created":"2019-10-24 00:13:57.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/25"}}},"commit":{"parents":[{"commit":"5a55f8d13a537bd11316cfcdcffdd7a6aaed6448","subject":"Add msgpack_(de)serialize_msg()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5a55f8d13a537bd11316cfcdcffdd7a6aaed6448"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-10-24 00:04:28.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system[1][2]. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version[3] of\nnats-python[4]. At first, I tried to use nats.py[5] for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module[6]\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.6+ because nats-python uses type\nhints.\n\n    [1] https://nats.io/\n    [2] https://bravenewgeek.com/dissecting-message-queues/\n    [3] https://github.com/yosshy/nats-python\n    [4] https://github.com/Gr1N/nats-python\n    [5] https://github.com/nats-io/nats.py\n    [6] https://github.com/aio-libs/janus\n\nIt is WIP because\n* no auto reconnection\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/98469d97c81a0f746808af43e096be1e47640b54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/98469d97c81a0f746808af43e096be1e47640b54"}]},"branch":"refs/heads/master"},"ab14a9f2b7b4e44bcbf2cc2c57449018ac878a26":{"kind":"REWORK","_number":26,"created":"2019-10-25 04:50:56.000000000","uploader":{"_account_id":659,"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","username":"yosshy"},"ref":"refs/changes/29/680629/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/29/680629/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/29/680629/26"}}},"commit":{"parents":[{"commit":"5a55f8d13a537bd11316cfcdcffdd7a6aaed6448","subject":"Add msgpack_(de)serialize_msg()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5a55f8d13a537bd11316cfcdcffdd7a6aaed6448"}]}],"author":{"name":"Akira YOSHIYAMA","email":"akirayoshiyama@gmail.com","date":"2019-09-06 08:22:21.000000000","tz":540},"committer":{"name":"Akira Yoshiyama","email":"akirayoshiyama@gmail.com","date":"2019-10-25 04:45:09.000000000","tz":540},"subject":"WIP: Add NATS driver","message":"WIP: Add NATS driver\n\nNATS is a high-performance messaging system[1][2]. It has message\nreplication capability. It has no message persistency like RabbitMQ\nbut it\u0027s not a problem for oslo.messaging.\n\nThis patchset adds nats driver using a customized version[3] of\nnats-python[4]. At first, I tried to use nats.py[5] for NATS python\nbinding but I found that asyncio-based binding was not suitable for\noslo.messaging:\n\n- asyncio has its own eventloop and thread handling and it\u0027s not\n  expected to use with another eventloop system like eventlet.\n\n- With unpatched modules like threading, queue and socket, it\u0027s\n  difficult/unable to communicate an asyncio eventloop and the main\n  thread because queue.Queue() or asyncio.Queue() didn\u0027t work between\n  inside and outside of asyncio eventloop. Yes, janus module[6]\n  provides a special queue for the environment but it also didn\u0027t\n  work.\n\nThis driver works with Python 3.6+ because nats-python uses type\nhints.\n\n    [1] https://nats.io/\n    [2] https://bravenewgeek.com/dissecting-message-queues/\n    [3] https://github.com/yosshy/nats-python\n    [4] https://github.com/Gr1N/nats-python\n    [5] https://github.com/nats-io/nats.py\n    [6] https://github.com/aio-libs/janus\n\nIt is WIP because\n* no auto reconnection\n\nChange-Id: I9720e8a284ce66b15a23e9881fa3a9aed4ca77ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ab14a9f2b7b4e44bcbf2cc2c57449018ac878a26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ab14a9f2b7b4e44bcbf2cc2c57449018ac878a26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
