)]}'
{"id":"openstack%2Foslo.messaging~681522","triplet_id":"openstack%2Foslo.messaging~master~If17c50324fcadb00a5aff16c433e22e6b15f156d","project":"openstack/oslo.messaging","branch":"master","topic":"lc-test","hashtags":[],"change_id":"If17c50324fcadb00a5aff16c433e22e6b15f156d","subject":"Test change, ignore me","status":"ABANDONED","created":"2019-09-11 15:50:14.000000000","updated":"2019-09-12 13:40:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e9dff44b7fe214d22b1c873632a59a330188f847","_number":681522,"virtual_id_number":681522,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-09-11 18:19:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 18:19:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e5de5b7877cab0b6082a653d2fb833d47a013b1","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-09-11 15:50:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa0e1d33d319f1c9322545115a71e987e09f82a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 18:19:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/884c6e23d66c402e962989c0b7c4ef50 : SUCCESS in 1h 37m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b43ce342ae045519103cf11ad8a948c : SUCCESS in 1h 30m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7318e55d8ca84919bc52e9c6c4f659e0 : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1256429220847c7be990456c78132f2 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b20fbacfff834b79aee100b10a47b976 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b184cf3a7444330b0f407236be828c1 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3112977f2b74166aaa6834ea5e60815 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42a33c05e26640438ce76e1ef5c76294 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f7ef618b5c949b08b7ce8f7ce1054b4 : SUCCESS in 3m 59s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/79dedb1aac79433880613a8d449789a1 : SUCCESS in 5m 11s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/c7a11e187103417c97a968ae2939a510 : FAILURE in 11m 33s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/ab1b855fcd62403faf8ef96705ef1af2 : SUCCESS in 5m 48s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/2107c979eeee4a6bbaa9e08bcc79f2e7 : SUCCESS in 6m 47s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/b7387b3a67ad44e0a4113546373c571e : FAILURE in 12m 05s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/87a7ceb91cd548e0a8f8046e2149b046 : SUCCESS in 5m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d6f5d52f450940c1adc0b82e91519803 : SUCCESS in 1h 38m 07s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/95c157f088b7486b9cc8fe234a67fba1 : SUCCESS in 1h 32m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/47ab4feeaa7a41c0a64dc1fd3fd1566b : SUCCESS in 2h 04m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/39d90e60594f4e8da6cbaea8b491ace2 : SUCCESS in 1h 35m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/0503ab06bcde4567a37d275e6e602b91 : SUCCESS in 2h 01m 34s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/3b096a2d0c8f474c8749892a02908de5 : SUCCESS in 1h 07m 55s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/15ffa2fe9cc04a66af09601052fecdb7 : SUCCESS in 1h 09m 58s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/cbb9c30f8a7e4c618451b35b25e8414e : FAILURE in 52m 13s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/d485597341844d09bcdef984fef8e362 : FAILURE in 55m 15s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/3bd4e92df157467d916681fd260edcf1 : FAILURE in 46m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9cf430330542d6f7893f9a60fd2a1ed05d556d93","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-09-12 13:40:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"194ca53db59df221ad8c0ea9ec8f9cb394a1e298","revisions":{"194ca53db59df221ad8c0ea9ec8f9cb394a1e298":{"kind":"REWORK","_number":1,"created":"2019-09-11 15:50:14.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/22/681522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/22/681522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/681522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/681522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/681522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/22/681522/1"}}},"commit":{"parents":[{"commit":"b7e9faf6590086b79f9696183b5c296ecc12b7b6","subject":"Add the mandatory flag for direct send","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b7e9faf6590086b79f9696183b5c296ecc12b7b6"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-09-11 15:49:41.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-09-11 15:49:41.000000000","tz":0},"subject":"Test change, ignore me","message":"Test change, ignore me\n\nChange-Id: If17c50324fcadb00a5aff16c433e22e6b15f156d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/194ca53db59df221ad8c0ea9ec8f9cb394a1e298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/194ca53db59df221ad8c0ea9ec8f9cb394a1e298"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
