)]}'
{"id":"openstack%2Foslo.messaging~682908","triplet_id":"openstack%2Foslo.messaging~master~Ibc40ad1ddedbe04d6ba614467a78e43b3aa6348d","project":"openstack/oslo.messaging","branch":"master","topic":"bug/1826281","hashtags":[],"change_id":"Ibc40ad1ddedbe04d6ba614467a78e43b3aa6348d","subject":"Fix help text for heartbeat_in_pthread option","status":"MERGED","created":"2019-09-18 14:30:15.000000000","updated":"2019-09-18 18:35:02.000000000","submitted":"2019-09-18 18:33:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682908-1568831616183-c3ff88a4","meta_rev_id":"6d368478676a6e919da6f632d38ae65c26095d94","_number":682908,"virtual_id_number":682908,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-09-18 18:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-09-18 15:52:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-09-18 15:52:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 15:52:30.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-09-18 18:33:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c22dc8a887e041007deff4a7d69e7a5ec475bd10","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-18 14:30:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62479a8f1b26558a5efee745d280001ba22340ea","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-18 15:52:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTrivial, obviously correct fix. Fast approving.","accounts_in_message":[],"_revision_number":1},{"id":"830a94b64b5f4d75cb9f16b0cccc77e9ac712e63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 16:48:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d4731028d53b4e0987e467bce846e403 : SUCCESS in 1h 53m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4a836879a964b72a9e48f62c1877403 : SUCCESS in 1h 30m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a0ea4e4b8b44733975643f7b81bf710 : SUCCESS in 5m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/249f9fca32ce43f0a3c51c0a3b40ee92 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4944b72e70c7482fa532d6df1e8e1be4 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/506577f46b8f419fbcb0411a0cf7a0d5 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff42c5f21eed4b6d803af9d4a19735b3 : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab51dfe4e90e46869ea3c7c4dae98f8a : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d08da43511140fbbc43086cec9ff9dc : SUCCESS in 3m 49s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/06eec3fab0164e08896b75d8a505e28e : SUCCESS in 5m 56s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/dcbd7aaf27d44fe5982007db377ca6d9 : FAILURE in 12m 36s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/10ee54cbc6ed4ef6a35ea636cc87a1e0 : SUCCESS in 6m 36s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/14641756781c4a20b46e85af67aa6bc3 : FAILURE in 6m 13s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/b70dc36e64454e1698e26989a372a138 : FAILURE in 12m 28s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/4a1c2ebd683b422b8b97c76208570214 : SUCCESS in 6m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c3cf82f44b8f4a95940679a3f30abea0 : SUCCESS in 1h 36m 30s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/85a01d23376e485392381403ef377752 : SUCCESS in 1h 43m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/c2fd4c76dc644c19ad04805a1a0e1505 : SUCCESS in 1h 54m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1bad06e9b953470ea5f5e8c7ad5f794f : SUCCESS in 1h 53m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/bc5a0590f455438489ca81c13ec79017 : SUCCESS in 2h 09m 18s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/e43721d3961945329bcf2eaeac576d81 : SUCCESS in 1h 23m 25s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/316ca59c73af4964a11290ade7308153 : SUCCESS in 1h 14m 44s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/477c40716a9f4525b1564d6bdc820870 : FAILURE in 54m 44s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/466e60a215e74ceeaa7c47d205a2157f : FAILURE in 47m 16s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/4755799eb99b49cb84f1ec7cfcc4b3bd : FAILURE in 1h 00m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23a07de87912917867248b8b24b03b4e542206d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 16:48:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5f06e40946e5d02451633c4b911942808c96b0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:33:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f358b82eb1e54930b404dbf16c5b1f34 : SUCCESS in 1h 34m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8e8283e768d43ba976d5dee7b1cf4cf : SUCCESS in 1h 25m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9936083ae1c540119f0cd834641cb6ff : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f90843aa0f2141daa94f9afb8d51a398 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/943e70a901cf4ccf8d12983720c5a4d5 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61fe08cf5fbb494eb8640d0b298a7828 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01c966322e1e44e7b97143da2b027a05 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acbb94731cbc4ed397e5b3fbb992bc6c : SUCCESS in 4m 33s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/38277ba830654d55b9301972297f6238 : SUCCESS in 6m 31s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ac6af973d7824c4c827870ab70bc45d3 : SUCCESS in 1h 38m 50s","accounts_in_message":[],"_revision_number":1},{"id":"edd3c433b588c4d451f9a9f1c66385831d4e5db7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:33:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"df6d79c61aa54250e69504d43841026aed95d67e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:35:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb6835923a8d433e8ff376b3d4c109b4 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"440357d74027650701d74a82437921fe946fa347","revisions":{"440357d74027650701d74a82437921fe946fa347":{"kind":"REWORK","_number":1,"created":"2019-09-18 14:30:15.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/08/682908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/08/682908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/08/682908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/08/682908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/08/682908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/08/682908/1"}}},"commit":{"parents":[{"commit":"b7e9faf6590086b79f9696183b5c296ecc12b7b6","subject":"Add the mandatory flag for direct send","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b7e9faf6590086b79f9696183b5c296ecc12b7b6"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-18 14:29:58.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-18 14:30:14.000000000","tz":-240},"subject":"Fix help text for heartbeat_in_pthread option","message":"Fix help text for heartbeat_in_pthread option\n\nChange-Id: Ibc40ad1ddedbe04d6ba614467a78e43b3aa6348d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/440357d74027650701d74a82437921fe946fa347"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/440357d74027650701d74a82437921fe946fa347"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
