)]}'
{"id":"openstack%2Foslo.messaging~686387","triplet_id":"openstack%2Foslo.messaging~master~Ie622b8a39514f87b60ff2e801843c5e3c06106e7","project":"openstack/oslo.messaging","branch":"master","topic":"fix-spacing","hashtags":[],"change_id":"Ie622b8a39514f87b60ff2e801843c5e3c06106e7","subject":"Fix spacing in help message","status":"MERGED","created":"2019-10-03 12:11:00.000000000","updated":"2019-10-07 18:18:53.000000000","submitted":"2019-10-07 18:17:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686387-1570472253320-001e37fd","meta_rev_id":"fb6856088e362d64347a739822b006d40d3922be","_number":686387,"virtual_id_number":686387,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-07 18:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-04 08:13:03.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-10-03 12:36:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-10-03 12:36:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 12:36:27.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-10-07 18:17:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5958c261e5856cda597a9b98f5fe346dc0dbf5e1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-03 12:11:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9885672a0545e967b2457a63ad368ad57a7575bb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-03 12:19:43.000000000","message":"Topic set to fix-spacing","accounts_in_message":[],"_revision_number":1},{"id":"bf45db62119998f86cbfb1e2adab8be99db45487","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-03 12:36:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTrivial, clearly correct fix. Fast approving.","accounts_in_message":[],"_revision_number":1},{"id":"40346c0d5ff4a458b621c19653e0527422c82ef9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 14:33:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/173ce4f0847949f08fd46f1e16290638 : TIMED_OUT in 2h 10m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91fac2142bc54516864919d6a8d01384 : SUCCESS in 1h 49m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f0a10cc431b4d49a5fcb522bd36bba8 : SUCCESS in 8m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10a224fc9126412fb5b81fbdfc55da0c : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27bfc23ac9714777bdbb63e1fd4f26ff : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85100ed8e73041a79418c94af3afa23f : TIMED_OUT in 41m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a79c342010f47d485fb69e2e2e59926 : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d1df2e1a39f4212998e3e1d302218c2 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a72e1da1b544464eb4012fee2345ca33 : SUCCESS in 5m 16s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/af7a9e1305c04dab9d8f49fbc582682f : SUCCESS in 8m 42s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/a2e4d13b649e4da5b2fb52057c164033 : FAILURE in 12m 10s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/4b5359d99cdf4cb887b8aaefe96a9338 : FAILURE in 7m 36s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/0c2ac8d2679e4aa18a2719200425f5fd : SUCCESS in 5m 50s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/536de434ef414667a10d500b85951434 : FAILURE in 12m 57s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/98b2ebd1f65444d8aa41929c29a08f03 : FAILURE in 7m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ba1b88dba75a4a8c90e8db007bde1d13 : SUCCESS in 1h 54m 06s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/572aefe2302d47faadebce08118bac31 : SUCCESS in 1h 53m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/20714cf04545444d8f796ce9ba52eaf9 : SUCCESS in 1h 56m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/46382d7342054773ab5cecc6c7c2873b : SUCCESS in 2h 09m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/0339662ff51b4a98b053371cc91e4a77 : SUCCESS in 2h 13m 41s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/e366d15ed5964ecdb72e46430fa194ff : SUCCESS in 1h 14m 37s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/9022209c39054d6b834ff7c17e72dc29 : SUCCESS in 1h 26m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/8e7b38a95c34469fab56fd02d7632d9f : FAILURE in 49m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/8fade2efab3d400b84370ee916454592 : FAILURE in 36m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/a51b96f6e3e84998818af1ab912a9a20 : FAILURE in 52m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78ed1dda71a268a2399e434ca213da05a3c58797","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-04 08:13:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7185eba367bc7c85f5857320a4d9b04cd7015407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 10:27:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1b8f28954e96471a8f6f1c1365d595b9 : SUCCESS in 1h 46m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7c8a1d4d5484051a4a8b625943613ee : SUCCESS in 1h 26m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7a7b2ad964048119798b4f4b0bc7958 : SUCCESS in 6m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7c92f57091c439ab787aacc7d0ff094 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/969fc0ad93e7428e95b793d1ec71e56e : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c28508db287a481da6302d8c4cf1c5c2 : TIMED_OUT in 41m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c71c4dc954f3450790cb6e6e27c59a4a : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5faec06bc31043aaa05b806876557c5f : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82931b93b0a747ce8cca6dedf481658d : SUCCESS in 3m 58s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/763d1d7414674746a93da451a2e4e7d4 : SUCCESS in 6m 15s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/8b863dbd353b4353838cf128dadb3399 : FAILURE in 11m 30s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/2935f5e244384b598d2b80b224525c45 : SUCCESS in 5m 25s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/0713465f403f4ef4a72d0119c1e05a09 : SUCCESS in 5m 26s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/81db95f106ca48548f2026345f15d99f : FAILURE in 14m 45s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/6337052ddb714b7dbfe52eaca6ab6c60 : SUCCESS in 5m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b499f2553db34c9bba864fb03395a920 : SUCCESS in 1h 43m 24s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ef35b2ab827046d6a03ab77a47fd1713 : SUCCESS in 1h 35m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/4d07d458ccfc4cb094c2fce1c1959bc6 : SUCCESS in 1h 53m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b8b3ab1cfb6d44d78ad6993357cd0ea3 : FAILURE in 2h 08m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/0539c0f82b9c410f92c4fe1d0179c56f : SUCCESS in 2h 07m 58s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/302925fb5bb840019ca81f4d302ff51c : SUCCESS in 1h 01m 00s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/c2fa67d04ede4b948e7dcef30467e5e7 : SUCCESS in 1h 31m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/7bf3a483d0fb46df8f238697af90022d : FAILURE in 53m 58s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/cc5b654e35e44ed88989c0f9c15e33bc : FAILURE in 45m 10s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/0d8d6646f5fe4ee48d555df724b04bd6 : FAILURE in 45m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85f783846c35a07c631d2efa99204aeb860aade5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-07 13:56:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"73fa6ede9997487bee4f9bce24fe0c9929cec458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 16:29:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/10d33b92612b40978d3a7beb51fa4f44 : SUCCESS in 1h 58m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e889b7f91a44462d8630c9913a137b78 : SUCCESS in 1h 34m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95e2b82a102940ba9e3503838c3e0c11 : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02f9920ac1304f6198d2519be5be5d97 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc9c3030339f4d759c51e780aece0d97 : SUCCESS in 3m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/859a16b88639409f80c1fa4238ee5789 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a45da83bba645dcb7a0290dac6ad559 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a49ea0caa849483495e23b91773fa706 : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3fa05f9a4344ad088c9dfcebc575aff : SUCCESS in 3m 28s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/2f9ac48bca794797857fbad7df3d70ee : SUCCESS in 7m 33s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/fd0cf2ee822e4299bae4bf198cff91e4 : FAILURE in 12m 01s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/e54c69bcb52d47dbbd6a655e643a5be4 : FAILURE in 6m 19s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/c0bdd715af2243369b38873f63968e7c : SUCCESS in 5m 52s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/d6b1add49f9344419385878ef27976a9 : FAILURE in 12m 42s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/bee93306e75749a5954f601e8b3351f1 : FAILURE in 5m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a55cb873e1244b0aa04672c9c798ae58 : SUCCESS in 1h 44m 28s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/bb8ff97a7c3a488abb4a66331c6afb02 : SUCCESS in 1h 57m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/729a4cb7315e417d924f81943a8498e1 : SUCCESS in 2h 09m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/4f85bf04bf3646709ea2f1a803a8d795 : SUCCESS in 1h 49m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/04fec9bedd924f45aaf6861e982955f0 : SUCCESS in 2h 27m 30s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f43d9bec772a45ff8a02ee87314b6841 : SUCCESS in 1h 00m 07s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/04d669d52e5f4f86b0e43720b5d5d22d : SUCCESS in 1h 21m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/33084916bc1b469ebb25281c4975f397 : FAILURE in 47m 41s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/3bea291287ab4ac298e80e41bc802b53 : FAILURE in 34m 08s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/9e2686521b474450928115b4098aeeb1 : FAILURE in 57m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c877f5e9d137c1d3fb6d8d4ecb8fe4f000395132","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 16:29:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef4e485dd43e95af577c21d3dc4d550cd6adacc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 18:17:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d36d94a8187d342d82e3d47d2e47a23b8ef5053a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 18:17:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b03cbe9baa7f4868bdb5f124f4f12131 : SUCCESS in 1h 37m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88c7ef43c88140c4b1b1989e5b4bd6ab : SUCCESS in 1h 21m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ecc4d8012b345f7a7ea69bb0447e943 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b58cae2d96034bbd9c74aa7956a79d23 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd637b79915d488bb5a68b1a5c9de81b : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d42f48f302c349af88e5d2448aa983bf : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f1ae203b91145d5a5bbe3ef13d3a7b6 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23eeb38ad708467d980bd0a7e850377d : SUCCESS in 3m 36s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/940e8661e5724f9fa3460fc445fb98d9 : SUCCESS in 6m 31s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/41aa9fb703804cef8410c80b0eb401d7 : SUCCESS in 1h 41m 41s","accounts_in_message":[],"_revision_number":1},{"id":"fb6856088e362d64347a739822b006d40d3922be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 18:18:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edb19bbc2e294eb88f6a147f45f84274 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b36c1529f15343f60ad64a04b17913ad3e85b61f","revisions":{"b36c1529f15343f60ad64a04b17913ad3e85b61f":{"kind":"REWORK","_number":1,"created":"2019-10-03 12:11:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/686387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/87/686387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/686387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/686387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/686387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/87/686387/1"}}},"commit":{"parents":[{"commit":"7734ac1376a1a9285c8245a91cf43599358bfa9d","subject":"Merge \"Align message serialization between drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7734ac1376a1a9285c8245a91cf43599358bfa9d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-10-03 12:10:28.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-10-03 12:10:28.000000000","tz":120},"subject":"Fix spacing in help message","message":"Fix spacing in help message\n\nChange-Id: Ie622b8a39514f87b60ff2e801843c5e3c06106e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b36c1529f15343f60ad64a04b17913ad3e85b61f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b36c1529f15343f60ad64a04b17913ad3e85b61f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
