)]}'
{"id":"openstack%2Foslo.messaging~691819","triplet_id":"openstack%2Foslo.messaging~master~I65f57973b66f353a1e6fef3d0bf7793f821c1c5c","project":"openstack/oslo.messaging","branch":"master","topic":"bug/1850259","hashtags":[],"change_id":"I65f57973b66f353a1e6fef3d0bf7793f821c1c5c","subject":"Remove unused variable WAKE_UP","status":"MERGED","created":"2019-10-29 08:04:39.000000000","updated":"2019-10-29 20:48:17.000000000","submitted":"2019-10-29 20:46:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691819-1572382013969-771dd57b","meta_rev_id":"aa4f9000e013984c81ad2e7963491f2be05608eb","_number":691819,"virtual_id_number":691819,"owner":{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"date":"2019-10-29 20:08:37.000000000","_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-29 20:46:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-10-29 17:46:46.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},{"value":2,"date":"2019-10-29 11:05:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-10-29 17:46:46.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"date":"2019-10-29 16:21:24.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2019-10-29 16:21:30.000000000","_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 11:05:20.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-10-29 16:21:24.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-10-29 16:21:30.000000000","updated_by":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"reviewer":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"state":"REVIEWER"},{"updated":"2019-10-29 17:46:46.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-10-29 20:46:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c585872815bb0ef75b492cbd4003c7a555c0622","author":{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},"date":"2019-10-29 08:04:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63add0ce48ca8add07805eab0e4526c8cf86f094","author":{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},"date":"2019-10-29 08:06:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d862adcdd3491b91ba847d022164346b75f543a2","author":{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},"date":"2019-10-29 08:09:42.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"50cc5c51967f03eef21b15be7c2d416cbe82c6d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 10:28:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/a28d3e52af8e4f1e98beb0180cada7c9 : SUCCESS in 1h 35m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d20a2bddbf754979b973890b283c3b8b : SUCCESS in 1h 26m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c57b5667ddc64efd85ad6edcfce25728 : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bdbb90c333e48018f5f76180bc4b82a : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/277be678eb5a4146bf141e3e61a5c99f : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a545a34d2bfa46758e998548c7f839e9 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05db7e9129374019a2c73aa425983f39 : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f5fe1b77fce421e9c26f50500146a92 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb5ae343f5947eeaa5a00f8c7ca497b : SUCCESS in 3m 31s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/a291ce0e95f9475a96d20163de3927fd : SUCCESS in 5m 21s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/25f5251bcb6647368f77e759fd4c9c97 : FAILURE in 11m 41s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/1a30ac73dee14b91b341651f6f643287 : SUCCESS in 5m 12s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/0b896abc14304ad9b85fafbbe738784f : SUCCESS in 5m 45s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/dddd378c920445dba5ca8b0ea0be693b : FAILURE in 12m 42s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/6f00a56026504c5aa1759171d974d0d1 : SUCCESS in 5m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e9089c5bb6bc48fbaa5aaf9450d9b7df : SUCCESS in 1h 37m 41s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/978295c58b3e48ce9037389c6bacba65 : SUCCESS in 2h 00m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/f0d00eca75a443e9976b72081f269c5a : SUCCESS in 1h 44m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/bf2757b7fb81466fa334ca581f1f5d37 : SUCCESS in 1h 30m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/c24ab24faba44fe99d90a2f73b5ac32c : SUCCESS in 2h 15m 14s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/e1b1a9325e5d479ab00f4e3adbb88935 : SUCCESS in 1h 06m 30s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/75b16adf99234f2b9ca064cef66511b6 : FAILURE in 1h 22m 10s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/97a116253f774a54bca17b890aa3215a : FAILURE in 38m 31s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/9c0211f82a25437da9f026441d61846e : FAILURE in 45m 07s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/958dc05e49f342869e219789c65abf8a : FAILURE in 48m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"369e7a7ebe1e0bc0de8f94872a49c7f35b6593a4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-29 11:05:20.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM\n\nPEP8/linters seem happy... it\u0027s weird they doesn\u0027t detect this one\nNo other occurances exists in the whole project and nobody seems to use it in the whole openstack...\n\nhttps://github.com/search?q\u003dorg%3Aopenstack+WAKE_UP\u0026type\u003dCode","accounts_in_message":[],"_revision_number":3},{"id":"d36eac372218de53a4bb0b816abb873423dbef52","author":{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},"date":"2019-10-29 12:41:00.000000000","message":"Patch Set 3:\n\nThat\u0027s true. \n\nMy local \u0027tox -e pep8\u0027 didn\u0027t report any problem, either. I\u0027m afraid to find unused variable we need some other tools. Like it\u0027s explained in this post https://stackoverflow.com/questions/693070/how-can-you-find-unused-functions-in-python-code","accounts_in_message":[],"_revision_number":3},{"id":"7599aa2788a1ff0906cf2385a9d583c160447c08","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-10-29 17:46:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nI\u0027ve added a couple of our messaging folks to this, but looking at the commit where this was added, it looks like it was probably just missed as part of the cleanup to stop using it: https://github.com/openstack/oslo.messaging/commit/aebe53f242f40df1b77c61ae60e2ba6a575a1329 In that commit it was referenced in a number of places, and since it isn\u0027t now I expect it\u0027s safe to remove.\n\nAs far as why this doesn\u0027t fail pep8, we do ignore some flake8 tests in this project so it\u0027s possible we just aren\u0027t checking that.","accounts_in_message":[],"_revision_number":3},{"id":"5048bc76b184377fe908f7252eaa290927e8f7cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 17:46:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e0f9d06a43ad84b905a0f57b0ae986b67036c69e","author":{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},"date":"2019-10-29 20:08:37.000000000","message":"Patch Set 3:\n\n\u003e I\u0027ve added a couple of our messaging folks to this, but looking at\n \u003e the commit where this was added, it looks like it was probably just\n \u003e missed as part of the cleanup to stop using it: https://github.com/openstack/oslo.messaging/commit/aebe53f242f40df1b77c61ae60e2ba6a575a1329\n \u003e In that commit it was referenced in a number of places, and since\n \u003e it isn\u0027t now I expect it\u0027s safe to remove.\n \u003e \n \u003e As far as why this doesn\u0027t fail pep8, we do ignore some flake8\n \u003e tests in this project so it\u0027s possible we just aren\u0027t checking\n \u003e that.\n\nHi Ben,\n\nwhere did you mean that in some places we do ignore flake8?\n\nIn the tox.ini, the pep8 env looks like below:\n\n[testenv:pep8]\nbasepython \u003d python3\ncommands \u003d\n  flake8\n  # run security linter\n  bandit -r oslo_messaging -x tests -n5","accounts_in_message":[],"_revision_number":3},{"id":"37ae85fb15ac7d85674648d7bc5c6cb36c54ab52","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-10-29 20:16:21.000000000","message":"Patch Set 3:\n\nThere\u0027s a separate flake8 configuration section that does the skips: https://github.com/openstack/oslo.messaging/blob/c01b03e87c66e964100d3d56d91b0f5f045e6190/tox.ini#L99","accounts_in_message":[],"_revision_number":3},{"id":"fd53711f53bc4346ccc4be33aff27e1203d13d13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 20:46:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5161bce79d5979cb93a5facc9742e41ee3dec8c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 20:46:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e2d4173c0dcc457cb47c86dafda116cb : SUCCESS in 1h 57m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b627f917f2b8485f9775803d93f0f5c9 : SUCCESS in 1h 38m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b7cc98f832b43c0ae1c4c6f965168c3 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e6c8ab6b4704b39a475fe81d4d33a43 : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae4e76f608bd4eaa94673c21e0866e4e : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c90c3adb67b84893af580cfbdc41d6a5 : SUCCESS in 6m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/339526bfe33241cb96a3c0d953b5d3e7 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c22c30909b459b9837bc354c4ec9e5 : SUCCESS in 3m 38s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/bca746c312964a87a5f304a15e49357c : SUCCESS in 5m 34s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bebe86b91158483cb29b2c3637ae9db1 : SUCCESS in 1h 47m 11s","accounts_in_message":[],"_revision_number":3},{"id":"5c551cb746a92e6e1a2ed87881767e861e0db609","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 20:48:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e854101f5cee480bbaeabc24103d488b : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6bca848f5b272149bb32353a62f9d37108fcbe15","revisions":{"a3cf6a688ee857ab7be7395069555f06b60cb9e2":{"kind":"REWORK","_number":1,"created":"2019-10-29 08:04:39.000000000","uploader":{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},"ref":"refs/changes/19/691819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/19/691819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/691819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/691819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/691819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/19/691819/1"}}},"commit":{"parents":[{"commit":"c01b03e87c66e964100d3d56d91b0f5f045e6190","subject":"Merge \"Update master for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c01b03e87c66e964100d3d56d91b0f5f045e6190"}]}],"author":{"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","date":"2019-10-29 08:03:33.000000000","tz":60},"committer":{"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","date":"2019-10-29 08:04:36.000000000","tz":60},"subject":"Remove unused variable WAKE_UP","message":"Remove unused variable WAKE_UP\n\nChange-Id: I65f57973b66f353a1e6fef3d0bf7793f821c1c5c\nFix: #1850259\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a3cf6a688ee857ab7be7395069555f06b60cb9e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a3cf6a688ee857ab7be7395069555f06b60cb9e2"}]},"branch":"refs/heads/master"},"0e4476f23cecd50dad14cf5358920066ef5cda3e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-29 08:06:30.000000000","uploader":{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},"ref":"refs/changes/19/691819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/19/691819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/691819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/691819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/691819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/19/691819/2"}}},"commit":{"parents":[{"commit":"c01b03e87c66e964100d3d56d91b0f5f045e6190","subject":"Merge \"Update master for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c01b03e87c66e964100d3d56d91b0f5f045e6190"}]}],"author":{"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","date":"2019-10-29 08:03:33.000000000","tz":60},"committer":{"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","date":"2019-10-29 08:05:56.000000000","tz":60},"subject":"Remove unused variable WAKE_UP","message":"Remove unused variable WAKE_UP\n\nChange-Id: I65f57973b66f353a1e6fef3d0bf7793f821c1c5c\nFixes-Bug: 1850259\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0e4476f23cecd50dad14cf5358920066ef5cda3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0e4476f23cecd50dad14cf5358920066ef5cda3e"}]},"branch":"refs/heads/master"},"6bca848f5b272149bb32353a62f9d37108fcbe15":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-29 08:09:42.000000000","uploader":{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},"ref":"refs/changes/19/691819/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/19/691819/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/691819/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/691819/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/691819/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/19/691819/3"}}},"commit":{"parents":[{"commit":"c01b03e87c66e964100d3d56d91b0f5f045e6190","subject":"Merge \"Update master for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c01b03e87c66e964100d3d56d91b0f5f045e6190"}]}],"author":{"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","date":"2019-10-29 08:03:33.000000000","tz":60},"committer":{"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","date":"2019-10-29 08:09:17.000000000","tz":60},"subject":"Remove unused variable WAKE_UP","message":"Remove unused variable WAKE_UP\n\nChange-Id: I65f57973b66f353a1e6fef3d0bf7793f821c1c5c\nCloses-Bug: 1850259\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6bca848f5b272149bb32353a62f9d37108fcbe15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6bca848f5b272149bb32353a62f9d37108fcbe15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
