)]}'
{"id":"openstack%2Foslo.messaging~694686","triplet_id":"openstack%2Foslo.messaging~master~I4b72d8feb85c2b9b4657510c356cd21e22fe40c2","project":"openstack/oslo.messaging","branch":"master","topic":"amqp","hashtags":[],"change_id":"I4b72d8feb85c2b9b4657510c356cd21e22fe40c2","subject":"Make sure minimum amqp is 2.5.2","status":"MERGED","created":"2019-11-17 10:59:43.000000000","updated":"2019-11-18 19:55:15.000000000","submitted":"2019-11-18 19:53:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694686-1574106838875-fe519b9b","meta_rev_id":"a0664731541c120d5ec8f78c9c2c27e4859c3e3e","_number":694686,"virtual_id_number":694686,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-18 19:53:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2019-11-18 15:14:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":2,"date":"2019-11-18 15:12:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":1,"date":"2019-11-18 15:14:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-17 11:02:01.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-17 13:29:59.000000000","updated_by":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"reviewer":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"state":"REVIEWER"},{"updated":"2019-11-18 15:12:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-11-18 15:14:05.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-11-18 19:53:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34c87b936e1163c1ef626dab4156996821e86a4e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-17 10:59:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b2b9a508743aa825bdb7d65b1c3d088b3689fac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 12:55:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d122200f5b294a709bbd3b35682fdad7 : SUCCESS in 4m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3a60b1cb49f74853acfe0ec17b9a8462 : SUCCESS in 1h 36m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab18a161f666404d8dd35dc1f7d562b8 : SUCCESS in 1h 21m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df70901f1476424481ff66b5243585e9 : SUCCESS in 8m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6005e8955545416b9887d8d9f753820c : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c14c45024f0434f8a3e2047e8ecd5e6 : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2c24b160b054e968651a5da184e8982 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3887fe01ff6d43a980306fa899c2f732 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6830216e0f44631ad0b8d46935bceea : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31a0397b9c0342b29a22d7af51e09a54 : SUCCESS in 2m 57s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/d5fa517e7fb441e9b2dfbef531e5429c : SUCCESS in 4m 52s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/a0afad5cb1364a9c8ea75748a1d1ddb7 : FAILURE in 11m 12s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/bbf7993bd50443a3a530d7097c04e978 : SUCCESS in 4m 56s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/c16cea51a050485aaf7be80a860ac36c : SUCCESS in 5m 07s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/7a5cfafb6f224a5cb5933600f9e1f359 : FAILURE in 11m 15s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/436d8f5666da443192dff317b1694658 : SUCCESS in 4m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d1833ad6454c42378450cc75f309d00a : SUCCESS in 1h 36m 00s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c60ab8b677454d10809fb5f5c6510fa9 : FAILURE in 30m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/fa30148e210c4172a4267282a83a5cd6 : FAILURE in 36m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/55050e36bd3d4b1a942eb2bb86a19dc9 : SUCCESS in 1h 39m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/92c52ccc532b48309603125c7ba96db5 : SUCCESS in 1h 50m 44s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f8f4c199d15641d1b30e1450c8807563 : SUCCESS in 1h 03m 50s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/22469917287448abbbdff19f8f9bf84b : SUCCESS in 1h 13m 00s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/0fd5532bc7184b9692e40ad5b936bc9c : FAILURE in 47m 05s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/c2ab4f44e1a54cc4ab66f6817150cda0 : FAILURE in 43m 19s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/5e0327a09fa94c839a4661ab77f32895 : FAILURE in 45m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01d91e99a75653dbf7ea3e0a21e1e2a3e0c1893e","author":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"date":"2019-11-17 13:29:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1c326b30283c519df2d352ca020db02c970b6398","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-11-18 15:02:09.000000000","message":"Patch Set 1:\n\nIs there a link to the particular bug that was fixed in amqp that necessitates this bump?  If so can that be included in the commit message?  thanks","accounts_in_message":[],"_revision_number":1},{"id":"2ce5b870118f013cb7838465482239f92d1d5ac5","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-18 15:05:37.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b40e49629e18b1ff4742614de0b72153ed170122","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-11-18 15:12:07.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks for these changes.\n\n(Originally discussed there \u003d\u003e https://review.opendev.org/#/c/693704/ )","accounts_in_message":[],"_revision_number":2},{"id":"11d3e4d3415a8ca0a0d020cf27b6d8e2e14fd1a1","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-11-18 15:14:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"88060c78d48fbc2e594a960f6fb411156d3db33e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-18 15:17:47.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 9c5e04ac77f2cba9e4e73a22083144343a0fdd45","accounts_in_message":[],"_revision_number":2},{"id":"562ce2b83da0c12dea764b8ded30c3a37f501bbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 17:36:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/005ed0cc18954fea90c5817f9f13acf6 : SUCCESS in 3m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3aa0f6dd586946f4b68269ef8b95912e : SUCCESS in 1h 51m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eef62f22f3904760b8bc684f594e7d8c : SUCCESS in 1h 40m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0aa6a8a4c70244ddad0d0cf99b6fa71c : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/593c64d8ca264114afc547be6f83fbe2 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4587dc4c0123428dacb3f1e3ad17d042 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6914a03364c4bd38f1e498b73a25999 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3958a374ce0462382429f74431f2a58 : SUCCESS in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22fcb667f84741429081d5fbc0f0a0ee : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/195eb9a8eceb4e0a9757b42eba4d9c79 : SUCCESS in 3m 56s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/82016dd3f0dc4e2986423253b19a5fcf : SUCCESS in 5m 17s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/df55ed30790345389cd0729a896a7bea : FAILURE in 12m 20s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/a643e09fffd547e3a7e22304b3f2f948 : FAILURE in 5m 23s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/0d36680b26f24a0baee466927d13c18e : SUCCESS in 5m 44s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/2d7b77cf78f6430eb17656a10532e8e6 : FAILURE in 11m 46s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/6a848bfe45724c3a9d04c6165908ff5d : SUCCESS in 5m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/dfb62ad05e844dda858e29815f1955ea : SUCCESS in 1h 35m 03s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/8b4f09168de94d07b7b6261bf5d6c314 : FAILURE in 34m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/f944d46975424f52b76b99fe4a576274 : FAILURE in 43m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9c87bd7275f049bfb2f192e4e5d10896 : SUCCESS in 1h 47m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/2869eab0e36f4806abce06e163bba37f : SUCCESS in 2h 09m 59s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/141837d981dc465ba4c04890b6792c47 : SUCCESS in 51m 22s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/4f2813e65fb3411a813db3daa8905d82 : SUCCESS in 1h 05m 53s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/ad0f478675664710bcc720f3aa870aba : FAILURE in 58m 41s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/0a6ba804cdb3486eaff66b8c8bd9de50 : FAILURE in 42m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/4a07ac4573be409e84d4a7e33eefbf7e : FAILURE in 52m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5e7ee83028698bf9aee5a1051123c34912c929d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 17:36:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bd218fc9775e8a9b3c6c4aa5705b3a9b9219f2d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 19:53:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c3b4c0f1bc62474ab847b8aea2bb90ff : SUCCESS in 3m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/484fc248a70a42259598fc2b49172196 : SUCCESS in 2h 04m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14942adfaee149f1bb0f27a87d8a9825 : SUCCESS in 1h 34m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a13e2c7b91f94e45ba7f258dea483471 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/648eff56483048f8abd785fb43fe4f4f : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/152f4e845ead4081918f17e562923bb4 : SUCCESS in 9m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e00f9774444347ea8833215b986e1da0 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db260c38bd854b4ba928c3c4e27dfc9b : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4113edba21b845d7a9232a66d49563c0 : SUCCESS in 4m 45s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/b49a871f5b3b4929ab06dc5c0be97c60 : SUCCESS in 5m 36s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/40ad63862a424d62bd6d7d9793445ca7 : SUCCESS in 1h 43m 30s","accounts_in_message":[],"_revision_number":2},{"id":"42647b9cefb06b91921329122d6bc88d879fecd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 19:53:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"716bdc902b64b504c13d0168fe854c59b05764f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 19:55:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f2468dc145148b89a8d941ec712ee07 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c8d6fed762607fad8f73ade531cb06d6b919b096","revisions":{"292c3c863f3a0c00e4c6e2e798f97c8640c283b1":{"kind":"REWORK","_number":1,"created":"2019-11-17 10:59:43.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/86/694686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/86/694686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/694686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/694686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/694686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/86/694686/1"}}},"commit":{"parents":[{"commit":"da3c3e7d1d79b71c87a7194cf4cb677d12efbfa9","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/da3c3e7d1d79b71c87a7194cf4cb677d12efbfa9"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-17 10:59:06.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-17 10:59:06.000000000","tz":120},"subject":"Make sure minimum amqp is 2.5.2","message":"Make sure minimum amqp is 2.5.2\n\namqp fixed a bug in 2.5.2 that is needed\nalso update kombu to support amqp 2.5.2\n\nChange-Id: I4b72d8feb85c2b9b4657510c356cd21e22fe40c2\nCloses-bug: #1852058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/292c3c863f3a0c00e4c6e2e798f97c8640c283b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/292c3c863f3a0c00e4c6e2e798f97c8640c283b1"}]},"branch":"refs/heads/master"},"c8d6fed762607fad8f73ade531cb06d6b919b096":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-18 15:05:37.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/86/694686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/86/694686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/694686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/694686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/694686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/86/694686/2"}}},"commit":{"parents":[{"commit":"da3c3e7d1d79b71c87a7194cf4cb677d12efbfa9","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/da3c3e7d1d79b71c87a7194cf4cb677d12efbfa9"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-17 10:59:06.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-18 15:05:33.000000000","tz":0},"subject":"Make sure minimum amqp is 2.5.2","message":"Make sure minimum amqp is 2.5.2\n\namqp fixed a bug in 2.5.2 that is needed\nalso update kombu to support amqp 2.5.2\n\nsee \nhttps://review.opendev.org/#/c/693704/\nhttps://github.com/celery/py-amqp/commit/86cb254dceab75e0240b4fa6b97249de70036a4b\n\n\nChange-Id: I4b72d8feb85c2b9b4657510c356cd21e22fe40c2\nCloses-bug: #1852058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c8d6fed762607fad8f73ade531cb06d6b919b096"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c8d6fed762607fad8f73ade531cb06d6b919b096"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
