)]}'
{"id":"openstack%2Foslo.messaging~697971","triplet_id":"openstack%2Foslo.messaging~master~I5dbfb32b95a91dd50edaeebd7f35cf5aab442888","project":"openstack/oslo.messaging","branch":"master","hashtags":[],"change_id":"I5dbfb32b95a91dd50edaeebd7f35cf5aab442888","subject":"Handle msg_id attribute exception","status":"ABANDONED","created":"2019-12-09 10:41:43.000000000","updated":"2019-12-10 08:42:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a8dc8eae7f48c2c4bb9af595c978b85ca95fc9f9","_number":697971,"virtual_id_number":697971,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-09 21:23:40.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"date":"2019-12-09 17:59:19.000000000","_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-09 15:06:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"date":"2019-12-09 19:42:21.000000000","_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 15:06:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-09 19:42:21.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2019-12-09 21:23:40.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"}],"messages":[{"id":"8fcea440f8520378213050000032ad908252901e","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2019-12-09 10:41:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02f23a9b75015efba8fad4fd55f30b3f198b6c82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 12:20:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/986346a978d744e89e728a316eee6ca4 : SUCCESS in 1h 33m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fda786b7625d44bdb7a265feb671b25c : SUCCESS in 1h 26m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84d7a8ebacbf4173b30b499919b827dd : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b2c501bdc94437eb177ad2f073c80d1 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cf3f1fc952f4e378602382ca25bc87b : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e85c1ef016948dfbf0bedf0a6d2b356 : TIMED_OUT in 41m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cd09c833a25412288effb4131029948 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3adf87426c6545699efeeaff7710586b : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43be9ef736aa4358bb5f685d03f9149e : SUCCESS in 3m 39s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/d6d06e10a5414816af9c1e80082edebe : SUCCESS in 5m 41s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/d502929b7b314c449c852586046e0234 : FAILURE in 11m 41s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/e72e6ecddc4546c79c0dad197d0bf8cb : SUCCESS in 5m 54s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/040c11f5ef964cd797c429789279ea65 : SUCCESS in 7m 08s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/4036de7d1f7a43c092a1e98af02b8b03 : FAILURE in 11m 35s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/c3b0ad0ffad446b9b795495d22cea736 : SUCCESS in 5m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8408c63385864fa1a712352f8def9d4f : SUCCESS in 1h 32m 51s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3ff1f1d590f5452eac29d86bbfb8abf5 : SUCCESS in 1h 29m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/4e82589c017a49d3982919ced5c06940 : FAILURE in 11m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f142476f426449e2a652d87b3d58b69b : SUCCESS in 1h 24m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/24f22b4659584fecb1d31e351652b47a : FAILURE in 11m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/10c4aeb0d9224b6f9cb09bb19b210e87 : SUCCESS in 1h 02m 22s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/452e1d0e64a24a83b9fdb9420810fdda : SUCCESS in 1h 28m 35s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/8baa84fc1da64069977b2491f497ab5c : FAILURE in 42m 19s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/26aa317c201a4a6f9f3ed7b3b8fdfcab : FAILURE in 42m 33s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/3111e85b3ee2416f9ee5c4aceba17ea5 : FAILURE in 44m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"018cc98172f48fc87b94771744378ae1ce20774f","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2019-12-09 13:19:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ea1c7b4c7be562309c0828f96a57a2b921e69f98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 15:06:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/7edb99b7921940f598f9b23dc0401071 : SUCCESS in 1h 21m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28058ef046ab447492c15aec1ada3013 : SUCCESS in 1h 42m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b16e2c6d790d4b4ab03437ae76d4df4a : SUCCESS in 6m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9809b50424164f63b60ad2dbbd1b59e5 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eeb5dbfb58c41c8b8a18f5366eeeee7 : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b36b368a7be7449c933696d6553622ce : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9baa3d71921f4bb39ca981bc8349a427 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/062fc15493c94cec8935d6efd714a859 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c52cf9fd43684894ba82071b9b5118b9 : SUCCESS in 3m 24s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/b2223594c85e423baab6da9f10578eff : SUCCESS in 5m 40s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/fb19226ed8be4b9c8b44cd85e033e08f : FAILURE in 11m 27s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/c2a883548dd94233a025a0ae9f5c6e40 : SUCCESS in 5m 22s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/5b591da424554f4882e095972b611be4 : SUCCESS in 6m 26s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/5c74e49728c44287af346c1a4c81cb89 : FAILURE in 12m 13s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/5a4512da6c9749b491afefca45f076b2 : FAILURE in 5m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/21b11ea6778642bb937263f119626cc5 : SUCCESS in 1h 33m 06s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/5bda6776b7924c70ad562a7a89690f82 : SUCCESS in 1h 45m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/8b694bc720b94ba9910915aef4648f8f : FAILURE in 8m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a51e56eef1fe4baf8688aca1830da765 : SUCCESS in 1h 27m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/c5b29c265bcb4980b672088ae52eeb72 : FAILURE in 13m 00s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/1518b859560f4c0e93c8dcd43cbbd342 : SUCCESS in 1h 05m 35s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/849614e70dac4a228fda5277fc19af32 : SUCCESS in 1h 17m 21s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/dd00750c914948f68e5fdf7f7661eed2 : FAILURE in 42m 03s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/c8c83cd629684b48a93cb672c4a0324b : FAILURE in 43m 06s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/d05fcfeb2b4547de85342c41a1f33086 : FAILURE in 43m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e599b92a47c5ace503b570cc9f97d049733efb0c","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-12-09 17:01:48.000000000","message":"Patch Set 1:\n\nI\u0027m surprised this fixes the problem since there are references to message.msg_id further along in the code.  In fact the AMQPIncomingMessage class constructor unconditionally sets up the msg_id attribute.\n\nThis needs to be debugged further - can you get a traceback of the failure as well as verify the class name of the message object?","accounts_in_message":[],"_revision_number":1},{"id":"fddf50d8690b2ffcfa74ac98dbd0e28f9e814f2b","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2019-12-09 17:59:19.000000000","message":"Patch Set 1:\n\n\u003e I\u0027m surprised this fixes the problem since there are references to\n \u003e message.msg_id further along in the code.  In fact the\n \u003e AMQPIncomingMessage class constructor unconditionally sets up the\n \u003e msg_id attribute.\n \u003e \n \u003e This needs to be debugged further - can you get a traceback of the\n \u003e failure as well as verify the class name of the message object?\n\nIndeed, when troubleshooting, I checked the class constructor as well and noticed it was \u0027None\u0027. I digged further into it but tbh i don\u0027t know the reason.\nBut, at the end, catching this exception fixes the issue. You can see in [1] that the job fails since the introdution of the log message. And it\u0027s successful with this patch.\n\n[1] https://zuul.openstack.org/builds?branch\u003dmaster\u0026job_name\u003doslo.messaging-src-dsvm-full-amqp1-hybrid","accounts_in_message":[],"_revision_number":1},{"id":"5ee32ee32b8fe460d185c908688e9066d6253d0c","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-12-09 21:23:40.000000000","message":"Patch Set 1:\n\nOh yeah - I see it now.  That msg_id attribute is specific to the rabbitmq incoming message class.\n\nThe change to fix:\n\nhttps://bugs.launchpad.net/oslo.messaging/+bug/1847747\n\nended up accessing a field (msg_id) that rabbit added as an extension to the base class.\n\nI\u0027ve opened up \n\nhttps://bugs.launchpad.net/oslo.messaging/+bug/1855775\n\nto fix this.\n\nFor now I\u0027ll propose reverting the breaking change:\nhttps://review.opendev.org/#/c/698090/","accounts_in_message":[],"_revision_number":1},{"id":"a8dc8eae7f48c2c4bb9af595c978b85ca95fc9f9","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2019-12-10 08:42:04.000000000","message":"Abandoned\n\nThank you Ken for clarifying that. Much better to revert and open up a bug, so I abandon this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31bd8ab07f6856e463a419e0210610c111a2483b","revisions":{"31bd8ab07f6856e463a419e0210610c111a2483b":{"kind":"REWORK","_number":1,"created":"2019-12-09 10:41:43.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/71/697971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/71/697971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/71/697971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/71/697971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/71/697971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/71/697971/1"}}},"commit":{"parents":[{"commit":"37fabdd895f03d51509aa16e6ae4b344fb909287","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/37fabdd895f03d51509aa16e6ae4b344fb909287"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2019-12-06 14:31:44.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2019-12-09 10:41:41.000000000","tz":60},"subject":"Handle msg_id attribute exception","message":"Handle msg_id attribute exception\n\nDownstream at RDO, with version 10.4.0, we hit an\nissue with Nova not being able to restart properly.\nAfter some troubleshooting, we figured out that the\nroot cause is the log message [1].\n\nSome incoming message does not have the msg_id\nattribute.\n\n[1] https://github.com/openstack/oslo.messaging/blob/37fabdd895f03d51509aa16e6ae4b344fb909287/oslo_messaging/rpc/server.py#L157\n\nChange-Id: I5dbfb32b95a91dd50edaeebd7f35cf5aab442888\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/31bd8ab07f6856e463a419e0210610c111a2483b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/31bd8ab07f6856e463a419e0210610c111a2483b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
