)]}'
{"id":"openstack%2Foslo.messaging~698090","triplet_id":"openstack%2Foslo.messaging~master~I17a448a768b544482b375b0076889db989e03e8c","project":"openstack/oslo.messaging","branch":"master","topic":"bug/1855775","hashtags":[],"change_id":"I17a448a768b544482b375b0076889db989e03e8c","subject":"Revert \"Add RPC incoming and reply log\"","status":"MERGED","created":"2019-12-09 21:22:18.000000000","updated":"2019-12-12 15:58:32.000000000","submitted":"2019-12-11 16:25:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698090-1576081514724-9875c5dd","meta_rev_id":"8b171dc439359abddee36957833bd421494cf973","_number":698090,"virtual_id_number":698090,"owner":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-11 16:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-12-09 21:38:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-12-11 14:13:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-12-11 14:16:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":1,"date":"2019-12-10 09:17:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-12-11 14:16:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 21:38:18.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-12-10 09:17:10.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2019-12-11 14:13:25.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-12-11 14:16:36.000000000","updated_by":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"reviewer":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"state":"REVIEWER"},{"updated":"2019-12-11 16:25:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"70407a41273c0e5ceb65aa2c2ba81b85c7f5a99a","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-12-09 21:22:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48cce4d18f0ea823379501bdfdd5402006e7433b","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-12-09 21:38:18.000000000","message":"Patch Set 1: Code-Review+2\n\nOops. Since this is just a revert of a breaking change, we can probably merge it without a second +2 if it passes CI.","accounts_in_message":[],"_revision_number":1},{"id":"1e8e8ca9afd0c80792ded0dbef375c47866edc3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 23:06:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/7a33d4cd69124bf08ef79818aa22cf7d : SUCCESS in 1h 28m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bdf1b47be5f4698818cb9d60d575709 : SUCCESS in 1h 34m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbaf325f6a3b422f907d84acb07fe2ae : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e7c114f94cf49ff8f77126c5a7f7be8 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03eb6a9e788e4a42bfbaf1f128f7e7be : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/670074d7c535477e907db494880fb3f5 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fafd55ead0a4e009a694af06f2318e8 : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7803c61325194443828b564c5f3e7a1c : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12457c9ee194414daa800fc1cd116118 : SUCCESS in 4m 45s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/d014990384254a06bafd8b62bda2d653 : SUCCESS in 6m 07s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/b8651cd28131429cb8f17237eb09a9f4 : FAILURE in 11m 57s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/46d55e72500945ccaf3fe5c268982cea : SUCCESS in 7m 52s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/cda6a3108b8c460b93e0be3dfd4a6bdb : FAILURE in 9m 00s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/d0ac71480a4946cb9aee72b9561e0b1c : FAILURE in 14m 30s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/fa97356b028541689d4ca1ea18d1aaf7 : FAILURE in 7m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9d0df2ca72b84656a47459c56420ad46 : SUCCESS in 1h 40m 02s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/cd086d27c0ae4cbab65c92a1931eeee1 : SUCCESS in 1h 23m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/bc0e0fd364294ea6aac2406e5e79afd8 : FAILURE in 13m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e651776634d84462b4fc532a8eaf7f58 : SUCCESS in 1h 34m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/77affde2b8284689a66cae3054de4c80 : FAILURE in 11m 50s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/1ea737d730234d1ca5eade999eeb95d9 : SUCCESS in 1h 09m 37s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/6158bc62a78245ddadcd99a2046a0b67 : SUCCESS in 1h 32m 10s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/e5d242f20fdb4865a34626fbdbbb1f9f : FAILURE in 58m 24s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/b01d954c9f844f65b5baa10a5ae0c1e0 : FAILURE in 49m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/af3c4491dd6d40008f356297b8892ef9 : FAILURE in 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4991830ddb8c5ee1f5243c2bf3e2961d65c8b5d1","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2019-12-10 09:17:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fdba01e985f28c752fb443179c61c246aadff7ef","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-12-11 14:13:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b535f60232e59068ab6d5444e7aee535f88ddb3b","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-12-11 14:16:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5b34a04fd08cd533ff7db4c36a111ecb6443a6b9","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-12-11 14:16:36.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"626a14f1da6a0f7c44311202ad821d789d81b066","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 14:16:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7522a5663f2488e784c8eb77520a6534d64d9179","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 16:25:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d54230e89d58b86fff0e0b4ae588484b2dd842ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 16:25:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2d2714b87cf0465eaa7b4fa9cbede878 : SUCCESS in 2h 07m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9976945594e4f0a9cebccdda6f6209f : SUCCESS in 1h 30m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96547fc5e30946d89a97ef3988b8ddff : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73812de4b7594b53ad805475c72f23a6 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c1674cf371a4af89990ff6582ee786f : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a27126e85f67497380a042ccf1cb85d4 : SUCCESS in 7m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd69aee9355944719d559633ea30ab21 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c46ef0421db2496e8e3a276f4f36f6ca : SUCCESS in 3m 49s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/9ae31ecea2884db1a90562a7e4d85210 : SUCCESS in 6m 11s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/59f6e9b7cf554af99af1d48d74727995 : SUCCESS in 1h 38m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4d7dc838550f03cdb09aba2f5138a68af8ba7ad7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 16:26:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33bf6d8a2ef84509917070367ccfaee5 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1},{"id":"d0a6ee4b0d9109756b21fbcbe7543cbbaa9a9fb5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-12-11 16:35:34.000000000","message":"Patch Set 1:\n\nCan we also get a bug fix releaese for this","accounts_in_message":[],"_revision_number":1},{"id":"19103d970a186f45126e4a22dbad7ba8ff0ce2a3","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-12-12 15:58:32.000000000","message":"Patch Set 1:\n\nOslo master branches are released ~weekly, so this should get picked up in the next set of those.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"859e0d4eaa45d75eb822f4a6be46c547187c6358","revisions":{"859e0d4eaa45d75eb822f4a6be46c547187c6358":{"kind":"REWORK","_number":1,"created":"2019-12-09 21:22:18.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/90/698090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/90/698090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/90/698090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/90/698090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/90/698090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/90/698090/1"}}},"commit":{"parents":[{"commit":"37fabdd895f03d51509aa16e6ae4b344fb909287","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/37fabdd895f03d51509aa16e6ae4b344fb909287"}]}],"author":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2019-12-09 21:17:07.000000000","tz":-300},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2019-12-09 21:22:06.000000000","tz":-300},"subject":"Revert \"Add RPC incoming and reply log\"","message":"Revert \"Add RPC incoming and reply log\"\n\nThis broke non-rabbitmq message bus drivers. See:\n\nhttps://bugs.launchpad.net/oslo.messaging/+bug/1855775\n\nThis reverts commit b104f254ab43d5e2589e845eff84b1bd096cb42d.\n\nChange-Id: I17a448a768b544482b375b0076889db989e03e8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/859e0d4eaa45d75eb822f4a6be46c547187c6358"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/859e0d4eaa45d75eb822f4a6be46c547187c6358"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
