)]}'
{"id":"openstack%2Foslo.messaging~700128","triplet_id":"openstack%2Foslo.messaging~master~I13e72ec6cbd006ca6eccfe21f2a0584e86a52669","project":"openstack/oslo.messaging","branch":"master","topic":"base_python","hashtags":[],"change_id":"I13e72ec6cbd006ca6eccfe21f2a0584e86a52669","subject":"tox: Trivial cleanup","status":"MERGED","created":"2019-12-20 09:44:25.000000000","updated":"2020-01-14 14:02:16.000000000","submitted":"2020-01-14 14:00:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700128-1579010447701-46a14a57","meta_rev_id":"db802b0497672524c7ce0dae4dc4de4c74d6310f","_number":700128,"virtual_id_number":700128,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-14 14:00:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2020-01-13 14:36:47.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2020-01-14 09:16:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2020-01-14 11:44:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2020-01-14 11:44:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-13 14:36:47.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-01-14 09:16:06.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-01-14 11:44:57.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-14 14:00:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e17798856bb81aae06c9b03250e71ce593056b2e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-20 09:44:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"380cc6123a06678ac50a76417ccaf59828ea3794","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 11:23:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c6f16c3e84404f9d88d6e137895aec72 : SUCCESS in 1h 30m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2603fc3c58db4a82ade1ddbe679c4e33 : SUCCESS in 1h 28m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84d4223e2f5d43aab8ea421603454f60 : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da03c451172c4462b518e09a1ed9e128 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a794bac813e4180815981b04eb1c960 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ab979341c6342acb4767b6902ef6710 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6abba5c8d09b45038b58ab035efb49e5 : SUCCESS in 7m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9550ff8ae4814c18b9396c38f428ff07 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e294348e05064eb4b32ec37cb327c279 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ec75a9707cc44609d926aef0114e31b : SUCCESS in 4m 24s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/5ea2f01317a04ca5833fb649be425c05 : SUCCESS in 6m 49s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/eb2f4f28d44e460f9018d1c2d99513a9 : FAILURE in 13m 23s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/d7a04e46146345df8fc82cf920b7383a : SUCCESS in 5m 40s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/902df73b9ac14e2fb15695d1c5a8a34f : SUCCESS in 5m 23s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/2285d40bb5c640ce9eda41641399af64 : FAILURE in 12m 01s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/7f929724bf464416a96d18cb484cceef : SUCCESS in 6m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/09280050c6e74dcbbfcd879cc262b555 : SUCCESS in 1h 38m 12s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/209caf9ebdd04de49d2e48e0772c297b : SUCCESS in 1h 25m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/907790e5c145424cb1e00bc0276b042b : FAILURE in 14m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b0ae3ca1cc6b4f18b0d2ef31a6078803 : SUCCESS in 1h 37m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/a954caeb2fdc471f87a05b19589f5f36 : FAILURE in 13m 13s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/534b81dbaf99421b9facecf6289122d9 : SUCCESS in 1h 10m 15s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/99f92d2101db4219a5a1b3354bb06566 : FAILURE in 1h 03m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"952231e6b155588f84657c0fcba8a76b89bde0dc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 12:37:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"adfcd0dc1a90e622f499f97fa9d7fbdb48c2c97e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-08 18:59:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9ad77f3d58f19066988e2294c1a7621d3b572458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 21:51:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/cb882e102a234b32823f79fe1c04ef16 : SUCCESS in 1h 37m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9023320f91ea4a65b1ec04d9212b453f : SUCCESS in 1h 33m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fea1ff580168444084eb4ec6648c95e4 : SUCCESS in 9m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88a1b7b1b90941beba5d2a2150d8b8bd : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cf48c3867ed465d9cee64e2a5307680 : SUCCESS in 6m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbd45150522745f28a0dd034c6f3cd9f : SUCCESS in 9m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddacbbd0ec4c4afd804db90724028640 : SUCCESS in 7m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d066f7bb21f545d0a28d9d0ee06cd3c2 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da03c2bb4c04af29fbb6114e28bed38 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d072d101eee342f6ad3392b675bd9ced : SUCCESS in 5m 28s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/e7af51837bcc45578961488f1ab62262 : SUCCESS in 8m 30s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/7bde3dcf1fb1464c9facf7a57526a9bb : FAILURE in 15m 35s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/69e84b39f7944884b1968179a5c816c2 : FAILURE in 9m 56s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/62634df88e0040aeabb6f6882c3a082a : SUCCESS in 7m 52s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/9736b5ff2a7f4fa788d6786f87fb61ad : FAILURE in 14m 40s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/ed476d4abd1b468180e10052ab89f8cd : FAILURE in 8m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3e608ea6b81346bab70f87bc502ac044 : SUCCESS in 1h 29m 14s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3e9f01c5deb1484cac387e736afc95cf : SUCCESS in 1h 35m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/c9cbc6d10d2242e2bb423d4c83827d60 : FAILURE in 8m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ec24d9ccfbc5496c85aa31eb8e7177ae : SUCCESS in 1h 38m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/b7b96c5166864fa2ae399f26f2186aab : FAILURE in 13m 06s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/131276c2c98d4671a198f0bdb4b1f2dc : SUCCESS in 1h 11m 24s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/813421d021844d84987c72829c4616dd : FAILURE in 1h 14m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a90f363581a81304264d598f728fb5e297e43dba","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-01-13 14:36:47.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"12a9d4c6588bf1783f535b3408acf8ea0436d4c7","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-01-13 15:28:33.000000000","message":"Patch Set 2:\n\nUpdate:  The oslo project has been given the green light to removing python2 support.  If you\u0027d like to simply remove the py27 tox entries entirely that would be fine as well.","accounts_in_message":[],"_revision_number":2},{"id":"d3ece12ac9a81933b29b52e9de768d90d8e82510","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-14 09:16:06.000000000","message":"Patch Set 2: Code-Review+1\n\nlooks good.","accounts_in_message":[],"_revision_number":2},{"id":"427373cd15784da0ed8a298b29d6bd1e703d89ba","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-14 11:44:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"86e74fefb86219093ee773fbd8336a79db5abf8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 11:45:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9f70fb1ffb5b8338cda3f3e50c5446357105545f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 14:00:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c7498317bc1144b6bf44ecad7192d19f : SUCCESS in 1h 28m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38b495fcc87b4621826ea599a89daef3 : SUCCESS in 1h 38m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cf29d5eec5c416183248bf142f264b0 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a26b0d06bc674f6898d84d65efdc57fa : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce90b3752150434c803e116e7a53fa67 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dac8852010974174919fb54326a63cb3 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bb6a37cd74f409b8385198e5bbb5197 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/681a45dc5df34e8faf77d433a3589a38 : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d89f71a7920d4ee2910c9114244fc149 : SUCCESS in 4m 33s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/662a9c217abc4eabbebaf700d304488a : SUCCESS in 5m 54s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c6d0079df8eb4ca7b7a3d9023d984a56 : SUCCESS in 1h 38m 52s","accounts_in_message":[],"_revision_number":2},{"id":"35af71da25699cc0be94c9df948093739293c72e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 14:00:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"650e0ed64cbbf6ae45e2c07f66c255946bf8bc6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 14:02:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af8ef1a1387647a786ebf1c5485c5684 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78cb9fbe759c47babb03762eb53bdf7e : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cc6cf12c3de704c470f15bc93564ef7ecdde5ad3","revisions":{"810a4b8a24728321bca9251214c9b031c50fe5d6":{"kind":"REWORK","_number":1,"created":"2019-12-20 09:44:25.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/28/700128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/28/700128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/28/700128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/28/700128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/28/700128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/28/700128/1"}}},"commit":{"parents":[{"commit":"41a44230434cf9037d57f5e26c1cc73559e198c7","subject":"Merge \"Ignore releasenote cache within git untracked files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a44230434cf9037d57f5e26c1cc73559e198c7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 09:42:49.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 09:42:49.000000000","tz":60},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nMove \u0027basepython\u0027 to the top-level \u0027testenv\u0027.\n\nChange-Id: I13e72ec6cbd006ca6eccfe21f2a0584e86a52669\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/810a4b8a24728321bca9251214c9b031c50fe5d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/810a4b8a24728321bca9251214c9b031c50fe5d6"}]},"branch":"refs/heads/master"},"cc6cf12c3de704c470f15bc93564ef7ecdde5ad3":{"kind":"REWORK","_number":2,"created":"2020-01-08 18:59:45.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/28/700128/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/28/700128/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/28/700128/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/28/700128/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/28/700128/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/28/700128/2"}}},"commit":{"parents":[{"commit":"41a44230434cf9037d57f5e26c1cc73559e198c7","subject":"Merge \"Ignore releasenote cache within git untracked files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a44230434cf9037d57f5e26c1cc73559e198c7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 09:42:49.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-08 18:59:34.000000000","tz":60},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nMove \u0027basepython\u0027 to the top-level \u0027testenv\u0027.\n\nChange-Id: I13e72ec6cbd006ca6eccfe21f2a0584e86a52669\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cc6cf12c3de704c470f15bc93564ef7ecdde5ad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cc6cf12c3de704c470f15bc93564ef7ecdde5ad3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
