)]}'
{"id":"openstack%2Foslo.messaging~700348","triplet_id":"openstack%2Foslo.messaging~master~I0ed3bd53d00f66e4253a60d61e63281c28736d95","project":"openstack/oslo.messaging","branch":"master","hashtags":[],"change_id":"I0ed3bd53d00f66e4253a60d61e63281c28736d95","subject":"tox: Trivial cleanup","status":"ABANDONED","created":"2019-12-22 15:52:38.000000000","updated":"2020-02-20 11:11:57.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"19c283b97a4cc5fde46561832b0b86769d15f1bb","_number":700348,"virtual_id_number":700348,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-23 13:44:32.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"date":"2020-02-20 11:11:54.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-23 15:27:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-23 15:27:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-20 11:11:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"3a8a603e27210e142cee4a24f91a98d1b8539d5b","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-12-22 15:52:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18d242ad2af13e0513c7e3544da5227a2f9bed07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 17:39:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8a2efe58ddf041d8a01d23a9f431c169 : FAILURE in 20m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d3ec56fbc014695bd3f4657918aa730 : FAILURE in 11m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be5895f50d124c6c8bc1dcb8fcf71b24 : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39bb84e12c444e95a0d54fe7841767cc : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f0138f09fe44d1aa3e5e4cd35d20ed : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7ba6123a9a44eba842e24490b19f942 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03de90459f434227bc7479199dfe16e9 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1cd9df9f11234a929ac5744b23512d46 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75808ff4bb904d25be6f99bfc0f699b0 : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b041fbd1617c46de80598891bb09992b : SUCCESS in 4m 14s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/0b9c2688a9524277a4deb777fb3815a7 : FAILURE in 7m 38s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/f9c5b7d2f07244d0aef77aa9e210ed00 : FAILURE in 11m 54s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/e59b265091b446e69b8ef2b8e740be15 : FAILURE in 5m 56s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/f9c87ecd9a6045908b107f4e05892489 : SUCCESS in 5m 41s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/a4d7ef60a6c649e88160db4c43f3d286 : FAILURE in 12m 21s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/58590174e6c443d095a574dd23b263e6 : FAILURE in 5m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e5ed3e8209354316ae67d6d8996206f8 : FAILURE in 20m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3ce2e514f57a4c258b4527853560375c : FAILURE in 20m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/180dcc137f3e43abac7272c83dd43407 : FAILURE in 15m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3dab09736719425ebb3901288786b68f : FAILURE in 19m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/f7cdb6a706654de28f6ecaa0a597ab43 : FAILURE in 11m 44s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/5fc41f0fedb84df4a342da2d2938a6c8 : FAILURE in 51m 19s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/b98513bdb57542b696e34b8e0fa4706c : FAILURE in 1h 39m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d8647f4017500a0ea4497e6970a5b9f3b0f6939","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 12:33:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6f3151780da6ee1f5bab92d1dbd3001be337e69a","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-12-23 13:19:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"03c3fa63ee4d1a586b82a1fb80c0b31ea8df5729","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 13:36:19.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6f0f44c88f14d5f762f55cbe046478171cf2f6ce","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-12-23 13:43:10.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"cf85a8e3ac79d65b886b87bf266437457e34420b","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-12-23 13:43:57.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"3b38e1af05ab6f46b293aa4193f152b747878592","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-12-23 13:44:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c8d8904f95712c398b2fa29eb8d8972e9edc4672","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 15:27:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e25b63ed5d254f788402f60e992ea612 : FAILURE in 11m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a82cd2cd0e794b77907db05e68f3f7b1 : FAILURE in 11m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e97c5a051c8e4897bac6259dafc2d698 : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/720492a2c30e4e8b9ab0d4ee3297e16c : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8e99d2a619b4648820550b46de0222f : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa9a00d08e0e4e018d98c3cc388de741 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbf42997269043eda2ef8d9c97a56193 : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/432a7236a61d451385eb1c326828c5cf : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a89aa32a93b445483450092eaaed38a : POST_FAILURE in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7961d17800eb453e95686547b8ffe2bd : SUCCESS in 4m 37s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/86b878b378224b6e8ea8904e0e14162b : SUCCESS in 6m 40s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/6d33b05346e84997906d8c8f020dc89c : FAILURE in 11m 54s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/0309bc9d8dbe4c96bc5d9721eb5f05aa : FAILURE in 6m 30s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/118847bcaee741458e21ed9f4ad93345 : FAILURE in 7m 39s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/0fbbd00abd214af1b91f3f912ec57e30 : FAILURE in 11m 48s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/7bbd5119ff5a4fce807ed3f62a406dae : FAILURE in 7m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f81c8e29aa4e4af6928fb95c077a9b28 : FAILURE in 11m 21s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4dfb9ac1ed8345cfb40c79464b769d82 : FAILURE in 11m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/8c020860293244e5a2488783d31e0982 : FAILURE in 10m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3c6a4cb79ec54bf39b77b0832abe5b9c : FAILURE in 9m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/47a337c06d3b4ccbb2f57b794f5cc892 : FAILURE in 6m 40s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/9164f05a3ead4cada3c8e7064dedfa56 : SUCCESS in 1h 00m 29s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/fecd10c877f34d4da18a967b8af6db01 : FAILURE in 1h 07m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"287cb06d09835cea94183f8f163a05756245bbbb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-15 15:22:10.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a336bebdcb7e3717cc021fc9b1b71d834d1ecdf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 17:13:04.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/aa23af7cd0f64d5c91136f7b4b2e6178 : FAILURE in 1h 14m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0710bb01868a45a3982e7bb05dea2945 : SUCCESS in 1h 23m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e619ef900a304b1083b724ed1180a277 : SUCCESS in 6m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8451f04583fd49fb8b77b2f7c84c8544 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abc4f4657f9540b9b3fca89895560ca5 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73850559ce6c4c66bfbdf66202839e93 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/893a614f10c947fe91a5345c36caeff6 : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2c07792da5d42a09a950e5e905f36f4 : SUCCESS in 8m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d36dfc7bf514205836b885cb7448ebe : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3195c9e271a3487e95e112ac76601a79 : SUCCESS in 4m 47s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/eb078547e3534870bf8f8d0e9b37c020 : FAILURE in 11m 19s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/97fa705ab45d45378b8be5fd5d6ebe17 : FAILURE in 15m 00s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/7de7e1d7c4bb4a409022f201e08995ca : FAILURE in 8m 32s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/54e771ba64f74c09a01839b9db740f6a : SUCCESS in 7m 28s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/7f68684c23724f37afc611d586b3ee16 : FAILURE in 14m 09s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/a5dc631694fb438487db870caf0d9238 : FAILURE in 6m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ba36a54467724f15858b9164d3d54807 : SUCCESS in 1h 43m 05s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/839ec3744e7d4693bb2e6bf5217c1901 : SUCCESS in 1h 37m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/562bba54a7334325a55ec46d48493c99 : FAILURE in 7m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/80bfd9b31e704a2186b3d72cb589c569 : SUCCESS in 1h 29m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/5b5a2e6b3c82478f951aaab9c1cdaea7 : FAILURE in 13m 34s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/1d4f83fbf09f42a593e3efa5081b8782 : SUCCESS in 58m 05s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/3da3fec9c6c745a686dd8c2d24509532 : FAILURE in 1h 02m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"12a24c6f302b4d843a95c78b50d896974896b0c0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-20 11:11:54.000000000","message":"Patch Set 4:\n\nThis has been resolved elsewhere so I don\u0027t think this is needed anymore. Thanks for the patch though :)","accounts_in_message":[],"_revision_number":4},{"id":"19c283b97a4cc5fde46561832b0b86769d15f1bb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-20 11:11:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"66f661b97423aac75d405b74b919dfb993653b3f","revisions":{"f51f02455ec4446f30f6eab9d63967d31c8a780e":{"kind":"REWORK","_number":1,"created":"2019-12-22 15:52:38.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/48/700348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/48/700348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/1"}}},"commit":{"parents":[{"commit":"41a44230434cf9037d57f5e26c1cc73559e198c7","subject":"Merge \"Ignore releasenote cache within git untracked files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a44230434cf9037d57f5e26c1cc73559e198c7"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 15:52:08.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 15:52:08.000000000","tz":480},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nmove \u0027basepython\u0027 to the top-level \u0027testenv\u0027\n\nChange-Id: I0ed3bd53d00f66e4253a60d61e63281c28736d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f51f02455ec4446f30f6eab9d63967d31c8a780e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f51f02455ec4446f30f6eab9d63967d31c8a780e"}]},"branch":"refs/heads/master"},"7061919252c2246299ba0b07be82008ce7590f9c":{"kind":"REWORK","_number":2,"created":"2019-12-23 13:19:28.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/48/700348/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/48/700348/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/2"}}},"commit":{"parents":[{"commit":"41a44230434cf9037d57f5e26c1cc73559e198c7","subject":"Merge \"Ignore releasenote cache within git untracked files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a44230434cf9037d57f5e26c1cc73559e198c7"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 15:52:08.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-23 13:19:23.000000000","tz":0},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nmove \u0027basepython\u0027 to the top-level \u0027testenv\u0027\n\nChange-Id: I0ed3bd53d00f66e4253a60d61e63281c28736d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7061919252c2246299ba0b07be82008ce7590f9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7061919252c2246299ba0b07be82008ce7590f9c"}]},"branch":"refs/heads/master"},"c1ca1541eb9ef123a336b7cb3beefcf44afe7965":{"kind":"REWORK","_number":3,"created":"2019-12-23 13:43:10.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/48/700348/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/48/700348/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/3"}}},"commit":{"parents":[{"commit":"41a44230434cf9037d57f5e26c1cc73559e198c7","subject":"Merge \"Ignore releasenote cache within git untracked files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a44230434cf9037d57f5e26c1cc73559e198c7"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 15:52:08.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-23 13:43:07.000000000","tz":0},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nmove \u0027basepython\u0027 to the top-level \u0027testenv\u0027\n\nChange-Id: I0ed3bd53d00f66e4253a60d61e63281c28736d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1ca1541eb9ef123a336b7cb3beefcf44afe7965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1ca1541eb9ef123a336b7cb3beefcf44afe7965"}]},"branch":"refs/heads/master"},"66f661b97423aac75d405b74b919dfb993653b3f":{"kind":"REWORK","_number":4,"created":"2019-12-23 13:43:57.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/48/700348/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/48/700348/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/48/700348/4"}}},"commit":{"parents":[{"commit":"41a44230434cf9037d57f5e26c1cc73559e198c7","subject":"Merge \"Ignore releasenote cache within git untracked files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a44230434cf9037d57f5e26c1cc73559e198c7"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 15:52:08.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-23 13:43:52.000000000","tz":0},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nmove \u0027basepython\u0027 to the top-level \u0027testenv\u0027\n\nChange-Id: I0ed3bd53d00f66e4253a60d61e63281c28736d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/66f661b97423aac75d405b74b919dfb993653b3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/66f661b97423aac75d405b74b919dfb993653b3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
