)]}'
{"id":"openstack%2Foslo.messaging~700384","triplet_id":"openstack%2Foslo.messaging~master~I8f6b4a669ec11721f76cf03abcd7d802c3c11eb5","project":"openstack/oslo.messaging","branch":"master","topic":"bug/1857319","hashtags":[],"change_id":"I8f6b4a669ec11721f76cf03abcd7d802c3c11eb5","subject":"Don\u0027t log NoSuchMethod for special non-existing methods","status":"MERGED","created":"2019-12-23 07:38:54.000000000","updated":"2020-09-23 07:03:55.000000000","submitted":"2020-01-27 16:35:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700384-1580142902877-71032318","meta_rev_id":"390eff8dc288960285a504db904da53253ec9b33","_number":700384,"virtual_id_number":700384,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-27 16:35:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-24 11:03:47.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2020-01-27 14:49:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-01-23 12:15:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":2,"date":"2020-01-24 09:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":1,"date":"2020-01-27 14:49:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-20 10:44:16.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-23 12:15:13.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"},{"updated":"2020-01-24 09:52:44.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-27 14:49:31.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-01-27 16:35:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a09c2ccdb14045f3a05a208744e3e8591adb8752","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-12-23 07:38:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7248dfd283c589d25105a22baa220425b46fddb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 08:52:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/be314193a39c435f8b2b06385b54b640 : FAILURE in 9m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31641bd506104b86a43fd60f13496c7f : FAILURE in 9m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61328a07975c461e83582890b706ffba : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bc18d7f20a14a66b4d8ded89315f8ed : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b94ca5f838e484db103c450c9d2f409 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc49360b5fc94d87b1099fd4ad7a778c : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/554ef0d6b8d74e8f849d9593b318bcfb : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/010bf05a8d7d4ec2899c9785b2440eb1 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a359483f7ba458791e25832454df2f3 : SUCCESS in 4m 47s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/489473cc5f4747ebbaf8fa164ff17e67 : SUCCESS in 9m 55s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/f30ea7738bda418ba14ed17e37b5c578 : FAILURE in 15m 01s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/d60c4799841c4e7c942df5cd74ffc495 : FAILURE in 8m 08s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/a35f9a3a02ba4e07b1b03d97e984149f : SUCCESS in 10m 06s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/093d5d6c53294a938622c7520c6af400 : FAILURE in 12m 41s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/0323833385e5474ba5b590e939e4b1cc : FAILURE in 5m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/eadb15c4b3c14bbe890915dd2f198120 : FAILURE in 12m 09s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/0131ac2e3ccb419ea0ffd3fb1819425b : FAILURE in 12m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/dfee0a14fc5348d080635785286d29fa : FAILURE in 8m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/742a1fffe9af44b497d76cb38556030d : FAILURE in 11m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/8bade67b64e748dfa9b0d56e0f64a100 : FAILURE in 8m 35s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/33ad2d2497934e6387b01c26de941539 : SUCCESS in 1h 02m 41s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/efd8d4577566482b8dde879561a2a53b : FAILURE in 1h 09m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16b7ab5a93209e55501c52db03250c2691e14819","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-12-23 09:08:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8bffc3b38c70d132e8737c582be6f52ddfae6a37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 10:57:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f14055d77e6647839854558c2c0b56b3 : FAILURE in 12m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c32ae7e9b524c12af168616864ce71f : FAILURE in 12m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/660470098bfb4a3c8a765da59571dd2b : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f56db472b37a4b66b1d6cc980381e3ef : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/906b8b6d00db4ad49e7cd4a6622aaae5 : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bf1b13fcacf46d0b46c7d0a67fbe975 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0141e73501e44e77a520978f89c136fe : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3fc386ad2044530886f65424b145e19 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f5cd19e65248eca24ea0107a20ccf6 : SUCCESS in 3m 43s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/5580c2d932ea4e1fa207cebd856d1ff1 : FAILURE in 5m 21s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/71ccbd4aa43c4605a04f28127afd3eaa : FAILURE in 14m 58s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/9293a942a16a4d78ae79425768d72e6c : SUCCESS in 5m 20s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/5402f873171d4185ba071ace68391017 : SUCCESS in 5m 54s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/5f83538dca894b889c5969269ae9f429 : FAILURE in 12m 02s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/b5a002d55a214ff7a10c60c61c8bab26 : FAILURE in 8m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f8a8207fc16542958e1045910ae7fdb0 : FAILURE in 12m 08s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e2dafba8d4cd4c9eb5fdb219431ab949 : FAILURE in 10m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/d0a5f724308346638de511c1204601ff : FAILURE in 6m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f7cc1b6b3f97441e8987f4f91767a5ce : FAILURE in 13m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/4d89c70267c5430ba599c1154b81416a : FAILURE in 12m 26s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/ae52e938e8c043148b0d5207bc0d3f6a : SUCCESS in 1h 01m 40s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/b102355a59f342779cca4c3efdd072c3 : FAILURE in 1h 45m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6fbedf2f6b1b896b3c007d9c6709d4dcc86959b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 12:42:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"afb2f041b8dfc1c1c12fa5cfe984f412fe94568b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 12:46:39.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"7f5dbbbd7fcdf74eba6b591f094c58da87761536","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 12:47:09.000000000","message":"Patch Set 1:\n\nrecheck let\u0027s see how badly the CI is broken","accounts_in_message":[],"_revision_number":1},{"id":"0012cfdf840712ac080485551b9b4c3b041c069b","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-12-23 12:48:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb47f0e9612e67a145ce6cdf0cb56c27a4bef148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 14:04:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/5b679f01666a41538ff7fba0727079c5 : FAILURE in 11m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/711231c17da34a4aa63dd6d8bb1b529f : FAILURE in 10m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5219169e52a4b3e8571bcd97077d0e0 : SUCCESS in 6m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1709b38eb1c9498c9ba17f9a41db1c4e : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50afdeeac7f342ac88243a8abc74fc64 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58a0f869f7da41ddb7d03a8c8d5b1eba : TIMED_OUT in 41m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dff208f7044644178bf55dc3395d1945 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ddeeb6d042444d529fd0758a21255ebe : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35ff5f1a2ef4eed9d07b495a01bdf24 : SUCCESS in 3m 38s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/c9131b9880a04610bf68e3330ab87a83 : SUCCESS in 6m 43s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/afe379463acd4521b02500311e7b3fe6 : TIMED_OUT in 40m 59s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/0f70d19b20634796bd8910641cc48bb8 : FAILURE in 6m 15s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/ee872323458245ed896160c95f4a8882 : SUCCESS in 6m 30s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/48e9e620d6094819a562ef5dc32703ad : FAILURE in 11m 16s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/8836b51c92f24d8b824aa5c7ef10f9aa : SUCCESS in 5m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6a69d135814c46c98469bcbd531e184c : FAILURE in 12m 23s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a34cc1a1e52d4491bb25bb6358ed28c6 : FAILURE in 11m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/4552c70108214146b90b01c460a0c972 : FAILURE in 9m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c8fa75b0f7f74405bd6ddd8a3102043e : FAILURE in 11m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/8e7c2c1b7eef496ea59128e8d126fc83 : FAILURE in 8m 38s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d97265cee3334db6b539c67b63a298b1 : SUCCESS in 55m 56s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/429d71cf455149588463b172f82cc0e6 : FAILURE in 1h 15m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1b776862d2f7119c168c99b180f7727821ed987","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-12-24 07:35:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f287d9713025d16e959488d81f2e3e73f1766726","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 08:45:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/05ebd69c0f75464891a33a72437d2c3f : FAILURE in 10m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5cdb937ad864a2c8ad09ce39fa20f09 : FAILURE in 14m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a4feaac0f6049f98e63d984b5ccb213 : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d864e207bab43b7b085163dc73f971b : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7018bd7fe5d34df3bdd47c9e649438c4 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fcb305ae8d7434783f15dd3322e1c02 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c51e0052ef24ecd9614729e8bb45589 : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86e341bd2af547d0b9021e264ae7ef2e : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca510e8d1bdb4aa898e6e13f6d3dc600 : SUCCESS in 4m 21s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/c9b8a3bf743d47748964cdab882c4912 : SUCCESS in 6m 24s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/6d60cdf6679e49bfba6782f4c6c750f3 : TIMED_OUT in 41m 07s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/9e71b8e428a444f58030b9a40e4beed8 : SUCCESS in 5m 54s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/6b23cef8c8c34eda9232fd20734cc32f : SUCCESS in 6m 08s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/8fb25faa53d84884bf293bf9c4e50c75 : FAILURE in 12m 01s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/e7cd3c9070224f93af51a386c6bfe808 : SUCCESS in 6m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/cbff318ecb8d4e7d9f69ae4a7cec2f62 : FAILURE in 14m 08s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/81ae8038f0ac4bf5a631a447360fc499 : FAILURE in 14m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/e565294b387b42e583b580b8d795d0db : FAILURE in 11m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b8f629a6761142a5a8aa005781df6e09 : FAILURE in 12m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/6205974f46db4f3b88c8033494b40263 : FAILURE in 11m 11s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/6f51feb9faa14fe89f94bd569f8cc173 : SUCCESS in 1h 04m 40s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/cb68009c42cb482b9b0c1a04557c22bc : FAILURE in 1h 03m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"83187f39c57ca2d9281fc1dd5adc0606d62edcee","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-01-13 08:23:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"20c7033dec1da0eb13787e887556c8496f7c7d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 08:41:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d37a6a7e0df8462b8c390174f4f9d337 : FAILURE in 7m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7827cd95dde342b3a95a96a1d64238da : FAILURE in 9m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2cca05d072044aa9cac349efef1a4c1 : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e7f742f72fa475d8af6f1c39016a6f4 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c644eb8ea694e6baf137825a0a4ec2e : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5577cc3ef1f04ff78fc6fc8356c17f62 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/616b6bf918d944d78e65604e6b3a6152 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ac545871e3c46d79b1472374a0e9e9a : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56172b5f4de942499b5f988454d18689 : SUCCESS in 4m 18s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/6869b42d784e46cb98d70f27095cb412 : SUCCESS in 5m 21s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/d0c43d31d3b24a4b9e5fca1502cc1aaf : FAILURE in 11m 58s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/a2b8f6b576e84abbbedde034d6d32323 : SUCCESS in 5m 45s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/3f57c9879da84248ba6da4d2e2b32d38 : SUCCESS in 5m 57s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/366087eb3238412b9c978ae5085cd3ea : FAILURE in 11m 38s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/04f6787c88dc430f9ead1d83b01d3e0e : FAILURE in 6m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4527bb2faca84b809ceede9d3fe425e6 : FAILURE in 11m 43s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/69df8df46ff346568a21512042e7e5ca : FAILURE in 11m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/3244b0a5d19f4c1a97364c705d08b17b : FAILURE in 12m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9a1a5acbf9eb407e91279fc77f634f5e : FAILURE in 11m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/e3dabdbf0a3c4438b0e8b99f2aa1e955 : FAILURE in 11m 25s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/5e507e9488b546bb90a2f30bbd1018ac : FAILURE in 5m 03s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/b85631220d6440f78fed59da1b9fc974 : FAILURE in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"96ccb3f642acabd0383b48481eef96e702c6972f","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-01-13 08:47:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a1ba25001d59584bd240d9dd0029140110df543b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 09:32:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/a12174d27b654f2596b2b00c04f73678 : FAILURE in 12m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70f976ac6cb94fce92391db5632db281 : FAILURE in 9m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3ebac0d85f743659fc08531642d678b : SUCCESS in 7m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6dbca38b97044f494052228e21ef9bd : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c6257c3e61247d78051054aea4c8dfc : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f17dbda41d0b4046bb3c50d4f2f05840 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f1f3149b710423ab6b9e5f848420939 : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22476235ac6146aa84f32f548222d616 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51cf1e77278f451fa5aae101c6fa42ae : SUCCESS in 4m 00s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/773090d8142e44f3bc59ae1370334399 : SUCCESS in 7m 25s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/3658253e4777487d96877773d4f7a7f6 : FAILURE in 6m 53s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/3d891e5b719a49e499cdd482eb1d51e4 : FAILURE in 6m 28s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/2f20919de78746dbba0fdd2c2038b161 : FAILURE in 8m 36s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/cbb1013f219e4545a6863106277d5ac8 : FAILURE in 15m 04s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/84d00e23c5c64c6da698d0eb01b924c7 : SUCCESS in 6m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/01113c677db148b884f54235e1b6834d : FAILURE in 12m 29s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/780513a1292f42beba1107dbe98b32fd : FAILURE in 9m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/4761c91f02044683ba6e7d175d8b85f3 : FAILURE in 11m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/0ae1915369304929bfe8ea22257246ec : FAILURE in 10m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/75065c721bee42feb33b8d27137b6bed : FAILURE in 11m 42s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/46a21372504d4f3b929e42f0e5f63cdf : FAILURE in 6m 12s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/f8df7904acb44a64bd9f33d3f5096233 : FAILURE in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff571f8a313a564ba54c5b30986c34d49b743ae5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-15 15:44:19.000000000","message":"Patch Set 2:\n\nThis makes sense, but would it be possible to update tests for this?","accounts_in_message":[],"_revision_number":2},{"id":"4327b7fec83c9ef99f805bde824f6034c100ecf1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-15 15:44:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"04ecff6ac198a3e2469eacffaf7528642fc9b4f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 17:56:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1192394a8e6f4e1ba2dd9659d1b2e822 : SUCCESS in 1h 36m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/930de78bdef1434f9e37036389c8423d : SUCCESS in 1h 33m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/beddc69b0a3040be81f8b66dc84f2a12 : SUCCESS in 6m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f08c49a6e78456aad27dc418d3905fd : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8ab2ead323e44928babeafc51e6b69e : SUCCESS in 7m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9431faef6a284b6f941bd0ac06f0ff1a : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2af4ba866f3c4defaf07358e97d0d106 : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a3e087f6a074a52ac69b0c608997c71 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feffb89b5a734adba82af9091c4653aa : SUCCESS in 5m 02s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/25be44bccdda4013bc6b5727b20743a6 : SUCCESS in 6m 05s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/3185b04a2d844369b77438a10c2f3a78 : FAILURE in 13m 33s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/dbdae654a0c84b699b1bc45989265c76 : SUCCESS in 5m 39s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/616cbdd3bf664aae86575e725aa2ec8d : SUCCESS in 6m 22s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/9ae1c19e12c247e485a14163fe528cfb : FAILURE in 12m 50s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/5f05fc57159740cc9ae07ebb97e56a70 : SUCCESS in 7m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c62e9a79b72647f9bac81ec91612914b : SUCCESS in 1h 53m 26s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/cfefc9b65fb64990b9d7645182c1d681 : SUCCESS in 1h 40m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/39e6d83b703b4546b559901c1209e1b4 : FAILURE in 13m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9d3959c507084404baba3eb9e317af07 : SUCCESS in 2h 08m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/0aa430ab0b6049febed365a6fdb73e0f : FAILURE in 13m 05s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/3f6cf73c01a64660a0c8778fd696bb72 : SUCCESS in 1h 01m 17s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/c532eeaf9910449c86f06cb56d8c50aa : FAILURE in 1h 13m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"08bcaeab863a51c84d262ff08d9dfd6e326c55c9","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-01-16 06:27:57.000000000","message":"Patch Set 2:\n\n\u003eThis makes sense, but would it be possible to update tests for this?\n\nI\u0027d like to, but didn\u0027t find an existing test for _process_incoming() to add to.","accounts_in_message":[],"_revision_number":2},{"id":"684a69b3d2da970d1994fb801de6fe02084d237d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-17 14:59:59.000000000","message":"Patch Set 2:\n\n\u003e \u003eThis makes sense, but would it be possible to update tests for\n \u003e this?\n \u003e \n \u003e I\u0027d like to, but didn\u0027t find an existing test for _process_incoming()\n \u003e to add to.\n\nI don\u0027t seen existing test too...\n\nI\u0027m ok to move forward with these changes even without tests.\n\nAlso it can be worth to add a release note to track these changes and the possible changes on server behaviour which can impact user and which would ought been noticed in version info.\n\n@Stephen thoughts?","accounts_in_message":[],"_revision_number":2},{"id":"3908525101d6877f662c1525403f2fd8a3b470f3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-17 15:33:19.000000000","message":"Patch Set 2:\n\n\u003e \u003e \u003eThis makes sense, but would it be possible to update tests for\n \u003e \u003e this?\n \u003e \u003e\n \u003e \u003e I\u0027d like to, but didn\u0027t find an existing test for\n \u003e _process_incoming()\n \u003e \u003e to add to.\n \u003e \n \u003e I don\u0027t seen existing test too...\n \u003e \n \u003e I\u0027m ok to move forward with these changes even without tests.\n \u003e \n \u003e Also it can be worth to add a release note to track these changes\n \u003e and the possible changes on server behaviour which can impact user\n \u003e and which would ought been noticed in version info.\n \u003e \n \u003e @Stephen thoughts?\n\nYeah, the addition of the release note WFM","accounts_in_message":[],"_revision_number":2},{"id":"8dd79332da75dd4ffa59d34c04d17e07e41f5574","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-01-20 08:33:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"39861d131fb9a6c488a65248e0d5b2a94baa90b6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-20 10:44:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d441c8882df3ed014c137cfb2272bcffb5986b3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 10:52:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9c212f3a35e74ff6adef89f0fe428b9c : SUCCESS in 2h 02m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/243ea1fabbb143a4bad353333fa5bb83 : SUCCESS in 2h 06m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afe182ac60c34ca2a22abfb3289c25c2 : SUCCESS in 8m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80a7f877a7ac4b8687c4a637c34bf1fd : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8cf5de5cc6d420a837d46c4611e3b38 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a081b9285994e60bb60d18e35a8bb9e : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c322600a1e5441e8a7e79ee9290de646 : SUCCESS in 6m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c6e1c16801f437ca40ceeab99564b74 : SUCCESS in 10m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/431a6e3f98d04c7dbf933199d958309d : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c3af4c765954894bd8a7ffdeeb068c4 : SUCCESS in 5m 33s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/8b507a4a037344ba96741e8df1603667 : SUCCESS in 10m 55s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/3c6bae5ac6f74a5c88d834031d429618 : FAILURE in 16m 12s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/292a4baeef454310a31da71fb9acde0d : FAILURE in 7m 30s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/cd6ac7d3c669454e817047b66a0def35 : SUCCESS in 9m 56s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/59d6e6441dbf409eba76042c012eb729 : FAILURE in 13m 33s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/c90a0bbe85094aacbcc800518e4436bf : SUCCESS in 5m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/22c12e17239742c5a89ba7ec444ef5cf : SUCCESS in 1h 27m 44s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/741e4ed0ef1349b4816eb2a1cb595b78 : FAILURE in 2h 14m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/6c8c2d79b20e43df9ccc96238294b592 : FAILURE in 13m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1e68fbc2f14e40329cb577a83110e4ff : SUCCESS in 1h 28m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/d5c89203fc654544a625657d537a5fdb : FAILURE in 11m 36s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/503cc6ca0dbf405ab5509f287a7bca2d : FAILURE in 1h 35m 20s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/e952c17daa34450b818bde94e6cdaf87 : FAILURE in 1h 06m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9c8ee111c27516151ef90ecc3a2374e8218c091c","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2020-01-22 11:56:47.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"57514e751cc957627585ddca0a5047ed31df277c","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-01-23 11:37:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ca1f66ab5cd5017d1af38c728d76cefa9d728761","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-01-23 11:38:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d62cf6de641e6b4c9cb23df301bd6eea0a765bdb","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2020-01-23 12:15:13.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d053a813fb46be6fc0f85d247cd0db00f3af14e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 13:30:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3b934ac76f584d9284e67063e511c79f : FAILURE in 1h 23m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4aa7f838635c4b3f8c4266c275984cb1 : SUCCESS in 1h 18m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c7c79d9bd6a4715b55bacd7b0ac7a13 : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f652be51d6f42b397bf48cd66777743 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2219dd4e019448a92ed87ab8b8722fa : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74840b4e568f474882893e72f4a215cf : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bf8987fa0d949b79fd811a7c50b9485 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b10879b345e4ce9b7bad38f89c26898 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/757b335c67394c60b5a85e2d491f485f : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82b8cbc6c1d041a1a22de68e022d9a3e : SUCCESS in 7m 14s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/9643423bcfb548f897d96a605bff58aa : SUCCESS in 6m 54s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/5e406d3316f0498da8b8590f5c9eb1df : FAILURE in 12m 08s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/7012fe63e4ec42d0b27f7aaddc03b949 : SUCCESS in 9m 04s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/149fc3d106624fc4841761ddc9cbc7d7 : SUCCESS in 6m 08s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/0808f41bd4cf48969115bb8357aa570e : FAILURE in 16m 53s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/0a36bfe1af284c23bba70b0a3ed85774 : FAILURE in 6m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/baf2db6ce2c5444ab40f9133320df54c : SUCCESS in 1h 35m 09s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a9264ff8eab84ea0b020dcce7ab5c422 : SUCCESS in 1h 45m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/b440e1ffe84240e9a75be44afd76ce5b : FAILURE in 13m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5cfc949dc53a4ca1b112bf593358a64e : SUCCESS in 1h 35m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/34bf6d259cea4da8a162571b440fa9a0 : FAILURE in 9m 56s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/b6f127e54a2742f7bcf3c72928b04c5e : SUCCESS in 1h 06m 09s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/38839d5f909744b1aa4c1f084c944684 : FAILURE in 1h 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4105539c63e1dfb4379c1c6f3376977c04ce184a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-01-23 13:34:49.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2af976f0c59d85f6665944a4475298323c92a5c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 15:36:38.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2403712bbd15479686caa6ddafe26726 : SUCCESS in 1h 28m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7aeb9b4b9abe4b42b50a9b36160b353c : SUCCESS in 1h 26m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49f91c416a1e414fb38bd6a06bf119d9 : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10228aba260b48b08da473ef4aad4a86 : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45aad370e3344bc2ba88bdb681a6f869 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96109bc484cf40eab43520c8c76d3bf8 : TIMED_OUT in 43m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51dc13e93f6b4a94bfe20da76ee4ae3b : SUCCESS in 9m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f204b01c31145fb9b4f7df81e8fdff8 : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa85bed7c6b416c8c5453e9e0bfc3be : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8983318817524db395179d0931747b79 : SUCCESS in 5m 55s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/e949c0e497bb4909b6041fcd6d29c588 : FAILURE in 9m 33s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/6102777431af4692b2c54233b5828456 : FAILURE in 16m 03s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/206d351c55de4ac3bfde332f5586502f : FAILURE in 9m 18s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/a2629f55d5aa430498538e2252e29ec2 : FAILURE in 10m 51s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/4f3e8f144e76410e8d44fd1bdb587f19 : FAILURE in 14m 56s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/7e4ba11e88fb4ad68cfa16c260a18c7c : FAILURE in 7m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1201de6368ad4dd5875b3667b62a998c : SUCCESS in 1h 35m 28s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a692d3c14e4e43c8967bef1605080172 : SUCCESS in 1h 54m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/fcb65abbf4494900a905bf1fdce102f5 : FAILURE in 16m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/46623126f5d14ccc87a47efcef95ba24 : SUCCESS in 1h 34m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/f77e7f86e19f4b85944fbc3c54b6afc3 : FAILURE in 11m 06s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/dbde9496f51545bdba212f93e2ce9823 : SUCCESS in 1h 03m 21s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/79c227653ae94551a3362f5e05fe6cc9 : FAILURE in 1h 09m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6b63907d71ed7abeb81e8b9ab8765d8cb9f8d2ea","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-01-24 06:07:47.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0ae50574b9a4eba9ef41f980d5c8128e662ee7e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 07:46:43.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2031c41766534ad4b9b4f4cf2025d9cf : SUCCESS in 1h 30m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50c6b6adc989438094bcf0d901c1ea6d : SUCCESS in 1h 23m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9f8fdb6ce4a413ca07f4ea4b11527f3 : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9607e4e7a4d640bd9fd05671b51ae40a : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbecfaf4322a4af78d0506985e387002 : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f5fb14cfa95423e9ef3298bc2392ad7 : TIMED_OUT in 41m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/072b58cf105a4d04b7ccde29a5fd7929 : SUCCESS in 7m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e68178c031cd4377933be34350ef5434 : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/326540f3959f44b1a3994b9385d27226 : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebad1a18ce454692a554c47cb8b7be6c : SUCCESS in 5m 34s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/6d12a2eeab804b5f9c474d14580079b2 : SUCCESS in 8m 58s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/11ae61c6d426418b853ce0ae4c88e015 : FAILURE in 13m 42s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/78b689b2f5b14911844bbed5b3920e4c : FAILURE in 9m 16s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/fbc081417382409e95bbc8de0d22b428 : SUCCESS in 9m 23s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/85fabd6ef87141f0b3336f149093c731 : FAILURE in 14m 37s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/0443db682d884193bf0a5fc5b3433f2b : FAILURE in 8m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5e076929ed994995bbfea27499435e5c : SUCCESS in 1h 30m 16s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4cc71184a8e5468cafc1fb08f8b11e15 : SUCCESS in 1h 31m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/de63252ee3874caa8f9a5ead405b5352 : FAILURE in 9m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b812eb797f3b421c997caa9e26ce28b7 : SUCCESS in 1h 27m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/e49ae0e6893e4180b68d1cf0471d874d : FAILURE in 14m 28s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/2c3f96465059431d94b2daaf1eb56f1b : SUCCESS in 1h 04m 47s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/c4ca1e51e4384144ad8823fef2f4c4f7 : FAILURE in 1h 03m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3d0815c259953c1fd33b1768f78f10d1cdd73296","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2020-01-24 07:58:03.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1432cb18068b803d0b682a512c270ed94f3cd3bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 09:42:42.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6d30e10c0b4b432482bc0683f51e587b : SUCCESS in 1h 30m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d23fa6ebaf6a4e88ac1e84aa47d7c745 : SUCCESS in 1h 25m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba9cfb94d64b44dc9e2ffdbe35b5a325 : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5318873ca6d742369025d85162db7afb : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f30e47fbca7c4f09b0b44901601a2cfb : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f0cee7c51a64641bd957884b6a0a425 : TIMED_OUT in 41m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac7d05d556f34be09145b07ea71f1a3a : SUCCESS in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fde4dd73c0da43b9841d0d57edab7ebb : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df2134bc52ba4310af747276d266bda2 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4037ea2de1fb4c18ba10a2257e06a06a : SUCCESS in 4m 42s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/21cf1e8523a54560ac596363223280d8 : SUCCESS in 7m 11s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/d2272f4aa878404996b4248004ab4b04 : FAILURE in 14m 09s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/f0e7edc1f14e4f80b356e54c3e7c6563 : FAILURE in 6m 33s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/ed09721af4e845e281685c82024f9d64 : SUCCESS in 6m 47s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/8dd84b9d044b4003aebfb5c8c5ba686b : FAILURE in 13m 05s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/0644eed4d012499aba90d657a01a0c5d : FAILURE in 7m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/256700185e86455e9c9eeb77c976df17 : SUCCESS in 1h 41m 33s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/492ee58021114de7bd9ead538eb20028 : SUCCESS in 1h 35m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/a2600ca807444e4ba6ae5ee67024b3f4 : FAILURE in 8m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3d40aa194cfb40488fffa88c758c273e : SUCCESS in 1h 40m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/511050655e1b461db316b2015e3a8bd6 : FAILURE in 12m 03s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/941d31be09fa4364ad59cc1d971c441e : SUCCESS in 1h 00m 22s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/35b5e32e5c5e4c50a13c3d78249747f3 : FAILURE in 1h 17m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bd8be1014342b4bc1837a306e8c635fd66ed6aa1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-24 09:52:44.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7cbfe737ea1b83eaeaea2efc528643378e269013","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2020-01-24 11:03:47.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b4846f2506362521ddcd5e3f243d93e4471e09f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 12:51:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e3d445c60c174cffa62dd194ef5f88de : SUCCESS in 1h 42m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f9e766a7eb44b78842da838025979fc : SUCCESS in 1h 34m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/199ae2961a22486097322908373c0c11 : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3ed19e06b8a44348529b613724da179 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deea082c6be34748868af27bcafbc844 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/65409b9c94fe4ad3972f0972b248a57f : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9e508fcc89240fdbacde1e73ea3b93f : SUCCESS in 8m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d51e98778a8d4cfe9605b9d169c057b4 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bceb502979264ef0bc521d3f1299f301 : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45c05dc893124b36987ca115411ef99c : SUCCESS in 4m 53s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/30484c37991c40e29d4c1d6f2158cf59 : SUCCESS in 5m 55s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/9b4c2140c3584e50b1aea9ad4d87f6c8 : FAILURE in 12m 19s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/b7d7cfddd9f94a4287eb25077b7531bc : FAILURE in 6m 28s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/35bc0f45284c4c4784e474166a9dcbe7 : SUCCESS in 6m 35s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/1462a7367ef84aee921db4e9970b4c3e : FAILURE in 15m 28s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/d4f242f8bc0840fb9b44b726bb2575a5 : FAILURE in 5m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/df42e2c816434f0989df855c02c65cb7 : SUCCESS in 1h 24m 46s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6865357988aa46579685701694adc313 : SUCCESS in 1h 36m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/9451cc3552e54bd39e926902cb8aae7e : FAILURE in 8m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e485a0b2b92645f884473da933c33395 : SUCCESS in 1h 25m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/d767202a0dc24c63b689b603d5a772c9 : FAILURE in 10m 35s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/0f83c8abb35949ce8aed432fd7af72ff : SUCCESS in 57m 14s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/588ac0caa9214ed68ef28e4530107939 : FAILURE in 1h 03m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d94d00f3452d2d847d72bb31cc33df42846469d1","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-01-27 14:49:31.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e7da8d5895bb02e49a58ebdb3db3afa3d814b4d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 14:49:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"97170c682197413c807aec5087131c61f565cf55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 16:35:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/199e725a986342bf9f71250b26a2fa6f : SUCCESS in 1h 41m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffddb4a0d65f413eb9291c448196832a : SUCCESS in 1h 23m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5020a55b0cc4133896e058d87f791d8 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c33f488479e4c46b8908ed2ce0fce93 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/24c1c026ee6a4dcbb48627b621a9c7e3 : SUCCESS in 10m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4167e8d61caa4ae5875994ee5f13bde1 : SUCCESS in 7m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/663efa1bb05640c09dad348395942c3d : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49e41e0f6d49408898f41e026adbd83e : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57f5bbf73a474501ba87de93d27db961 : SUCCESS in 4m 42s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/4c7415bd477f45e8ab51d31c5b911008 : SUCCESS in 6m 23s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ec2cbd7c0b80454292c57176d4106724 : SUCCESS in 1h 35m 56s","accounts_in_message":[],"_revision_number":4},{"id":"5795f8534cb21f3c9427d713c6d40b704391e93f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 16:35:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"24a490102809b9ff69de8af6dd8082695dcba216","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 16:36:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65fab502322943f88fe7ccc19efa0fb0 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97871cd71e194e64bbcceff9bac57e2e : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":4},{"id":"10b37e00f7273b535ef7554765de7a9ba855565d","author":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"date":"2020-09-23 07:03:55.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 3c2a66e8c78d7c4325734b55186f6aead52029e6","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d7eb89eac8f76c46849f5e0d18a7ea00dc2d07bb","revisions":{"79bf507877b67e19dd77bd7a796aadd86c45dcf7":{"kind":"REWORK","_number":1,"created":"2019-12-23 07:38:54.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/84/700384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/84/700384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/1"}}},"commit":{"parents":[{"commit":"859e0d4eaa45d75eb822f4a6be46c547187c6358","subject":"Revert \"Add RPC incoming and reply log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/859e0d4eaa45d75eb822f4a6be46c547187c6358"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-12-23 07:32:28.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-12-23 07:38:09.000000000","tz":240},"subject":"Don\u0027t log NoSuchMethod for special non-existing methods","message":"Don\u0027t log NoSuchMethod for special non-existing methods\n\nSuch non-existing methods are used as health probes.\nPlease see bug for details.\n\nChange-Id: I8f6b4a669ec11721f76cf03abcd7d802c3c11eb5\nCloses-Bug: #1857319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/79bf507877b67e19dd77bd7a796aadd86c45dcf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/79bf507877b67e19dd77bd7a796aadd86c45dcf7"}]},"branch":"refs/heads/master"},"728097dd8ac56589e2d43fe28a1f37862a49a487":{"kind":"REWORK","_number":2,"created":"2019-12-24 07:35:26.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/84/700384/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/84/700384/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/2"}}},"commit":{"parents":[{"commit":"859e0d4eaa45d75eb822f4a6be46c547187c6358","subject":"Revert \"Add RPC incoming and reply log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/859e0d4eaa45d75eb822f4a6be46c547187c6358"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-12-23 07:32:28.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-12-24 07:33:40.000000000","tz":240},"subject":"Don\u0027t log NoSuchMethod for special non-existing methods","message":"Don\u0027t log NoSuchMethod for special non-existing methods\n\nSuch non-existing methods are used as health probes.\nPlease see bug for details.\n\nChange-Id: I8f6b4a669ec11721f76cf03abcd7d802c3c11eb5\nCloses-Bug: #1857319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/728097dd8ac56589e2d43fe28a1f37862a49a487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/728097dd8ac56589e2d43fe28a1f37862a49a487"}]},"branch":"refs/heads/master"},"5d5b537a5bedc998e058d495bfa32d40fdf433aa":{"kind":"REWORK","_number":3,"created":"2020-01-20 08:33:44.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/84/700384/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/84/700384/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/3"}}},"commit":{"parents":[{"commit":"04b2b5d451df7d5aa4047a6b38aa364c400602a3","subject":"Merge \"Adding debug logs on AMQPListener poll\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/04b2b5d451df7d5aa4047a6b38aa364c400602a3"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-12-23 07:32:28.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2020-01-20 08:33:29.000000000","tz":240},"subject":"Don\u0027t log NoSuchMethod for special non-existing methods","message":"Don\u0027t log NoSuchMethod for special non-existing methods\n\nSuch non-existing methods are used as health probes.\nPlease see bug for details.\n\nChange-Id: I8f6b4a669ec11721f76cf03abcd7d802c3c11eb5\nCloses-Bug: #1857319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5d5b537a5bedc998e058d495bfa32d40fdf433aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5d5b537a5bedc998e058d495bfa32d40fdf433aa"}]},"branch":"refs/heads/master"},"d7eb89eac8f76c46849f5e0d18a7ea00dc2d07bb":{"kind":"REWORK","_number":4,"created":"2020-01-23 11:37:45.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/84/700384/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/84/700384/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/84/700384/4"}}},"commit":{"parents":[{"commit":"04b2b5d451df7d5aa4047a6b38aa364c400602a3","subject":"Merge \"Adding debug logs on AMQPListener poll\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/04b2b5d451df7d5aa4047a6b38aa364c400602a3"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-12-23 07:32:28.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2020-01-23 11:37:31.000000000","tz":240},"subject":"Don\u0027t log NoSuchMethod for special non-existing methods","message":"Don\u0027t log NoSuchMethod for special non-existing methods\n\nSuch non-existing methods are used as health probes.\nPlease see bug for details.\n\nChange-Id: I8f6b4a669ec11721f76cf03abcd7d802c3c11eb5\nCloses-Bug: #1857319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d7eb89eac8f76c46849f5e0d18a7ea00dc2d07bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d7eb89eac8f76c46849f5e0d18a7ea00dc2d07bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
