)]}'
{"id":"openstack%2Foslo.messaging~700505","triplet_id":"openstack%2Foslo.messaging~master~Idda79dc87bcd0e3367a6abd1b52104c000ad1dcd","project":"openstack/oslo.messaging","branch":"master","topic":"trivial","hashtags":[],"change_id":"Idda79dc87bcd0e3367a6abd1b52104c000ad1dcd","subject":"bindep: Add \u0027librdkafka-dev\u0027 dependency","status":"MERGED","created":"2019-12-24 11:30:00.000000000","updated":"2021-04-16 13:13:39.000000000","submitted":"2021-04-16 13:11:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700505-trivial","meta_rev_id":"208f6b1fa047143205629533cadff37150655d64","_number":700505,"virtual_id_number":700505,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2021-04-16 10:26:58.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-16 13:11:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-08 15:04:01.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-06-03 12:54:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-08 15:04:01.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-05 12:57:45.000000000","updated_by":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"reviewer":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"state":"REVIEWER"},{"updated":"2020-06-03 12:54:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-08 15:04:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-08 18:14:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd3c09d6ba05b33df1b01a6fe40b37d27ce84e5c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-24 11:30:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87efec5252d03a6647706475cc7dbee55009716f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 12:37:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3c270c8bb9ac4b07833a1f552edd58fd : FAILURE in 13m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04cd616e161d4c7aa82375677cf1efd6 : FAILURE in 14m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92fc96ac0839445089120a067f2888a0 : SUCCESS in 5m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c21dfafa6ce74dd6856fef100f52da9a : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb445c636d95472293956270b0f559af : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/acacd1d74883461e9e46c1e731ad1a0f : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45231a329ffb43aba879efca8f6ade01 : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d11c87d771164918b0e1b4a9cf9ff735 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57b4692126eb440aba6fb5ac42e52af2 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1b7e4e2e9d34a2a827a47ef236d9a1a : SUCCESS in 4m 05s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/78f3069739874ce7aa38da1d6872ce04 : SUCCESS in 6m 15s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/9b70b06610ba4f3a9ee2bd6c0d9b3a93 : FAILURE in 11m 55s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/144bb9d8be2d4e778de42227af733b61 : FAILURE in 5m 34s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/328e10075a39498584f9b6dafd774a89 : SUCCESS in 5m 53s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/17c92f297231447fbccef2be98e03acf : FAILURE in 11m 57s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/45e7c6fc3bce44d4a1e4786969593038 : SUCCESS in 5m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/00c34a9fce374a0f85c93b0a01b5a6dc : FAILURE in 10m 07s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ced9075f2f184dc6b65ad8f77acadb2d : FAILURE in 9m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/509fbb5175ab4bbc8c4f53b98b0031f2 : FAILURE in 14m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c7818b2446a2494f95b9ea911bdf8745 : FAILURE in 10m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/d4be9ff5bcc04f7c9ba08ed1cf06b6c7 : FAILURE in 8m 00s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/78f0ada30f8742718e9763a0aa4d83cf : SUCCESS in 59m 14s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/0b4dc8c6334d4c22ad294c02befc9948 : FAILURE in 1h 05m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8965d9a90aa8ad37c8f5aaa7c1c839fd1845cb29","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-02-05 12:57:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eb5f85ec3755b5d9844425710bc5f904d3e432f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 14:47:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9964ba5bea3543339d6d7f1b825261bb : SUCCESS in 1h 19m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7854cc1ec394fa58adf66e7f6d5afd1 : SUCCESS in 9m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b23c93f4ebb46658572670f02bcde8c : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44b687e3a73f4b85b0c46fbcf7cd1103 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb14bc16464545da9681acbb50706bd6 : SUCCESS in 9m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e71adde64b534ca693a63e75626a945d : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d3f00a2587243a391471f5560e6560b : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e256842a9040ac9f76942511ab02a6 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7902b1046d14985b2560db08588f3e6 : SUCCESS in 5m 55s\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/6734ace747094821858f7de1638fdea7 : SUCCESS in 6m 42s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/0ebea445a3c14774a0b9a581962d6cbd : FAILURE in 12m 56s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/90b0e11fb4834904bf6eb98985f14167 : FAILURE in 8m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0a10ea19029c41169202538ae6f4cf6c : FAILURE in 1h 27m 00s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/174e71fceadd4bdaa5d47a5c220a69a0 : FAILURE in 1h 24m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/9e6a21294b2c4db9b23d71539c570083 : FAILURE in 12m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e4a069d6de1c429c950e526ed127fdfa : SUCCESS in 1h 42m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/9077ad4f71b74b4c8f2d8bed05c4b992 : FAILURE in 8m 58s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/858928d979764431b029f2f147db4a8f : SUCCESS in 1h 04m 01s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/6f99a92457d8464d90ca2be4f46e44ac : FAILURE in 27m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"293656e03edd4b04453028298e005ade101dae94","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-02 16:07:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"41cc3e5377d6d4118df27944b6d822f5c2ddb6fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 17:47:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/738c782fdc6046b4aa3977ad37b7fc80 : SUCCESS in 1h 20m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0be87390d6084f7a83f690fdaf3fba86 : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da01f0eee394406fbecc21c272072f8d : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fae84ca3a33f44b9bb382463ed06c4c6 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff7b5b895ff347309c18cc91d3d3505c : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/692032d3b1de497d91cc53071cb2082d : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffcde43161a04bdcad608226507f3d4f : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f5aed8aa2804493a92ca287777aa261 : SUCCESS in 3m 48s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/1b89b399cc554bd9b8440dd8753d954d : SUCCESS in 5m 46s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/acc629c245674f28931aab282b3a394d : SUCCESS in 7m 30s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/527a89d0394744c3876fbe8b961f6616 : SUCCESS in 5m 43s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/1b2cd77282cf4509b92ed2bd14d9425a : SUCCESS in 6m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/aaa4bac9920d4fe7bdffc2b533250897 : SUCCESS in 1h 31m 23s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1e83764375434ea7a3591fed6007f610 : SUCCESS in 1h 32m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/098a60d417a742ce9ade94fae33211e2 : SUCCESS in 1h 34m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/e2c295f495054bf6a99c55b2d2dc9e7d : SUCCESS in 57m 52s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/e17fb5436a12484fa8e8d25d4aac9ba5 : FAILURE in 14m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"740896f362108e4787601153e466fbbd94820eb5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-03 12:54:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e96bbfb2c697195581299a08571a916bb7997cfa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-08 15:04:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"36b31d000287b839645b78162702c5e82924db2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 15:04:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"354e80afba4086166a15e89ee2454148005f1ef6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 16:31:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7171b6273f54140bc5966fe2d5f25ef : SUCCESS in 1h 22m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f441e866076543edaba7f5ab54db9e3e : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfa9c8f147194438ab6ddb9f2ce61a8d : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99ca263683dd492685da74dfed206998 : TIMED_OUT in 41m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00419e38e2a94b998cbc3256274445a5 : TIMED_OUT in 42m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc5a763916094c398e3805995a222e9e : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a98f7959b4db47c1b465722a06f28b8d : SUCCESS in 4m 16s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/303ed2493eb449b693a85405f2ec2752 : SUCCESS in 6m 45s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/57e1157d44f34602a1ac43b8865acc9a : SUCCESS in 1h 25m 30s","accounts_in_message":[],"_revision_number":2},{"id":"2f6dceed73c081e33e759e3552c35a2571104c29","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-08 16:32:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"83bc16b6d99610fe44ee40497ebb4fc183996024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 18:14:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad8b614b88b24ce3bf881ae30528c7da : SUCCESS in 1h 20m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96a9e7585b9e406b8536021d09c436fe : TIMED_OUT in 51m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c63bbdbe88f64271b5412fa5becb8ed5 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e0aa50c887d4977888c8ae11cbe9c59 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52b0be8d15364788b3a98fc977d6425a : TIMED_OUT in 41m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad375c7e87f943b58950698ca0534670 : TIMED_OUT in 41m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fcf23292c0049c89cb14251c5674c34 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32c364a980b34a478a4ead520aa37806 : SUCCESS in 5m 53s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/7951a4d3747b4f618107bbbda42e1046 : SUCCESS in 8m 27s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/951efdd3a3eb46cd8b6de46253bc5c16 : SUCCESS in 9m 02s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/50f9898819c341f687b15de92915dd02 : SUCCESS in 8m 44s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/837dfa84e1b94b6b9b96537631438cff : FAILURE in 7m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f9c84dd17a0c4db0a8bf8999b002fd60 : SUCCESS in 1h 20m 04s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6fd7d36d40fc4fc2a65de2c017aeae83 : SUCCESS in 1h 25m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a21f9514ee47432f8dab97974121d988 : FAILURE in 1h 37m 38s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/130008ee0927416ab0a7a82da673a489 : SUCCESS in 58m 08s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/d4510718818a4911a351cf5abdda3d4a : FAILURE in 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c22b6c5807adc941f62b91c92c1b21d113c2da9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-11-04 17:13:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fcd86349fcf362ce957d2a5c4226c29decfba940","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 19:14:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f3eda7b87884931915bf699a95bc9ff : POST_FAILURE in 1h 12m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50a75076aa6942c8bd626e76cf2f36d2 : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df1781b37fd240918d3e478648587712 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57c7a89cfecd488fa7b1f27f0a230389 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66d2189c8a2645bfb55327ae3622582e : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37d7bc62db514c7394311942165371dd : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b18bbdfb533846c99f897ad50a249f3f : FAILURE in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72cb88cedda4c22b9fbc2a8b8645039 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/696e49df683543a9bfa35222d678a6ea : SUCCESS in 4m 12s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/e374ec927cff45b2be6f864f43df899c : SUCCESS in 5m 32s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/47c81ef8a2e34c6cb8479db6815cafaa : FAILURE in 6m 46s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/c4cd3dbd8c564a378d4daf75d7fec75f : SUCCESS in 6m 22s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/31c56d3235664192b2c63021aaca0280 : FAILURE in 7m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7630081c27844f2f9fdb095de0005b4b : SUCCESS in 1h 36m 18s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/bfed486c8d75442c8a888326d3350c2d : SUCCESS in 1h 28m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/11a729ee1cea4a748c7fa613cd73e59f : SUCCESS in 1h 24m 14s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/6eebdc89a4284048b81f4f7c37fa4b31 : SUCCESS in 1h 07m 18s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/edfbd952d8f442feafd18b2f63370c1d : SUCCESS in 1h 13m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"59945c325132cee56dd7dc0e5c31a7bf32fdf1af","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:26:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1902187f6be097e19ff0caed65f8fbd008f32520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:03:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e01fc09501ff4e829a3810619dfe5fc6 : SUCCESS in 1h 21m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2ea1b7c9ae94f04a2be8c983af3eb61 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faad375d61984199b6f9d8de971d8d16 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcc5c91840b749ffbece236be68efaac : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e901317532144b88bfe4ab031cdb6883 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e90f6067de0b4083908fd1144fd67877 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0c301120074f8bbc03b47f9588e2a5 : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b915dd86e184db58db18d4236b0f7e7 : SUCCESS in 4m 32s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/4204cccf8df542a7a7d3fc4f08408147 : SUCCESS in 6m 19s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/095e44ba284546e1a3b4bb61e53c4764 : FAILURE in 6m 31s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/7da79495688f485094558886482f7b89 : FAILURE in 5m 56s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/912c1e0da2e24d7cb7aaf7710f2463a6 : FAILURE in 5m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c57661ed4a0c4b79b10a90a5b9109dbf : SUCCESS in 1h 18m 22s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/163c831f181d48fdb919178b4318c4cb : SUCCESS in 1h 22m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7a3384eee608424896f7e9061cdb1561 : SUCCESS in 1h 28m 58s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/119b0fb2ee4e40bca6c8bd4cfb980509 : SUCCESS in 1h 00m 10s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e6be3f9f3cac4ed6b7c4229a5c6699f8 : SUCCESS in 1h 12m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd693007abe50ff64e41f9db0670a67205d1baef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:04:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b92731c43ca8a55475263a66e81fce8984898908","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:11:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/818f15e48e4e48c5884049b22a4e8c7f : SUCCESS in 1h 05m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43e6c64e5042494ebdcb1e413ceaf351 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1c6bf83a9ed4c3c90df702514e8d430 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65da1e93b1f54ec2877a42b530645b9c : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/964443f1dbc142aba1632a46cd8595b4 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fae81f1e50143b8a8a9f39ef3ad557c : SUCCESS in 4m 12s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/c03bc85259a443c9b2592dd2e8fa9ec7 : SUCCESS in 6m 12s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e888f44fe11f434fb4eae7c73c471a34 : SUCCESS in 1h 02m 34s","accounts_in_message":[],"_revision_number":2},{"id":"482f20d824ab0c8da03cd590a3857c5d06f6e4dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:11:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"208f6b1fa047143205629533cadff37150655d64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:13:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13564bac199d444e8de4ee2f9daa27e1 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64683efdb3a646308f0dbe297a897a00 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8d29ab3d2c4a1a6e76ad0f27fa5234a2b5e28996","revisions":{"567ab9eb051fd6468d86585f0636865e96dc2ac0":{"kind":"REWORK","_number":1,"created":"2019-12-24 11:30:00.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/700505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/05/700505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/05/700505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/05/700505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/05/700505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/05/700505/1"}}},"commit":{"parents":[{"commit":"41a44230434cf9037d57f5e26c1cc73559e198c7","subject":"Merge \"Ignore releasenote cache within git untracked files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a44230434cf9037d57f5e26c1cc73559e198c7"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-24 11:28:34.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-24 11:28:34.000000000","tz":0},"subject":"bindep: Add \u0027librdkafka-dev\u0027 dependency","message":"bindep: Add \u0027librdkafka-dev\u0027 dependency\n\nWe\u0027re seeing the following in our build logs:\n\n    In file included from confluent_kafka/src/confluent_kafka.c:17:0:\n    confluent_kafka/src/confluent_kafka.h:22:10: fatal error: librdkafka/rdkafka.h: No such file or directory\n     #include \u003clibrdkafka/rdkafka.h\u003e\n              ^~~~~~~~~~~~~~~~~~~~~~\n    compilation terminated.\n    error: command \u0027x86_64-linux-gnu-gcc\u0027 failed with exit status 1\n\nResolve this by installing the development headers.\n\nChange-Id: Idda79dc87bcd0e3367a6abd1b52104c000ad1dcd\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/567ab9eb051fd6468d86585f0636865e96dc2ac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/567ab9eb051fd6468d86585f0636865e96dc2ac0"}]},"branch":"refs/heads/master"},"8d29ab3d2c4a1a6e76ad0f27fa5234a2b5e28996":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-02 16:07:00.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/700505/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/05/700505/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/05/700505/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/05/700505/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/05/700505/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/05/700505/2"}}},"commit":{"parents":[{"commit":"4fc9e70aa11f8cb0476a927ee471adbd296bba8f","subject":"Merge \"Print warning message when connection running out\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4fc9e70aa11f8cb0476a927ee471adbd296bba8f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-24 11:28:34.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-06-02 16:07:00.000000000","tz":0},"subject":"bindep: Add \u0027librdkafka-dev\u0027 dependency","message":"bindep: Add \u0027librdkafka-dev\u0027 dependency\n\nWe\u0027re seeing the following in our build logs:\n\n    In file included from confluent_kafka/src/confluent_kafka.c:17:0:\n    confluent_kafka/src/confluent_kafka.h:22:10: fatal error: librdkafka/rdkafka.h: No such file or directory\n     #include \u003clibrdkafka/rdkafka.h\u003e\n              ^~~~~~~~~~~~~~~~~~~~~~\n    compilation terminated.\n    error: command \u0027x86_64-linux-gnu-gcc\u0027 failed with exit status 1\n\nResolve this by installing the development headers.\n\nChange-Id: Idda79dc87bcd0e3367a6abd1b52104c000ad1dcd\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8d29ab3d2c4a1a6e76ad0f27fa5234a2b5e28996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8d29ab3d2c4a1a6e76ad0f27fa5234a2b5e28996"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
