)]}'
{"id":"openstack%2Foslo.messaging~718650","triplet_id":"openstack%2Foslo.messaging~master~I5ca930cada523bc77c644703a0f02b9160816231","project":"openstack/oslo.messaging","branch":"master","topic":"bug/1871813","hashtags":[],"change_id":"I5ca930cada523bc77c644703a0f02b9160816231","subject":"Print warning message when connection running out","status":"MERGED","created":"2020-04-09 10:07:27.000000000","updated":"2020-05-29 14:07:15.000000000","submitted":"2020-05-29 11:38:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718650-1590752318194-48d602ed","meta_rev_id":"10f795322b2c888349cab541287eb2edb220e3bc","_number":718650,"virtual_id_number":718650,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-29 11:38:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2020-05-28 09:42:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-05-13 04:10:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2020-05-28 09:42:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 14:26:49.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-05-12 14:05:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-28 09:42:06.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-05-29 11:38:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5099a329fee00dabcfe3d6504e5f8b17949c42b3","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-04-09 10:07:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a7a05ff0f9ef7836362521c6e0d14b24fcef06e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 12:28:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfa53413b56a47bcb1119d8d58dc34ac : SUCCESS in 1h 24m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3201c5c0ad5347af99b4a2311c1a643e : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cdc78fdb375482f987dedc6521e31f9 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9d85ff51d5141f48d1039e2ab6bd5e3 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e455ffb5b1a3419ca3e8440cacec6801 : SUCCESS in 7m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d05b1a8de194ab0be1fe8f38db1b6dd : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33421fabcc5d451fb6ee3274048bf269 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b484b521d1f4f5982a33e6291c6a873 : SUCCESS in 5m 09s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/2497c95ed1ea4db0be8db447b858ccf1 : SUCCESS in 6m 41s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/b3990d454d0c4002aea6c2d2364fb9c9 : SUCCESS in 9m 21s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/5ec85593d1e24f76aa8c5af3790b6495 : SUCCESS in 6m 42s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/c00939ead9bc4c55ae8a43e3a43be345 : SUCCESS in 7m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/957eb32b622147e9a4f127d1c64e5e22 : SUCCESS in 1h 41m 01s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/bf3fd769858748a997fdf6bd603f98ca : SUCCESS in 1h 25m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/215e1178524a4e6e8a8591131af5caed : SUCCESS in 2h 13m 01s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/97beaa3c97b8494f8219ceae8441171f : SUCCESS in 1h 12m 15s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/56cbf12806d248e9b3f7f267c949c74a : FAILURE in 23m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f259a227134f9b2d06c0c592aecaaeca5ee5f4b4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-16 12:06:20.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nHello\n\nI\u0027ve some minor concern inline","accounts_in_message":[],"_revision_number":1},{"id":"b18b7683460e594257aadcb439416414115a6542","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-04-20 14:26:49.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nJust a suggestion on the warning text:","accounts_in_message":[],"_revision_number":1},{"id":"f26d8ea18a4b4e66723a312c20c234e01b9bc880","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-05-11 13:03:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ce4ddce2d78895f43a94da35c2b3414ff839d81","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-05-11 13:04:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4e7781d4f064d8d7b34ce7d5e5ee51624b13c3ac","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-11 13:34:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1685784256d82e0d7274bb1cbbcceb35a93d8266","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-11 13:36:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"00ae09ff07ed174ecc52fdd056455ec2ee0ceb76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 14:47:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69a9449b56ae4b61a5e57755139a8c78 : SUCCESS in 1h 32m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/612bc244933e46968e235d6f2a7c43af : SUCCESS in 9m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e716bd2299494d43addf46f92eb159a0 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03529c1408f8414a81406c509e3c1978 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af2ef6d3e1444ae2aa93bd2bd02246a6 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27ab39ea8001441bba8aabc0f5b34f07 : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b6135cf0904e6f9c7fdf100769f65a : SUCCESS in 5m 54s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/e24066628e6c4a4d876023b5aaa978e7 : SUCCESS in 8m 43s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/9bb547b8515b4d10b96d50add8cb697a : SUCCESS in 8m 36s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/60009648603e40518e8000ae79fe037c : SUCCESS in 9m 48s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/8e3f382fa31c439a8b110ebbdd7a7e89 : SUCCESS in 10m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/da233a32ada9402abda037c953ed74e1 : SUCCESS in 1h 37m 28s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/87ea0e09c3404032b8316b46123ecc9d : FAILURE in 30m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/447f8ed17a614607884e5fe6507769e2 : SUCCESS in 1h 37m 16s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/dab7aed1cfc64760bcbd913ee717eaa7 : SUCCESS in 1h 00m 24s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/95ded37d7ded4a18a170fc7babcdf1d3 : FAILURE in 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"13bcab144d57e8654f20b0c995a58a5fac084eba","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-05-12 00:09:01.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"dd747c3b4c9b32b5dfa82265b023620d437fef89","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-05-12 00:10:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"35474934d837b464bcb3cef1cac05e389b7d5b80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 01:53:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff5305fca44e4c33a4a3f252a9f86f92 : SUCCESS in 1h 31m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bc76af5e9204f10bc4a6bb66f6f8d8c : SUCCESS in 6m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24386f392dbb463782b8bd84584f9aba : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23811150553947b687310a26f037712f : FAILURE in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c2984f1d03c4dc5b77bccf7798fa0f0 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbe076a6a67f484f830a8014da3c7050 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dff5bc1a764345148b78a328bef2b792 : SUCCESS in 6m 03s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/b1a4843203b540f488b6cfb5b1e55f51 : SUCCESS in 6m 16s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/14b4789f0a2a4c289935643255c49423 : SUCCESS in 6m 52s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/c54fef5b12bd489e9029ef31dd456dfa : SUCCESS in 6m 46s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/fe3fca25acfd4eaea0af40842950a1e8 : SUCCESS in 7m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2b6aa6140d724395a19e1584d51f838c : SUCCESS in 1h 34m 51s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/df41ff3aa52841c0b89d74f1fb19bddc : FAILURE in 37m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/33d1a8018e984f63bc9bd59e94b4ad4f : SUCCESS in 1h 39m 40s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/49331272e565492c9c1cb7839f38857e : SUCCESS in 57m 10s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/78d99e1a4a5b469b9b0a7afba19907bc : FAILURE in 15m 43s (non-voting)\n\nWarning:\n  Comments left for invalid file oslo_messaging/tests/drivers/test_amqp_driver.py\n  Comments left for invalid file tools/simulator.py\n  Comments left for invalid file oslo_messaging/tests/functional/utils.py","accounts_in_message":[],"_revision_number":4},{"id":"99a376e3496ed22a0e55ce7e1a605691717d5ff5","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-05-12 13:34:14.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"4633106a6c6a74e432933be226e2bf77ade3472c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-12 14:05:41.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":5},{"id":"d7355541390910314d10adcc1504eb524814cf2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 15:19:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ff091f0cb0b47d19ee0b1cff362444e : SUCCESS in 1h 21m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e65620d98ad743b5a4f8d302dba0f7e0 : SUCCESS in 6m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39c2496cd59442739c93aeabaca34b82 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f31f3bbc5d7b49f7bbab3e3cd025d9a4 : FAILURE in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8c5e1f4a11a48699db723fd7695670b : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e862b51d81b94f07b61cca28926f06de : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e0105e25f63467489bc328d9061ace8 : SUCCESS in 6m 00s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/4ce78480c38f49569c5777fc4c29323d : SUCCESS in 6m 30s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/cd58b93a109e4405a2fbcd6ff5819960 : FAILURE in 8m 53s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/f7cb1641d95d48b6ba1b0c002a753f29 : SUCCESS in 9m 26s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/0492dae8f22448f6a0b2db4e564a55bc : FAILURE in 7m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/60af13ae43ee4aec9033fba9fb5e3d59 : SUCCESS in 1h 38m 21s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/15d9d7908d0b4364a1b9a96c2281afb9 : FAILURE in 30m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1cb2f4a82f944b3aab7d91acbc4b80d8 : SUCCESS in 1h 22m 03s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/338585b668374f6c9894a2c7eca07a60 : SUCCESS in 1h 04m 19s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/df5b64576aa7401f838407d169fc3ca7 : FAILURE in 17m 54s (non-voting)\n\nWarning:\n  Comments left for invalid file oslo_messaging/tests/functional/utils.py\n  Comments left for invalid file oslo_messaging/tests/drivers/test_amqp_driver.py\n  Comments left for invalid file tools/simulator.py","accounts_in_message":[],"_revision_number":5},{"id":"fa91f93130fe1c0335e87b801f8c4b42f1f9a569","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-05-13 04:10:11.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"f05ba27a4675357467f03de856cb70722ca66063","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 05:56:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f51b7d85e60c4e268c99fdfe8cbce599 : SUCCESS in 1h 14m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/990df192cf8f43e19495102220258a9b : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08d743b4941f47a981e79e55531ccb02 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e050bfa99d994b0c909a2b97228365fa : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df8b43592413497881cafaafd7497ca7 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2232c4144ebf4278b6b72e152910788d : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c7126b6e06448b3a84d15555d803552 : SUCCESS in 5m 39s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/01ab9f98e951463489007d8bc8ca35ed : SUCCESS in 6m 25s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/7d6a1d678baa47a39be5cda884c3e6d5 : SUCCESS in 6m 52s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/8be8c8b7586140eeb7d34d6ed7e14349 : SUCCESS in 6m 36s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/81eef6c36a38436491090411855d19ae : SUCCESS in 6m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/85e1f066ed804608bf9fb8fc8edde058 : SUCCESS in 1h 30m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4f5abbeb39fe42159e7be248b6116903 : FAILURE in 37m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8490574bec2c432aa113a7d796c1fd87 : SUCCESS in 1h 39m 33s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/86724ade1b4348929e243adb7d323a26 : SUCCESS in 1h 00m 39s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/345c39c9780d45ec965d256fadc2a1fb : FAILURE in 11m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c6ba835094705a7b881e9a9a5f80bd37f87c64d8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-28 09:42:06.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nSeems reasonable","accounts_in_message":[],"_revision_number":6},{"id":"7a09f53f008973cdb019e51cd6caa21e71406725","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:42:18.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0b634e18a94033cb7e0b75b8f942cfb02dfbb8f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 11:03:10.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6755175c40b04733b4d56d40630ec0d7 : SUCCESS in 1h 17m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1156b32e48b0457f9f3ff504460b0823 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7548b494bf824a8bbe7cac78dea64d29 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1d0e382b439482ab7ef358f6e749897 : SUCCESS in 8m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c02a10e6f9fe4e768d7d46416aec2cd4 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b97d66c9325431aaa98aab01ece030f : SUCCESS in 4m 29s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/e05b253852144ca4b5bf9add8a5c3956 : SUCCESS in 7m 08s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2846723094014da5a3444d71a71d77be : FAILURE in 1h 05m 54s","accounts_in_message":[],"_revision_number":6},{"id":"9702584ed292f92884c20ff6f4da9d9e50b67b7c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-29 08:38:21.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"dd69eebd3b48dd0d630f85531ade81458a2a2e45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 10:08:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d8f53ebe26247c5a8350982e084ec40 : SUCCESS in 1h 15m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f17a1e00698346b89acb8a3d4d0dae27 : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06a6225431f949c99a9baf8ed4b5f6f6 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da7c8c0e6bc54de39a63d7418fd7da18 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f921347c6bec48958942781020bac343 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/165965a1747845f3b45f72f74417caf5 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41d1bb2abf24287a2d7e82af524ae96 : SUCCESS in 4m 34s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/6331dc65cb9b460ebc10c87057277f25 : SUCCESS in 6m 25s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/ad3c818d37a741af952061e6d01639ad : SUCCESS in 7m 37s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/f6aea12fdfce4bffb222c19022fa8149 : SUCCESS in 6m 42s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/6360bf628de5423dad55186ee0d978d8 : SUCCESS in 7m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f1cd920e8c7a42d09af73e68b89bd694 : SUCCESS in 1h 24m 51s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a63d4806afbf4a36b9e17b5a6a8c0715 : FAILURE in 1h 20m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f73e1efc9249471a99b394e37a0fc4f2 : SUCCESS in 1h 25m 35s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/141264dbee624c20b4fa43a8bd4186a8 : SUCCESS in 54m 02s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/6807c66b76444ffdb48fa252de2634a7 : FAILURE in 16m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b5fad63111df0c03eec00054b5b776cd0981100b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 10:09:07.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"6aa3857c81219e4140df7b32875c3943f56858c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 11:38:37.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50b0d7bbe7e444a3a114987e092b1cf8 : SUCCESS in 1h 27m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a33463971ec44cdcae7e886b3ef09897 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b58a9e52f3644629895f98b44ca144cd : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f17459ba7e81427b9bec418c80c05062 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e368a8c09a5444f8a43aa7517c9a9815 : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ad9e48c5c449499107a1fafff0a5c5 : SUCCESS in 4m 18s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/cedb1dad17954e85be5f75c77b9dfa11 : SUCCESS in 6m 45s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6fef980436484926a7cae001498a6a2c : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":6},{"id":"4a7dfddc41cf21474f9201c2c535d1e99672dc55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 11:38:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"97bc7667c1e79177af7336143b9adef5ee32252d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 11:40:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be8838dda7af4056b23d584348d7a2ce : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":6},{"id":"5f32a79337c0c65bca090c44d30348450c69eac3","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-05-29 14:07:15.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 39f770f55b3331b7a99b3ea42a34982a20635cf2","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"96300a32213d1a9bf114b90bef6971d69b839d1d","revisions":{"c89cd7eba156b56393e0ae4b4b87ca486362f4fc":{"kind":"REWORK","_number":1,"created":"2020-04-09 10:07:27.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/50/718650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/50/718650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/1"}}},"commit":{"parents":[{"commit":"7f638bb493571f2a6d30d3679945119d6596cfa7","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f638bb493571f2a6d30d3679945119d6596cfa7"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-04-09 10:00:39.000000000","tz":480},"committer":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-04-09 10:07:15.000000000","tz":480},"subject":"Print warning message when connection running out","message":"Print warning message when connection running out\n\nI believe when connection pool running out of available\nresources, there will be all kinds of weird consequences\nfollowing. So let\u0027s print warning logs when we are going\nto waiting indefinitely for the lock\n\nChange-Id: I5ca930cada523bc77c644703a0f02b9160816231\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c89cd7eba156b56393e0ae4b4b87ca486362f4fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c89cd7eba156b56393e0ae4b4b87ca486362f4fc"}]},"branch":"refs/heads/master"},"6a18f8811a215e79da470f1be4229d84aeacfe93":{"kind":"REWORK","_number":2,"created":"2020-05-11 13:03:33.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/50/718650/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/50/718650/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/2"}}},"commit":{"parents":[{"commit":"7f638bb493571f2a6d30d3679945119d6596cfa7","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f638bb493571f2a6d30d3679945119d6596cfa7"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-04-09 10:00:39.000000000","tz":480},"committer":{"name":"ushen","email":"yshxxsjt715@gmail.com","date":"2020-05-11 13:03:07.000000000","tz":480},"subject":"Print warning message when connection running out","message":"Print warning message when connection running out\n\nI believe when connection pool running out of available\nresources, there will be all kinds of weird consequences\nfollowing. So let\u0027s print warning logs when we are going\nto waiting indefinitely for the lock\n\nChange-Id: I5ca930cada523bc77c644703a0f02b9160816231\nCloses-bug: #1871813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6a18f8811a215e79da470f1be4229d84aeacfe93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6a18f8811a215e79da470f1be4229d84aeacfe93"}]},"branch":"refs/heads/master"},"7298feefe18e2529c298ad2efed199498abaa8b4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-11 13:04:03.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/50/718650/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/50/718650/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/3"}}},"commit":{"parents":[{"commit":"c1768401f7a47bd88c044687a00f2e139f1985b4","subject":"Remove monotonic usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1768401f7a47bd88c044687a00f2e139f1985b4"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-04-09 10:00:39.000000000","tz":480},"committer":{"name":"norman shen","email":"yshxxsjt715@gmail.com","date":"2020-05-11 13:04:03.000000000","tz":0},"subject":"Print warning message when connection running out","message":"Print warning message when connection running out\n\nI believe when connection pool running out of available\nresources, there will be all kinds of weird consequences\nfollowing. So let\u0027s print warning logs when we are going\nto waiting indefinitely for the lock\n\nChange-Id: I5ca930cada523bc77c644703a0f02b9160816231\nCloses-bug: #1871813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7298feefe18e2529c298ad2efed199498abaa8b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7298feefe18e2529c298ad2efed199498abaa8b4"}]},"branch":"refs/heads/master"},"eea3c1c679f53ff9d07104e9a85d2243c8b5738b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-05-12 00:09:01.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/50/718650/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/50/718650/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/4"}}},"commit":{"parents":[{"commit":"c1768401f7a47bd88c044687a00f2e139f1985b4","subject":"Remove monotonic usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1768401f7a47bd88c044687a00f2e139f1985b4"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-04-09 10:00:39.000000000","tz":480},"committer":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-05-11 23:56:16.000000000","tz":480},"subject":"Print warning message when connection running out","message":"Print warning message when connection running out\n\nI believe when connection pool running out of available\nresources, there will be all kinds of weird consequences\nfollowing. So let\u0027s print warning logs when we are going\nto waiting indefinitely for the lock\n\nChange-Id: I5ca930cada523bc77c644703a0f02b9160816231\nRelated-bug: #1871813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/eea3c1c679f53ff9d07104e9a85d2243c8b5738b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/eea3c1c679f53ff9d07104e9a85d2243c8b5738b"}]},"branch":"refs/heads/master"},"3fd21bdd2d45a6f6aad411a4d5e6ab0e065953a7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-05-12 13:34:14.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/50/718650/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/50/718650/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/5"}}},"commit":{"parents":[{"commit":"7f638bb493571f2a6d30d3679945119d6596cfa7","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f638bb493571f2a6d30d3679945119d6596cfa7"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-04-09 10:00:39.000000000","tz":480},"committer":{"name":"ushen","email":"yshxxsjt715@gmail.com","date":"2020-05-12 13:33:25.000000000","tz":480},"subject":"Print warning message when connection running out","message":"Print warning message when connection running out\n\nI believe when connection pool running out of available\nresources, there will be all kinds of weird consequences\nfollowing. So let\u0027s print warning logs when we are going\nto waiting indefinitely for the lock\n\nChange-Id: I5ca930cada523bc77c644703a0f02b9160816231\nRelated-bug: #1871813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3fd21bdd2d45a6f6aad411a4d5e6ab0e065953a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3fd21bdd2d45a6f6aad411a4d5e6ab0e065953a7"}]},"branch":"refs/heads/master"},"96300a32213d1a9bf114b90bef6971d69b839d1d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-05-13 04:10:11.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/50/718650/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/50/718650/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/50/718650/6"}}},"commit":{"parents":[{"commit":"c1768401f7a47bd88c044687a00f2e139f1985b4","subject":"Remove monotonic usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1768401f7a47bd88c044687a00f2e139f1985b4"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-04-09 10:00:39.000000000","tz":480},"committer":{"name":"norman shen","email":"yshxxsjt715@gmail.com","date":"2020-05-13 04:10:11.000000000","tz":0},"subject":"Print warning message when connection running out","message":"Print warning message when connection running out\n\nI believe when connection pool running out of available\nresources, there will be all kinds of weird consequences\nfollowing. So let\u0027s print warning logs when we are going\nto waiting indefinitely for the lock\n\nChange-Id: I5ca930cada523bc77c644703a0f02b9160816231\nRelated-bug: #1871813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/96300a32213d1a9bf114b90bef6971d69b839d1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/96300a32213d1a9bf114b90bef6971d69b839d1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
