)]}'
{"id":"openstack%2Foslo.messaging~719838","triplet_id":"openstack%2Foslo.messaging~master~I76094aecc710a44f43b6d4052851b87a0f596932","project":"openstack/oslo.messaging","branch":"master","topic":"add-victoria-python-jobtemplates","hashtags":[],"change_id":"I76094aecc710a44f43b6d4052851b87a0f596932","subject":"Add Python3 victoria unit tests","status":"MERGED","created":"2020-04-14 10:15:34.000000000","updated":"2020-04-14 15:48:32.000000000","submitted":"2020-04-14 15:41:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719838-1586878898156-eaa21e4c","meta_rev_id":"9ca4bf1ca2c69eb13c765a0b12ea9ee880ec22ab","_number":719838,"virtual_id_number":719838,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-14 15:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-04-14 10:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-04-14 10:52:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 10:52:52.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-04-14 15:41:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd6a28b83f55ff430e59e460e2a90916f1a54a63","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-14 10:15:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2eabb37ba0b2f62fe307c67d454ac14f9df0bfd9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-14 10:52:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"984b52759a707a617b2f6a02595a9d5b4e67c508","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 14:09:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0dc4cb5bfa314d37b0a10dbbcac1286d : SUCCESS in 1h 31m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58ef267d484e44369a8cb45e78e7ed1e : SUCCESS in 7m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef8059a6b1314c76adae586200432f69 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cedc15a73ef4056aef747e02cf1920e : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32ddcca70b8a4fd69270d6bbcb1d40e7 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb2125932b0a4cb187c8a1c564614e6e : SUCCESS in 8m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f24201bd19b42049fd40b0c9e4daf44 : SUCCESS in 4m 42s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/d409ce47f7d6480793f320eac990a329 : SUCCESS in 5m 37s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/b937961c5ac5411eb9784c37be9c946a : SUCCESS in 7m 14s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/3f06bce9adb344fd8c8871c8e5ccce50 : SUCCESS in 5m 51s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/a0325548760c4e19a8c934a311285aac : SUCCESS in 7m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/79516a0ee085415e897d4ebf5117c9ed : SUCCESS in 1h 46m 07s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ee3f8da409274d9493053c4f27531838 : SUCCESS in 1h 27m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f61de98464064f43bcbf9f4120b07747 : FAILURE in 1h 31m 44s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f296f9c11e3943c1a91ddeae6a4b3bcf : SUCCESS in 1h 16m 08s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/a7e9762b5b8441419f34dff8c4e2ec21 : FAILURE in 25m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"802a7f80265e5e5ed2fa68c7911ae726d829ac41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 14:10:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b6f538d496f2fe752ea56d26dc2772fe423cc30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 15:41:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8964115f2b04473a45852e6843ebe2d : SUCCESS in 1h 13m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c570a6813cfb4c3c80b4bd500d37f76a : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c590b2d6404144b7326362f5947588 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2efd457cec494933bc850d47531402e3 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6fcffb53126471da8af2a5ac191f4c6 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12fe388d825b4779809b380d0c7f4ace : SUCCESS in 4m 30s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/48e7528e312b4a9db0388d419bfa1575 : SUCCESS in 6m 56s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a98cd0105ee14cbd965546c64499222c : SUCCESS in 1h 25m 16s","accounts_in_message":[],"_revision_number":1},{"id":"092bb32becfeec7e3b98be1973ddf1ef22ce7085","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 15:41:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b506c20ffd87b542b81e3479a4e6d102f4fd87c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 15:48:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/983cf88dbcd346159890bad126ad5d6d : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a66fe2abd8ce64dcdbbf64c6da3c820590cad41","revisions":{"9a66fe2abd8ce64dcdbbf64c6da3c820590cad41":{"kind":"REWORK","_number":1,"created":"2020-04-14 10:15:34.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/38/719838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/719838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/719838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/719838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/719838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/719838/1"}}},"commit":{"parents":[{"commit":"434ec937f9e75d5e79226c0e39c63f0b41465e00","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/434ec937f9e75d5e79226c0e39c63f0b41465e00"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-14 10:15:32.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-14 10:15:32.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I76094aecc710a44f43b6d4052851b87a0f596932\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9a66fe2abd8ce64dcdbbf64c6da3c820590cad41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9a66fe2abd8ce64dcdbbf64c6da3c820590cad41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
