)]}'
{"id":"openstack%2Foslo.messaging~726707","triplet_id":"openstack%2Foslo.messaging~stable%2Frocky~Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72","project":"openstack/oslo.messaging","branch":"stable/rocky","topic":"bp/transport-options-stable/rocky","hashtags":[],"change_id":"Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72","subject":"Add transport_options parameter","status":"ABANDONED","created":"2020-05-11 00:29:26.000000000","updated":"2022-04-05 09:58:45.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"908a59cc1bc135b2d041277686275c39837be266","_number":726707,"virtual_id_number":726707,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"date":"2021-08-05 14:35:38.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-05 16:00:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-05-12 06:08:03.000000000","_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-1,"date":"2020-06-22 15:41:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-05-11 00:29:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 00:29:26.000000000","updated_by":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"reviewer":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"state":"REVIEWER"},{"updated":"2020-05-11 02:00:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-11 08:32:13.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-13 14:51:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-22 15:41:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-08-05 14:37:09.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REMOVED"}],"messages":[{"id":"f7ec851baaff8ba5bf43a34e4afa95c04161ecb5","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-05-11 00:29:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ab3afee13d894f080caae18bfb4a2d4e521efd7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 02:00:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0939742c55b45f5b20812587c4b7b4d : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e03b660cc83a45168b657b6a70eefb28 : SUCCESS in 5m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0caff9516a7b4b3c97f02563b04b432c : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246301cbe09241e98ad174ea94a4e865 : SUCCESS in 3m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bd1dfe7765f74ac385b8a623a8ccb9c8 : SUCCESS in 1h 30m 02s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/c28801767542417897ba607eff4aba50 : SUCCESS in 5m 05s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/d182369224da473493beaddd870cbcd9 : FAILURE in 5m 08s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/49cac21ad34b4463b2de6cf4222990b6 : FAILURE in 4m 40s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/8144a199d46240478519d99d8f8d5910 : SUCCESS in 5m 39s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/752094a6575a421292f33ca9ac1733bb : FAILURE in 5m 12s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/8ffce8ae5d7a4a2d97a0910f842dbae7 : FAILURE in 5m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3c5550d32c3448048986a88c05a5085a : SUCCESS in 1h 23m 21s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d3ab81d970e643a7b53bf2915a6daaf5 : POST_FAILURE in 42m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/d398b601e38242c38f5697b7e9927165 : FAILURE in 37m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9f57c25544ec413582dfc2f662b48b4d : FAILURE in 31m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/d124dad6807f4660a992b68b12a78bfc : FAILURE in 35m 34s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/265f1257e1934bf2b9b90fe8bb8e4c3e : SUCCESS in 52m 16s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/edd3101aabd7488785aef049461e379e : SUCCESS in 1h 06m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/8b32951d0d96496fa61681547aa5348b : FAILURE in 36m 44s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/d1c97d0730334156ade9d887c5e1c6e4 : FAILURE in 36m 45s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/45c4db21b2d84007947c02aa5336c25c : FAILURE in 33m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27dd256bd7d549d3b8d53d70b29f08b7 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d083c155c3a9145769c68a4c1f1045e9d9d3e61b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-11 08:32:13.000000000","message":"Patch Set 1:\n\nNew features should not be backported.\n\nand if you backport, you first need to backport to newer branches.","accounts_in_message":[],"_revision_number":1},{"id":"3dbc8c9cf7b1723c762a0abaa6c2fab33e93cf9f","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-05-12 06:08:03.000000000","message":"Patch Set 1:\n\n\u003e New features should not be backported.\n \u003e \n \u003e and if you backport, you first need to backport to newer branches.\n\nthanks, I think you are right. But although this is a new feature, I believe back port this flag to previous release is really must for identifying and resolving a wide range of MQ related issues.","accounts_in_message":[],"_revision_number":1},{"id":"7182c534342d285da2a9d0361d23177fc023ffc7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-13 14:51:56.000000000","message":"Patch Set 1:\n\n\u003e \u003e New features should not be backported.\n \u003e \u003e\n \u003e \u003e and if you backport, you first need to backport to newer\n \u003e branches.\n \u003e \n \u003e thanks, I think you are right. But although this is a new feature,\n \u003e I believe back port this flag to previous release is really must\n \u003e for identifying and resolving a wide range of MQ related issues.\n\nHello,\n\nI agree with Andreas backport a feature to a stable branch isn\u0027t something we usually do, especially on oslo [1].\n\nThese changes was mainly added to allow us to use the `mandatory` flag [2][3] through transport option.\n\nIIRC you can\u0027t pass additional config manually that will be consumed via this mechanism, so in other word I don\u0027t think you can use it without adding additional features to oslo.messaging in a second time.\n\nI mean, I think these changes used standalone are not really useful.\n\nThoughts?\n\n[1] https://docs.openstack.org/project-team-guide/stable-branches.html#active-maintenance\n[2] https://blueprints.launchpad.net/oslo.messaging/+spec/transport-options\n[3] https://review.opendev.org/#/c/667902/","accounts_in_message":[],"_revision_number":1},{"id":"b79e8ac91e3685576cef3bdeb8571d100d93d8bc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 15:41:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6e8962ecc294aa4f7842dbdae60668f6c110cf1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 14:25:47.000000000","message":"Abandoned\n\nRocky is no longer maintained (https://releases.openstack.org/). Thank you.","accounts_in_message":[],"_revision_number":1},{"id":"2e99869e4c76a582849fc54a105ae8216f3c9a73","tag":"autogenerated:gerrit:restore","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 14:35:22.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"de99f756ade5fd8c5d461003f5a9af2e100e9101","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 14:35:38.000000000","message":"Patch Set 1:\n\nSorry ignore my previous comment","accounts_in_message":[],"_revision_number":1},{"id":"2bd6f4518765ebd8f0b3b88c137c31d7134ae626","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-08-05 14:37:09.000000000","message":"Removed reviewer Andreas Jaeger.","accounts_in_message":[],"_revision_number":1},{"id":"caa64f7bee0b9bcfb75e960c5ee946debed7e12f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 16:00:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce2c5d94b6ec44c28ceacd5f55405bc2 : FAILURE in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ef4f02ac01b4f1588f11c5e4a04bd51 : FAILURE in 5m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/50345682b32b4173b4bf120bb7696f62 : FAILURE in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac3336ee49e49fda6089292f04377a1 : FAILURE in 3m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9756927612244f318ecc2775ec629d4a : SUCCESS in 1h 21m 16s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/d4753e814e1a4f37b47d4a7c71727dae : FAILURE in 4m 22s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/aa5419c89a9b435f99b1b093a42822a6 : FAILURE in 4m 02s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/8ed07c36813041b4841ea967defb57c6 : FAILURE in 4m 53s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/c6d049b9e7d84b4298f2e0b3baa4f7f2 : FAILURE in 4m 23s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/f86855b2c52a465fbdc44af976b524f9 : FAILURE in 3m 54s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/54e6be1e15734205816d39a645d6c693 : FAILURE in 5m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/dd581cba3c8b4f88849ec59ebe2f5453 : SUCCESS in 1h 19m 10s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/12c84195ca1a4bfd9eb20faaf37d3838 : FAILURE in 29m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/788394d2fcf64a92bcd5c658ec2372f3 : FAILURE in 17m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/0aee51acd67140b2be6a9ae6bc6efec0 : FAILURE in 27m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/2b05dc229c0443089692d35b63a39068 : FAILURE in 18m 41s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/ff302c058294457ba5abf6333b020bdc : SUCCESS in 50m 00s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/f4b677927102422f86460ecbb03f3532 : SUCCESS in 59m 17s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/14f11c3cf8584846acdbf1d64a12653a : FAILURE in 33m 49s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/4673331165954a47bb2466fd0b4d25f3 : FAILURE in 34m 06s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/2433c6360f014d8ca3e40b84b3eb72c3 : FAILURE in 30m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c424523d2c4e4ca6ab805e96061ae4ae : FAILURE in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"908a59cc1bc135b2d041277686275c39837be266","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-05 09:58:45.000000000","message":"Abandoned\n\nI think it\u0027s probably too long since this was proposed to merge now. Please let me know if anyone disagrees","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9931198ab4b9b80fd76eb619c4dbda9fd96615a7","revisions":{"9931198ab4b9b80fd76eb619c4dbda9fd96615a7":{"kind":"REWORK","_number":1,"created":"2020-05-11 00:29:26.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/07/726707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/07/726707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/07/726707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/07/726707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/07/726707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/07/726707/1"}}},"commit":{"parents":[{"commit":"b67a457e4fa71e9220c149087dce013c7f81144f","subject":"Use default exchange for direct messaging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b67a457e4fa71e9220c149087dce013c7f81144f"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"committer":{"name":"norman shen","email":"yshxxsjt715@gmail.com","date":"2020-05-11 00:29:26.000000000","tz":0},"subject":"Add transport_options parameter","message":"Add transport_options parameter\n\nWith this new parameter is possible to pass other parameters\nfrom the client to the drivers.\nSo it is possible to tune the driver behavior.\n\nFor example can be used to send the mandatory flag in RabbitMQ\n\nNote:\n  - The transport_options parameter is not actually used (yet).\n  - This part of blueprint transport-options  (first part)\n\nImplements: blueprint transport-options\nThe blueprint link is\nhttps://blueprints.launchpad.net/oslo.messaging/+spec/transport-options\n\nChange-Id: Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72\n(cherry picked from commit 83266cc6ea7dbe682f8fdb960a548c68fd1235b5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9931198ab4b9b80fd76eb619c4dbda9fd96615a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9931198ab4b9b80fd76eb619c4dbda9fd96615a7"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
