)]}'
{"id":"openstack%2Foslo.messaging~726746","triplet_id":"openstack%2Foslo.messaging~master~I40522c4accb4aaf8115d11fee8b081e2d991cb4d","project":"openstack/oslo.messaging","branch":"master","topic":"drop-six","hashtags":[],"change_id":"I40522c4accb4aaf8115d11fee8b081e2d991cb4d","subject":"Remove six usage","status":"MERGED","created":"2020-05-11 08:21:38.000000000","updated":"2020-05-14 11:33:18.000000000","submitted":"2020-05-14 11:30:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726746-1589455819562-d9d82c02","meta_rev_id":"bee6302dc9caa565c98f9f43eb13b8982ca27c62","_number":726746,"virtual_id_number":726746,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-14 11:30:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2020-05-11 15:14:06.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2020-05-14 09:59:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2020-05-14 09:59:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 15:14:06.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-05-14 09:59:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-14 11:30:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a7d759e7b5b326f7b6d344f01fae6b4fe2293ab","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-11 08:21:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96166ed6ee141d14ec38e8a8344f7a4e93b575e0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-11 08:22:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b60da6df12e0fc14d7e4aa820e2bdf5580a28ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 10:05:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d5d3a7402197463487875995ed2a90fb : SUCCESS in 4m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96746e5a0207482f8ddba5cb1a3bf2e1 : SUCCESS in 1h 32m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0bd54807c934b05b15dc3d637a74215 : SUCCESS in 8m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f87a80ec70c1404f9f7ddcf86a5cf3df : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c340df604aac44948c8254b63327ab8f : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9daeb9a1a64e499890078cbb99a4ffce : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87172aaadf79463ea2aea588848d5ffe : SUCCESS in 10m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c2fe42b17f64de89e73f1d1b66cddb5 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/372a4fabfa704a1a9956e5d416cbba22 : SUCCESS in 6m 01s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/76f66d113fae4283a7a3393acbee5675 : SUCCESS in 7m 46s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/2b69ed5b3e594850937a82187e88563c : SUCCESS in 7m 52s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/ff415ac04473444294c41c429b2fdfc5 : SUCCESS in 10m 58s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/0edcfcaaf0694e45ae9b9a25673aaf63 : FAILURE in 8m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/62d3ec67804f41bbae3b7c7c2e9e1c8f : SUCCESS in 1h 34m 10s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/304b2d00cd4248a0a3737560292cc919 : FAILURE in 32m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b481dcce7aff4525a10932fd3c241d69 : SUCCESS in 1h 41m 51s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/28dc654db3324f81a2b4d5521978ddf4 : SUCCESS in 59m 59s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/37d948ab6c3e48a0977610a88e42f431 : FAILURE in 16m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df1921416409d1425d6ed7f7675007abed1ca27e","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-05-11 15:14:06.000000000","message":"Patch Set 2: Code-Review+2\n\nAdieu six!","accounts_in_message":[],"_revision_number":2},{"id":"5bc286f4f23806365df3359b290373d4cd2364ed","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-14 09:59:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThanks Andreas!","accounts_in_message":[],"_revision_number":2},{"id":"99e53e558e264a819eab8a051915d45f36855997","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 09:59:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6b02884fcda05ef2f773857894e5dc2947e57b3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 11:30:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/648f736b4ba24420a00f2e58c58f6cc2 : SUCCESS in 4m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5a4a2f513d347de8d3625bd9e7f59dc : SUCCESS in 1h 25m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e9dd9a96f0a4a1caa4bfa7b4feed266 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ee0a4365dc4fe1a30e120f44a77540 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ccbf789e824457990cc9fc85b498095 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/279eb305a5f044009cb17a136aaf1b9f : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/822a91d324bd4e528fa4a3c91544ca2f : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9fff0e4b85448de851094e27579b22b : SUCCESS in 4m 53s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/bc4e79f726ea4cf7b4374ecd0c427611 : SUCCESS in 8m 15s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f10e026e9cdd433b97bb909e98fea0bc : SUCCESS in 1h 25m 22s","accounts_in_message":[],"_revision_number":2},{"id":"4afc9f704acd4beacb653d4150e7f0f0f87b85c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 11:30:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fb4e39db5dccdd02d1b327fa0923abc082496b3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 11:33:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/103f5658b1aa4949bbbe1f749ce1f2c3 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f71914cc4d504471b6520ae62a42151f : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e44c9883066d9b2d081a594b97aac3d598d491c9","revisions":{"6f0ce00f6f06d2af90224ce498dcf849e78ec708":{"kind":"REWORK","_number":1,"created":"2020-05-11 08:21:38.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/46/726746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/46/726746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/46/726746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/46/726746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/46/726746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/46/726746/1"}}},"commit":{"parents":[{"commit":"c1768401f7a47bd88c044687a00f2e139f1985b4","subject":"Remove monotonic usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1768401f7a47bd88c044687a00f2e139f1985b4"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-11 08:17:47.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-11 08:20:07.000000000","tz":120},"subject":"Remove six usage","message":"Remove six usage\n\nRemove six, the python 2/3 compatibility library. It\u0027s not needed\nanymore since the repo is python3 only.\n\nRemove a now unneeded hacking test.\n\nChange-Id: I40522c4accb4aaf8115d11fee8b081e2d991cb4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6f0ce00f6f06d2af90224ce498dcf849e78ec708"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6f0ce00f6f06d2af90224ce498dcf849e78ec708"}]},"branch":"refs/heads/master"},"e44c9883066d9b2d081a594b97aac3d598d491c9":{"kind":"REWORK","_number":2,"created":"2020-05-11 08:22:05.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/46/726746/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/46/726746/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/46/726746/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/46/726746/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/46/726746/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/46/726746/2"}}},"commit":{"parents":[{"commit":"c1768401f7a47bd88c044687a00f2e139f1985b4","subject":"Remove monotonic usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1768401f7a47bd88c044687a00f2e139f1985b4"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-11 08:17:47.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-11 08:21:58.000000000","tz":120},"subject":"Remove six usage","message":"Remove six usage\n\nRemove six, the python 2/3 compatibility library. It\u0027s not needed\nanymore since the repo is python3 only.\n\nRemove a now unneeded hacking test.\n\nChange-Id: I40522c4accb4aaf8115d11fee8b081e2d991cb4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e44c9883066d9b2d081a594b97aac3d598d491c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e44c9883066d9b2d081a594b97aac3d598d491c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
