)]}'
{"id":"openstack%2Foslo.messaging~727787","triplet_id":"openstack%2Foslo.messaging~master~I5e7e3f84519770e39754338ea4968fa6cd3ee6f6","project":"openstack/oslo.messaging","branch":"master","hashtags":[],"change_id":"I5e7e3f84519770e39754338ea4968fa6cd3ee6f6","subject":"Move jobs to py38","status":"MERGED","created":"2020-05-13 14:58:41.000000000","updated":"2021-02-03 16:58:47.000000000","submitted":"2021-02-03 16:52:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"727787","meta_rev_id":"b7f52381d0eb14051c8f5d84a4b6fb35848f1d62","_number":727787,"virtual_id_number":727787,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-03 16:52:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2020-12-16 17:25:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-02-03 11:46:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-02-03 11:46:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 15:24:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-06 18:38:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 15:53:25.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-12-16 17:25:53.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"}],"messages":[{"id":"d1fce94779a511f3023d99b6704c9b7721cafcab","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-13 14:58:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c67d72981349031651014a1172f2b6619bcf613a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 16:54:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96d62f256ea74c768dd534cfbec422b6 : SUCCESS in 1h 21m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3663e1b1539948c1bd62a2fd49743ef8 : SUCCESS in 7m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4226d59a1ba34c70948d985df02d2303 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/973418c8676e4212bb31e986fd036975 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ee4f4bf32c84922bd1c09caf4314947 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8db413cc94104a519ed4bd417c5c19d3 : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9302e40f5b0c4b79887a4555b3bd5245 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/149341b1025d4d8e9bee6480977cbded : SUCCESS in 5m 58s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/8e402c7996924f2096b23396642ee4b3 : TIMED_OUT in 41m 24s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/e6616b8fc64240ef8b401453f2b90610 : TIMED_OUT in 41m 24s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/c60af0c5dd3b4b5fb5510186024a9944 : FAILURE in 6m 55s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/b5eb33bb44c1451d89ee50c4620703b2 : FAILURE in 6m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/941444f0760742be842b5eea2056ed19 : SUCCESS in 1h 52m 18s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1b3629902c9344bf95f766f56ba2f4df : FAILURE in 28m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1178a29ff2204461968f113b3b636b8c : SUCCESS in 1h 26m 14s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/ef032b9fa5e045469dbfeb6a9ead9626 : SUCCESS in 55m 38s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/b21453dad1274f6ebf40dd2fe3cb3da9 : FAILURE in 15m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f007c5511515b1c479011467e1c252c87bb4d408","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-14 10:05:55.000000000","message":"Patch Set 1:\n\nrecheck\n\nErrors seems related to pifpaf...","accounts_in_message":[],"_revision_number":1},{"id":"5c9c6242935b43ba1adbe17ab8e38402759463fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 11:36:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3acc1b7d54ba4f5ab8f06b8b93e6b765 : SUCCESS in 1h 14m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f7395b47fc24ac798b8dfa6f9a68731 : SUCCESS in 8m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6f36fca1a574a998ab51115975cdfaa : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7127e76f90448ff9b941d7ac7835e8a : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8146dd31a2442aa8335107718bb34f7 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5301e66a4ab44223946329850b791122 : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6335f481ecca42c99d133872606feaaa : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53fc58672dd543478e5e2fdc978bfc5c : SUCCESS in 5m 34s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/e2c90df2eb76433fbf27343a28e4545a : TIMED_OUT in 41m 25s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/d1948c8f9263424d8efc09dbb93770d9 : TIMED_OUT in 41m 12s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/054995206df64dc0ba69c554e1cbfb37 : FAILURE in 7m 13s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/b56bf88aa9864ef28b90af24af0b0d2b : FAILURE in 7m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ee6a9992022443e4850c7273ed777eb9 : SUCCESS in 1h 29m 04s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/27a9096c3b1e45bfa7cb46eb60a0ec0a : FAILURE in 33m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e67d4745812a44e8af672736abe75898 : SUCCESS in 1h 27m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/a220ba0940764e06904f30be9e2424ea : SUCCESS in 1h 00m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/8b1c74902f344cebad78215ef6f45ec8 : FAILURE in 13m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f52037ea655b0815eb192051a6ecdeaab76f19c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-01 08:02:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2f0c305321073e843876418c2ed2b72e65a25c16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 09:45:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3cfc5203339443b394dd9b0422da410d : SUCCESS in 1h 34m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df547a7bb73648fd95175fa8fd3b76f7 : RETRY_LIMIT in 5m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f7fadade50b49f98ee40a0fcf73f296 : RETRY_LIMIT in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aae26a6bc3d44e4aa8f5d8bccf74b1a1 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb6d15675a5a4c29a7a2d9159f76dfea : RETRY_LIMIT in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b07185094a048aeb6e3acf9b2e8ff97 : RETRY_LIMIT in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b923e8afcd6436fa816a9bf2dc982ae : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeade0b1cf5e4315bcd3b3444a7bb24c : SUCCESS in 6m 40s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/9d72ff509206436984dae3a1e03fcb9f : RETRY_LIMIT in 4m 05s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/c2f42e7483c743f6befbbb5170445e92 : RETRY_LIMIT in 5m 16s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/8ba8aef745cc444a92b331cf78a859c5 : RETRY_LIMIT in 4m 05s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/1a8005ba341944e4979c777726a4c27b : RETRY_LIMIT in 4m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f623fa820dff4881b614703e5ff38548 : SUCCESS in 1h 24m 38s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/17c8f24f79ff4bd5abdd12d233d08459 : SUCCESS in 1h 39m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f16cb36fa12f4d84ba62c1e2bf1221b7 : SUCCESS in 1h 29m 02s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/3963ab3d5c4340ca9fe83c2654c20cb2 : SUCCESS in 1h 03m 48s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/49c80012c0b84a728e1d4cc79501671d : FAILURE in 18m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f29f2047684743f23f9b5b02e9b7bd812f2c0c6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-06 16:40:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"41eb6b76d15de489e753d12fc6d0948e78a516fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 16:40:27.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"d340cd99ff46607c6eca89fc60a6ad5c2de556c2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-06 16:44:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dc250a2b3680455e8db5a6ac6d3ea4660eb70bc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 18:38:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6f41bda383046fea1f91add1458db90 : SUCCESS in 1h 32m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f235702b1d96406fa88c29c254a01f08 : SUCCESS in 9m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad181282bf3f41629ae9384c36b01fb1 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818aa593092a4c0fa515a5b1071bc0d9 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54bbbf6e0603412bbee4d4450143cc07 : SUCCESS in 11m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e3e7bb3e88e4168b39fa36b45dacf2e : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d567a301180d4c7195d382c321fc5a46 : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e6472e2ba264091acc2ba826e34ac73 : SUCCESS in 7m 25s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/cb102d6eb4884cf6a23a40fbc4123b5d : TIMED_OUT in 42m 16s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/defed1d3e25b441ea67c70bc10526856 : TIMED_OUT in 42m 13s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/6d62e8ea7cb24998b541ed3ece78055d : FAILURE in 11m 06s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/cbaa792dfb1f4c61a6a60b3d02dcfa7c : FAILURE in 10m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ce75ab635b934b5aa820d5b68e49fce4 : SUCCESS in 1h 24m 09s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/0af47b7febb948bb8a56cc2485a59f7e : SUCCESS in 1h 31m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f451b4ee8c6a44dc872c9f81929f778f : SUCCESS in 1h 43m 56s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/ac1943f6a9684a75b31ea91fbb8b13c3 : SUCCESS in 1h 14m 41s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/5f7ebe55f4bc4a6bb27dff1677fc6646 : FAILURE in 22m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9e8f6e70ddf2715639668a652ed0e8f3b557b39f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-11 15:52:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"53443cb22b9b3effc4e421f01e3593a7bc484553","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-11 15:53:25.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4cae4a704f2fa57b5678233d24b6dafbeb59b510","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 17:59:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5b968c74339402790fd3514522b51c9 : SUCCESS in 1h 35m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc50f9c18e454081aecd7e843dd457a8 : SUCCESS in 7m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e0f21e1be344b68beafef9b8fbaa972 : FAILURE in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e5e7403e95f443e88a41c8facb63815 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a96ff6941dd144eebb512916f81f76ba : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffd6ff7f71dc46a18ff323a6fe5c5267 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28b9545b5ba0430c9371ea079b7aba83 : FAILURE in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1234c4e9d94ac18ce3caa65f127d39 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/094a5872e1674a5a852055d8b17e59ba : SUCCESS in 5m 04s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/b7ee9e74fc514efc834f1117bda01dea : SUCCESS in 6m 16s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/48c6ceec8b014edeb8ebca2faa5dc9a4 : FAILURE in 7m 32s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/cbac6e0f0a6e47069c64d448dacc884e : FAILURE in 5m 34s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/d73bac130ec14e22a750a25a2b3ddf67 : FAILURE in 5m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/18c0bd3ff28b4192a007c97a0600574c : SUCCESS in 1h 26m 36s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c13ca184901d458caab1e5cc3bc0096b : SUCCESS in 1h 38m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/19db34f346974b7caff96167ad22b913 : SUCCESS in 1h 28m 50s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/1c7c41ca6cd044ecb777e1bb5f177856 : SUCCESS in 1h 11m 43s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/80a49a7458a844e5848124b73acee691 : SUCCESS in 1h 45m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8d11b3edadf6b4e5ec6252b71ad08d1aac4f9a42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-16 10:05:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4f11d565e80b387661edc56959848b675d6cc792","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-16 10:06:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8cd61713cabeb09a175f99430836799c571c5317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 11:49:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64e6a184441d41f0b46c843a3bf05c04 : SUCCESS in 1h 36m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4af35a19acc0443abe2f584fe33a053b : SUCCESS in 7m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/990286ef8c8b4ffb833471fa74b7e300 : FAILURE in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02339633659c4b439440487b69f68723 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b04dddc20fb48739ab776fa24278fb9 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6257083ec9045ba87d8af29a7519015 : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b84b34e6afaf4f2bbe9ab5f48528b5e0 : FAILURE in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28363929f3f14cf4a3637cc28ca30ef9 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd9d2ab34f574d649807ba46a4d27826 : SUCCESS in 4m 20s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/fd66c5464460458ab262e221ea38007a : SUCCESS in 7m 13s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/34057e2d99734c9db609c3199f54f825 : FAILURE in 8m 46s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/15c16d9efebe41fcbdbaaa61252c7028 : FAILURE in 5m 27s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/7c43e56375f04e13997faaf5da336ef5 : FAILURE in 5m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5c138d93e81c4803be6e37de50624d00 : SUCCESS in 1h 34m 05s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6d4f91f3098f407cb7d212babf6b7bb2 : SUCCESS in 1h 40m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/df0e9c25a8dd444d9c8c61802d8b4445 : SUCCESS in 1h 27m 00s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f144cfcfc7c84765912e61a66accdb71 : SUCCESS in 1h 06m 21s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/58258f57861248e198f8fe7b1904ed21 : SUCCESS in 1h 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"83f59fd0e63567cbf188c6c1e3a46ab1af600c72","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-12-16 17:25:53.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck\n\nThis doesn\u0027t touch lower-constraints, so I have to assume that was a spurious failure.","accounts_in_message":[],"_revision_number":4},{"id":"d84dc508cf08dad3ad81c54a44b1f4bcb352ebd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 19:18:41.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7509612502774f8b868be874f55b8ba0 : SUCCESS in 1h 42m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e92a7aea48984a819f71f3c55901128d : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a5dfcde47d04efab387013ca37f49e4 : FAILURE in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92377cf2f7f341beab0ef8a0f8e61bf5 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f025cec3ae194ecbb333619cbc7f1351 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f85a8aba605545f58034b08e2de46006 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d867577fdc35472283221ef4df6c1337 : FAILURE in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd32e0065e0f45f6bf40b2948914fa16 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7a0664d3a6849549f7636c7e79a4ce5 : SUCCESS in 3m 26s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/b860138c11f14bf8add0497289354910 : SUCCESS in 5m 09s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/90acbeb4659245a0bc0c48a181a6b61e : FAILURE in 6m 06s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/1370d98f11994ab2bc69377599cd86f7 : FAILURE in 5m 03s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/697d2a8a107a4948b60bfda013e57fc1 : FAILURE in 5m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d863925b15fd42ed8920910679f1e37c : SUCCESS in 1h 20m 44s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d0958452d0db4add80935037a4e1adf9 : SUCCESS in 1h 11m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/74d8dcf6ba25440c96c09828607b3aa5 : SUCCESS in 1h 25m 14s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b4f86cbaddb34040b163887a2b264b0a : SUCCESS in 44m 37s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/f609ce2a2b5347039231b9dc62dd1c57 : SUCCESS in 1h 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"17663594bbc3b2b571c4cbf42813a0847fa66d3a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-08 10:22:34.000000000","message":"Patch Set 4:\n\nrecheck\n\u003e Patch Set 4: Code-Review+2\n\u003e \n\u003e recheck\n\u003e \n\u003e This doesn\u0027t touch lower-constraints, so I have to assume that was a spurious failure.\n\nYes I think this is related to the LC issue. \n\nWorth to recheck without it now.","accounts_in_message":[],"_revision_number":4},{"id":"9ecb8fb6660d9e5cd380af491f00d655833644a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 12:29:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52272f6d40a443e2b519a1cf7c2309a5 : SUCCESS in 1h 31m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/201940ca6ccf43acaefb96ceaef8ab87 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44378ac367344ca5bb9c42bb8233e6d1 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/855b8a4c29bb46d288bca6e70f871748 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee06b7b99e1d4aeb98d9af48de3b742c : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e3a457eba824ea9b0d09877112884d1 : FAILURE in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b656f61ccc8410db73327a59693ae79 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6af2a5f2f9584e4b81cea6138ab1ec3a : SUCCESS in 4m 09s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/92508b90038b4a5bb14537d1437a6d7c : SUCCESS in 5m 07s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/a2fdfcafa74d4803b2c0b0a02b1fd5af : FAILURE in 7m 11s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/e3323ee98c6a482b83cb665d2ae9bc21 : FAILURE in 5m 29s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/1d29252d94a6435abcf786d8c3cefa01 : FAILURE in 5m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8162a5986bab46aa9c0d2e80a1bc8874 : SUCCESS in 1h 27m 51s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d9bebea867c84dd3a176a05ecac8e012 : SUCCESS in 1h 22m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3e7fe3e7d1d94fa2bfbb46d6f870de82 : SUCCESS in 1h 52m 01s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/64d62a9e002d4dbb8a6b788aa2a0857b : SUCCESS in 1h 00m 28s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2cdb3f16a5d4e809cf6029bff3bfc42 : SUCCESS in 1h 14m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2f65a1e995cbbf6eda4234a415cdfecb60b71431","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-03 11:46:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"61851c7034ba9e74e07a219f605b3474d35fb511","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 11:46:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2f114b00acde504df1fc105b606e6cdf4406020c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 13:24:13.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05f080708e604414b284823291a0dabc : SUCCESS in 1h 31m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c06f25768fd47068429e0bcf7511a51 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e923e6b2a9d14f2d848bf351da659ea1 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e6a0ea72db646f5922b385471ef7aa2 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceeb56e2488a4aa7a302bf9478acb82f : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/665f9d9c37894c1b8e76b440aaefd6ab : SUCCESS in 4m 40s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/85969c6e425e45778f43c480a9046d6d : SUCCESS in 5m 23s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b493aa319b41466fb2cd1bd63c9a76af : FAILURE in 1h 14m 38s","accounts_in_message":[],"_revision_number":4},{"id":"1b112a4004b282e3419e5d04e01fddce2322594e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-03 13:34:21.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6be0242f6b1809eec822bcff7c253316d46a53cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 15:19:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bbb397df57d145b3be4b63435368b5c4 : SUCCESS in 1h 28m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdca8e515d2c4f679fa8651b5e6106cb : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cd81eecd53c4849a5c0ab5f1264171c : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77be872cb4ac4e7787112504fea193a0 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37c1588d28b0496baf9eb9b9b636d8d1 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8499652dd263447f9a2b921bab03ce36 : FAILURE in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c4e9c6656da410d8d4af080aacca3a5 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/022658558004462f8518d31d27532e16 : SUCCESS in 4m 03s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/b9faccca389f41eaaf8d66bebd6bc36b : SUCCESS in 5m 20s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/f4b224aeb1084408ac8e326eed70e90f : FAILURE in 7m 18s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/c4483cd3f7e44d2e870d04dc868ee7bf : FAILURE in 5m 15s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/a581fe9dcfbf41ada8fe0aa323154dab : FAILURE in 5m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c69a0449275f413b89e6da36895d6391 : SUCCESS in 1h 21m 59s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c899cbc33f754356b3691294f8bf5908 : SUCCESS in 1h 24m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/6f62bab61ee0425b8768ee80a3212a0e : SUCCESS in 1h 35m 36s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/3362663989b44f578fafabb1506badb8 : SUCCESS in 58m 46s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/f5a4edb1a12a4639872bb0629b1d1a32 : SUCCESS in 1h 32m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"46b87527c3eb8e49b7c099617b223a07c11dbb3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 15:20:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"36afac57ccd91d2b9c5403208299b1ae86ee1ba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 16:52:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96f05abead8e459080394f152b666baf : SUCCESS in 1h 24m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e95a0aa21a34014ab25a5bf3e4f74df : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45651f1871f64daab8425b011c675d9a : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6f802315ca94bc9b08fb3e992e9d01e : SUCCESS in 9m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba4d9e9d0c541d29ae74a426b68f8d4 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e217c2d3dbe242e389afff6de5454a61 : SUCCESS in 3m 43s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/9ec685260188488fa870607167d95973 : SUCCESS in 6m 47s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/011d007b8f4f4821ad40dde3773d1b47 : SUCCESS in 1h 29m 38s","accounts_in_message":[],"_revision_number":4},{"id":"521995d3a27a149af8543d5dcac8ccde307f540c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 16:52:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b7f52381d0eb14051c8f5d84a4b6fb35848f1d62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 16:58:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c68810ba1574e17a8f805d8bb6d8b86 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9a1a49f428b4dd79742bceba38c3936 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f16892ea481082dfcdc8369092dacf9bac912704","revisions":{"467947e0ce0af6cec81a29e6605c46c6fdfd96d7":{"kind":"REWORK","_number":1,"created":"2020-05-13 14:58:41.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/87/727787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/87/727787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/1"}}},"commit":{"parents":[{"commit":"c1768401f7a47bd88c044687a00f2e139f1985b4","subject":"Remove monotonic usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1768401f7a47bd88c044687a00f2e139f1985b4"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-13 14:58:15.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-13 14:58:15.000000000","tz":120},"subject":"Move jobs to py38","message":"Move jobs to py38\n\nChange-Id: I5e7e3f84519770e39754338ea4968fa6cd3ee6f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/467947e0ce0af6cec81a29e6605c46c6fdfd96d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/467947e0ce0af6cec81a29e6605c46c6fdfd96d7"}]},"branch":"refs/heads/master"},"c7339fe930bf951251fe466a74f255b38b655bb3":{"kind":"REWORK","_number":2,"created":"2020-07-06 16:44:21.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/87/727787/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/87/727787/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/2"}}},"commit":{"parents":[{"commit":"afb035d97185fa9c622bdb9b4c31c71f01160370","subject":"tests: Resolves issues with kombu \u003e 4.6.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/afb035d97185fa9c622bdb9b4c31c71f01160370"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-13 14:58:15.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-06 16:44:06.000000000","tz":120},"subject":"Move jobs to py38","message":"Move jobs to py38\n\nChange-Id: I5e7e3f84519770e39754338ea4968fa6cd3ee6f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c7339fe930bf951251fe466a74f255b38b655bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c7339fe930bf951251fe466a74f255b38b655bb3"}]},"branch":"refs/heads/master"},"f26a377c3a744d20a266f6366b59ec7da31aec25":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-12-11 15:52:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/87/727787/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/87/727787/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/3"}}},"commit":{"parents":[{"commit":"54f8724be281b1acfae175d7ec1c14e439269117","subject":"Fix doc title rendering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/54f8724be281b1acfae175d7ec1c14e439269117"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-13 14:58:15.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-12-11 15:52:10.000000000","tz":0},"subject":"Move jobs to py38","message":"Move jobs to py38\n\nChange-Id: I5e7e3f84519770e39754338ea4968fa6cd3ee6f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f26a377c3a744d20a266f6366b59ec7da31aec25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f26a377c3a744d20a266f6366b59ec7da31aec25"}]},"branch":"refs/heads/master","description":"Rebase"},"f16892ea481082dfcdc8369092dacf9bac912704":{"kind":"REWORK","_number":4,"created":"2020-12-16 10:05:37.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/87/727787/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/87/727787/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/87/727787/4"}}},"commit":{"parents":[{"commit":"54f8724be281b1acfae175d7ec1c14e439269117","subject":"Fix doc title rendering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/54f8724be281b1acfae175d7ec1c14e439269117"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-13 14:58:15.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-12-16 10:05:26.000000000","tz":60},"subject":"Move jobs to py38","message":"Move jobs to py38\n\nChange-Id: I5e7e3f84519770e39754338ea4968fa6cd3ee6f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f16892ea481082dfcdc8369092dacf9bac912704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f16892ea481082dfcdc8369092dacf9bac912704"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
