)]}'
{"id":"openstack%2Foslo.messaging~728396","triplet_id":"openstack%2Foslo.messaging~stable%2Fpike~Id3603f4b7e1274b616d76e1c0c009d2ab7f6efb6","project":"openstack/oslo.messaging","branch":"stable/pike","topic":"bug/1789177-stable/rocky-stable/pike","hashtags":[],"change_id":"Id3603f4b7e1274b616d76e1c0c009d2ab7f6efb6","subject":"Use default exchange for direct messaging","status":"ABANDONED","created":"2020-05-15 09:24:03.000000000","updated":"2022-08-01 16:11:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3e4be48d9a6a1d4f152f7c4d29fe2ffbe3738f50","_number":728396,"virtual_id_number":728396,"owner":{"_account_id":22595,"name":"Stefan Nica","email":"snica@suse.com","username":"stefan.nica"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-05-16 17:04:38.000000000","_account_id":22595,"name":"Stefan Nica","email":"snica@suse.com","username":"stefan.nica"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-15 10:47:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22595,"name":"Stefan Nica","email":"snica@suse.com","username":"stefan.nica"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":22595,"name":"Stefan Nica","email":"snica@suse.com","username":"stefan.nica"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-05-18 16:09:21.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22595,"name":"Stefan Nica","email":"snica@suse.com","username":"stefan.nica"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22595,"name":"Stefan Nica","email":"snica@suse.com","username":"stefan.nica"}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 10:47:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-18 16:09:21.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2021-08-05 14:29:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"}],"messages":[{"id":"7f69094dbfed521f3ad6eb68155fc9de450c6f59","author":{"_account_id":22595,"name":"Stefan Nica","email":"snica@suse.com","username":"stefan.nica"},"date":"2020-05-15 09:24:03.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"e19381bc5851e880f79e1c32c2b50593c3607e59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 10:47:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48c7360bb1444c8e8d84831c8b820948 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/355b8cacb020414da49bb24bbff8ca68 : SUCCESS in 5m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/57df545165714773b1640ee23955c830 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da515f43f20491281d005aae7bf1695 : FAILURE in 3m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/136c359060f5471888fb1f2b0c7375f6 : FAILURE in 1h 17m 28s","accounts_in_message":[],"_revision_number":1},{"id":"fdbddd9354e6e4456c9aaea062721f5018506d23","author":{"_account_id":22595,"name":"Stefan Nica","email":"snica@suse.com","username":"stefan.nica"},"date":"2020-05-16 17:04:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"58c79c22472bc8cff914b46451d117f68f5217b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 18:42:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ffd821748d749c191b81c3fed213066 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a9e5f43c0cd045f291922a4a8e2195ba : FAILURE in 6m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b7406ad85603492a99f107cf5a7dd8fe : FAILURE in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ecdcb37de244c73904e609984d24770 : FAILURE in 3m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/518a14203e8147b785bc2755ac8a4108 : SUCCESS in 1h 36m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0566b5077f6da7685898dea553852032a061a322","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 14:20:06.000000000","message":"Abandoned\n\nstable/pike is no longer maintained (https://releases.openstack.org/). Thanks for your understanding","accounts_in_message":[],"_revision_number":1},{"id":"781992d52db9157f595481261f238d857394f543","tag":"autogenerated:gerrit:restore","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 14:29:28.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"9882d84c1b5183317229da4b5fc30a6be25114c7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 14:29:52.000000000","message":"Patch Set 1:\n\nrecheck\n\nSorry ignore my previous comment","accounts_in_message":[],"_revision_number":1},{"id":"e188af03f2bf472677c906f9a7cd4829762e3942","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 15:59:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d335a806fa80434e84df30086920140e : FAILURE in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c47f8f970b7b4f4d9ba9ff19fbcb7284 : FAILURE in 4m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d36d3bf02c234dc092bee198da46cc2d : FAILURE in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b58c4c9e6fa42918a4da67415e67beb : FAILURE in 4m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/80d88cc3cc074a398f1489b02f7dd580 : SUCCESS in 1h 27m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3e4be48d9a6a1d4f152f7c4d29fe2ffbe3738f50","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-01 16:11:28.000000000","message":"Abandoned\n\nstable/pike has transitioned to End of Life for oslo, open patches need to be abandoned in order to be able to delete the branch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a971972e46533a9c3af26bde1c85d8dacf36ad5","revisions":{"2a971972e46533a9c3af26bde1c85d8dacf36ad5":{"kind":"REWORK","_number":1,"created":"2020-05-15 09:24:03.000000000","uploader":{"_account_id":22595,"name":"Stefan Nica","email":"snica@suse.com","username":"stefan.nica"},"ref":"refs/changes/96/728396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/96/728396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/728396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/728396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/728396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/96/728396/1"}}},"commit":{"parents":[{"commit":"42275b21b405bbe76847eb242b1c5166c1c4a0a6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/42275b21b405bbe76847eb242b1c5166c1c4a0a6"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2018-08-27 08:18:58.000000000","tz":240},"committer":{"name":"Stefan Nica","email":"snica@suse.com","date":"2020-05-15 09:24:03.000000000","tz":0},"subject":"Use default exchange for direct messaging","message":"Use default exchange for direct messaging\n\nLots of exchanges create problems during failover under high\nload. Please see bug report for details.\n\nThis is step 1 in the process: only using default exchange\nwhen publishing. Consumers will still consume on separate\nexchanges (and on default exchange by default) so this\nshould be (and tested to be) a non-breaking and\nupgrade-friendly change.\n\nStep 2 is to update consumers to only listen on default exchange,\nto happen in T release.\n\nChange-Id: Id3603f4b7e1274b616d76e1c0c009d2ab7f6efb6\nCloses-Bug: #1789177\n(cherry picked from commit b67a457e4fa71e9220c149087dce013c7f81144f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2a971972e46533a9c3af26bde1c85d8dacf36ad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2a971972e46533a9c3af26bde1c85d8dacf36ad5"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
