)]}'
{"id":"openstack%2Foslo.messaging~733460","triplet_id":"openstack%2Foslo.messaging~master~I287fd718f9283e5d7d6951b9e55d5a6b48e07e3c","project":"openstack/oslo.messaging","branch":"master","topic":"sphinx","hashtags":[],"change_id":"I287fd718f9283e5d7d6951b9e55d5a6b48e07e3c","subject":"Fix pygments style","status":"MERGED","created":"2020-06-04 07:29:30.000000000","updated":"2020-06-08 09:07:25.000000000","submitted":"2020-06-08 09:05:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733460-1591607134385-68f2f97b","meta_rev_id":"42bfdfa70daa910aa4de7c0ac9c151f515f64da9","_number":733460,"virtual_id_number":733460,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-08 09:05:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2020-06-04 13:22:10.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2020-06-08 08:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2020-06-08 08:56:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 13:22:10.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-06-08 08:56:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-08 09:05:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd489de98b84b4dde22035071200d1be829b8989","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-06-04 07:29:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbc0777885607930fe05eba4eb7bc6047f822c12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 08:43:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f7ffa5eedc04027acfd95e21409159e : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d69b9e3fad4aacbae8c379907e6fcc : SUCCESS in 5m 16s\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f3fa3481301d40a2bb0009b88d63f5db : SUCCESS in 1h 07m 50s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/7fffffebae2043e68119a6799de83123 : FAILURE in 19m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c00d2024378aa06074c92a9ed3f0274395a79817","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-06-04 13:22:10.000000000","message":"Patch Set 1: Code-Review+2\n\nGenerated docs LGTM","accounts_in_message":[],"_revision_number":1},{"id":"ca359f52b4469179ca657ee9bc564f87b97d72c5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-08 08:56:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e55d4e582c26972d059f1bfa5d71574fd32f1c44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 08:56:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58e2e9eedd0fb3fef647ad4ab39019bd9260c1b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 09:05:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0b39e3335feedbf2b0862a9110f3e7892dbb1cce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 09:05:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06e8c360118842dc8f8f1c859f050892 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1dc4d54c76a42ac87afdd49c8e15821 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9adbae0ac84941ab9a3590c254fb1695e88c8c19","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 09:07:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c922c4ab6f64234964c1fa8bb69c160 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2074e47604199e9802520dbc4dfb54da778e090","revisions":{"c2074e47604199e9802520dbc4dfb54da778e090":{"kind":"REWORK","_number":1,"created":"2020-06-04 07:29:30.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/60/733460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/60/733460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/60/733460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/60/733460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/60/733460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/60/733460/1"}}},"commit":{"parents":[{"commit":"4fc9e70aa11f8cb0476a927ee471adbd296bba8f","subject":"Merge \"Print warning message when connection running out\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4fc9e70aa11f8cb0476a927ee471adbd296bba8f"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-06-04 07:27:58.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-06-04 07:27:58.000000000","tz":480},"subject":"Fix pygments style","message":"Fix pygments style\n\nNew theme of docs respects pygments_style.\n\nmore info: http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: I287fd718f9283e5d7d6951b9e55d5a6b48e07e3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
