)]}'
{"id":"openstack%2Foslo.messaging~734489","triplet_id":"openstack%2Foslo.messaging~master~I49eeb9819bc49b73f3d58e700e46dfe0f6e7ae83","project":"openstack/oslo.messaging","branch":"master","topic":"drop_mock","hashtags":[],"change_id":"I49eeb9819bc49b73f3d58e700e46dfe0f6e7ae83","subject":"drop mock from lower-constraints","status":"ABANDONED","created":"2020-06-09 09:45:39.000000000","updated":"2021-04-16 10:27:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9968368118f9e5b0a47f20ffce22706e1973986f","_number":734489,"virtual_id_number":734489,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"date":"2020-06-11 11:44:48.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-09 11:22:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":1,"date":"2020-06-09 11:07:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 11:07:07.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-06-09 11:22:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"402e2273902d4b6c4db12507c62b9495abaaf314","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-09 09:45:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd153f66164ac98f3ca26cb7acdde5d88a0d5b06","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-06-09 11:07:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9d20cf459758346ac7972679a86eb202c9b2e9ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 11:22:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/188f531477a848efab66bd4d4ffae60d : SUCCESS in 6m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/674193b379f3446f8f90046da5bcf92e : SUCCESS in 1h 29m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ec07cb171e943d6bbfb38b0fd19efb0 : TIMED_OUT in 52m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/368d6aeca119462d8fdcd2011834bc11 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1a149a4c00d416bbeb291cbbb9d43ef : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0e64f56afd1408c9e7fcafc3aa01cbb : TIMED_OUT in 42m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c9f9b2a9567441e9e475b911c28280f : TIMED_OUT in 42m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fbad371f1134a608ed88ff6c256fd50 : SUCCESS in 6m 43s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/07e2e5c56fd8453e9bc646647ca23acf : SUCCESS in 6m 13s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/f4329fba95cd408192fe600249822dff : FAILURE in 6m 26s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/71ff324bb6804ac28ac68d31bb4dd503 : SUCCESS in 8m 41s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/864a221cb4a44789960310191506e9f1 : SUCCESS in 6m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b815d194cc5644bc97b0be4852446c32 : SUCCESS in 1h 24m 17s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2712645bd9b84e3fad5c1121bde121cf : SUCCESS in 1h 29m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c49f6c729f26482cba1b973b7d9dbffb : SUCCESS in 1h 08m 51s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/c6aabee333344623bb177e279b0f1448 : SUCCESS in 1h 03m 58s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/a7d54de5fc2647ceabf3a7fc2d92d03c : FAILURE in 17m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba184ec9a793d362a1347f62cb61bbf29ef016f8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-09 18:03:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"030625b8bd8cbb5575dfd524aedf76909df0655b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 19:46:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6e8e4f4724364ef097ad2ca360757e24 : SUCCESS in 5m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fdc894f803046fa95465d3d0a58caf5 : SUCCESS in 1h 16m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/240d57151c4b40e4b262c46ab4fcc8d1 : TIMED_OUT in 51m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60d60cb72bfc4e6d81f28ef4a13f1a4f : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c85e49c3356043928e309c8d863d850e : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/288c70d1decd4c09afe1b543cad43414 : TIMED_OUT in 41m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42ec9e16488e4bccbf8e7eee15f4430d : TIMED_OUT in 41m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/219895b03c4b48dab1ecc973ac9dc21a : SUCCESS in 4m 39s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/845f0024382945f08ab076a5b643f500 : SUCCESS in 6m 11s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/b3926fdebdf64c6381b1ddde8dd37aad : SUCCESS in 6m 43s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/5c240c505d364ccfb338e069dc093bbc : SUCCESS in 7m 04s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/be38e5ab2e64444aa42d380341bf95b9 : SUCCESS in 9m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/65f2d13527294f3ab84525e8f72d7117 : SUCCESS in 1h 35m 29s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/5c8fe387b51343f3a5afe10d88f41203 : SUCCESS in 1h 26m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e5894dd52e5543c9acd5b09753d70f5b : SUCCESS in 1h 27m 10s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/a3dc0e1f0c24466da920c6aed72a9c25 : SUCCESS in 1h 01m 48s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/22cbe48c2afd4d04893d0c67e4f9cc44 : FAILURE in 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7fe3b7d2f7c7e5a8044e8e31589b92f1ecb753a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-10 08:40:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6b99de442f24e7f6aed81490f4a322b1f74674fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 10:27:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/469a5df0db184fdd91db2c6d3f324864 : SUCCESS in 5m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36f8f2c9d0f44e11a09fe908b0debf33 : SUCCESS in 1h 42m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27455e15666248eeaf4fe964621c59e3 : TIMED_OUT in 51m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ff653d114e14117814ce6a7abe992b3 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71ffc4ece4e54c93ae136086dc57b659 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42303c69c98f4400b0efefa94ddcff37 : TIMED_OUT in 41m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecc2cfa7381743f799fd40b12c7ed449 : TIMED_OUT in 41m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/405724c0456248d3a0d2fcc37cab7449 : SUCCESS in 7m 23s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/871be52afef04ec399590c85be155ad5 : SUCCESS in 9m 17s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/f7ad49bc367d46ef90d7c2fdb3577841 : SUCCESS in 8m 24s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/b33eb6497fef4f1aab1010891f7facf2 : SUCCESS in 7m 54s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/06c2bb81205c4a8da95eb38a073c84f6 : SUCCESS in 8m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ad217b3998514d0e8e6ec75fd489b86e : SUCCESS in 1h 14m 09s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/09f7553430db411f88d7052e37f80361 : FAILURE in 1h 37m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8d91a3452427448593b605444bde3048 : SUCCESS in 1h 40m 21s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/394f452722f54edd84cb51f37a76cc50 : SUCCESS in 1h 00m 08s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/1edb18945734452a967235fb4df024fb : FAILURE in 15m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49cadfbbf946e068461ce45377eb455485eb7809","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-11 11:44:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"278af48b56e960d40bf38bc1e7d1e0a8166dc583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 13:12:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8e6a2bad871f41aaae16eab42f37857c : SUCCESS in 5m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/804d413e632f48c5b0b1cbdfbe5cf390 : SUCCESS in 1h 12m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11f73b1c24e4496bb9553fd60abf9c35 : TIMED_OUT in 51m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2a4f9fd233c441aa4260493a2a82492 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3ad7727b2b64e489ab3565aa88bb175 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad9f2d18174d4cca8be1a2ad9830b7b1 : TIMED_OUT in 41m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f754cf2bb185458db1294de673d1a21f : TIMED_OUT in 41m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9b9f41533c74793b1175b67aa95e94d : SUCCESS in 5m 20s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/f43c78fb7bd04c3aa727f1147b658d83 : SUCCESS in 8m 02s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/377b92207ab4432bbb439869ee6244c2 : SUCCESS in 6m 26s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/7397c7a6394c418db838ad361b2700f9 : SUCCESS in 6m 52s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/f194271647144cd582e1ae50ba35b4c4 : SUCCESS in 6m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f68b27399c2b4da4afce0ec5c72a7f17 : SUCCESS in 1h 19m 25s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a7ab081dc1cc4e2a962bb8a7f2f4c87e : FAILURE in 1h 26m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a89eaff1965e477684a747375ee48cd6 : SUCCESS in 1h 20m 10s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/9be968da516943d7a79f2b332a8a0133 : SUCCESS in 56m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/1378ec149e22464ca2cb71f42a5c1a8f : FAILURE in 15m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9968368118f9e5b0a47f20ffce22706e1973986f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:27:09.000000000","message":"Abandoned\n\nWe\u0027ve dropped LC so this is no longer necessary","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9da16dd4e324c9e925cc2579b192e1150ab975d","revisions":{"c9da16dd4e324c9e925cc2579b192e1150ab975d":{"kind":"REWORK","_number":1,"created":"2020-06-09 09:45:39.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/734489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/89/734489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/734489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/734489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/734489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/89/734489/1"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-09 09:45:32.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-09 09:45:32.000000000","tz":120},"subject":"drop mock from lower-constraints","message":"drop mock from lower-constraints\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we don\u0027t need it\nin lower-constraints.\n\nThese changes will help us to drop `mock` from openstack/requirements\n\nChange-Id: I49eeb9819bc49b73f3d58e700e46dfe0f6e7ae83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c9da16dd4e324c9e925cc2579b192e1150ab975d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c9da16dd4e324c9e925cc2579b192e1150ab975d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
