)]}'
{"id":"openstack%2Foslo.messaging~734740","triplet_id":"openstack%2Foslo.messaging~stable%2Ftrain~I5ca930cada523bc77c644703a0f02b9160816231","project":"openstack/oslo.messaging","branch":"stable/train","topic":"bug/1871813-stable/ussuri-stable/train","hashtags":[],"change_id":"I5ca930cada523bc77c644703a0f02b9160816231","subject":"Print warning message when connection running out","status":"MERGED","created":"2020-06-10 04:26:31.000000000","updated":"2020-07-08 20:55:46.000000000","submitted":"2020-07-08 11:13:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734740-1594206800068-2afef63d","meta_rev_id":"c309f55c86cd5f5713c4f55778a6c72b2aae879d","_number":734740,"virtual_id_number":734740,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-07 07:57:06.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-08 11:13:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-06 23:55:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-07-06 23:55:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2020-07-08 09:24:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 15:59:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-07 07:57:06.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-07-08 09:24:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-08 11:13:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4849944e243be7a8cea9fd277bbc8c54c15c6854","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-06-10 04:26:31.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"d8aceb3185c38a45f04f046a5e6fb13db5b03aa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 07:22:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/04b456ff6856434e8ae53774e016a842 : SUCCESS in 1h 01m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/503c050fcbbf489083a297e48a647721 : SUCCESS in 1h 10m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cf8a3290e8c406a8ea378b232cc110b : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50814df764814611ab99227f8729253b : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/395679565ab2471bb19b3c39653ed197 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a33cc0e19f84aa4829150fdfd5c1aaa : TIMED_OUT in 41m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e51e34ebfcd40b2b075f99025d0a2f8 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc0252839a4e4551901b9b418a64af3b : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af46858b34134bd4b9fcb5631ef614a9 : SUCCESS in 3m 55s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/a4083684dc3c4457a1594e563d067f97 : SUCCESS in 6m 24s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/8912bc3b72374686af0836a82fbe52f7 : FAILURE in 12m 04s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/b94cce4edf2746719e43982e55a1fc97 : FAILURE in 7m 41s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/62a72dff45124cfe9198c1542b85d162 : SUCCESS in 7m 23s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/f82a26b6822a4324af37280f66c8c9fc : FAILURE in 13m 42s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/79124b8d778e40f88b91cb08493dfb2b : FAILURE in 6m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ab8ada5bed624c3d9766e735c2ae9236 : SUCCESS in 1h 29m 01s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a358ab51717a41cebff163e1783b7994 : SUCCESS in 1h 23m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/ed4e26cbf55d407fabf701c5081ff23b : SUCCESS in 2h 54m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8801c7d10120470cb873c840156ee5e8 : SUCCESS in 1h 05m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/c87524b1699a41a59c72845fd84c6e77 : SUCCESS in 1h 43m 39s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/5b99bc2cfe2c4fcc8f23f3db815cfc68 : SUCCESS in 47m 52s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/35c73e1e83034ef59a8672877a99b92b : FAILURE in 20m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/a819750d2dc446cbab075bdc6d86975f : FAILURE in 43m 30s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/ed518cfbbb0b4a6d84cede8daca8c028 : FAILURE in 48m 03s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/2f93e570f94e49efa8d9a056109ea1c9 : FAILURE in 31m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fac77dd2df7e2140e90f766cbc784474afa11533","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-06-10 07:49:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c35049cc831aa56c1636a5aa3cc7693e9e2d3894","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 10:07:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/524c97b60ed043eeb4bb9ebc4558bece : SUCCESS in 1h 43m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11a6d1988cb64598816ba546ebe0f64f : SUCCESS in 1h 23m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65f433a6c9144d5e9775804227d6527b : SUCCESS in 8m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a62b5aadf64d46edaa9fe53ca0552397 : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ebf1c8890d411097996485b3f30727 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b4252360c6f47cfb1af3cc06ce2727f : SUCCESS in 9m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a59602bae3a40359826fd4cdf28e2cb : SUCCESS in 10m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec43819720744a5da0747749195cbb3a : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d362df6ba3f548beb0afa9cea4c3a3f4 : SUCCESS in 5m 55s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/434e0be205fe49b2888f84be7e1c9595 : SUCCESS in 7m 18s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/2774c62c09084e63b914f9ad4ede66eb : FAILURE in 12m 29s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/8026f6ce73b0410992eddfc536ca51eb : SUCCESS in 8m 21s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/5f8d32ebe49a465fb7a457829201f7d3 : SUCCESS in 6m 35s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/12df3b3ed27b42b28dc3301c41f4089f : FAILURE in 16m 04s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/f9ae51b062e14b6697676d136d055b3c : SUCCESS in 6m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e9cecfa154004c34ab4845079818d6e8 : SUCCESS in 1h 37m 39s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c4400185c23745e89d607b2a0238248e : SUCCESS in 1h 48m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/a0defc14fffa49ae8a76c0755379f29e : SUCCESS in 2h 12m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ada842e5a13b4516a69790061cf7d3f2 : SUCCESS in 1h 37m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/b3eb54f5f41f419c8069825378071478 : SUCCESS in 1h 46m 59s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/61251488381740cd8730d2371bda13f5 : SUCCESS in 1h 05m 11s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/ff017e7c707d4f13a8dd95870fb753a5 : FAILURE in 29m 08s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/1e418c49669f4a2bb20359bd8d0eef63 : FAILURE in 46m 49s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/d39949c6992149a09fde78bdf45336ae : FAILURE in 36m 07s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/f5c0fcfd22e84de68c9977eb83289075 : FAILURE in 52m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d50962e76b00176bd4bb22f0d49396bc8d01578b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-10 12:29:15.000000000","message":"Patch Set 1: Code-Review+2\n\nClean backport","accounts_in_message":[],"_revision_number":1},{"id":"b735a1e9c8adee2743d32ac1ece5e83bbd1575ee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 15:59:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMerged to stable/ussuri, good to backport.","accounts_in_message":[],"_revision_number":1},{"id":"f866feef6e65bbbbf400f7aebafe9f6df02b681b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:00:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc86115ce1c72ace55e7f6d9416ca7cc92851435","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 17:38:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f4351629dceb43d793b109a8c1af3966 : SUCCESS in 1h 32m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a823e55dd9984abca43909c62bad84b1 : SUCCESS in 1h 26m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/348dbf3018fd42c1a2736697acb247bc : RETRY_LIMIT in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/309d6bfc591f4cc4a9dd009f8d398774 : SUCCESS in 7m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68cbf01a823d4287952b161518b3e697 : RETRY_LIMIT in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a3c8cb5215a4ce1950c64d08cee634f : RETRY_LIMIT in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42e2b567064b4f92bee4e61ad38c7a84 : RETRY_LIMIT in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecffb109e4074b8ba3fe24dda4a7c6f0 : SUCCESS in 7m 21s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/c4f792ca4f964d6ca157e0c76662a299 : RETRY_LIMIT in 6m 03s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3597e1636bd24c789b320d9dda068996 : SUCCESS in 1h 28m 13s","accounts_in_message":[],"_revision_number":1},{"id":"82fa26c46ed24a4b6b4278e935eb993578db18c3","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-06-22 23:53:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"412fb906a9151cbad8aa19dc1c6dc63433b045f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 01:38:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/77b4fbc2da7d4833803ab357060e6d58 : SUCCESS in 1h 24m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5966981d979437c9f0bde41a11a4ed1 : SUCCESS in 1h 19m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51b23d8f95e641d0b9cee0e2c0697e9b : RETRY_LIMIT in 3m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd014b60653c4639af827440b374b069 : RETRY_LIMIT in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c55cec29d9e4576b55f5abc98a95548 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9163ed87988541f4b8775efac8eb788e : RETRY_LIMIT in 2m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3db2a5c590246089957784389b0a999 : RETRY_LIMIT in 2m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/969c4a69359e457d9d76c50701fcc412 : RETRY_LIMIT in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/175cde83de354e7198622ab6ec4dab6d : SUCCESS in 4m 19s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/39bae88e76ec43ca815d803d7d71fed0 : RETRY_LIMIT in 3m 26s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/65316fd79f9c43c29ff1a366bbea03aa : RETRY_LIMIT in 3m 26s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/5418fb8d6ae943b89eec1346213ac09a : RETRY_LIMIT in 2m 59s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/78ac8b28272049908152e5dafd2d1f96 : RETRY_LIMIT in 3m 35s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/e67afeeebfd54a429aa8eefc85f9bb57 : RETRY_LIMIT in 2m 34s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/3e8a7d3b0d0c47aeb9e34efcf50bc9b1 : RETRY_LIMIT in 2m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4c26da606d724c799b6d039cc6c7695f : SUCCESS in 1h 35m 39s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/43bcdc3d975c45249ab8ba68b18c683f : SUCCESS in 1h 38m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/7f396b26627543f3867d252e2fdc337b : FAILURE in 14m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2c7241b495064407bd23b06f874ee4be : SUCCESS in 1h 35m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/7a51c125ecf949cc81d557bc096c48d5 : FAILURE in 17m 04s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/1ca7c524954040698a263b093ec172e0 : SUCCESS in 1h 00m 45s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/e9d4a01971d24e4aae2a379982395013 : FAILURE in 36m 47s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/58d281097f0b43409b7be36e3267155b : FAILURE in 52m 08s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/00966efefdca4c74acf4b960ff6a803a : FAILURE in 54m 47s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/0944f21861254a4aabfc44faf049ff28 : FAILURE in 52m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a20f58c6e72887ad31d256a721d55e9d018eddb0","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-06-25 22:50:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"011a61f08a318879ab603a52a5f67639959ce70d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 00:50:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/077af1ec5ca4464499472e35f05f8688 : SUCCESS in 1h 40m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36eec7f355234641a102ee3a44c71f4a : SUCCESS in 1h 25m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f06244a82de34f2c9ffb86b33f30a28b : RETRY_LIMIT in 3m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15492ac1a1a14e8eabba0c9490560e4e : RETRY_LIMIT in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb95a97e48f2457584c7da84b026af50 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75b4dcf15bee4051b2ef43d007f7f890 : RETRY_LIMIT in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/175ade42b0af43b38a05edbf405af5bc : RETRY_LIMIT in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04037911dfd541e2a9499c37a5ec7a54 : RETRY_LIMIT in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec897c52cab74c48908c634f54dad0d5 : SUCCESS in 5m 14s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/5f83deede2424ede84bfe595fe53a918 : RETRY_LIMIT in 4m 17s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/81af8e4fce384deaae8b0ec5921fa7f0 : RETRY_LIMIT in 3m 02s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/cd896173292b4e0cafbfdda2cb3fb8ca : RETRY_LIMIT in 6m 19s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/1451012025be4c91a4e61d57d2639d00 : RETRY_LIMIT in 3m 29s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/b66f104ac6ec407fb157881c48d70bf4 : RETRY_LIMIT in 3m 12s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/d6ff26e73be649f0be19acd25604295c : RETRY_LIMIT in 4m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8dad55b2f26c4c89968058da5a83fe1d : SUCCESS in 1h 29m 43s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/b2fe2edf03104d9ab1becfa2e59c7a34 : SUCCESS in 1h 27m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/55249fe75a974d6f8de96b5151aa657f : FAILURE in 12m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7d14173c666d449598ba552fa8f27d1e : SUCCESS in 1h 45m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/177725f41bc54aa99fb6806c83f88229 : FAILURE in 16m 15s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/89206321b0c24843956a6a68f92acf6e : SUCCESS in 1h 04m 39s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/1d98cb656c1b4dce8a1abd7f07573562 : FAILURE in 32m 13s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/bb7b26c84ba04d5eb5fc8c1cc9e83aa0 : FAILURE in 42m 35s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/6a3348115095484db816d0f15851ee1b : FAILURE in 44m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/a911a3f4cf40451d9c07cc3ab18cf600 : FAILURE in 42m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a82b85921660573e3e711e4bd67307b87d9fd46","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-06 15:20:41.000000000","message":"Patch Set 1:\n\nrecheck - test_setup.sh is now fixed in Train.","accounts_in_message":[],"_revision_number":1},{"id":"4de93128e7637061cd62ed08f76cee1a5349f2e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 17:34:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d77e83e9e3be45f9b8b0dc9941a0d6d4 : SUCCESS in 1h 42m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2fc5390648b45dfa7420857c27d72f5 : SUCCESS in 1h 38m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/185aa086e9e94b0a8b93f75ef37afaa5 : SUCCESS in 9m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a34b8d724804637afcedf39cf357d05 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dca2f5eb2b7a4f5cb012842c972cb77b : SUCCESS in 7m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/beb6968cb2ae419ebf9658554866be98 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f91274d7fd34d8d9e7bdf5b8260a878 : SUCCESS in 9m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb3f35214f8f4379a37088d641d0d31e : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40f04cf7c9f146a5bde53f4819db388c : SUCCESS in 4m 44s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/71febd37241c4205b2c78c98d0420177 : SUCCESS in 5m 45s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/5a3c1c4321ad43e08ffe55725c9bc8f3 : FAILURE in 13m 24s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/6efc296525614f9ca81b6c179002fd40 : FAILURE in 8m 42s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/59ed9e467e884f7b829b04005e9be2fa : SUCCESS in 5m 51s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/8ee686be226a42c9a31176e7488650a6 : FAILURE in 14m 22s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/423e8584c15c48c4ac8328c0dc7ff519 : SUCCESS in 6m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7301c9805c8c4f41867054333ea5ce65 : SUCCESS in 1h 16m 18s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/b9f26406a2fc455eb84f697f6621d201 : SUCCESS in 1h 11m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/becf19b4ed274890a923c135eed295ec : FAILURE in 15m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/4029f666c17546688930440289bc57b4 : SUCCESS in 2h 03m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/763d37f896964291a28413f0648c3012 : FAILURE in 14m 03s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/2bde9963b64b4955902bdb3994cd6062 : SUCCESS in 1h 03m 14s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/776961f6fcff4996a5f841d6c25f569d : FAILURE in 21m 37s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/1fd4db09dae449c89fc97253de4fe322 : FAILURE in 45m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/cec5bc7fa79b4d869bb6dca0694d518d : FAILURE in 56m 29s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/6eb0205d2ee54a2a9bd8c2c9cfa301a3 : FAILURE in 56m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57e3787bd390142bb56c85593a715e231cf8d7df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 17:34:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3785ebda8a88005984beb1bbc8b3b35b7f402818","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 19:35:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/169a7d01b9ca4b43aeafb28a9a67d640 : SUCCESS in 1h 55m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e821a52997ae4be8a2c50bb7cd7fff60 : SUCCESS in 1h 46m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b0aca226a72479fb5746bda796e95bf : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/076f08de1f2b464a9653932c0e8353be : SUCCESS in 9m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e39acda0b56e45b598529fc5355e2f8b : TIMED_OUT in 43m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f343c644f2ec403f9c4e048d097c2f57 : SUCCESS in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2b8fc1489434d58b836780293a8b517 : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8372adff926a4ad981de85aa1982a5a8 : SUCCESS in 8m 33s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/705a2e96a1b84749b92f1708e6d497f6 : SUCCESS in 8m 39s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/cb684bcf5e9a4dc8b34a53ae2915f788 : SUCCESS in 1h 39m 33s","accounts_in_message":[],"_revision_number":1},{"id":"d62fe8c3ff117c14a37b36850abddaba0b634b4e","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-07-06 23:55:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9cb8e6c0a45573f83bbaa1d87d172fd8842939a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 01:46:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d76ae7a7de6d401b873bf31fd781d767 : SUCCESS in 1h 27m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac5f24e3995e458ba8aa294072643132 : SUCCESS in 50m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f305954eb1b46f9873c77d9adc7beaa : SUCCESS in 5m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9420e404e13e458f8b2292f451dd7ab6 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06cff45ff5c7441db2b04f490c2c7f33 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99000234d5384f0bbba1788ab271191c : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc3761d27dc340eda85f48b96e642606 : SUCCESS in 7m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea8aa87daef7491890ff8839f3812015 : TIMED_OUT in 41m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6832245f78e4408aa4b707bb966630d6 : SUCCESS in 4m 05s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/30e35d8eeb4e4ed39ebe585363528ddd : SUCCESS in 6m 08s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/49707e267f784c4f983e2d2ad7d5c177 : FAILURE in 11m 57s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/9e906db5bfed42dd8fee66e628f457a3 : SUCCESS in 5m 53s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/83fe63c7f3704023b6894187c248654b : SUCCESS in 5m 07s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/56d39f3f4b234a9484963379c96b3c62 : FAILURE in 13m 05s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/9d172898b4d942e8b88740cb810e91b7 : SUCCESS in 5m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/81f455dfcef844408c6c7a9692b8a950 : SUCCESS in 1h 32m 44s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7342a8f6d15e43cb9beea37c9d70de94 : SUCCESS in 1h 27m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/5b332360774145e08791c91036b0e68b : FAILURE in 17m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/99f33bbd452e44bba3598263e9ebef32 : SUCCESS in 1h 44m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/204a31283a1541d1ac9a62a755f544c8 : FAILURE in 18m 54s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f364866c2f41432a9ec1e20db09eeeee : SUCCESS in 1h 04m 55s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/2a2bac8513b84df2908fd537d97f4e9e : FAILURE in 25m 36s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/27e460b45075426282c47fc9235c66a4 : FAILURE in 1h 17m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/005541d4c5be4352992e88794ceee723 : FAILURE in 46m 53s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/c59edec515ce4ad4be1987752b2cb1a7 : FAILURE in 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1ff3bb2998005da344ef86af5e28245b8b55ed71","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-07 07:57:06.000000000","message":"Patch Set 2:\n\nrecheck - timeout @ tox installdeps. weird. \n\nhttps://zuul.opendev.org/t/openstack/build/ea8aa87daef7491890ff8839f3812015/log/job-output.txt#678-682","accounts_in_message":[],"_revision_number":2},{"id":"d6437796af2c112079496e2e5855be7e031d6163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 10:03:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6608a69ed251422aa91c0e3c85d70b02 : SUCCESS in 1h 35m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/61f470a0b9df4bcb92c1b21f762ab2f0 : SUCCESS in 1h 20m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69ab011007544e3aa78ec78bd794907c : SUCCESS in 6m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04117ee5869b45d2aa5e09f46539e194 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36fb0c39771647cea95d4f40c44601a7 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd8de31d3f564ff2a7f1bcfd388af768 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e05f40fb0b4449788b4bef347e7419d2 : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f097f7cbed304266bf0e0b433a0ecae6 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba73bf836c346b0b8a3ffaea80274d8 : SUCCESS in 5m 20s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/da1b1f5a13f942cabdd4599be66df857 : SUCCESS in 8m 15s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/d33bb7e055344f1ea8a32c8070478e26 : FAILURE in 14m 19s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/240739cee257498fa40d445da1f55d84 : SUCCESS in 5m 55s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/6ba96be8468346dfa1dfc7194462412a : SUCCESS in 5m 46s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/0a605ab5168c4b88a78b8488cb05c6b1 : FAILURE in 15m 59s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/2925693e6ef048a48ab4868388d857b6 : SUCCESS in 7m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/91409360c3e846a8932021ee4308d448 : SUCCESS in 2h 02m 19s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/8cac167871124609bde1583e8d32d7ed : SUCCESS in 1h 35m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/0d3687ba53e7486ebe21e3abfb59c757 : FAILURE in 17m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/cddb8a5ebf4d4c94887b3c0a498cfdbb : SUCCESS in 1h 10m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/7cbe90488bf04f209f3a01cadd4aa052 : FAILURE in 18m 12s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/87d1d04b369f4de0b167a0ff65f82beb : SUCCESS in 1h 23m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/788ea8e2424143e39eb75040507182b0 : FAILURE in 24m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/cf7354c5d430493e90ee24b53de7d252 : FAILURE in 48m 33s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/2c21f16007994b34a616e1be65a03c27 : FAILURE in 47m 58s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/5e09a62d0eb2426db81edcc779ec3330 : FAILURE in 58m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"514f21872f949385f3da2d79fbb274248df14642","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-08 09:24:27.000000000","message":"Patch Set 2: Workflow+1\n\n\u003e Merged to stable/ussuri, good to backport.\n\nUssuri is now merged","accounts_in_message":[],"_revision_number":2},{"id":"52e2d5b2bb79e06613509d61e6c9b418530b44ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 09:24:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3fc6fc210482e63f4a954c48da75a67dc45dd8e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 11:13:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/62c115b5542b4703b65c4b73a0d62ad5 : SUCCESS in 1h 33m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e56a6eace74e49fd9082419c50af4660 : SUCCESS in 1h 19m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec83d9f78bb74c188483300034aac15a : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a540093549cf44c3a08be278922d81f0 : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e19221e1008e4d3497d28f99337efed0 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6509184f05ce42ec99ddb498e64aa123 : SUCCESS in 7m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/470b131de62a4e2cab90261b84053f5e : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d9a4eb314143369cc5d1a175534a63 : SUCCESS in 6m 03s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/8b2cfb16dfc54a309cc85cce57fc9b9f : SUCCESS in 7m 18s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8f3fef2b7d7d4d969e3d831d7cab63de : SUCCESS in 1h 45m 48s","accounts_in_message":[],"_revision_number":2},{"id":"31f26790dd5de17a9960301eed8af95242a6956d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 11:13:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d194254e8ae0a4ade4ba47ac29b846c59e4b53e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 11:15:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/023af3534ab44ff5888b2be41395ef00 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2},{"id":"190e46ea6b8d37a4d5cc0c3def44281445630be5","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-07-08 20:55:46.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit fd795bbb8f5aaa740fa810f2248646f6ec3a211e","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1613b7a968f33ef7dc683d05fde05be0c458c8fc","revisions":{"4b603657b7b2f04ac8b612b7ae1caeea0034d2df":{"kind":"REWORK","_number":1,"created":"2020-06-10 04:26:31.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/40/734740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/734740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/734740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/734740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/734740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/734740/1"}}},"commit":{"parents":[{"commit":"29c1f3990adaa62651ec96754c49e8043c97cc61","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/29c1f3990adaa62651ec96754c49e8043c97cc61"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-04-09 10:00:39.000000000","tz":480},"committer":{"name":"norman shen","email":"yshxxsjt715@gmail.com","date":"2020-06-10 04:26:31.000000000","tz":0},"subject":"Print warning message when connection running out","message":"Print warning message when connection running out\n\nI believe when connection pool running out of available\nresources, there will be all kinds of weird consequences\nfollowing. So let\u0027s print warning logs when we are going\nto waiting indefinitely for the lock\n\nChange-Id: I5ca930cada523bc77c644703a0f02b9160816231\nRelated-bug: #1871813\n(cherry picked from commit 96300a32213d1a9bf114b90bef6971d69b839d1d)\n(cherry picked from commit 39f770f55b3331b7a99b3ea42a34982a20635cf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4b603657b7b2f04ac8b612b7ae1caeea0034d2df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4b603657b7b2f04ac8b612b7ae1caeea0034d2df"}]},"branch":"refs/heads/stable/train"},"1613b7a968f33ef7dc683d05fde05be0c458c8fc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-06 23:55:06.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/40/734740/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/734740/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/734740/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/734740/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/734740/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/734740/2"}}},"commit":{"parents":[{"commit":"88cf5aebef790ff40ab8ca7219d537ccfc54d12e","subject":"Simplify tools/test-setup.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/88cf5aebef790ff40ab8ca7219d537ccfc54d12e"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-04-09 10:00:39.000000000","tz":480},"committer":{"name":"norman shen","email":"yshxxsjt715@gmail.com","date":"2020-07-06 23:55:06.000000000","tz":0},"subject":"Print warning message when connection running out","message":"Print warning message when connection running out\n\nI believe when connection pool running out of available\nresources, there will be all kinds of weird consequences\nfollowing. So let\u0027s print warning logs when we are going\nto waiting indefinitely for the lock\n\nChange-Id: I5ca930cada523bc77c644703a0f02b9160816231\nRelated-bug: #1871813\n(cherry picked from commit 96300a32213d1a9bf114b90bef6971d69b839d1d)\n(cherry picked from commit 39f770f55b3331b7a99b3ea42a34982a20635cf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1613b7a968f33ef7dc683d05fde05be0c458c8fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1613b7a968f33ef7dc683d05fde05be0c458c8fc"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
