)]}'
{"id":"openstack%2Foslo.messaging~735385","triplet_id":"openstack%2Foslo.messaging~master~I51cf67e060f240e6eb82260e70a057fe599f9063","project":"openstack/oslo.messaging","branch":"master","topic":"dev/am/ping","hashtags":[],"change_id":"I51cf67e060f240e6eb82260e70a057fe599f9063","subject":"Add a ping endpoint to RPC dispatcher","status":"MERGED","created":"2020-06-12 19:05:37.000000000","updated":"2020-08-28 19:59:26.000000000","submitted":"2020-08-28 19:55:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735385-1598644530183-69a35627","meta_rev_id":"15c9fff59ec50b5796652f4202283e378a3f400d","_number":735385,"virtual_id_number":735385,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 19:55:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"date":"2020-08-20 09:51:53.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":7321,"name":"Prakash Ramchandran","email":"pramchan@yahoo.com","username":"pramchan"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":2,"date":"2020-08-20 13:55:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2020-08-19 09:53:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":7321,"name":"Prakash Ramchandran","email":"pramchan@yahoo.com","username":"pramchan"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":1,"date":"2020-08-28 18:14:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":7321,"name":"Prakash Ramchandran","email":"pramchan@yahoo.com","username":"pramchan"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":-1,"date":"2020-08-19 09:53:59.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":7321,"name":"Prakash Ramchandran","email":"pramchan@yahoo.com","username":"pramchan"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":7321,"name":"Prakash Ramchandran","email":"pramchan@yahoo.com","username":"pramchan"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 09:12:08.000000000","updated_by":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"reviewer":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"state":"REVIEWER"},{"updated":"2020-06-15 09:12:08.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2020-06-15 09:12:08.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2020-06-15 09:12:08.000000000","updated_by":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"reviewer":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"state":"REVIEWER"},{"updated":"2020-06-15 09:12:08.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-06-15 09:12:08.000000000","updated_by":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"reviewer":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"state":"REVIEWER"},{"updated":"2020-06-15 09:12:08.000000000","updated_by":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"reviewer":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"state":"REVIEWER"},{"updated":"2020-06-23 11:28:40.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2020-07-10 15:37:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-27 12:09:13.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-07-27 20:54:08.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-08-06 14:48:50.000000000","updated_by":{"_account_id":7321,"name":"Prakash Ramchandran","email":"pramchan@yahoo.com","username":"pramchan"},"reviewer":{"_account_id":7321,"name":"Prakash Ramchandran","email":"pramchan@yahoo.com","username":"pramchan"},"state":"REVIEWER"},{"updated":"2020-08-19 09:53:59.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-20 09:51:53.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-08-28 18:14:26.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-08-28 19:55:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02633ad7f5b3cc6c0c3d1b6d747088bbd150ab59","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-06-12 19:05:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c608fb2a015c38c8fd49edda1ff2cb04cfa4743b","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-06-12 19:07:05.000000000","message":"Patch Set 1:\n\nHey oslo team, I know that I missed the tests in this change, but I\u0027d like you to review, talk about that.\nI will be happy to read your concerns.\n\nCheers","accounts_in_message":[],"_revision_number":1},{"id":"3d32dbc9d12e7e61bcf15f69cf78cce752a693c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 20:47:08.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb0b47972e474d9ca46d49b0aaed1951 : SUCCESS in 54m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d155783eec1499a8e81c487b1f820c4 : TIMED_OUT in 51m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6ee35c75b564f03bfaba784a9341c88 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd79fb637fd54e50964e27f379eb9b3c : FAILURE in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a65d75b9abc14fd68bb1d499d410e5b0 : TIMED_OUT in 40m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8843d57b6f844cba99164243b1a8448b : TIMED_OUT in 40m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c31f4c36ae54b43ae1f9a20980fb994 : SUCCESS in 3m 47s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/226ee09c19cb4f1b99c10fcde6d0fceb : SUCCESS in 5m 39s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/9d31765489e94725bcfc8b5a6801351b : SUCCESS in 7m 28s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/382f88df800543a5b82cfd2ffb41be13 : SUCCESS in 5m 13s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/21a55ddae00b49b58142f7d09ac23d03 : FAILURE in 6m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1c6881e323a3451195b0ae6dd52f979e : SUCCESS in 1h 40m 56s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e46ffa4fb2f94149a2c09f47d3bda785 : SUCCESS in 1h 27m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b18cae201bb94ebc9b8dace1e5ed077b : SUCCESS in 1h 18m 50s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d68631e469b7483491f1994062991d85 : SUCCESS in 1h 01m 12s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/ebe1a92d90b94ecfbac300f50b203ef3 : FAILURE in 17m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3077adb4dcacac539c300dfd73c00f1a21b2b1f","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-06-15 08:06:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"de0f3979c13c75627978b2b27db7d6171d4ec116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 08:59:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffb14c8d22bf499c9ceb79c60f40c884 : FAILURE in 22m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32b54cdf70c94bf9be0aed7bb977c046 : TIMED_OUT in 52m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e017d1e9925c4528916dd14b870c6d35 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e46c38dee684ab7ac82a0847090ccde : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73c56edc4a184aa4ac7fcf355bcb8cb3 : TIMED_OUT in 41m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ef9160cd7524f7bbe8c93addbc45b4c : TIMED_OUT in 43m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e6a6b92868b42bea9f6482130b4eebd : SUCCESS in 4m 59s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/b580447ad63841609b151f6e6f7446d5 : SUCCESS in 16m 39s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/9a03724471574f4991e04a017db50e4f : FAILURE in 18m 53s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/15dad158dd554b4196f0e46555a5feab : SUCCESS in 15m 55s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/4ede520aaf6642148deb7ee0c413131f : FAILURE in 16m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/285265c32fa64f7e8205189715b1b5e2 : FAILURE in 24m 06s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/f9150860eb4d45d486d60b981d8c83c3 : FAILURE in 27m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/4cfbf5cf07d74fa69a2123f67e7ee722 : FAILURE in 26m 52s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/2fc216827c9c4d6c8f9eda8cbee2d1a3 : FAILURE in 27m 57s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/57d97aa33aae45b18f6d1fa41da3ee7e : FAILURE in 21m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"060742591104a40b8e660aeebda2b54a62427e12","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-06-16 12:42:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"441adcbbead3a2a38a2691dc2185c506c43e1fd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 14:18:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2dd6e60637f040499d24a04e60574e3e : SUCCESS in 1h 30m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d756a3e1ad84c039d2551e568de1f20 : TIMED_OUT in 52m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a29941f991040f9835165c19999bb0a : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afc052509d3b43a79e155a4c01630ab4 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/600ab4b1d25d4a38b2f63c8bb08958a7 : TIMED_OUT in 42m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06716906bfe443e59c26497749d4e74e : TIMED_OUT in 42m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7eddb82808d4814b0fc3a2de77ec76d : SUCCESS in 7m 49s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/c040029f5bb6486e939d89f119e8fdaf : SUCCESS in 8m 06s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/03406c00eb33439ba9b67141d859e0d2 : SUCCESS in 7m 58s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/c013dd03e3ea4cee9e161c8b75cefc27 : SUCCESS in 7m 09s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/2999e72ecf934be08e2b9c0c0fef2278 : SUCCESS in 8m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/63e91b315d564d1ca39247077229a82c : SUCCESS in 1h 04m 01s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d2f37ad4f4794b669d2971ca0dbc4d43 : SUCCESS in 1h 30m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/dd2871ec847a4ab2bdc097d69cb46579 : SUCCESS in 1h 27m 33s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/38d81b00e2bc42c69703e85d1aff0242 : FAILURE in 18m 27s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/4c4e1428673843189371968ca16d0982 : FAILURE in 19m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3bde9236727e5eb34610660a0e2a17ddac0452f0","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-06-17 07:34:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a53a36b8d82ac930c425998e342d34f126d4ac91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 09:25:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffd65af8f6804222882ecd399391c697 : SUCCESS in 1h 30m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/868e8fec421748a886f85999d98cb216 : TIMED_OUT in 51m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b906a1397f894c76bc9c6fddeb1f6796 : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0925237516934cf8939bb71c6e3d1d30 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b57763a0e0e4cd79ef57e41390a5ea9 : TIMED_OUT in 42m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fe420acd9cb461cb04bd16dc3aa5d4f : TIMED_OUT in 42m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a50a46b4280492b99f9226ee1777ff4 : SUCCESS in 6m 35s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/ff17e839264941bc820f65c246a0b9de : SUCCESS in 10m 33s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/02d1b8eeda0d421c8aa58f542ee7cdba : RETRY_LIMIT in 6m 26s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/fc5446f5c4b241df81c2f2056d01745e : SUCCESS in 7m 06s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/b3574956db8443cbb2587951f667d803 : SUCCESS in 8m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8ba43f2ee8094d2bb4a04bb6ab11fda4 : FAILURE in 1h 27m 12s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/b66b0c3307974614b02ca54a876b2dec : SUCCESS in 1h 33m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/181888970d5d4ad3b588bce57ecc163c : SUCCESS in 1h 42m 57s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/2ab4a350626e45c1b1c58b23e8b4ead0 : SUCCESS in 55m 32s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/14a524cf5576472f801fd02483e91a5c : FAILURE in 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"04acc606867f711201cbd890afef7377e7548631","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-19 12:47:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"175048c08c604dc823af37d9d59bf1d19e648423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 14:43:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff4cf5b23e384a2dbbaa8110a5091075 : SUCCESS in 1h 05m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/572a8ece1b104f03ab5feaecbea62956 : RETRY_LIMIT in 3m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e769bf5e9f34e0fa6fe7964eb6ad479 : RETRY_LIMIT in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbcc359c1bb74df79583ff2a830063e3 : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfe85cf729ce4d25bdf742345c473d68 : RETRY_LIMIT in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7471f45bc21432b9c01b07266138586 : RETRY_LIMIT in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c861f231d0e4403865d20c2fee149bc : SUCCESS in 8m 27s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/88615b7adc2f4872b408018777efe361 : RETRY_LIMIT in 3m 03s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/e365640d0c0a490ba916dcd6b4091ca8 : RETRY_LIMIT in 4m 59s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/d543403d0ca944f3bc45e9dcfad54921 : RETRY_LIMIT in 3m 07s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/442c41268fd040918200d53d757bd942 : RETRY_LIMIT in 6m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/36ae5a8c72184c87bd770e320719d26c : SUCCESS in 1h 52m 00s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/38c17462ac9049b49c966dc2f0b115c2 : SUCCESS in 1h 35m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/efec8ee3070848a9a2adc54aa138f45b : FAILURE in 1h 22m 44s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/6d8256175db44c27bfd823c34aba2d9d : SUCCESS in 1h 01m 56s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/34e9505ea68f45fe9101e00af7111933 : FAILURE in 20m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"83974ef6e191799af20b0bb20ec9a1fa883d3777","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-06-23 11:28:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c5a43762a8369d78ca77e30654f45313c472eff","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-07-02 15:31:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e6fd653e1b25442039b4085701832b5662d52b3a","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-07-02 15:32:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"88842e68c6dacac5d7039899f6891c648d00412f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 17:20:19.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5896053ca07e4be1870e63f298e560e3 : SUCCESS in 1h 38m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49736bbfcf434861aa35862faafe3e26 : SUCCESS in 8m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5bee4aca86a44be8a6f8eaa33c77487 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22bb3daea6c5484a91a5b7b50909dc9a : FAILURE in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/598a44a336c445f2893c10787541a322 : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e5be030663c4d1383cd6c42c7791279 : SUCCESS in 14m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400df0cf7ba849e085720140895598cb : SUCCESS in 7m 02s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/5beba387dd8a448783475066cecbb1ec : SUCCESS in 8m 57s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/20489509fbe4497a84ce05c5627da085 : FAILURE in 12m 39s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/74dd614c9058447f96b69d859e77760f : SUCCESS in 8m 02s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/0b8091f7c4ac4ffc8d20cf87fb4e22e8 : SUCCESS in 9m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bce1f567474a4bb8b1aa0fc8b6784b7e : SUCCESS in 1h 33m 15s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e0320320dcec4eee856d25cf87ed441f : SUCCESS in 1h 06m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b15bcc4b1ec34d9ab0bda3269f3acd1b : SUCCESS in 1h 34m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/8879f6ca074d4237bceeaa5ababf2e8b : SUCCESS in 1h 08m 01s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/8853357b738944e6abd8ed19b39c5b22 : FAILURE in 16m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e9983308f38aafabe778952cc9c2ddb03ad4c176","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-07-03 07:22:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f4147f91218725b636ec1b831d2194d974f1e439","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-03 08:29:49.000000000","message":"Patch Set 4:\n\nHello,\n\nThanks, this could be helpful.\n\nCan you add some related test, especially to test the loaded endpoint, and maybe some functional tests to check the ping feature.\n\nAlso please add a release note to highlight this new feature.","accounts_in_message":[],"_revision_number":4},{"id":"da00cdd48f5d905ead299e7a30c741a73f0914ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 09:19:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d30e31ecea54f88ab0978ffca1a14bc : SUCCESS in 1h 21m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fd193de22c445e7909e7131bbeba547 : SUCCESS in 7m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e05c7925ebfb47e7be21aa59e9230623 : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96f2bc50e7b346019b79df685fe9f400 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24867432d9f24b3cb2ceef4d603b660a : SUCCESS in 8m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/193cb53e04bb4bf7bade87479dfffd10 : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3186df8f0541edb57797fdfbe72208 : SUCCESS in 6m 49s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/ef00ab8d83d04e8186d877947e7c8175 : SUCCESS in 7m 31s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/b8bcc3e0c8d74254b7300f3945816ba2 : SUCCESS in 8m 53s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/cce46506aad24f05aba792f13b3ffa8a : SUCCESS in 7m 41s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/bc1cd71be4d74a92aeec0a1c041d9193 : SUCCESS in 9m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ac96e028c9004ef49182cdeaa49e0721 : SUCCESS in 1h 38m 39s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a267432199394fb5a9ac5b5e87d52c65 : SUCCESS in 1h 36m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/4f7437b70dec4318a0a6dcd73bc61a9b : SUCCESS in 1h 50m 17s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/a926be0f571243edb2f51aec59f3c60b : SUCCESS in 40m 29s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/2c991e4fc7c84c949e6acf217c215f84 : FAILURE in 21m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8211f60361134e462593395de606963e9c086ecc","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-07-06 15:11:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c3e9f627c3e97ca1235e02f834f108261c18b914","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-06 16:19:00.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)\n\nWe already have healthcheck. Is there a different use case or need that the healthcheck method would not work?\n\nhttps://docs.openstack.org/oslo.middleware/latest/reference/healthcheck_plugins.html\n\npep8 issue needs to be fixed. Make sure to run tox locally before submitting.","accounts_in_message":[],"_revision_number":5},{"id":"a89a787e991511c6bd1cf1c1eab1ec80aa84b807","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-06 16:33:39.000000000","message":"Patch Set 5:\n\n\u003e (3 comments)\n \u003e \n \u003e We already have healthcheck. Is there a different use case or need\n \u003e that the healthcheck method would not work?\n \u003e \n \u003e https://docs.openstack.org/oslo.middleware/latest/reference/healthcheck_plugins.html\n\nGood point it could be a bit duplicated with oslo.middleware\u0027s healthcheck features. If I correctly understand the needs here are to monitor openstack services (like nova, neutron, etc...) so oslo.middleware\u0027s features could be enough for our needs (related to the explaination of the commit message).\n\nHowever, in one hand an advantage of these changes could be to allow us to check the state of connections over RPC where oslo.middleware, seems to monitor things over http, on the other hand we introduce some sort of duplicated features.\n\n \u003e \n \u003e pep8 issue needs to be fixed. Make sure to run tox locally before\n \u003e submitting.","accounts_in_message":[],"_revision_number":5},{"id":"7bacf7401222dcd7e57e3b89087e5e45d486022c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 16:47:13.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0792fd8c5fd48368cc83c3144e0de3d : SUCCESS in 1h 25m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d1f84aae2674069b5e4223a13de81c7 : SUCCESS in 7m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e410ebdfe6c44689652db0fbce7223e : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09218eb6e2964d0ab27a5900fce43e47 : FAILURE in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23842e1b5efe4bf1974fb2deedf7d27d : SUCCESS in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63309bede9374890b1231c10ba77458a : SUCCESS in 9m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a091ec0c4ec34d51b75fcd42b195d850 : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb3df8833a4c47cf9e39fa48882ed80c : SUCCESS in 5m 18s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/28de6b76bbf246c78a3fafd43a3c7fa0 : SUCCESS in 6m 08s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/7fa166f0ad2f4a4c9eb54989398ac8fe : FAILURE in 9m 15s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/596f9724db5645b99863b03d71bfdc18 : SUCCESS in 8m 55s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/d57f271f847f41dc878d8c17e08c1839 : FAILURE in 7m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/da182c03c1e34adcb5799a20b455ac7a : SUCCESS in 1h 25m 12s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/06649c8837fe4725b925478093df5366 : FAILURE in 1h 24m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/40ced2d068a145aab5394b577c4f1f34 : SUCCESS in 1h 30m 40s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/50471cb019024114a6df941b69651067 : FAILURE in 10m 15s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/33add82f5afe40cca995a6f7ab566ef0 : FAILURE in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8a02724c50b374648deddb3fd9f779bda98891be","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-07-07 07:57:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ee5617eafb815340acc8f790a1058da33936fe2e","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-07-07 08:03:17.000000000","message":"Patch Set 6:\n\nHey Sean, Herve,\n\nSorry for the pep8 issue, I forgot to execute it on this specific file.\nI did all suggested changes.\n\nAbout the oslo.middleware healthcheck, this is different.\nAs herve said, it\u0027s answering on HTTP while this patch is for RPC.\nAlso, it works for all RPC endpoints (so nova compute, neutron agents, etc.).","accounts_in_message":[],"_revision_number":6},{"id":"df8ff90f6f892d0d5800505b0ce4be27bd039462","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 09:40:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8fcbc07f8a7446cdb919c93e2e70aaba : SUCCESS in 1h 27m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc7fc747564246f8a9cd93eb289fd90b : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2672b5ffcdea41d28f5c384f79d168f8 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75740c37eed04f16a7655a54b53f385c : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/336f54e8d06a41acaab7390780143b1a : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba6e9c9ae0074dcc89a3597d31b16cd1 : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e93eb1743a4b9d933d0b9970b6ebc1 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/613455b5614840c680e9765ea75e5e2a : SUCCESS in 6m 07s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/355c98a9930d4f8ebc7b3d514b96b2c4 : SUCCESS in 9m 31s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/7c045db770d147e1bece17c797a6419d : SUCCESS in 8m 00s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/63e963169f394212b86442e1dd5191dc : SUCCESS in 10m 04s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/316b545f637846dfb51b4920dd6ed084 : SUCCESS in 6m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/fa88271720fe4db98b9609f1a036470d : SUCCESS in 1h 31m 49s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3744c1e0c89f4c788f18099cc3f51e11 : SUCCESS in 1h 20m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1f443427ed514c4ca02659e65adcd5b1 : FAILURE in 1h 30m 51s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/285c63418bbd4d5dacf9c30c79e9112e : SUCCESS in 1h 03m 51s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/6e70ce1327e84b4faee374f4472ea836 : FAILURE in 18m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1ed6655b0090ab0d9a67ecec8549ed03a6295a45","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-08 09:55:02.000000000","message":"Patch Set 6: Code-Review+2\n\n\u003e (3 comments)\n \u003e \n \u003e We already have healthcheck. Is there a different use case or need\n \u003e that the healthcheck method would not work?\n \u003e \n \u003e https://docs.openstack.org/oslo.middleware/latest/reference/healthcheck_plugins.html\n \u003e \n \u003e pep8 issue needs to be fixed. Make sure to run tox locally before\n \u003e submitting.\n\nAs Sean\u0027s comments have been addressed and due to the fact that I\u0027m fine with the rationale behind these changes (PING over RPC) I +2ing this patch. I\u0027ll lets other core developer decide if we want to continue with that and if make sense for them.","accounts_in_message":[],"_revision_number":6},{"id":"f8a9340c27d799cbfecb59b648db2dc6179785e2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 15:37:46.000000000","message":"Patch Set 6: Code-Review+2\n\nThis makes sense. Thanks for the explanation.\n\nI will hold off on approving to give others a chance to review since once we add something like this, it won\u0027t be so easy to revert it if there is some issue.","accounts_in_message":[],"_revision_number":6},{"id":"26ef51a0d26b90e6c228290f3feb732e2519d1ea","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-07-10 16:16:20.000000000","message":"Patch Set 6: Code-Review+1\n\nI really like this capability.\n\nSomething to consider for a future addition would be a simple command line pinger tool?","accounts_in_message":[],"_revision_number":6},{"id":"eb0a61d55968236dea51d4b2953340b5d62ba1e7","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-07-27 12:09:13.000000000","message":"Patch Set 6:\n\nI am glad to see someone tackling this, it\u0027s been a while we\u0027ve been discussing this topic now... :)","accounts_in_message":[],"_revision_number":6},{"id":"af3b02655383c8ea6c7ac95185428cacb9fd9223","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-07-27 20:54:08.000000000","message":"Patch Set 6:\n\nHas this been tested with Nova? Based on [0] it sounds like it may break there.\n\nThe other part about whether we should be doing healthchecking directly on the rabbit bus or via some other mechanism is a bigger question, so it would be good if people can leave thoughts in the thread on that too.\n\n0: http://lists.openstack.org/pipermail/openstack-discuss/2020-July/016106.html","accounts_in_message":[],"_revision_number":6},{"id":"6acf801cb30f821bd35db8a4b714d083c3073073","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-11 10:38:44.000000000","message":"Patch Set 6:\n\nAgree we should check if there is naming collision, and if needed pick a more unique name to avoid it.\n\nRegarding whether it\u0027s the best way to monitor this, I\u0027d argue that the patch just adds one door to monitor the health of the system -- the door is closed by default, and even open, it does not create any load by itself.\n\nIt\u0027s an operational choice to decide whether you want to use that door to monitor your deployment, how actively you want to use it, and accept the trade-off (extra load on Rabbit vs. more visibility).","accounts_in_message":[],"_revision_number":6},{"id":"81db3827d1bcac7f94a2a5d70ea0b1feeac5ed90","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-08-17 20:03:40.000000000","message":"Patch Set 6: Backport-Candidate-1\n\n(2 comments)\n\nLet\u0027s move this forward but it will need to rename the method to keep from colliding with existing applications.\n\nSee inline.\n\nthanks","accounts_in_message":[],"_revision_number":6},{"id":"31679e51e0eee745c06788b9bc4d0302ddadfa64","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-08-17 20:03:56.000000000","message":"Patch Set 6: Code-Review-1","accounts_in_message":[],"_revision_number":6},{"id":"dc330a78928ab8667b2d92e23be34df24053c6f3","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-08-18 13:05:53.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"34600ef9392af6f32474f2a7d90fe84acc0c7f96","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-08-18 13:07:09.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ee25cedf69fa6677e2842fb0b0fb7682aa138ad4","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-08-18 13:10:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"433cbf80bad4781d873a024acb5f6e415486db8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 14:48:04.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95e2a08c0599476ba80fe239e91e2796 : SUCCESS in 1h 30m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb04720564a943adb062be88d02bf933 : SUCCESS in 7m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3072e9478f33484094ca8f53846e363f : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f48479b59864c7c9c9953ff6fe6995f : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07217cf312f34f729cc3abb0371afb7e : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe84c9f1a5954437b866b17be30e9d6b : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36be76454e254f56a89b43f4a2c31d64 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0f6860ec29d4239bf1c4eaa5087aacf : SUCCESS in 4m 22s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/8c77c0be33ec4ec3a2b6364f18a19c32 : SUCCESS in 7m 22s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/9e9ec1269d7749b1a39f2868201f9945 : FAILURE in 8m 26s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/732952806d1f4394b75b43aa9d7dc7c7 : SUCCESS in 6m 31s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/3925ecd9ed8a4194bab30715ef148178 : FAILURE in 6m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2f30aa3c978f43d38637ab4e03d336da : SUCCESS in 1h 04m 48s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/91b33ca8ab344609940327eb7270897b : SUCCESS in 59m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8ed27cbfc6ae4db19fe915d94a7f034f : SUCCESS in 1h 17m 33s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/5d51c26bcdb8463ab3b72053989a401c : SUCCESS in 50m 38s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e289dc959a484d4c8aff4acf5a4e0cbe : SUCCESS in 1h 01m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8b24bb5badbd30a7150710b1abc3e6809423cad3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-19 09:53:59.000000000","message":"Patch Set 8: Backport-Candidate-1 Code-Review+2\n\nLGTM, Ken\u0027s comments have addressed and we avoid colliding with services\u0027 endpoints.","accounts_in_message":[],"_revision_number":8},{"id":"0ba9aa6e9f8334dff76c4648157388d9de6aa5fe","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-20 09:51:53.000000000","message":"Patch Set 8:\n\nLGTM, but before approving this, we should loop back on the ML thread.\n\nArnaud: could you post on that thread and confirm the assumptions I made in http://lists.openstack.org/pipermail/openstack-discuss/2020-August/016477.html ? Also any precision you could give to Sean on the failure mode that this feature will detect that you have trouble to detect otherwise... Then I think it should be good to go.","accounts_in_message":[],"_revision_number":8},{"id":"f438da15a381c938c46ca9281b74a39131055837","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-08-20 13:55:28.000000000","message":"Patch Set 8: Code-Review+2\n\nNicely done.\n\nIt would be good to have a functional test for this feature as well so we can have test coverage against a \"real\" deployment, but that can be done in a follow up patch after this patch lands.\n\nsee oslo_messaging/tests/functional/test_functional.py for functional test examples.","accounts_in_message":[],"_revision_number":8},{"id":"723c56f2ed2f894a298f0a2db47bf690d22bc858","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-08-28 18:14:26.000000000","message":"Patch Set 8: Workflow+1\n\nArnaud has followed up on openstack-discuss as requested by Thierry.  It has been quiet for over a week so I\u0027m going to hit the big red button on this...","accounts_in_message":[],"_revision_number":8},{"id":"d8621156691d6d8cd71f50cbe269aa51b40b5c9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 18:14:41.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2be866dc173311e7957af29c4fe86b4153fa0bb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 19:55:29.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9db5f3186a9402ba9dc8d9cc4aeb419 : SUCCESS in 1h 24m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/558a33ca16484c2281294b08f47c700e : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db88aaea3e5a47d7b9f92880f21279bc : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d91e1bc1c9041ed8f6e8fa571aa5adc : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b44239742a647079d837dbc0342d2bc : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2325c2813aca4386bb75229c42d1edba : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e768a2d3043448718de63f12289984fc : SUCCESS in 4m 14s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/b646dcfc0e62433481eb985a36aba11d : SUCCESS in 5m 57s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/88fa3310b33048d28985e7c6f39e2842 : SUCCESS in 1h 32m 33s","accounts_in_message":[],"_revision_number":8},{"id":"4b93197a077cd46347926f6620d97a8d20137319","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 19:55:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"f8db0d881fbcd26e828730b580c2a8167695f313","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 19:59:26.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc65660fe8c948d59718fcdfe513ee2d : SUCCESS in 2m 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44c0a20200544c7ca7d9d66beb0bef50 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"82492442f3387a0e4f19623ccfda64f8b84d59c3","revisions":{"5780b6107c5186179cdbb48cb189bb875e478d45":{"kind":"REWORK","_number":1,"created":"2020-06-12 19:05:37.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/85/735385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/85/735385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/1"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-06-12 18:34:28.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-06-12 19:03:42.000000000","tz":120},"subject":"Add a ping endpoint to RPC dispatcher","message":"Add a ping endpoint to RPC dispatcher\n\nThe purpose of this patch is to add an endpoint directly in RPC\ndispatcher, so this endpoint will always be available, in a cross\nproject manner, without the need for projects to manage it by themself.\n\nThis endpoint stay disabled by default, so this change is harmless\nwithout a specific configuration option.\n\nTo enable this ping endpoint, an operator will just have to add a new\nparameter in the [DEFAULT] section, alongside with rpc_response_timeout\n[DEFAULT]\nrpc_ping_enabled\u003dtrue  # default is false\n\nThe purpose of this new endpoint is to help operators do a RPC call (a\nping) toward a specific RPC callback (e.g. a nova-compute, or a\nneutron-agent).\nThis is helping a lot for monitoring agents (for example, if agents are\ndeployed in a kubernetes pod).\n\nChange-Id: I51cf67e060f240e6eb82260e70a057fe599f9063\nSigned-off-by: Arnaud Morin \u003carnaud.morin@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5780b6107c5186179cdbb48cb189bb875e478d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5780b6107c5186179cdbb48cb189bb875e478d45"}]},"branch":"refs/heads/master"},"2eec739bddad20bb0599c393033aba9eada5c832":{"kind":"REWORK","_number":2,"created":"2020-06-15 08:06:24.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/85/735385/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/85/735385/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/2"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-06-12 18:34:28.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2020-06-15 08:05:30.000000000","tz":0},"subject":"Add a ping endpoint to RPC dispatcher","message":"Add a ping endpoint to RPC dispatcher\n\nThe purpose of this patch is to add an endpoint directly in RPC\ndispatcher, so this endpoint will always be available, in a cross\nproject manner, without the need for projects to manage it by themself.\n\nThis endpoint stay disabled by default, so this change is harmless\nwithout a specific configuration option.\n\nTo enable this ping endpoint, an operator will just have to add a new\nparameter in the [DEFAULT] section, alongside with rpc_response_timeout\n[DEFAULT]\nrpc_ping_enabled\u003dtrue  # default is false\n\nThe purpose of this new endpoint is to help operators do a RPC call (a\nping) toward a specific RPC callback (e.g. a nova-compute, or a\nneutron-agent).\nThis is helping a lot for monitoring agents (for example, if agents are\ndeployed in a kubernetes pod).\n\nChange-Id: I51cf67e060f240e6eb82260e70a057fe599f9063\nSigned-off-by: Arnaud Morin \u003carnaud.morin@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2eec739bddad20bb0599c393033aba9eada5c832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2eec739bddad20bb0599c393033aba9eada5c832"}]},"branch":"refs/heads/master"},"1905255fbbbbf77511c56543be9cc7ec2543f554":{"kind":"REWORK","_number":3,"created":"2020-07-02 15:31:36.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/85/735385/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/85/735385/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/3"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-06-12 18:34:28.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-07-02 15:31:24.000000000","tz":120},"subject":"Add a ping endpoint to RPC dispatcher","message":"Add a ping endpoint to RPC dispatcher\n\nThe purpose of this patch is to add an endpoint directly in RPC\ndispatcher, so this endpoint will always be available, in a cross\nproject manner, without the need for projects to manage it by themself.\n\nThis endpoint stay disabled by default, so this change is harmless\nwithout a specific configuration option.\n\nTo enable this ping endpoint, an operator will just have to add a new\nparameter in the [DEFAULT] section, alongside with rpc_response_timeout\n[DEFAULT]\nrpc_ping_enabled\u003dtrue  # default is false\n\nThe purpose of this new endpoint is to help operators do a RPC call (a\nping) toward a specific RPC callback (e.g. a nova-compute, or a\nneutron-agent).\nThis is helping a lot for monitoring agents (for example, if agents are\ndeployed in a kubernetes pod).\n\nChange-Id: I51cf67e060f240e6eb82260e70a057fe599f9063\nSigned-off-by: Arnaud Morin \u003carnaud.morin@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1905255fbbbbf77511c56543be9cc7ec2543f554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1905255fbbbbf77511c56543be9cc7ec2543f554"}]},"branch":"refs/heads/master"},"e4fb55bea58860893295b723c460cd398d53e431":{"kind":"REWORK","_number":4,"created":"2020-07-03 07:22:07.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/85/735385/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/85/735385/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/4"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-06-12 18:34:28.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-07-03 07:21:58.000000000","tz":120},"subject":"Add a ping endpoint to RPC dispatcher","message":"Add a ping endpoint to RPC dispatcher\n\nThe purpose of this patch is to add an endpoint directly in RPC\ndispatcher, so this endpoint will always be available, in a cross\nproject manner, without the need for projects to manage it by themself.\n\nThis endpoint stay disabled by default, so this change is harmless\nwithout a specific configuration option.\n\nTo enable this ping endpoint, an operator will just have to add a new\nparameter in the [DEFAULT] section, alongside with rpc_response_timeout\n[DEFAULT]\nrpc_ping_enabled\u003dtrue  # default is false\n\nThe purpose of this new endpoint is to help operators do a RPC call (a\nping) toward a specific RPC callback (e.g. a nova-compute, or a\nneutron-agent).\nThis is helping a lot for monitoring agents (for example, if agents are\ndeployed in a kubernetes pod).\n\nChange-Id: I51cf67e060f240e6eb82260e70a057fe599f9063\nSigned-off-by: Arnaud Morin \u003carnaud.morin@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e4fb55bea58860893295b723c460cd398d53e431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e4fb55bea58860893295b723c460cd398d53e431"}]},"branch":"refs/heads/master"},"2f82334a6f410836a58913a54146e2e5eb8ce2c8":{"kind":"REWORK","_number":5,"created":"2020-07-06 15:11:23.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/85/735385/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/85/735385/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/5"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-06-12 18:34:28.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-07-06 15:10:54.000000000","tz":120},"subject":"Add a ping endpoint to RPC dispatcher","message":"Add a ping endpoint to RPC dispatcher\n\nThe purpose of this patch is to add an endpoint directly in RPC\ndispatcher, so this endpoint will always be available, in a cross\nproject manner, without the need for projects to manage it by themself.\n\nThis endpoint stay disabled by default, so this change is harmless\nwithout a specific configuration option.\n\nTo enable this ping endpoint, an operator will just have to add a new\nparameter in the [DEFAULT] section, alongside with rpc_response_timeout\n[DEFAULT]\nrpc_ping_enabled\u003dtrue  # default is false\n\nThe purpose of this new endpoint is to help operators do a RPC call (a\nping) toward a specific RPC callback (e.g. a nova-compute, or a\nneutron-agent).\nThis is helping a lot for monitoring agents (for example, if agents are\ndeployed in a kubernetes pod).\n\nChange-Id: I51cf67e060f240e6eb82260e70a057fe599f9063\nSigned-off-by: Arnaud Morin \u003carnaud.morin@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2f82334a6f410836a58913a54146e2e5eb8ce2c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2f82334a6f410836a58913a54146e2e5eb8ce2c8"}]},"branch":"refs/heads/master"},"4f52bca1e0b2fc12e3c1b625d00bc5b18b206a92":{"kind":"REWORK","_number":6,"created":"2020-07-07 07:57:10.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/85/735385/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/85/735385/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/6"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-06-12 18:34:28.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-07-07 07:55:10.000000000","tz":120},"subject":"Add a ping endpoint to RPC dispatcher","message":"Add a ping endpoint to RPC dispatcher\n\nThe purpose of this patch is to add an endpoint directly in RPC\ndispatcher, so this endpoint will always be available, in a cross\nproject manner, without the need for projects to manage it by themself.\n\nThis endpoint stay disabled by default, so this change is harmless\nwithout a specific configuration option.\n\nTo enable this ping endpoint, an operator will just have to add a new\nparameter in the [DEFAULT] section, alongside with rpc_response_timeout\n[DEFAULT]\nrpc_ping_enabled\u003dtrue  # default is false\n\nThe purpose of this new endpoint is to help operators do a RPC call (a\nping) toward a specific RPC callback (e.g. a nova-compute, or a\nneutron-agent).\nThis is helping a lot for monitoring agents (for example, if agents are\ndeployed in a kubernetes pod).\n\nChange-Id: I51cf67e060f240e6eb82260e70a057fe599f9063\nSigned-off-by: Arnaud Morin \u003carnaud.morin@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4f52bca1e0b2fc12e3c1b625d00bc5b18b206a92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4f52bca1e0b2fc12e3c1b625d00bc5b18b206a92"}]},"branch":"refs/heads/master"},"8b3ee79cf4a9df2235545573c2da74610e598d0e":{"kind":"REWORK","_number":7,"created":"2020-08-18 13:05:53.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/85/735385/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/85/735385/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/7"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-06-12 18:34:28.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-08-18 13:05:39.000000000","tz":120},"subject":"Add a ping endpoint to RPC dispatcher","message":"Add a ping endpoint to RPC dispatcher\n\nThe purpose of this patch is to add an endpoint directly in RPC\ndispatcher, so this endpoint will always be available, in a cross\nproject manner, without the need for projects to manage it by themself.\n\nThis endpoint stay disabled by default, so this change is harmless\nwithout a specific configuration option.\n\nTo enable this ping endpoint, an operator will just have to add a new\nparameter in the [DEFAULT] section, alongside with rpc_response_timeout\n[DEFAULT]\nrpc_ping_enabled\u003dtrue  # default is false\n\nThe purpose of this new endpoint is to help operators do a RPC call (a\nping) toward a specific RPC callback (e.g. a nova-compute, or a\nneutron-agent).\nThis is helping a lot for monitoring agents (for example, if agents are\ndeployed in a kubernetes pod).\n\nChange-Id: I51cf67e060f240e6eb82260e70a057fe599f9063\nSigned-off-by: Arnaud Morin \u003carnaud.morin@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8b3ee79cf4a9df2235545573c2da74610e598d0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8b3ee79cf4a9df2235545573c2da74610e598d0e"}]},"branch":"refs/heads/master"},"82492442f3387a0e4f19623ccfda64f8b84d59c3":{"kind":"REWORK","_number":8,"created":"2020-08-18 13:10:02.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/85/735385/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/85/735385/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/85/735385/8"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-06-12 18:34:28.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@corp.ovh.com","date":"2020-08-18 13:09:29.000000000","tz":120},"subject":"Add a ping endpoint to RPC dispatcher","message":"Add a ping endpoint to RPC dispatcher\n\nThe purpose of this patch is to add an endpoint directly in RPC\ndispatcher, so this endpoint will always be available, in a cross\nproject manner, without the need for projects to manage it by themself.\n\nThis endpoint stay disabled by default, so this change is harmless\nwithout a specific configuration option.\n\nTo enable this ping endpoint, an operator will just have to add a new\nparameter in the [DEFAULT] section, alongside with rpc_response_timeout\n[DEFAULT]\nrpc_ping_enabled\u003dtrue  # default is false\n\nThe purpose of this new endpoint is to help operators do a RPC call (a\nping) toward a specific RPC callback (e.g. a nova-compute, or a\nneutron-agent).\nThis is helping a lot for monitoring agents (for example, if agents are\ndeployed in a kubernetes pod).\n\nThe endpoint is named oslo_rpc_server_ping.\n\nChange-Id: I51cf67e060f240e6eb82260e70a057fe599f9063\nSigned-off-by: Arnaud Morin \u003carnaud.morin@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82492442f3387a0e4f19623ccfda64f8b84d59c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82492442f3387a0e4f19623ccfda64f8b84d59c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
