)]}'
{"id":"openstack%2Foslo.messaging~736598","triplet_id":"openstack%2Foslo.messaging~master~I4768cc4485aa885d06092a56abe307487cd36710","project":"openstack/oslo.messaging","branch":"master","hashtags":[],"change_id":"I4768cc4485aa885d06092a56abe307487cd36710","subject":"Add new restriction for kombu","status":"ABANDONED","created":"2020-06-18 11:16:26.000000000","updated":"2020-07-01 14:06:24.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5ba98e7d37a81107d021362b06447ae5cc3fc642","_number":736598,"virtual_id_number":736598,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"date":"2020-07-01 13:52:46.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-22 18:03:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2020-06-23 17:30:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":-1,"date":"2020-07-01 13:38:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 11:18:21.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-22 15:48:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-22 18:03:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-23 17:30:47.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-07-01 13:38:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"ce54500df943b1c67eaac251d896630b9e5a6b9f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-18 11:16:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecd3ac887e1a7b20ac037b281d9f442c064b80d8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-18 11:18:21.000000000","message":"Patch Set 1:\n\nThis entry needs to be added to global-requirements file in requirements repo so that the requirements check work.\n\nIs there a bug upstream?\n\nTo merge this: Create requirements change, and make this change Depends-on that one.","accounts_in_message":[],"_revision_number":1},{"id":"82d6325b43b16fe67caa3777352804dbb561ed6f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-18 11:44:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d77f0709d420dc50ef4f3fb670f21e547d3a8497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 13:38:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0bc0c04380004a1d93958e4e200a77f1 : RETRY_LIMIT in 2m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67b3e8f95bb140899d80306c31e45d58 : SUCCESS in 1h 15m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57ff83c2fdec491696d10b6d3a60469b : SUCCESS in 7m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef57804472a3422593df17a01175bf6a : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7677cc4d508747c7b7d05cf45945f28c : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ef79083c6df4725b76ffe0965475ecb : SUCCESS in 9m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d77a2aef8519428883effad2a5fa0342 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a6704405bbe4791a99b78fbb80de49d : SUCCESS in 8m 21s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/f5b3c3e76c5a45bdbdcc1190b5c92440 : SUCCESS in 9m 28s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/3fc2510c97234a349e6ffc2c71213fd6 : SUCCESS in 8m 27s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/87855f22c6144fc8a10efb9e211a47e1 : SUCCESS in 11m 08s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/8d84497df46e4f71a4c9c70d852796ba : FAILURE in 8m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0fa90885a173431095492864ba764e76 : SUCCESS in 1h 36m 14s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a760330492bc421ca158f5f79e15d09b : FAILURE in 1h 36m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/78673342d8dd461db6fbfbb0fa148f96 : FAILURE in 1h 48m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a90907255dfdf31a9c272e093e2af99239d7a788","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-18 13:46:05.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9cd0bf0004dee5dce38c717288ddfcd43de854b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 15:32:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c9eee437110544fba58512e6921f40d1 : FAILURE in 3m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa9bef1798244c0bae98fdd3804c38b9 : SUCCESS in 1h 29m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/def20d721a694a3fa1390dc56410bba6 : RETRY_LIMIT in 4m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18d9d6b8e69644efb4bf68ae1025794a : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2d0b6a3a0de4b02b55f62e46580cbf0 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2871d91293e44d7a849d560ac6e3d421 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2149c291f11b442fa3ce7895335ab682 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/957a85c4b0664887bea68528e9b89e9d : SUCCESS in 5m 16s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/2e9af4d3c8dc45f9adf63fd7de92c010 : SUCCESS in 6m 46s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/694a23253b7648928ffcd1e05ae64ca5 : RETRY_LIMIT in 7m 00s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/f0b411b7556f48458fbfcb137cbcb3a8 : SUCCESS in 10m 06s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/5620d91e667249d89c085ccb443048db : SUCCESS in 6m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3eb3ed88e0b14c718cb933719ba320b7 : SUCCESS in 1h 35m 12s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c0a49bf6e6a34b7f8d41cf6cc90023dc : SUCCESS in 1h 17m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/dc2e4ae3c03d4e12b2044ed536e424bd : SUCCESS in 1h 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b2974928ad29a427c59b9dc3d3d80e5e0e79afa5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-19 15:57:31.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"2f23c9a0d0cf75e6381743799b9fee1f857f888f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-19 16:12:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5d5cbf3c1742b92cd2c958727e9ad0923378c2b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 17:39:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c87a3441918c4e72bf299c63371ac8a6 : FAILURE in 5m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6de522f78f294651a187062a7fe605e5 : SUCCESS in 1h 14m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62336500767a4a93956d605e8916198d : SUCCESS in 9m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c2e4e38927f4199972af2ffdc8cc22f : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eeb9037b5bf40a8be4107c382e587ea : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2740f917a4b448d89623b78178c1435 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adde7491a4874e4ebdbb0c909d25d77d : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8543916283f42b295d490de797e6246 : SUCCESS in 4m 54s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/fc57bbacac934559b15e39427188975a : SUCCESS in 8m 19s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/5ef4ca588d0646d39ca5d1c675d91700 : SUCCESS in 11m 10s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/8b31f4b980954f0db857a3aed2240814 : SUCCESS in 8m 28s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/a413b3e1d8ed4ea3bbbf5dcd8243309e : SUCCESS in 6m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/55675a68960c4b01955a35bd7e635739 : SUCCESS in 1h 35m 07s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6c79934e72364d3cae914b8ab1213754 : SUCCESS in 1h 36m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/98a70f018cec4c509391ac6ac4c911e7 : SUCCESS in 1h 18m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9c165b6de3992dd52a3dc1dc5656e694ca5cd1af","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-06-22 15:46:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b248b5c861c402d7a22f84d845edadd65375fc13","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-22 15:47:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0984bc6b36768a852172398df6782e3f78d2c578","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-22 15:48:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8b840b8fb2f8481b95f7de9291cc9a2252fc0870","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-22 16:22:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ae6c05becc96f01a0756633a1fba5be4a6266165","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-22 16:28:52.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nThanks for the comments, Ben, Hervé!\n\nCreated the patch to requirements[1] (also added as Depends-On here)\n\n[1] https://review.opendev.org/737307","accounts_in_message":[],"_revision_number":5},{"id":"b2ceb9e437f433ddbd20b5ea30007c6d116492e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 18:03:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/aa17ab058dd745d58e43c0b11a0bdbd2 : SUCCESS in 3m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e10a30031dea4e62b5e8f3e7d052959a : SUCCESS in 1h 31m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f38a5e5f0e8a4b79bdd5b4a31cb5b2f2 : SUCCESS in 12m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d392e20707e7455289506b4285757529 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9ae287a5ee0461090fc2cdcb3fbc545 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/505835a8654d4a0d9bebd788b3e6ea31 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e4d938d9d03472fa8c3cad469dfb63c : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f6930cd490144c79908fc54ab4f5217 : SUCCESS in 5m 49s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/c0fde10e942f4637b5e5c2e1bd958ea1 : SUCCESS in 10m 32s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/1382df9da462467db6aca0634e25130a : FAILURE in 7m 19s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/077f6cddc94745eb99598634f3c74fda : SUCCESS in 7m 54s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/bcb46b13ead04d7c89f3bd50eb7670ff : SUCCESS in 7m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c7cdc75ebb2c4b05b7d57d890544f992 : FAILURE in 1h 32m 25s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3b67593924ba490681a07a4220123d8c : SUCCESS in 1h 30m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/42ead621d55f4a7f8c1a587bf2fbdf50 : SUCCESS in 1h 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ea8c87f9209c0809c195a9bb6dbc053eb91cf7ff","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-06-23 17:30:47.000000000","message":"Patch Set 5: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"2ba0f5b52d6221a86b33e5d5bdc5767bd19b4b1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 18:56:49.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"25373a1456a7b80efc9ff57f37ec6f15be604b87","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-23 18:59:41.000000000","message":"Patch Set 5:\n\nrecheck - depends-on patch updated","accounts_in_message":[],"_revision_number":5},{"id":"a67e0177f94d573afaacd8ad46c61b5ab1aa6710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 21:06:34.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0fd1b0781fce4a00b733bf10628f06b7 : SUCCESS in 4m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1a6ebb0e1894b84925bb779a77a203f : SUCCESS in 1h 09m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd39879fdc4f421985f27be9d22e91fa : SUCCESS in 6m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3915d01dfd46412a8517c8535e077300 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06f9ff0662514bd895139e87359b9911 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b43caa76707b4c438742b6b2b1a98654 : FAILURE in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e320954f442e44d4a2c18c9fca26180b : SUCCESS in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49cb6ec8afbb4079850386ecd2f0a999 : SUCCESS in 4m 06s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/3a74ff5b152447a99dd2728f40b138a5 : SUCCESS in 5m 03s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/3ae60866ba0c410d8874f3372ca0be00 : SUCCESS in 10m 33s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/8356646f26d549feb444f69981d15cb8 : SUCCESS in 5m 40s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/797f030c8fc44f6996c70c8092327852 : FAILURE in 8m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1ee52db02a0047fab7543dc8532ec9db : SUCCESS in 1h 09m 49s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ad1e5ffa5ecf4c1fa4c83fa198d82bfa : SUCCESS in 1h 53m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c5c16bf802f94effbac86d1d08d4ef05 : SUCCESS in 1h 19m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8b35a265ae4cfe58f01be23eda695d5f6e660d72","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-01 13:38:41.000000000","message":"Patch Set 5: Code-Review-1\n\nI don\u0027t think the solution here is to blacklist things. Rather, we should resolve the change in behavior. It\u0027s clearly intended as a feature of kombu - not a bug that they\u0027re going to fix.","accounts_in_message":[],"_revision_number":5},{"id":"614e78307b4dcbe1c1529ebf9890cb7ff4212fdf","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-01 13:52:46.000000000","message":"Patch Set 5:\n\n\u003e I don\u0027t think the solution here is to blacklist things. Rather, we\n \u003e should resolve the change in behavior. It\u0027s clearly intended as a\n \u003e feature of kombu - not a bug that they\u0027re going to fix.\n\nYes, I got the same feedback from Matthew in the patch in requirements repository (see depends-on).\nActually I investigated the stable job failures when I came across this issue, and found that kombu brought in a change that is the cause of the failures. But I don\u0027t really have knowledge about oslo.messaging and have some other things as well that I was/am working on, so if anyone has any idea how to fix this in a correct way, then it is welcomed. :)","accounts_in_message":[],"_revision_number":5},{"id":"5ba98e7d37a81107d021362b06447ae5cc3fc642","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-01 14:06:24.000000000","message":"Abandoned\n\nin favor of https://review.opendev.org/#/c/736248/5","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c60853682b908cf7b8c6d365497f5a5a39456968","revisions":{"30b5a12f9c31c8083f3d695149e1a0b0443526c1":{"kind":"REWORK","_number":1,"created":"2020-06-18 11:16:26.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/98/736598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/98/736598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/1"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-18 11:11:52.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-18 11:11:52.000000000","tz":120},"subject":"Add new restriction for kombu","message":"Add new restriction for kombu\n\nUnit test gets timed out with new kombu (4.6.10).\n\nChange-Id: I4768cc4485aa885d06092a56abe307487cd36710\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/30b5a12f9c31c8083f3d695149e1a0b0443526c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/30b5a12f9c31c8083f3d695149e1a0b0443526c1"}]},"branch":"refs/heads/master"},"8a5c73649f20079bb633c0d69a34c37643823934":{"kind":"REWORK","_number":2,"created":"2020-06-18 11:44:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/98/736598/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/98/736598/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/2"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-18 11:11:52.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-18 11:43:25.000000000","tz":120},"subject":"Add new restriction for kombu","message":"Add new restriction for kombu\n\nUnit test gets timed out with kombu newer than 4.6.8.\n\nChange-Id: I4768cc4485aa885d06092a56abe307487cd36710\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8a5c73649f20079bb633c0d69a34c37643823934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8a5c73649f20079bb633c0d69a34c37643823934"}]},"branch":"refs/heads/master"},"9d6154db495dcc0cca7f62c778b021f988aafa42":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-06-18 13:46:05.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/98/736598/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/98/736598/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/3"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-18 11:11:52.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-18 13:45:33.000000000","tz":120},"subject":"Add new restriction for kombu","message":"Add new restriction for kombu\n\nUnit test gets timed out with kombu newer than 4.6.8.\n\nDepends-On: https://review.opendev.org/736402\nChange-Id: I4768cc4485aa885d06092a56abe307487cd36710\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9d6154db495dcc0cca7f62c778b021f988aafa42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9d6154db495dcc0cca7f62c778b021f988aafa42"}]},"branch":"refs/heads/master"},"82a9e98e31e821e2179245ed655cf729a91c6a0c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-06-19 15:57:31.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/98/736598/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/98/736598/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/4"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-18 11:11:52.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-06-19 15:57:27.000000000","tz":0},"subject":"Add new restriction for kombu","message":"Add new restriction for kombu\n\nUnit test gets timed out with kombu newer than 4.6.8.\n\nDepends-On: https://review.opendev.org/736402\nDepends-On: https://review.opendev.org/736248\nChange-Id: I4768cc4485aa885d06092a56abe307487cd36710\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82a9e98e31e821e2179245ed655cf729a91c6a0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82a9e98e31e821e2179245ed655cf729a91c6a0c"}]},"branch":"refs/heads/master"},"c60853682b908cf7b8c6d365497f5a5a39456968":{"kind":"REWORK","_number":5,"created":"2020-06-22 16:22:31.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/98/736598/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/98/736598/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/98/736598/5"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-18 11:11:52.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-22 16:21:14.000000000","tz":120},"subject":"Add new restriction for kombu","message":"Add new restriction for kombu\n\nUnit test gets timed out with kombu newer than 4.6.8.\nThis patch is identified as the cause of the timeouts:\n\nhttps://github.com/celery/kombu/pull/1193/commits/398aa5b8cd1fe1fc189ab35a2cb0bececf51bf37\n\nDepends-On: https://review.opendev.org/737307\nDepends-On: https://review.opendev.org/736248\nChange-Id: I4768cc4485aa885d06092a56abe307487cd36710\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c60853682b908cf7b8c6d365497f5a5a39456968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c60853682b908cf7b8c6d365497f5a5a39456968"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
