)]}'
{"id":"openstack%2Foslo.messaging~736622","triplet_id":"openstack%2Foslo.messaging~master~I4c459d8c947dac213a1866c0d37e8f3d547aa82e","project":"openstack/oslo.messaging","branch":"master","topic":"connection-error","hashtags":[],"change_id":"I4c459d8c947dac213a1866c0d37e8f3d547aa82e","subject":"Drop a python 2 exception management","status":"MERGED","created":"2020-06-18 11:27:24.000000000","updated":"2020-07-17 16:29:52.000000000","submitted":"2020-07-17 16:28:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736622-1595003294915-c1f85032","meta_rev_id":"1556988785c178655f45811be45926dce9ba18ed","_number":736622,"virtual_id_number":736622,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"date":"2020-07-01 09:35:15.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2020-07-06 12:00:25.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-17 16:28:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-17 14:49:40.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-10 16:20:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-17 14:49:40.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-01 09:35:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-07-10 16:20:59.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-07-17 14:49:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-17 16:28:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4402d6d3f19f5ab75dea94e5c67e8dada4751f3e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-18 11:27:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9eb6f75a1b54ddb97468e770c0b0758ad9f673f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 13:09:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82f8d4fb2320409ba252d33dd33ef105 : SUCCESS in 1h 24m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70c82d9d6f3a4b2a9f4652320388877c : RETRY_LIMIT in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a2c20b56d5d441a84ac872bb20ecab3 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcc3dd54479a49c8aa80a87afd5972a3 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6a68586a14b44f1954f0095546f580e : TIMED_OUT in 41m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47b455083e0044f7b13aa0d7441ad061 : TIMED_OUT in 43m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbc64d8bf3e9441799932fb8c4368831 : SUCCESS in 5m 16s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/6a2704f6a12640a7a59e05b595d50d03 : RETRY_LIMIT in 6m 19s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/86cf4b71626e4973a00b142538955011 : SUCCESS in 10m 53s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/7e93a1436f6042b391ca4e3ecdde8e23 : SUCCESS in 8m 08s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/2342b6679b4240a995daed373e533776 : SUCCESS in 10m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/27022a6c65774fbe9158070457735854 : SUCCESS in 1h 29m 54s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7f7eba986c2a4402ac72800a50c9eb07 : SUCCESS in 1h 29m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/42da5876c85c47f0b9db08c41e21fa52 : SUCCESS in 1h 12m 21s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/886e1419ee474e73b484a5ac6fe26a4f : SUCCESS in 59m 19s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/2922fa8e92f94210babd3cf96d0f29a9 : FAILURE in 17m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d1afbd6a212dc1082e2236fc0e6a50d632f1df9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-18 14:45:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c0b1f5c0d540d3cc29f565f41877825cfb4c0261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 16:21:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e538192cf1814dae943dee30f2c89bf7 : SUCCESS in 54m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21a2b4b9a05f4ccc90e4f09b9e3cff7b : TIMED_OUT in 51m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92599118edf14abaa363678b950a937d : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b9ea88aa9984d90b073d66fbff0853d : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/400dea1e92724b33b84fcfc0f984db23 : TIMED_OUT in 41m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aeeeb042702b4fd4b84725e7b3c3c65f : TIMED_OUT in 41m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14bba33dab314c40938b4b5ec763148d : SUCCESS in 8m 26s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/66de8ded9e2a456c99e6fe143f9569d7 : SUCCESS in 9m 12s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/629691defee6405095636339427d3569 : FAILURE in 8m 55s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/c98d57dcef5b43c2b0418a8e420043d9 : SUCCESS in 8m 58s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/eb0fd699fdac420daf1fac67eb649b42 : RETRY_LIMIT in 5m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/eed84a7c88bb48b9bd83d60fee7bcf2f : SUCCESS in 1h 22m 45s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/0250fd769bfe4d2fa301b5c8559b6da0 : FAILURE in 1h 28m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/578a84f765ce4a1882bee1bc2599cb93 : FAILURE in 1h 33m 37s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/7db2b4258c1c4ce39d24975376a47cad : SUCCESS in 1h 13m 14s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/2c7d53c6ef1147e99bd4777ea286ec8c : FAILURE in 14m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10026d37382c76a23c6b79d2d287f8d380765ade","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-19 10:50:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6926659f1b2ed289ed77800ac6edfd5632ececdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 12:26:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef46608cf7d4459dbbb58a963dd4155c : SUCCESS in 1h 21m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e3ac9ab49f04b6ea1b9da5549f112e2 : RETRY_LIMIT in 4m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5990dc16472646299d42a465ffd965a5 : RETRY_LIMIT in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872609f57f1b4e12bc7248f29f807760 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9aa7550a583b418ea5d3d396356d354a : RETRY_LIMIT in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a79b296033df4b158047b1224a2ea399 : RETRY_LIMIT in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50baa7a319b64b25b466928b5ebc89b8 : SUCCESS in 4m 33s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/7d9cbf12bc234330ab237e55f78087e5 : RETRY_LIMIT in 3m 13s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/79c58d565b9849d99054ce61fe14b2ba : RETRY_LIMIT in 4m 44s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/c218c9859035480b9e30dcef79053147 : RETRY_LIMIT in 3m 08s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/fc5d016a690e4ab7b104c2ed9970038c : RETRY_LIMIT in 4m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4393995974fa490da9f0cc3bb99b49fa : SUCCESS in 1h 24m 59s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7a2dcea147714b3ca904cddba32c3a0f : SUCCESS in 1h 27m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9fbd7197f8cf4606b2360574d208fea7 : SUCCESS in 1h 34m 21s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/bda397d76f9843babd8f86df72beae5e : SUCCESS in 1h 20m 35s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/ee36625019294b238116dfe28b03ac24 : FAILURE in 17m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"967985991e75d0984b14935d7f85ad84ec153571","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-19 14:21:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"72e2b2807eb6ab25e41a16c1a9d8e98f40a526ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 16:20:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d4b7de81684458e97ebf50db933e078 : SUCCESS in 1h 35m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5010970f269c425d8173cf47b0e5f750 : RETRY_LIMIT in 4m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac684eb41e8f4402b8b570a6d2a4f864 : RETRY_LIMIT in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d696bbf63e2c44b28f24e5807a22a390 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0131557755a4b8992a47279e9381969 : RETRY_LIMIT in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dc12fd2481a4d3387a14ea055ad65e1 : RETRY_LIMIT in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce50d59eceae4e0fbcf0bcf7341d21d3 : SUCCESS in 7m 15s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/2645afad67db4eec9a96841078cf5c29 : RETRY_LIMIT in 2m 56s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/b5a0f427ad5c493c87bbf9a1d366d1c7 : RETRY_LIMIT in 4m 03s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/b5157d4d1c844f218edf8f9d5189dc0b : RETRY_LIMIT in 5m 23s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/0f15332ee0cb45529d4f76df26e08074 : RETRY_LIMIT in 3m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6e7dabec270848f691e5acc5909224b3 : FAILURE in 1h 42m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7d1e9b53ef5c4fe9837402596f552fa7 : SUCCESS in 1h 55m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/375cbd2d6056461b94f5b267c9cead32 : SUCCESS in 1h 03m 15s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/98b2b913ddc8490dbcbeb51703bded2f : SUCCESS in 59m 48s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/fe3c14f3639043609b9f9d1ff8041af0 : FAILURE in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b2bb4fe7391ed40c4274d65e6432f68d551b153","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-22 08:38:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"516ee8d47244499694e12741b392c491b8d33b69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 10:33:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9354777f61734983bcda93ff7306010d : SUCCESS in 1h 18m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e907e4d59e814882abc0de2fb7c2d7b8 : RETRY_LIMIT in 4m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/296b07c599be4c88a4ac44568dcbe516 : RETRY_LIMIT in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d88378f46c9f42d8b9a12f64f1fb167e : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0a1c42a7bda479f810d117c8b7c618a : RETRY_LIMIT in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45ee2e0016ec40ce822b9c4516ab255b : RETRY_LIMIT in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c2347629a0f4817a3ffdf4bce0f7796 : SUCCESS in 4m 55s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/03126870da494b2d92e39dcbe9fe82c7 : RETRY_LIMIT in 3m 15s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/526f1202e94e4f4aa52e36dd90a0da22 : RETRY_LIMIT in 3m 28s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/62786e9d39184e41a0ce6f45e73d96e2 : RETRY_LIMIT in 3m 40s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/85ab290e7d4646aea3a1d81ee57a8008 : RETRY_LIMIT in 3m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6f3b0c49d5814976acad97d6ac99abf2 : FAILURE in 1h 19m 07s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3433cd3f953e4b5b80f8ca7be82d7b3b : SUCCESS in 1h 47m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c9f307b37fba426c99b0f856a368cbbd : SUCCESS in 1h 33m 02s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/ad3c46a49c1b4051a95481586eef17a2 : SUCCESS in 1h 03m 54s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/04753baa7d69470cbc421bcf9024267f : FAILURE in 14m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96587c4c304d0d22830b70ec4838d1beca882705","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-22 12:32:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a9a4e1032a9585f0b09ce889c9418d9609c7abaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 14:32:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93c5ddcc239f4de6bfc85f8685c967a6 : SUCCESS in 1h 56m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd56bb2f427247738a7ffed4c098ed23 : RETRY_LIMIT in 2m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5836d42671114bfc9887d156e3433e0c : RETRY_LIMIT in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20d8538e11484f458b783c5935c24f8f : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3460a856c7444feb13e4e629eeb1bd2 : RETRY_LIMIT in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b1a0ed9efbf4fc5bb0fbf14314b4506 : RETRY_LIMIT in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb715e71bbb4cf19aa36f25af0995b1 : SUCCESS in 4m 04s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/1548d56eee5c4715b44523ad9ebd96ec : RETRY_LIMIT in 4m 15s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/7c025dc7dd5942cd8c57eb9fe367ace0 : RETRY_LIMIT in 4m 16s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/a5634baf1a644f42945bc750438271f4 : RETRY_LIMIT in 4m 13s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/d795a1b620174fb2b359d4f9d9458a6c : RETRY_LIMIT in 3m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/452e6ec4181547798a7dce8091acfe48 : SUCCESS in 1h 27m 29s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3445e22cea1d407ab1a343cf1c444e5e : SUCCESS in 1h 31m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c837e6e6b6fe4889b903f2b1be4df79c : SUCCESS in 1h 28m 17s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/c3f94f379368436ba046b382a0ba38af : SUCCESS in 59m 45s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/5a9a467f95344a7787499ccccf851bb7 : FAILURE in 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99d080f0e9d89ef4838680b4eea6223e0895eda5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-30 15:52:51.000000000","message":"Patch Set 1:\n\nrecheck hopefully the CI is happier now","accounts_in_message":[],"_revision_number":1},{"id":"675c9f0327f3c9046c3d07418b1294f6f69c3840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 17:57:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04fbe684d530412ea4547461bdb66d3a : SUCCESS in 1h 16m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccf0b98a89cd49f9a1535958ea6d8058 : RETRY_LIMIT in 5m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0db5f9a8dec4f7faf16fa3055ed6443 : RETRY_LIMIT in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7413bda280824b96ba8bd8c7e3217e45 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d16ff31be6a4503912f5200e5624c36 : RETRY_LIMIT in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbe52ed4ea7a4b17a93d0f86e4b236ca : RETRY_LIMIT in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14277e6df73449caa20ac97f5603685d : SUCCESS in 6m 25s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/ad875f68b2e14f3eb785902597bbb3c6 : RETRY_LIMIT in 4m 59s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/fab68a4abea9437893ff5d92cab2e1f5 : RETRY_LIMIT in 5m 50s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/eecc3ff704e24d288786d09a3a34cf28 : RETRY_LIMIT in 5m 11s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/f4c698520d2d494291d0996d93bef522 : RETRY_LIMIT in 5m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8078739dc13840b18b75763be50addb3 : SUCCESS in 1h 56m 27s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2ac615bab5904c7a95e640390c5b4b97 : SUCCESS in 1h 01m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9a16cd1b22074cedbd2b564dcce756d8 : SUCCESS in 1h 00m 14s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/1d6bddb6cda54c068975232d18d11643 : SUCCESS in 1h 12m 40s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/49781242546244889b41651223f95a73 : FAILURE in 19m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"659b6f29cd2ca821a621d2fed3c2809d0a6e3bd9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-01 09:35:15.000000000","message":"Patch Set 1:\n\nThis seems to be failing because tox is missing from the host, but I don\u0027t know why that is :-\\","accounts_in_message":[],"_revision_number":1},{"id":"438ff59cf8d74d40f5354d509af0d15a3e480d0b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-06 12:00:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3f191359f86437bdc0c314c3c5e211f1b37c0c32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 13:55:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/433246cf949648e095b0b5cf9b062b7e : SUCCESS in 52m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3707e3a24fa4e3ba505a2474abf5694 : SUCCESS in 6m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2259d3a82134b3d8386e35b16299cec : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66e93d0a75fc4b259cca5665b7455de0 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38deaf72dda6433ab7467cd0aa91a3bb : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c27f4f2b0f84552bc54f0aba35fecbf : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/731e6dd15ba24f9abb3ea60dae731a81 : SUCCESS in 5m 24s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/79b65e844eae4c318f07fb365135045a : SUCCESS in 6m 08s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/f12ab8e7e4204e3e8d55a83f20868022 : SUCCESS in 8m 02s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/e707e93d775d4bc4af2b0dc9e25b1082 : SUCCESS in 6m 26s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/6046e1ef01e04c8bbc7e6879f2cb1761 : SUCCESS in 7m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/30e3efa4e70b4d7b9f8cd7413d3723a5 : SUCCESS in 1h 35m 25s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1d9e311400414b778db2c4cadab05b4d : SUCCESS in 1h 48m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3ec88623093048b2b921c2f11686a703 : SUCCESS in 1h 47m 05s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/b092dd967720463192b60600f84a016c : SUCCESS in 1h 15m 18s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/9ca3675e022048d98c7c6a64bca5c2d2 : FAILURE in 20m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af31bcbba187aa8e1ae3a2fc523dd2eb1f8595a4","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-07-10 16:20:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ec32accbaea09630745ed7ad99f2ad70c0638e80","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-17 14:49:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cda10f8631496e2e9963b5f57c600671bacec167","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 14:49:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46c7b3c0fa78c9c14194f1a6d028e974950af508","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 16:28:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3a7d4bac658d8cb1dff42cd7a00e343bf5209af4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 16:28:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/547bc1f7df3d405982a16688116b4b79 : SUCCESS in 1h 14m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dba262336df4a578f995bc011bb0892 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e50a7fd9e714ccdb43258fbfd5d3ced : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f19460f6899412bad6f89cc689fd18c : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b15bb71da844e0aa5730ffd05370ac0 : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed9ed76f192e4672b4df3adc8c631b34 : SUCCESS in 4m 10s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/a798a6d97c164f419948f22500e31851 : SUCCESS in 5m 13s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/fbf51497432f4f29b5a780d1d1d978a6 : SUCCESS in 1h 31m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ab2f26bb50c605c7c60219a434a73369f096fa0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 16:29:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd0df53b4b3e4d9db4c10a5bc4e4a72f : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"661bdd7f41d1e253e76c9d516019eb6aafe90f15","revisions":{"661bdd7f41d1e253e76c9d516019eb6aafe90f15":{"kind":"REWORK","_number":1,"created":"2020-06-18 11:27:24.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/22/736622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/22/736622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/736622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/736622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/736622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/22/736622/1"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-18 11:04:17.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-18 11:25:24.000000000","tz":120},"subject":"Drop a python 2 exception management","message":"Drop a python 2 exception management\n\nPreviously (when python 2 was supported) we introduced some specific\nexception management to detect connections issues when a rabbitmq\ncluster node disappearing.\n\nThe original issue was that a ConnectionRefusedError is thrown and not\nmanaged by oslo.messaging to detect heartbeat issue and call ensure_connection\nfor switching the connection destination (cluster node).\n\n`ConnectionRefusedError` is only a python 3 exception [2].\n\nNow that we only support python 3 we don\u0027t need to continue to\nwrap/emulate this kind of exceptions (ConnectionRefusedError) [2] and\nso we can drop the python 2 compatibility code to only support python 3 code.\n\n[1] https://github.com/openstack/oslo.messaging/commit/9d8b1430e5c081b081c0e3c0b5f12f744dc7809d\n[2] https://docs.python.org/3/library/exceptions.html#ConnectionRefusedError\n\nChange-Id: I4c459d8c947dac213a1866c0d37e8f3d547aa82e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/661bdd7f41d1e253e76c9d516019eb6aafe90f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/661bdd7f41d1e253e76c9d516019eb6aafe90f15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
