)]}'
{"id":"openstack%2Foslo.messaging~738538","triplet_id":"openstack%2Foslo.messaging~master~I5c468d10230f05438df780e045f1e92662edf070","project":"openstack/oslo.messaging","branch":"master","topic":"bug/1883038","hashtags":[],"change_id":"I5c468d10230f05438df780e045f1e92662edf070","subject":"Catch ConnectionForced Exception","status":"MERGED","created":"2020-06-30 01:01:19.000000000","updated":"2020-07-29 00:04:00.000000000","submitted":"2020-07-28 16:25:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738538-1595953533519-594063c4","meta_rev_id":"5fdcbf48617775bfdcf10e602ad1aaba3dd679d8","_number":738538,"virtual_id_number":738538,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 16:25:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-18 23:43:28.000000000","_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2020-07-28 14:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2020-07-20 07:54:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-07-28 11:18:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":1,"date":"2020-07-28 14:32:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2020-07-28 14:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2020-07-18 23:41:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 07:54:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-28 11:18:34.000000000","updated_by":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"reviewer":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"state":"REVIEWER"},{"updated":"2020-07-28 14:32:28.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-07-28 16:25:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fad6975e11ed89582e12a7f96ccf0678f5a9dc32","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-06-30 01:01:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b40f957b5c52351c93f73a07c50242c201e7e4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 02:33:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca81907464ae4c658f0eb83e76f92bdb : SUCCESS in 1h 19m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bb5f89987db478296bae04714d1043b : RETRY_LIMIT in 2m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b041c12321af4c63afb391a853d870d7 : RETRY_LIMIT in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07e805eb5a0246a0984b805dd9b4f913 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc3f780ed15742ebbf425ec55436886a : RETRY_LIMIT in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd6c3f3359294c46bf23b01e9f4edd08 : RETRY_LIMIT in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9f4d76d83e9404fa3cc97dd0a64b5e7 : SUCCESS in 4m 25s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/5911c5f4e86347e186ee6c9c7e26517c : RETRY_LIMIT in 2m 59s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/bcd75c15951d4938a3fb2bf383ae5f0e : RETRY_LIMIT in 4m 09s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/50a3ddf943df42408fde9b0f7f2ae306 : RETRY_LIMIT in 2m 55s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/0920f36561184e4c97ae7fd81ea61157 : RETRY_LIMIT in 3m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/feb5ea13f4ef4ac7bc910acd03b2ec5d : SUCCESS in 1h 23m 11s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ae66b07c2e414c1aa587e6928c256658 : SUCCESS in 1h 26m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8cf5b17352754a99976481503835971b : SUCCESS in 1h 19m 27s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/adfcabac8e114aad93421ea5a9ba3245 : SUCCESS in 56m 44s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/5878e64b448d44c89212d13cae38cf10 : FAILURE in 16m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da3055b82d7d53b8863c9a67a5691859c4eeb5a3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-30 12:08:01.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review-1\n\n(2 comments)\n\nHello,\n\nThanks for this patch.\nHave you seen this scenario in a rabbitmq clusters?\n\n\n-1 is for the nits in the commit message.","accounts_in_message":[],"_revision_number":1},{"id":"30af886f75dc1d34cd19f021b072747e2a66ec80","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-07-18 23:41:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8f6988310919869953aee1e1e6136e1688e82e5b","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-07-18 23:43:28.000000000","message":"Patch Set 2:\n\n\u003e (2 comments)\n \u003e \n \u003e Hello,\n \u003e \n \u003e Thanks for this patch.\n \u003e Have you seen this scenario in a rabbitmq clusters?\n \u003e \n \u003e \n \u003e -1 is for the nits in the commit message.\n\nHi, thank you for reply. I do saw this error message couple of times in production. If this publish connection is not used, it will continue printing error messages.","accounts_in_message":[],"_revision_number":2},{"id":"74748c03c61e567ba1092201d7b6624573cbc895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 01:22:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25021cb7674b4dc6b1ae3272e8c784a3 : SUCCESS in 1h 19m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab729235b50945dc857da024d48d8dc1 : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f37f368925247f682f6946cea0bb75c : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21351e1c35954209beda785908c0645a : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f10fb660c58443abdaf1d1f74a16aec : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e7ab6713b3240d09a9c09b99ad00fa1 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4095ba3a138c4ff4accf6bd1824e16b8 : SUCCESS in 3m 40s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/6c4f148fa6754ad5b7983d5c0ee31b08 : SUCCESS in 4m 41s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/e790adf0286c4deda9df63398076d11d : SUCCESS in 6m 02s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/92aab58ac2c24f1fa8d2d5a1ec07248f : SUCCESS in 8m 09s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/ac0dd0d4c33a48619554ada4f6af926a : SUCCESS in 5m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d0cca2dd5f494f13a6a7c1eff3d9c862 : SUCCESS in 1h 39m 35s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7e5811c1e5d841029c0da434b79ab801 : SUCCESS in 1h 34m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/445a177bc3c54faebb699dc857340ed4 : SUCCESS in 1h 23m 13s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/789c87d5705a4ac292088d7a82b1fac9 : FAILURE in 1h 08m 36s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/5fe593e1972e4da68cfdec361349c49c : FAILURE in 15m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0fe7bb8d191a713ae9da876541d50c44b5b4767c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-20 07:54:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"06ce6ba133542c15298cb8d381f01825bb62f012","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-28 11:18:34.000000000","message":"Patch Set 2: Code-Review+1\n\npyamqp raises this when \u0027Too many heartbeats missed\u0027 - https://github.com/celery/py-amqp/blob/64a2313dd28a969ad923045b58199bfc90003761/amqp/connection.py#L752","accounts_in_message":[],"_revision_number":2},{"id":"8843247eede642ebee54d0447f2d988511ba976d","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-07-28 14:32:28.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"61d74379924b15a7fdf3fe6110c43e59a498a27f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 14:32:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e451a08aeac1c590caabf86d0e2e9cb08647b734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:25:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5dc06768be5f485da57787cc412a3c18 : SUCCESS in 1h 14m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f3903ad78624b0f96c12bc04260e6dc : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/200123cf29794867ab9fa15039527d46 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/636e7dbec0df43dfa71080cb30ea0343 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54744ae544324a4fbe8163496338bfee : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/864d63f5e52c4aa9a113cb0dfe926247 : SUCCESS in 4m 16s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/0daeacd78e064fee8b158cd6c8f93e6d : SUCCESS in 5m 33s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c8f818d02dff489e960ee541e8185690 : SUCCESS in 1h 42m 51s","accounts_in_message":[],"_revision_number":2},{"id":"5af0f10e78a570a371568bb2151144ae75702e2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:25:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"37d243d1bdbb068d9e2f89ecf8545f03b2b18797","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:31:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/028ad712d0c840ec9a5f64d43f7523e1 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2},{"id":"a4bd67f462e0cfc451fd47283777c7bbcd42a282","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-07-29 00:04:00.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 360deb89a7e95d332267564b486219e389a793f7","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"26e8bb4cc9097fddf3c2e34599b131e3f5d73867","revisions":{"540e128c2f83ee3c67df34fb665bfe1086fd3f4f":{"kind":"REWORK","_number":1,"created":"2020-06-30 01:01:19.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/38/738538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/738538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/738538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/738538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/738538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/738538/1"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-06-30 00:56:26.000000000","tz":480},"committer":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-06-30 00:56:26.000000000","tz":480},"subject":"Catach ConnecionForced Exception","message":"Catach ConnecionForced Exception\n\nOccassionally I saw large number of connectionforced exceptions.\nsuch exceptions cannot be healed automatically and requires a reboot.\ncatch it and ensure connection established.\n\nChange-Id: I5c468d10230f05438df780e045f1e92662edf070\nCloses-Bug: #1883038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/540e128c2f83ee3c67df34fb665bfe1086fd3f4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/540e128c2f83ee3c67df34fb665bfe1086fd3f4f"}]},"branch":"refs/heads/master"},"26e8bb4cc9097fddf3c2e34599b131e3f5d73867":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-18 23:41:22.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/38/738538/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/738538/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/738538/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/738538/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/738538/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/738538/2"}}},"commit":{"parents":[{"commit":"c2074e47604199e9802520dbc4dfb54da778e090","subject":"Fix pygments style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2074e47604199e9802520dbc4dfb54da778e090"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-06-30 00:56:26.000000000","tz":480},"committer":{"name":"ushen","email":"yshxxsjt715@gmail.com","date":"2020-07-18 23:41:17.000000000","tz":480},"subject":"Catch ConnectionForced Exception","message":"Catch ConnectionForced Exception\n\nOccasionally I saw large number of connectionforced exceptions.\nsuch exceptions cannot be healed automatically and requires a reboot.\ncatch it and ensure connection established.\n\nChange-Id: I5c468d10230f05438df780e045f1e92662edf070\nCloses-Bug: #1883038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/26e8bb4cc9097fddf3c2e34599b131e3f5d73867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/26e8bb4cc9097fddf3c2e34599b131e3f5d73867"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
