)]}'
{"id":"openstack%2Foslo.messaging~738850","triplet_id":"openstack%2Foslo.messaging~master~Ibbcf21a57ab1e3f90c21901296e5c088b645127c","project":"openstack/oslo.messaging","branch":"master","topic":"test_setup.sh","hashtags":[],"change_id":"Ibbcf21a57ab1e3f90c21901296e5c088b645127c","subject":"tests: Resolves issues with kombu \u003e 4.6.8","status":"MERGED","created":"2020-07-01 14:03:31.000000000","updated":"2020-07-24 13:02:32.000000000","submitted":"2020-07-03 13:58:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738850-1593784687000-b4231495","meta_rev_id":"73c2e3b36db622de27ef0ba6a6f14dbef5f1116b","_number":738850,"virtual_id_number":738850,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-03 13:58:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2020-07-02 20:08:56.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-07-02 11:22:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-07-03 12:27:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2020-07-03 12:27:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2020-07-03 12:27:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-01 14:04:40.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-07-02 11:22:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-07-02 20:08:56.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-07-03 12:27:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-03 13:58:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad44cde5f1de793ccc8e54938fed461851333d9f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-01 14:03:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbf4df833bd0bf095b177ce7d7f65255a9751e30","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-01 14:05:47.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0c42e14efd82f027dfb55b97c9140b38c015ba37","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-01 14:15:10.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM, thanks Stephen","accounts_in_message":[],"_revision_number":2},{"id":"50d5e14bbe783ea007a4bbb53b22df04b6d05506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 15:37:40.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c4b0654c5444e958785d7c04dee9e4d : SUCCESS in 1h 27m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82e37724dc0b46c88f19b918521e7390 : SUCCESS in 5m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebb50527eed242b780923c3dc5cb150a : FAILURE in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ab575fcbb43472980c924595c0e5689 : FAILURE in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c1329727fd240a39577e5284eb57c52 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c93572b9f1647778b7f8e67bf644ea1 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2474ef385e9c4e1d9534300dc9471cf8 : SUCCESS in 4m 55s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/e177940761cc49689bf3889355fa212f : SUCCESS in 6m 31s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/9d56b1bf11cd4c83acac15f8dbb311c0 : SUCCESS in 8m 02s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/268cabe9fec241c4b8be40bc8825d8ec : SUCCESS in 8m 09s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/338e5af6b34b46a4813dcf6a544398f9 : SUCCESS in 6m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2f3b5f23bb4d4274b20cf80e7058b461 : SUCCESS in 1h 27m 11s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a64730648b364badb78e4ca852f1bac8 : SUCCESS in 1h 27m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/80bfa60111a04e08861c0fafd653ed82 : SUCCESS in 1h 21m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/e4ec63af3d864253adac0c33a65a543b : SUCCESS in 54m 07s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/83840ed9aec748a09dfbaba57767b9ea : FAILURE in 13m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6c3782bf4895d960db7d0e8e478e90abfb758b36","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-02 09:37:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5879ccb6b6a24ab43691039856ff3d013ed83a16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 11:13:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3bd20c32e50e48bbbda4a29fc871e59d : SUCCESS in 1h 20m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cac59e5623f4e5eafbeede4b73038c1 : SUCCESS in 9m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5be55db798da40e7ab04af48387a4bee : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29864b0c66fa44aeadd51f46890c15c8 : SUCCESS in 9m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ea52a966c0147ba833e57f5a94ea797 : SUCCESS in 10m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a10c6420b5e8431a9164f87a4e372f46 : SUCCESS in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb47e14c0ba04e35892c715e5f5f7290 : SUCCESS in 6m 22s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/e991bda5a9e345848e04c79a1a93dc55 : SUCCESS in 9m 55s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/94c04f95fb054efe8838c17986baa222 : SUCCESS in 9m 23s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/e1c3faf5a6bf45dcb4aa86657f20572b : SUCCESS in 6m 29s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/399136c81c3d4408b100a92cd99401d7 : SUCCESS in 7m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4bc6b6215d4c4f02b6ac545b96b886d7 : SUCCESS in 1h 21m 53s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/85bd73bcdb834b6ba3cbeb47f91966ba : FAILURE in 1h 16m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/99a0f2a7afb74bed97c25e55a5fb6c42 : SUCCESS in 1h 29m 04s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/469ae860b2664b0dadd7af2dcd1d6222 : SUCCESS in 57m 35s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/98e4d03adc694345940538ab7f7aec8e : FAILURE in 21m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"608f49145bda7a4899ff8314beda4fc5c7851876","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-02 11:22:25.000000000","message":"Patch Set 3: Backport-Candidate+1\n\nThanks Stephen, the fix makes sense! LGTM.","accounts_in_message":[],"_revision_number":3},{"id":"6f3b757d2a0b3fb72415479666dae54d1ac87a63","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-02 11:22:47.000000000","message":"Patch Set 3: -Backport-Candidate Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"75287ad6e67a6288a3ad9d45b74e0e29f71bf057","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-07-02 20:08:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"96895e1717cf779200bbe2a6c773c8ad1ee324a3","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-07-02 20:08:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"af33627323ec56022caeec99010a8025d3f7ffe1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-03 12:27:29.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2 Workflow+1\n\nI think we should backport these changes, thoughts?","accounts_in_message":[],"_revision_number":3},{"id":"639600dc4701785caea64702b57baec62e3098b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 12:27:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e89b48ada10ac33af795a6d742e4c5130cadfc4d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-03 13:14:34.000000000","message":"Patch Set 3:\n\n\u003e I think we should backport these changes, thoughts?\n\nHi Hervé. No need to backport as kombu\u0027s upper constraint is 4.6.8 in ussuri, so we won\u0027t see the issue there.","accounts_in_message":[],"_revision_number":3},{"id":"c2bfc7e247ac30c7334117595dc21e934a6ef6fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 13:58:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fe1a147d45441a6b8d4b494148f7915 : SUCCESS in 1h 28m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/adcafb44bd3a467f8fb12fa69b4f15b8 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c8c3b0ff7d64e97bde1a79567bc47d7 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c7afbb0010547ad8a3f353abfe6e4f7 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc434c8d07b74302ab1bb196d43dc889 : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a597a11cdb0045ce8eddc351cb5be2b2 : SUCCESS in 5m 23s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/d64ac7cc0d99437a836e6511c288c23c : SUCCESS in 6m 24s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ef35070d2fe74dfe825cef970dfc603e : SUCCESS in 1h 16m 32s","accounts_in_message":[],"_revision_number":3},{"id":"f1792f3070c2709bde82896ed17e2ed95c6eefad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 13:58:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"80da600618ec7f049aecd064814da6e11d956c12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 14:00:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c11d4a4d04ee4717ba296b4f15e04f03 : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":3},{"id":"78e9449a059b54d74148e8766a7588d8c1067d4c","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-23 08:09:47.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 6727be7890f64446f1658d79fac196d30b6f1a18","accounts_in_message":[],"_revision_number":3},{"id":"25671091914a2301132aac125b546c9cc73aa777","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 11:43:36.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 06257c378778c93e11a607798babc1b570ece623","accounts_in_message":[],"_revision_number":3},{"id":"e4b7c523d07de13f74f146e645bd83b7f73fe416","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 11:44:02.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d73ea6a41463a5b21fbf5d4913c0dec26baabd44","accounts_in_message":[],"_revision_number":3},{"id":"308eca8b03689016cb81961d7277a9d2f49fa1c4","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 11:45:22.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit b4ffe7ab22c01d2abac92746dfe4ff380d370902","accounts_in_message":[],"_revision_number":3},{"id":"4fd27dc79e98530ad4a2c0dd70cc1019edf99073","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 11:46:08.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 8d6f3edf94173ff096246e074d496cb4f2cbc352","accounts_in_message":[],"_revision_number":3},{"id":"e9c2567d6d127ffae7dfca897e8cde2c74873dfe","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 11:49:23.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/pike as commit 41c8b31c6ad2eb458d79284f934c622b6148399f","accounts_in_message":[],"_revision_number":3},{"id":"f2a423da8be21baad16addbd054ce0eda6f143f0","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 11:59:23.000000000","message":"Patch Set 3:\n\n\u003e No need to backport as kombu\u0027s upper constraint is 4.6.8 in ussuri, so we won\u0027t see the issue there.\n\nWhile that is the upper constraint, it is not always easy to follow due to what a distribution already ships with. I ran into this with ussuri packages where the existing kombu package is newer, but not with train which was built here only for an earlier distribution.","accounts_in_message":[],"_revision_number":3},{"id":"5b144f2c196cfefbefdf2c797b53c66afe972cbb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 13:02:01.000000000","message":"Patch Set 3:\n\nShould be reverted. 4.6.8 was not supported in this branch.","accounts_in_message":[],"_revision_number":3},{"id":"d8444f2db2c62ef782c3e28afb871fac7731b58f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 13:02:32.000000000","message":"Patch Set 3:\n\nDisregard last comment. This appears to be the only correct patch.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"afb035d97185fa9c622bdb9b4c31c71f01160370","revisions":{"059d54098bbc25cb07dea5f3032884b970ef7a27":{"kind":"REWORK","_number":1,"created":"2020-07-01 14:03:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/738850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/50/738850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/738850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/738850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/738850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/50/738850/1"}}},"commit":{"parents":[{"commit":"7e406c312a6514e7ae377edb52b9e02b5bf37a7d","subject":"Simplify tools/test-setup.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7e406c312a6514e7ae377edb52b9e02b5bf37a7d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-01 14:01:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-01 14:01:00.000000000","tz":60},"subject":"tests: Resolves issues with kombu \u003e 4.6.8","message":"tests: Resolves issues with kombu \u003e 4.6.8\n\nThe \u0027kombu.connection.Connection.ensure_connection\u0027 method has changed\nfrom calling \u0027retry_over_time\u0027 on \u0027self.connect\u0027 to calling it on\n\u0027self._connection_factory\u0027, meaning our mocks are outdated. Address this\nchange.\n\nChange-Id: Ibbcf21a57ab1e3f90c21901296e5c088b645127c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCloses-Bug: #1885923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/059d54098bbc25cb07dea5f3032884b970ef7a27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/059d54098bbc25cb07dea5f3032884b970ef7a27"}]},"branch":"refs/heads/master"},"b1245029e17f317b8735cf92bd65e8f04ca68d70":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-01 14:05:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/738850/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/50/738850/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/738850/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/738850/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/738850/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/50/738850/2"}}},"commit":{"parents":[{"commit":"7e406c312a6514e7ae377edb52b9e02b5bf37a7d","subject":"Simplify tools/test-setup.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7e406c312a6514e7ae377edb52b9e02b5bf37a7d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-01 14:01:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-01 14:05:44.000000000","tz":0},"subject":"tests: Resolves issues with kombu \u003e 4.6.8","message":"tests: Resolves issues with kombu \u003e 4.6.8\n\nThe \u0027kombu.connection.Connection.ensure_connection\u0027 method has changed\nfrom calling \u0027retry_over_time\u0027 on \u0027self.connect\u0027 to calling it on\n\u0027self._connection_factory\u0027 [1], meaning our mocks are outdated. Address\nthis change.\n\n[1] https://github.com/celery/kombu/pull/1193/commits/398aa5b8cd1fe1fc\n\nChange-Id: Ibbcf21a57ab1e3f90c21901296e5c088b645127c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCloses-Bug: #1885923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b1245029e17f317b8735cf92bd65e8f04ca68d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b1245029e17f317b8735cf92bd65e8f04ca68d70"}]},"branch":"refs/heads/master"},"afb035d97185fa9c622bdb9b4c31c71f01160370":{"kind":"REWORK","_number":3,"created":"2020-07-02 09:37:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/738850/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/50/738850/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/738850/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/738850/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/738850/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/50/738850/3"}}},"commit":{"parents":[{"commit":"7e406c312a6514e7ae377edb52b9e02b5bf37a7d","subject":"Simplify tools/test-setup.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7e406c312a6514e7ae377edb52b9e02b5bf37a7d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-01 14:01:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-02 09:37:09.000000000","tz":60},"subject":"tests: Resolves issues with kombu \u003e 4.6.8","message":"tests: Resolves issues with kombu \u003e 4.6.8\n\nThe \u0027kombu.connection.Connection.ensure_connection\u0027 method has changed\nfrom calling \u0027retry_over_time\u0027 on \u0027self.connect\u0027 to calling it on\n\u0027self._connection_factory\u0027 [1], meaning our mocks are outdated. Address\nthis change.\n\n[1] https://github.com/celery/kombu/pull/1193/commits/398aa5b8cd1fe1fc\n\nChange-Id: Ibbcf21a57ab1e3f90c21901296e5c088b645127c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCloses-Bug: #1885923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/afb035d97185fa9c622bdb9b4c31c71f01160370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/afb035d97185fa9c622bdb9b4c31c71f01160370"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
