)]}'
{"id":"openstack%2Foslo.messaging~738976","triplet_id":"openstack%2Foslo.messaging~stable%2Fussuri~Ic3a9e2c873619670edfbf71022d593f3cb5f70f2","project":"openstack/oslo.messaging","branch":"stable/ussuri","hashtags":[],"change_id":"Ic3a9e2c873619670edfbf71022d593f3cb5f70f2","subject":"Simplify tools/test-setup.sh","status":"MERGED","created":"2020-07-02 07:50:55.000000000","updated":"2020-07-02 14:35:42.000000000","submitted":"2020-07-02 14:30:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738976-1593700209736-2d1bbc0e","meta_rev_id":"94610be57fd5e09e71c8066c4aa8f7b09e39f666","_number":738976,"virtual_id_number":738976,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-07-02 11:39:25.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-02 14:30:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-02 12:07:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-07-02 12:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-07-02 12:39:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-02 07:50:55.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2020-07-02 11:37:02.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-02 07:50:55.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-07-02 11:37:02.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-07-02 12:07:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-02 12:39:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-02 14:30:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a58ddccf9f1754416b29d2e8bc483d6ec891aca6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-02 07:50:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cf099a298ae0ac2782034a577ca7d87c62aa32b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 09:52:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37602c2dc4934a3dbf918ca37d540b90 : SUCCESS in 1h 37m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12df6a52b50f4092ac642fe5c111e8da : SUCCESS in 9m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dd9149aadbd42f5a8c63bdb953ac28e : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e79a6fc6c2a47bda05c3cb9b5c66bf3 : SUCCESS in 8m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ead4bac988d04e31b6e7a47a6f26b0b4 : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/742b75776e0344fca8a97184ff347354 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac17b208e49f4820a1125a2dbc6c893f : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ed4d5f67d1044418f83acabd9dc6d56 : SUCCESS in 4m 53s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/4ababa6dca554c37a0967df2ee18ff18 : SUCCESS in 9m 40s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/24149308a504409898eb7a2f0e29fde2 : SUCCESS in 8m 02s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/325dbc1c1e4543df80cfebe02252a7a1 : SUCCESS in 8m 15s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/e2494d87f9944949a1abacd577a861dc : SUCCESS in 6m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e6e0af445e20498094d5bd4dc6e539fb : FAILURE in 1h 44m 55s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/44fd4c5e4322465b8c251bcde083a6c4 : SUCCESS in 1h 56m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/997d1734dee44e4c831dbafcf3875564 : SUCCESS in 1h 36m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/ddf8ce154de5436a85d6606082aecb2c : SUCCESS in 1h 13m 03s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/a38fc27dcb404703a3761e46924f2b81 : FAILURE in 31m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57d241e319dcbe3a0e51394719eb269bc5de4249","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-02 09:55:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6e1fdccfe798521df3e5f1f0bafe124f21d56298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 11:31:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da4bc2256b744aa9ad25d8f520c290dc : SUCCESS in 1h 12m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc0cdf6cdc304bbb92c8e532ed90a43d : SUCCESS in 7m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63cea6b958714e4a9b6a10944cef4d09 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6a8fa34cb2043b7a437d30882327541 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a53c749a80b64cf297fddf3714a70b53 : SUCCESS in 8m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4eefe08448a4e0a9055352d8ce3e5d0 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd48abf0cba4050983043b9a3465b7e : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a79bdf160b304cc3942e15de28ee9112 : SUCCESS in 6m 47s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/572fe88ae1fb44fd9d6f59f66867e3f3 : SUCCESS in 8m 27s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/7b331c714e5840be899068d34d657652 : FAILURE in 7m 25s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/72988c31258a45e3866f8eaf3ff00c7d : SUCCESS in 7m 08s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/635b59c429594d48af19085675b92e5d : FAILURE in 7m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/cd6f445388c541bfbe34289ccc8e9fa9 : SUCCESS in 1h 26m 22s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/544403a109ba4d259439280fc5c56722 : SUCCESS in 1h 26m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d58c7d3d9d9c4008988bdd5cbcdd98f6 : SUCCESS in 1h 28m 26s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/b446895eb4ce42778a10434268e104f4 : SUCCESS in 1h 13m 57s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/adedb90ba4504639bf21cc7ff81ae784 : FAILURE in 29m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34595d9a805e64054353c150bd241fe8d1190bcd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-02 11:39:25.000000000","message":"Patch Set 1:\n\nThis patch fixes gate. (See periodic failure: https://zuul.opendev.org/t/openstack/build/6c73f71a8d284134959d4799fd54ea41)","accounts_in_message":[],"_revision_number":1},{"id":"473340ce34ffb99fedf0403b07f1e0b30ef3d42c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-02 12:07:46.000000000","message":"Patch Set 1: Code-Review+2\n\nThis is needed to get stable check jobs passing.","accounts_in_message":[],"_revision_number":1},{"id":"8347dde5aaf3b6bc203aeb356e9579caebf03478","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-02 12:39:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4e9b4bfc76bae8203a8fa9e06de5976c77613fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 12:39:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21b7a6b3cb59f0561b3e06b5364699e7fbd83cc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 14:30:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e780dff1a6f411b961b80cfe4e7f5f1a9db1dd41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 14:30:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6c73d12ec4147e5b4538c42518485e3 : SUCCESS in 1h 44m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b16cee8dacfe425ea3f392746cb63e83 : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f754e7a93134946964a2c77147ff005 : SUCCESS in 9m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3b5f28a1bb144da8a1bb0345124d4fc : SUCCESS in 10m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a96716cc99834ecc886741907915d89d : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be760f5a8ad9451d86b5b738d6d90bcf : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad57e3ff8ccb406385ff89477fff0192 : SUCCESS in 7m 35s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/a963432920d340f5bd5eb0547f903110 : SUCCESS in 9m 10s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7a3a75956a9e4aa69bcfea209277b755 : SUCCESS in 59m 26s","accounts_in_message":[],"_revision_number":1},{"id":"94610be57fd5e09e71c8066c4aa8f7b09e39f666","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 14:35:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d1eb1360d44ad2bd5c05d8c5fcf74f : SUCCESS in 1m 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccd25f269ba749eaaac474f60e022f0e : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dfcc4b0010d39d1f16be413bb1c3a4808718cd19","revisions":{"dfcc4b0010d39d1f16be413bb1c3a4808718cd19":{"kind":"REWORK","_number":1,"created":"2020-07-02 07:50:55.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/76/738976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/76/738976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/738976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/738976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/738976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/76/738976/1"}}},"commit":{"parents":[{"commit":"39f770f55b3331b7a99b3ea42a34982a20635cf2","subject":"Print warning message when connection running out","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/39f770f55b3331b7a99b3ea42a34982a20635cf2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-17 14:53:03.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-07-02 07:47:37.000000000","tz":120},"subject":"Simplify tools/test-setup.sh","message":"Simplify tools/test-setup.sh\n\ntest-setup.sh is called after bindep is run in CI, and tox is not setup.\n\nSimplify the script so that it only does what\u0027s needed - install\nqdrouterd for Ubuntu from qpid/testing.\n\nInstall the other packages using bindep, add test to amqp1 profile.\n\nAdd amqp1 to bindep_profiles where needed for scenario tests, previously\ntest-setup.sh added it everywhere.\n\nNOTE(elod.illes): skip annotation removed from test_impl_rabbit.py, as\nkombu is constrained in ussuri with the version that does not need any\ntest to skip.\n\nChange-Id: Ic3a9e2c873619670edfbf71022d593f3cb5f70f2\nRelated-Bug: #1885923\n(cherry picked from commit 7e406c312a6514e7ae377edb52b9e02b5bf37a7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dfcc4b0010d39d1f16be413bb1c3a4808718cd19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dfcc4b0010d39d1f16be413bb1c3a4808718cd19"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
