)]}'
{"id":"openstack%2Foslo.messaging~739265","triplet_id":"openstack%2Foslo.messaging~stable%2Fstein~Ic3a9e2c873619670edfbf71022d593f3cb5f70f2","project":"openstack/oslo.messaging","branch":"stable/stein","hashtags":[],"change_id":"Ic3a9e2c873619670edfbf71022d593f3cb5f70f2","subject":"Simplify tools/test-setup.sh","status":"MERGED","created":"2020-07-03 16:49:44.000000000","updated":"2020-07-06 13:36:41.000000000","submitted":"2020-07-06 13:31:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739265-1594042270378-1a0f072d","meta_rev_id":"3e95ca555bcab535ee6a5092109d2f3b63e881b5","_number":739265,"virtual_id_number":739265,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-07-06 07:27:29.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-06 13:31:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-07-06 09:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-06 11:48:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-06 11:48:15.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2020-07-06 09:08:59.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-03 16:49:44.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-07-06 09:08:59.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-07-06 09:50:19.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-06 11:48:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-06 13:31:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41e99ac94acec85afd19340b91ecd4a7b368f4c3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-03 16:49:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"503defdfe9c3f0c422ceebd1399e32b8000a0757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 18:26:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9e14290e58ed43a9807c6e55a474912d : SUCCESS in 1h 27m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b25930a995347ed8820f74b17012204 : FAILURE in 20m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/463e45058d074ee4bc52c1e0c2a5c1ed : RETRY_LIMIT in 2m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7be2b920cfc414b908910e8ef957b71 : RETRY_LIMIT in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceda84ac257d44e380977aaa2cfff906 : RETRY_LIMIT in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae291e1c177641bab9972b8c86f9ef50 : RETRY_LIMIT in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7748cc9a464d494a9a9f9cecf2c5c001 : RETRY_LIMIT in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbdb0ae182e54f47b0b53e80bdd48802 : RETRY_LIMIT in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bc07baad0f44b20aa74d4a94dcdd2ac : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e325ce052f844bcb2647df3bfe601da : SUCCESS in 4m 36s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/c2204279b12b40ddbb9d0413faaed2a2 : SUCCESS in 5m 37s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/473fac015e0a4eb09b69201f84c8cb10 : RETRY_LIMIT in 3m 46s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/6d69971cd13d4b85a20076ea8f287941 : FAILURE in 6m 49s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/4aedee37b5df4d0b991a3e4d18589aea : SUCCESS in 5m 35s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/c5869d558a0f44dd92cc4f88cd413138 : TIMED_OUT in 42m 17s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/52331ea3e36f42cabc66e1a8029be4fe : FAILURE in 4m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5cabde25c4d246c0ba8f98a280788680 : SUCCESS in 1h 26m 35s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/21512edd09d543b6ac8c21545b2442d0 : SUCCESS in 1h 27m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/9bd23ccb82ca447fb616aba3270d296d : FAILURE in 10m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ad7644c0ff5c4d78a0b0255fddb1e591 : SUCCESS in 1h 31m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/c81d9e5dea19417186cd4b642738588b : FAILURE in 15m 56s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d5729e07e30143cd95891e62821479c5 : SUCCESS in 55m 59s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/90a8f378276d403290d47f4e2872a975 : FAILURE in 1h 03m 24s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/5d85b40076c84819b135bc063c422bc9 : FAILURE in 43m 30s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/dc2de9cdcaa14537a21869ec9060434c : FAILURE in 44m 02s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/0bede14c996243089b5cb85ab1cd4a1d : FAILURE in 44m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b126223883d0e676b79a9dc0ce8c7e669112ee5d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-03 18:27:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b0d7af6f44d568e83725bb8717db621ae53c45ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 20:03:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/82e6c2ad23034282990267ff00d7243b : SUCCESS in 1h 20m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1216301589f4382be01e7095efe7c96 : FAILURE in 17m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcdd3120db464359b3d7d57f1bf602d5 : SUCCESS in 5m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d48702ef035445f68793af95075110d6 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d32e8bfab5c4149b0dd3597e1d97999 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21df738571934989a3703bdbf94dbe5f : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc58443da77945a1afd5ba5933c82989 : SUCCESS in 8m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fb9ee6bf50140b996bb16a23ad9a820 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3e88e9dce764ea3b73e13250f975d99 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28216cb8c1f740e983111d007a836f49 : SUCCESS in 3m 57s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/6337a382ef864a71afbacdec04bb9a02 : SUCCESS in 4m 34s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/075f783b0e594d2bb7f231c5a09b3dc3 : SUCCESS in 4m 15s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/c30821881ae042728d2d6cf2a6c73edb : FAILURE in 6m 31s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/43c587e5100d421fbc786a811b168dd0 : SUCCESS in 6m 54s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/76ed1bf6b6c1479483002643ee9571cc : SUCCESS in 5m 41s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/e726ec4dff8d42c4adf51f237277a10f : FAILURE in 4m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/033c46a555174173bc85d5ab41a23b97 : SUCCESS in 1h 30m 44s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e967db78c5594b1eabf8441212c73292 : SUCCESS in 1h 31m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/15f1ca5a25b44b4ea1304e3308ed0a82 : FAILURE in 16m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c011630dffca4f50a5082d0b6176789c : FAILURE in 1h 30m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/cf498447f7c64a69869e55a8bc1214dd : FAILURE in 9m 50s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/6f4404e928af4a5ba3838f0472233ff0 : SUCCESS in 58m 42s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/edb8155fe76b428b8b9d64ee43ecf059 : FAILURE in 1h 03m 29s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/a4d187a5e7594cf39a374d20dbfd7cae : FAILURE in 45m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/5b5687fa80a549f693ad29ef7aa47285 : FAILURE in 43m 57s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/58406a18f97e4a30819322ab186d71fb : FAILURE in 1h 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"04ddba7ad1b457128e26169c43ad1f48c3f5ad99","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-06 07:27:29.000000000","message":"Patch Set 2:\n\nrecheck - \"/usr/local/bin/keystone-manage: No such file or directory\" - solved with setuptools 49.1.0 release","accounts_in_message":[],"_revision_number":2},{"id":"d2183e9e30768062729fc106ea8bb58d6c23be8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 09:06:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/51857333c3a940eca73a128dcf88a3ca : SUCCESS in 1h 17m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b23760ab2cd947258f53d5223c213b0b : SUCCESS in 1h 21m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/017c7c4364554696a9e5dcd17bae6a08 : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e41600ed64464cd3a949197ea05c7c51 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf2642ecf1b427eb4b248b955855ce5 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76c078cac21b4df4800e241b103ca57c : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af87983054e64c07825d351fbe638a4a : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df6e2dc4dfcc4daabf908674891da7cf : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be08e1fc14f4e7593c4c13ed0857cc6 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87a5c3571f3e448a9b7e7bbbbef71762 : SUCCESS in 3m 50s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/40fb51c0557f4edc92bb76cebdff6db8 : SUCCESS in 5m 12s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/09bbd1ff14ea45e082e10d2aa5e32b97 : SUCCESS in 4m 42s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/e4ad56b6c5df45929e3c1d038634cbdf : FAILURE in 4m 54s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/276b643616114c70b3a1d5c9d9a4f445 : FAILURE in 6m 51s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/226b1211df0c41baabfe2a7c61cbb0bf : SUCCESS in 5m 27s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/086b49bee83a4b959f6c23b80b8691bb : FAILURE in 5m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/722c5cb50089493798e5bf66f9b76ff8 : SUCCESS in 1h 32m 39s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1ba85952f7b64b52b8fe664a517f8f4e : SUCCESS in 1h 29m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/bd56da8d3589430ebc42699166cba82f : FAILURE in 18m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/6719c3614b0644d8a75c07e490b626d9 : SUCCESS in 1h 33m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/3f882c0b656042e286195220bc4778af : FAILURE in 20m 35s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/025f8b8c243e4c9abf860f5a918d9ed8 : SUCCESS in 1h 02m 01s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/e929c960c72d42d1970772c068b3da51 : FAILURE in 1h 05m 04s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/c6f33bdf29ff45218f4155cc95ee3236 : FAILURE in 49m 45s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/ca275fb0e5584d2f8a8fd5eaaa3f458c : FAILURE in 45m 34s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/8ba0cac4750443999b63c14b7c91b8c8 : FAILURE in 50m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0002b074be3250b5b48ee07feae4ad9a9b5b8f8b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-06 09:50:19.000000000","message":"Patch Set 2: Code-Review+2\n\nClean backport LGTM","accounts_in_message":[],"_revision_number":2},{"id":"ededc97356fd16d4b5b5927972711baec832845d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-06 11:48:15.000000000","message":"Patch Set 2: Backport-Candidate+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"228fea64769fdaa3c08e712d7544cfc6e934e15f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 11:48:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"32d49ed8e0c88b82864b0a21e5e0c6c673050b63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 13:31:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/58be90f22b9e441e88736c35506baae9 : SUCCESS in 1h 27m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/367454e661984100bd0cd67f481bb968 : SUCCESS in 1h 31m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80e8a7ee8bac4c91b62ae599e55b65b7 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c79998c54449cfba4c0bc7eb23725e : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b91a5ec74fb42a5803a91c899f649b2 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57cf497e4aa44a4cae14e5c8d51d506f : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef99ce91c7d746adb60baab7e50166c4 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b92d2cac2ec44c1be81d4dbad716b46 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bb706bbdde9412bb1bca3be9a193980 : SUCCESS in 5m 17s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/3767aab661dd4063a3f7ab0bb200e087 : SUCCESS in 6m 26s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e721a77a2a0e4600baf5871f9a6ed6bf : SUCCESS in 1h 34m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2059ef16ff14fb5de7db350dd1329d6bc8d7576f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 13:31:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3e95ca555bcab535ee6a5092109d2f3b63e881b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 13:36:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be249e91860849de87bd26cef59dd10d : SUCCESS in 2m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/056498b0669841bc84574d835b1c9809 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c5d17905de32a29352a7ad31fd1c7dcde8887017","revisions":{"e3875180626239509e8979f73d8647eeb3ed0bf7":{"kind":"REWORK","_number":1,"created":"2020-07-03 16:49:44.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/65/739265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/65/739265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/739265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/739265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/739265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/65/739265/1"}}},"commit":{"parents":[{"commit":"76fce54b9e693203f268774f876fa478ab44412c","subject":"fix typos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/76fce54b9e693203f268774f876fa478ab44412c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-17 14:53:03.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-07-03 16:46:46.000000000","tz":120},"subject":"WIP: Simplify tools/test-setup.sh","message":"WIP: Simplify tools/test-setup.sh\n\ntest-setup.sh is called after bindep is run in CI, and tox is not setup.\n\nSimplify the script so that it only does what\u0027s needed - install\nqdrouterd for Ubuntu from qpid/testing.\n\nInstall the other packages using bindep, add test to amqp1 profile.\n\nAdd amqp1 to bindep_profiles where needed for scenario tests, previously\ntest-setup.sh added it everywhere.\n\nConflicts:\n    .zuul.yaml\n    bindep.txt\nModified:\n    tools/test-setup.sh\n\nNOTE(elod.illes): conflicts are due to the following patches are\nmissing from Stein:\n* .zuul.yaml: I1874f96f78cb403e6f3a56a49cb83df40d531f8d\n  (dropping py35 testing)\n* bindep.txt: I9fb56450346fca84bc1573fa5e9b6fe81c14e33a\n  (messaging intermediaries update; the lack of this patch also\n  requires the modification of test-setup.sh)\n\nChange-Id: Ic3a9e2c873619670edfbf71022d593f3cb5f70f2\nRelated-Bug: #1885923\n(cherry picked from commit 7e406c312a6514e7ae377edb52b9e02b5bf37a7d)\n(cherry picked from commit dfcc4b0010d39d1f16be413bb1c3a4808718cd19)\n(cherry picked from commit 88cf5aebef790ff40ab8ca7219d537ccfc54d12e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e3875180626239509e8979f73d8647eeb3ed0bf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e3875180626239509e8979f73d8647eeb3ed0bf7"}]},"branch":"refs/heads/stable/stein"},"c5d17905de32a29352a7ad31fd1c7dcde8887017":{"kind":"REWORK","_number":2,"created":"2020-07-03 18:27:06.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/65/739265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/65/739265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/739265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/739265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/739265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/65/739265/2"}}},"commit":{"parents":[{"commit":"76fce54b9e693203f268774f876fa478ab44412c","subject":"fix typos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/76fce54b9e693203f268774f876fa478ab44412c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-17 14:53:03.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-07-03 18:25:21.000000000","tz":120},"subject":"Simplify tools/test-setup.sh","message":"Simplify tools/test-setup.sh\n\ntest-setup.sh is called after bindep is run in CI, and tox is not setup.\n\nSimplify the script so that it only does what\u0027s needed - install\nqdrouterd for Ubuntu from qpid/testing.\n\nInstall the other packages using bindep, add test to amqp1 profile.\n\nAdd amqp1 to bindep_profiles where needed for scenario tests, previously\ntest-setup.sh added it everywhere.\n\nConflicts:\n    .zuul.yaml\n    bindep.txt\nModified:\n    tools/test-setup.sh\n\nNOTE(elod.illes): conflicts are due to the following patches are\nmissing from Stein:\n* .zuul.yaml: I1874f96f78cb403e6f3a56a49cb83df40d531f8d\n  (dropping py35 testing)\n* bindep.txt: I9fb56450346fca84bc1573fa5e9b6fe81c14e33a\n  (messaging intermediaries update; the lack of this patch also\n  requires the modification of test-setup.sh)\n\nChange-Id: Ic3a9e2c873619670edfbf71022d593f3cb5f70f2\nRelated-Bug: #1885923\n(cherry picked from commit 7e406c312a6514e7ae377edb52b9e02b5bf37a7d)\n(cherry picked from commit dfcc4b0010d39d1f16be413bb1c3a4808718cd19)\n(cherry picked from commit 88cf5aebef790ff40ab8ca7219d537ccfc54d12e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c5d17905de32a29352a7ad31fd1c7dcde8887017"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c5d17905de32a29352a7ad31fd1c7dcde8887017"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
