)]}'
{"id":"openstack%2Foslo.messaging~739490","triplet_id":"openstack%2Foslo.messaging~stable%2Frocky~Ic3a9e2c873619670edfbf71022d593f3cb5f70f2","project":"openstack/oslo.messaging","branch":"stable/rocky","hashtags":[],"change_id":"Ic3a9e2c873619670edfbf71022d593f3cb5f70f2","subject":"Simplify tools/test-setup.sh","status":"MERGED","created":"2020-07-06 13:45:00.000000000","updated":"2020-07-08 18:01:10.000000000","submitted":"2020-07-08 17:51:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739490-1594230687659-326f21e0","meta_rev_id":"06af770d09c4edb0911e1a590156665ff70c4504","_number":739490,"virtual_id_number":739490,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-07-06 15:41:54.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-08 17:51:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-08 16:03:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-07-08 15:55:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-08 16:03:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-06 13:45:00.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 13:45:00.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-07-08 15:55:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-08 16:03:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-08 17:51:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"411b0d2229318f1ee21a6e3001868d542b453df5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-06 13:45:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3b805d44003ac00f994d97584edcc878bef5fc9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-06 13:52:08.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"5ed835f0a7224f70dbfbcb3202f6422dc18bd95a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 15:30:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a4fe60a69ec4de88c8e253a1aa64a16 : SUCCESS in 8m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3be441ac76c54cfcafce70ae61da8d3e : SUCCESS in 8m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f5994608646446a78b268e791c15873d : FAILURE in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1107d5e044d6430f9627fdc6f3c4a6da : SUCCESS in 6m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5a97a8f0997845bcb92611121b2c0ce1 : SUCCESS in 1h 42m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32df57bb1fc8454ab68a6640410b2414 : SUCCESS in 11m 05s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/57bdd0b7d6944b04ba909ad83b089b81 : SUCCESS in 9m 03s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/b6d003f51fb5475f8a36ddf17e18e730 : SUCCESS in 5m 34s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/3950c274512147a79a98528f8ff7745c : FAILURE in 7m 34s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/2975149caf38402286b3ba8ee201317d : SUCCESS in 6m 01s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/e1dfc4bfde164beebcc8cad0a3cdc223 : SUCCESS in 7m 58s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/75fa5d1f84e24138a26fe70bb98c5e77 : FAILURE in 5m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c0268e40ef4a4c6c953b7adc24c91d39 : SUCCESS in 1h 37m 32s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7335ed21b1c4484fb2ca2e60ebeed4f0 : FAILURE in 39m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/6d0be3097a8d42958c209fd87738c3a6 : FAILURE in 14m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/fa7f2ca81d294e2c8e424eaba2887f27 : FAILURE in 22m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/f9b840888df840ba8b668f1401d7f071 : FAILURE in 19m 47s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/716f5d6648b843c2b294e4ac31fd1d7e : FAILURE in 17m 09s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/2efab6b7e6674bf69e5eb75002c53847 : FAILURE in 18m 46s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/bbbffa7f2ba6481fa3ad2108c8d8dfb4 : FAILURE in 48m 26s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/7771d3f5d9054e6494459e068265262b : FAILURE in 41m 29s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/d9af0f25f11f479886ff900b1c4fdaa3 : FAILURE in 41m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebebc32f15e2435dad6d838ad1fb8f86 : SUCCESS in 9m 19s","accounts_in_message":[],"_revision_number":1},{"id":"80f05e92b1627dd4f0d7031c3c841c3f268c5871","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-06 15:41:54.000000000","message":"Patch Set 1:\n\nrecheck - py35 fail. strange. locally it\u0027s passing. :S","accounts_in_message":[],"_revision_number":1},{"id":"671d4803be100806f20b698e265f15e11b1255fb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-06 16:34:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5e1d6d1ff1bed4823b3ecc7b67d9868bfa936b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 17:37:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/358e998a1d414fc88d261a1a188fd18f : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1881b9eacdc449a08acdccee2deb2aee : SUCCESS in 8m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9946875abc0245268285bd3ddd372bcd : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f101b80e68004224969b0f061574cd04 : SUCCESS in 4m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ee21208d13db405887f289483c08162b : SUCCESS in 1h 18m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22e8b99040c34bbe975550142dbd241e : SUCCESS in 6m 05s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/e4b99817be5a42f3909655022ca00380 : SUCCESS in 8m 08s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/ed63d85dca814171943d96c2204676aa : SUCCESS in 9m 37s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/2abec823fde34117ac85bb08beec51ac : FAILURE in 6m 28s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/725ce7827d944e66a8ef128c457bfb1c : FAILURE in 6m 38s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/993dd136af5b4fd2a2fb644e2247b89e : SUCCESS in 6m 26s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/db6c934cfb3b41b5ae9501da44358678 : FAILURE in 5m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/34a791c9a8a84290a9e0839b614c5837 : SUCCESS in 1h 28m 38s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4dd2b482d06e4cc4b46b8a357326f112 : FAILURE in 40m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/7398d3e8be4043aba1001c062d287b89 : FAILURE in 19m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1735f69719dc4b93a093098de769d92a : FAILURE in 33m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/f97911862a1a4c39a031ef4423cbfd5a : FAILURE in 11m 47s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/74515210991944d4b9b2103a0b4b7113 : SUCCESS in 35m 47s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/8a5015fd61584e35b12a4c46b318280e : SUCCESS in 1h 21m 35s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/87282fe9474b4075b132f67657f7c615 : FAILURE in 36m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/93c3393fc3ea43eca0d51d11903ecea6 : FAILURE in 47m 03s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/8b92309fe41146a28366460a24d4e520 : FAILURE in 47m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36679191768446fa8d37a0071a48e731 : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2ef471abf7b6e784f805afaa71812ff9dd9036b6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-08 15:55:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"efdcca69cba2e4d93507867ff66cb614c1de2747","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-08 16:03:36.000000000","message":"Patch Set 1: -Backport-Candidate Code-Review+2 Workflow+1\n\nDarn buttons. :)","accounts_in_message":[],"_revision_number":1},{"id":"38b904643106f744a6770af8482aaee10fb90d62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 16:08:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6436b27ecc4bb6ba7b231e709e15d7f5819df99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 17:51:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ee2ae83f531b05d18b7579e67a1abfd7399c42e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 17:51:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/749573ef354446e3b309d46d5b995473 : SUCCESS in 7m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6c0982668e0443181e1757d7021a103 : SUCCESS in 9m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2b3b5018dee5436bb92c52f07661d7c1 : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/783e5ae2d77c4ba4b5b9cc1e12b11d0a : SUCCESS in 6m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/813ab27eaaea4b33b2b741f83969c83e : SUCCESS in 1h 29m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11b2d491aad2460782ffc710ebe565bc : SUCCESS in 6m 40s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/bd819b6f68d5457ca1f13a9744eebf66 : SUCCESS in 8m 01s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c3772b68682f43519a23cbbfd9d8dcd3 : SUCCESS in 1h 22m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d36749fca9b94dbf994fbdb787a116c5 : SUCCESS in 9m 31s","accounts_in_message":[],"_revision_number":1},{"id":"06af770d09c4edb0911e1a590156665ff70c4504","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 18:01:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a263e0fe1b1b413e833222c81ac689a7 : SUCCESS in 2m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6a5013f8edd413ab08cc3cd144855eb : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41a7da012e0f0df2a5ddb689897448fd02b6175c","revisions":{"41a7da012e0f0df2a5ddb689897448fd02b6175c":{"kind":"REWORK","_number":1,"created":"2020-07-06 13:45:00.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/90/739490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/90/739490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/90/739490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/90/739490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/90/739490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/90/739490/1"}}},"commit":{"parents":[{"commit":"b67a457e4fa71e9220c149087dce013c7f81144f","subject":"Use default exchange for direct messaging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b67a457e4fa71e9220c149087dce013c7f81144f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-17 14:53:03.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-07-06 13:44:00.000000000","tz":120},"subject":"Simplify tools/test-setup.sh","message":"Simplify tools/test-setup.sh\n\ntest-setup.sh is called after bindep is run in CI, and tox is not setup.\n\nSimplify the script so that it only does what\u0027s needed - install\nqdrouterd for Ubuntu from qpid/testing.\n\nInstall the other packages using bindep, add test to amqp1 profile.\n\nAdd amqp1 to bindep_profiles where needed for scenario tests, previously\ntest-setup.sh added it everywhere.\n\nChange-Id: Ic3a9e2c873619670edfbf71022d593f3cb5f70f2\nRelated-Bug: #1885923\n(cherry picked from commit 7e406c312a6514e7ae377edb52b9e02b5bf37a7d)\n(cherry picked from commit dfcc4b0010d39d1f16be413bb1c3a4808718cd19)\n(cherry picked from commit 88cf5aebef790ff40ab8ca7219d537ccfc54d12e)\n(cherry picked from commit c5d17905de32a29352a7ad31fd1c7dcde8887017)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a7da012e0f0df2a5ddb689897448fd02b6175c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a7da012e0f0df2a5ddb689897448fd02b6175c"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
