)]}'
{"id":"openstack%2Foslo.messaging~740115","triplet_id":"openstack%2Foslo.messaging~stable%2Fstein~I5ca930cada523bc77c644703a0f02b9160816231","project":"openstack/oslo.messaging","branch":"stable/stein","topic":"bug/1871813-stable/ussuri-stable/train-stable/stein","hashtags":[],"change_id":"I5ca930cada523bc77c644703a0f02b9160816231","subject":"Print warning message when connection running out","status":"MERGED","created":"2020-07-08 20:55:46.000000000","updated":"2020-07-11 03:27:35.000000000","submitted":"2020-07-10 17:27:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740115-1594402074807-1939c49e","meta_rev_id":"3e260cc70a807de3f876dbd4dd2319dfe39c4404","_number":740115,"virtual_id_number":740115,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 17:27:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-10 15:41:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-10 14:14:38.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2020-07-10 08:39:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-10 15:41:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-10 08:39:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-10 14:14:38.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-07-10 15:41:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-10 17:27:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58ccf58b832eaf7df21570686573239c4066659c","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-07-08 20:55:46.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"9b0b2a92cd3e7e30a07fc926791b580c2e73533c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 22:43:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8d9092932c9a41fca5691020ed82726a : SUCCESS in 1h 34m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95eef0efd7664a73aa4e49224b664d9c : SUCCESS in 1h 37m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b79c5733a354632b1c29426f5be7c4e : SUCCESS in 8m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30f2b75bd8e342c69a6fc827bfc7bf7d : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df7772aa26cf4b7f896293cc9c23fa2e : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa6437efccca4899a47ebc58491328df : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ddf4d89d393475ea40a6caa4e933af1 : SUCCESS in 10m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d1aae55c7774844bbf698531dbd7d9f : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb888ee98dc4e02b3158b24b4ee84aa : SUCCESS in 6m 56s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/e191ac20c2d4407eb4fa5265466c3d5f : SUCCESS in 6m 22s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/1b1a2ffb49f54f3381d57c80488de2a1 : SUCCESS in 10m 53s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/8531b4f98c49424f8c526e3c6f6548e4 : FAILURE in 4m 47s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/f7c65c3b848a4812866f3f72411d990d : FAILURE in 6m 39s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/01adf1a7049b46529e8912f8574311c6 : SUCCESS in 5m 42s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/55bb662aa2884dd0885d901d555ebbbd : FAILURE in 5m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3550c2a3a4d34a6aa7bb4445e2071b47 : SUCCESS in 1h 07m 41s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d399793ac9794eb0b904682f90477157 : FAILURE in 1h 44m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/5378c5f312a647e4bc7515bb19017fe9 : FAILURE in 17m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/4ef1e7aecf86463f902c42477a95bb8c : SUCCESS in 1h 23m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/c4273a143c484d568ea38900e3665542 : FAILURE in 17m 42s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/ffaa5a603328496f9a63c6879ce79f63 : SUCCESS in 1h 00m 35s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/5156b385bd1e4718be3fcaf5e7680d28 : FAILURE in 1h 07m 36s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/3c6437a9d96b4f0aa6e29a4473c090f5 : FAILURE in 47m 41s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/91868e3767bc4915b6df0481d0548208 : FAILURE in 49m 21s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/80f1fb022e4348e08792e63591a05af6 : FAILURE in 46m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"babd708f12513a91c4e9613a5a1ec0263b14ca10","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-10 08:39:04.000000000","message":"Patch Set 1: Code-Review+2\n\nAlready backported on more recent stable branches, clean backport, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"4970348549e1eac3da4be82af46f493323b4ea79","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-07-10 14:14:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2a4feb4f33729e0fb97054c2206c41338dd4ef56","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 15:41:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMerged to stable/train, low risk, and good to backport.","accounts_in_message":[],"_revision_number":1},{"id":"46b0e2f33c8b083fa9429fd54d3b255282ddc09f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:41:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"265bca3477b2a2319cdd62e4eb97fc9c63182c73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 17:27:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4b0351394e0713bffbf15568e4beff27caf810a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 17:27:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ae6f0739e805431a9f9870e98aacaf82 : SUCCESS in 1h 27m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae2e4a727c8d4672a5e87b95a2354dc2 : SUCCESS in 1h 24m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/162365dce4fd47ff84f3ecbb59363797 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/230d3ad0f66148debf15b07bfaac3ebb : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c7e02c41252f4e8c808322d2d340584f : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2f36351f1b14c57a071b2ffa562dd77 : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8f8a67e046e4e26956363763522a779 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b342fb3362446afae15388bc3757cc6 : SUCCESS in 4m 57s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/e9e40ea606f242d6a919ed5519f9d8da : SUCCESS in 5m 35s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c54b448d51504fa7b8ad861fd8f83ae8 : SUCCESS in 1h 39m 54s","accounts_in_message":[],"_revision_number":1},{"id":"2799832885b490d24ac7be67f6848f9a89e2e6d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 17:32:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5876ff92a36540a08a3c460ee548fec6 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"eeb297cdcd0d03193ebf0a4897ea4cfd76d12550","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-07-11 03:27:35.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 171132ff69bb0edcc0ebe1044ae59d1a6d572fd3","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd795bbb8f5aaa740fa810f2248646f6ec3a211e","revisions":{"fd795bbb8f5aaa740fa810f2248646f6ec3a211e":{"kind":"REWORK","_number":1,"created":"2020-07-08 20:55:46.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/15/740115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/740115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/740115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/740115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/740115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/740115/1"}}},"commit":{"parents":[{"commit":"c5d17905de32a29352a7ad31fd1c7dcde8887017","subject":"Simplify tools/test-setup.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c5d17905de32a29352a7ad31fd1c7dcde8887017"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-04-09 10:00:39.000000000","tz":480},"committer":{"name":"norman shen","email":"yshxxsjt715@gmail.com","date":"2020-07-08 20:55:46.000000000","tz":0},"subject":"Print warning message when connection running out","message":"Print warning message when connection running out\n\nI believe when connection pool running out of available\nresources, there will be all kinds of weird consequences\nfollowing. So let\u0027s print warning logs when we are going\nto waiting indefinitely for the lock\n\nChange-Id: I5ca930cada523bc77c644703a0f02b9160816231\nRelated-bug: #1871813\n(cherry picked from commit 96300a32213d1a9bf114b90bef6971d69b839d1d)\n(cherry picked from commit 39f770f55b3331b7a99b3ea42a34982a20635cf2)\n(cherry picked from commit 1613b7a968f33ef7dc683d05fde05be0c458c8fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fd795bbb8f5aaa740fa810f2248646f6ec3a211e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fd795bbb8f5aaa740fa810f2248646f6ec3a211e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
