)]}'
{"id":"openstack%2Foslo.messaging~740537","triplet_id":"openstack%2Foslo.messaging~stable%2Frocky~I5ca930cada523bc77c644703a0f02b9160816231","project":"openstack/oslo.messaging","branch":"stable/rocky","topic":"bug/1871813-stable/ussuri-stable/train-stable/stein-stable/rocky","hashtags":[],"change_id":"I5ca930cada523bc77c644703a0f02b9160816231","subject":"Print warning message when connection running out","status":"MERGED","created":"2020-07-11 03:27:35.000000000","updated":"2020-07-17 16:18:07.000000000","submitted":"2020-07-17 16:14:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740537-1595002456088-a8fb295d","meta_rev_id":"5c9dd136d1ff063c4afb0f6f44c474033f5a40d6","_number":740537,"virtual_id_number":740537,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-17 16:14:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-17 14:48:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-15 09:34:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-17 14:48:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-15 09:34:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-17 14:48:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-17 16:14:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20f69cec61587f94f0fdacef9ba222d78963ee33","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-07-11 03:27:35.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"18eca86fba2c697b2f4da7dbfdf76084a729329e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 05:07:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d859103e18f046549e204a8de11cbe87 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf290d2e349a42f6b0d916ef370dff38 : SUCCESS in 5m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7978e587d808412bb0ec51c8c167895a : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4871be8a28c48e69b24908a9a951dbd : SUCCESS in 4m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/998df6a34e774b04962ffe4fbaa3b49c : SUCCESS in 1h 14m 46s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/c3b468d9fa924735b89a53dac6ac8601 : SUCCESS in 4m 56s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/0b86b35c9e194342bb4bac47ddc52a35 : SUCCESS in 5m 14s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/bdee03096b654513bda7b36228f10f4a : FAILURE in 4m 07s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/e6e2846768e842f3bcc06625e77346e9 : FAILURE in 5m 46s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/58c923b566c645e4b89ffa61e5653b9a : SUCCESS in 5m 21s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/a5b937d19d7e496ea0da778acfb07651 : FAILURE in 5m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3058d7fab9174c75935e9037cef8431d : SUCCESS in 1h 30m 59s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ee30367ac4fd45c28b5cbe4293ef9c52 : FAILURE in 32m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/47c0c3edd25643cb94bd05642a3d15dc : FAILURE in 12m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7491d1bf3d024dc4827d3d49b861af32 : FAILURE in 20m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/aa4d2b4ac5d949f4add710710971ef04 : FAILURE in 13m 15s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/ecff013b5a4e49ff9fbd8665ea9277f0 : SUCCESS in 55m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/4f8715336f7f49aa9492524338dacc3e : SUCCESS in 52m 01s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/0a1c89912cd143a0b9b97bcc4d096d36 : FAILURE in 38m 54s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/cb236c0d67404958980baa42c0ac085c : FAILURE in 24m 08s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/f9856e2834cb4ad2a5f2724b76e7b1a9 : FAILURE in 37m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9622ed0821a463c892a996112eec633 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"bc6b500806f670c056de593ecc2d182202ad930d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-15 09:34:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bd89c425ebad6ee4e55cef6a7b120cb8347cad04","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-17 14:48:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMerged to more recent branches.","accounts_in_message":[],"_revision_number":1},{"id":"983aa588f9bd70701e02e3bc8ed91145c91b11fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 14:49:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81a61aae55342bcbe2abbf9be8aafbecef336ccc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 16:14:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0449f0de416f47b2a55a51adf4a7feea : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/07928d814b01482fb4e42c3d08257523 : SUCCESS in 6m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fe9d7c3a90f447bb85304348184102b4 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b962da3300604c9ca61ae8a8655fd31f : SUCCESS in 4m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/48db8f42d9c7422587ff2b21cc3c7e5f : SUCCESS in 1h 21m 01s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/11c281ed0d644c38a1a30b68f2e3db88 : SUCCESS in 5m 44s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/264dd447ed8441f4ad560fee5d2d413f : SUCCESS in 1h 22m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8091136c4c7e40ee9e8f8a0750e276df : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bc2a74b6c5ca1db1d01e85d76b3808d81b4172cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 16:14:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cd06e4410b6f0a67dcfd9a9a62b822a2f8037182","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 16:18:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f66f691108f94bf2ae565d990d460b7e : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"171132ff69bb0edcc0ebe1044ae59d1a6d572fd3","revisions":{"171132ff69bb0edcc0ebe1044ae59d1a6d572fd3":{"kind":"REWORK","_number":1,"created":"2020-07-11 03:27:35.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/37/740537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/37/740537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/740537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/740537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/740537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/37/740537/1"}}},"commit":{"parents":[{"commit":"41a7da012e0f0df2a5ddb689897448fd02b6175c","subject":"Simplify tools/test-setup.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a7da012e0f0df2a5ddb689897448fd02b6175c"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-04-09 10:00:39.000000000","tz":480},"committer":{"name":"norman shen","email":"yshxxsjt715@gmail.com","date":"2020-07-11 03:27:35.000000000","tz":0},"subject":"Print warning message when connection running out","message":"Print warning message when connection running out\n\nI believe when connection pool running out of available\nresources, there will be all kinds of weird consequences\nfollowing. So let\u0027s print warning logs when we are going\nto waiting indefinitely for the lock\n\nChange-Id: I5ca930cada523bc77c644703a0f02b9160816231\nRelated-bug: #1871813\n(cherry picked from commit 96300a32213d1a9bf114b90bef6971d69b839d1d)\n(cherry picked from commit 39f770f55b3331b7a99b3ea42a34982a20635cf2)\n(cherry picked from commit 1613b7a968f33ef7dc683d05fde05be0c458c8fc)\n(cherry picked from commit fd795bbb8f5aaa740fa810f2248646f6ec3a211e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/171132ff69bb0edcc0ebe1044ae59d1a6d572fd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/171132ff69bb0edcc0ebe1044ae59d1a6d572fd3"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
