)]}'
{"id":"openstack%2Foslo.messaging~741947","triplet_id":"openstack%2Foslo.messaging~master~Ibbb811e988cc048e6c39a0c2336c716a1a0ddb30","project":"openstack/oslo.messaging","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"Ibbb811e988cc048e6c39a0c2336c716a1a0ddb30","subject":"Move legacy grenade jobs to Zuul v3","status":"MERGED","created":"2020-07-20 13:52:35.000000000","updated":"2020-07-22 14:05:04.000000000","submitted":"2020-07-22 13:58:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741947-1595426299304-bfbdc47d","meta_rev_id":"2ca0dc8c51fba0b51cb424868307f9ee074733a9","_number":741947,"virtual_id_number":741947,"owner":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-22 13:19:35.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-22 13:58:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2020-07-20 16:12:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2020-07-22 12:16:23.000000000","_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-07-22 12:16:23.000000000","_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"date":"2020-07-22 13:17:38.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 16:12:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-22 12:16:23.000000000","updated_by":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"reviewer":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"state":"REVIEWER"},{"updated":"2020-07-22 13:17:38.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-07-22 13:58:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10e4b9e052e88d5c5c095c53ccb7858c494efe6f","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-07-20 13:52:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e35012e57cae05a92dba518c0bff1d5a79145731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 15:29:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ba5a4a8646747398097f4b71825aca8 : SUCCESS in 1h 32m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/743a6e21d23c40e8aeb33e45289c9951 : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/437a3d9a424946b8b6fccb26c9783490 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3dcc77608104cce9b83bd0792811023 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90838b76df3244b79359c0e229e9d015 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5b436df397849b699d6d6bef034ef1d : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0976dec6a77d42548a38905cf966b65b : SUCCESS in 4m 39s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/19717a1bf72349a68248f48417c6b4c2 : SUCCESS in 5m 07s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/3fffbcf6eaa44c7c825a996cbe667012 : SUCCESS in 6m 13s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/3e5a8885fdab432ba6ab2578337c5a2a : SUCCESS in 5m 25s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/e3e5ff809a4d48a391f377a21ec76ab1 : SUCCESS in 6m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5fa7d678998b4afc8eeb660ed80a4456 : SUCCESS in 1h 24m 36s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3e68abe1b272401fb7962ac2535f75d3 : SUCCESS in 1h 26m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d8c8e730f1df4f3083af2957c9affc49 : SUCCESS in 1h 34m 00s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/8da7ee569b64454cb1a931a3bf7accc4 : SUCCESS in 1h 07m 29s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/55d2d874b5ad47e9a9dd267d731145d5 : SUCCESS in 1h 19m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0796de430057dea7a13dcb226a1c90cf83df86c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-20 16:12:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ec908d8a8940f8b0a065e5993c33507c19a84db5","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-07-22 12:16:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c4c07f38b05c217308a4831833848e3a02a5558","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 12:16:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2ebe11b8df5b30b04778fa98b46c37efad056c4","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-07-22 13:19:35.000000000","message":"Patch Set 1:\n\nHi Luigi,\n\nCan you also review these changes?  Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"72b3da7f3ca521b44e8e5bcb426a699244770e1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 13:58:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2cb53eb854640838ae3f1ac78580cdf : SUCCESS in 1h 14m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f84fbc869f6437183a7e9c62c91e1f6 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e75f29a4ea4c19963bfc180a8661fa : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb33a8e3f1444f23b144dced685bfb81 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89e0571b96ab456ebd1739727f3520f1 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd53f34a0a624b25af3bb7f21c1db493 : SUCCESS in 4m 48s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/05188bc61dbf4b9992982235be41e675 : SUCCESS in 6m 32s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5bf7775e5ec546bd970526b72f0f7dcd : SUCCESS in 1h 34m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b7ed20303e851105dcc9564e6d31737415b5b63b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 13:58:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7ad9936680c9dc5bd257c8929b30d93b04cfd156","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-22 14:01:05.000000000","message":"Patch Set 1:\n\nUps, sorry. In retrospective, oslo.messaging is installed from source as expected (from the tip of stable/ussuri on the \"old\" environment, from the tip of master after the upgrade).","accounts_in_message":[],"_revision_number":1},{"id":"27a108cecd97d92345abf5e60fdfb9de11765a63","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-22 14:02:38.000000000","message":"Patch Set 1:\n\nAs usual: it\u0027s not requested by the goal, but it may improve your and QA future maintenance work if you backport this on all the branches where the new grenade jobs are available (currently ussuri and train; the normal devstack jobs should work back to pike).","accounts_in_message":[],"_revision_number":1},{"id":"576f872ca009ee1f6e7700d3cbb81b79ccc7a290","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 14:05:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25ddeaf30255439e844e9cd186ff6a7f : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5ae9f4aa0e18787b4bb676b3e725c225deae96d","revisions":{"a5ae9f4aa0e18787b4bb676b3e725c225deae96d":{"kind":"REWORK","_number":1,"created":"2020-07-20 13:52:35.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/47/741947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/47/741947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/741947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/741947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/741947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/47/741947/1"}}},"commit":{"parents":[{"commit":"f988751c5c4a10fddbaac1bfcbdac923ab36e712","subject":"Merge \"Drop a python 2 exception management\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f988751c5c4a10fddbaac1bfcbdac923ab36e712"}]}],"author":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2020-07-20 13:51:15.000000000","tz":-240},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2020-07-20 13:51:15.000000000","tz":-240},"subject":"Move legacy grenade jobs to Zuul v3","message":"Move legacy grenade jobs to Zuul v3\n\nAlso deletes obsolete playbook files for telemetry tests that have\nbeen removed.\n\nChange-Id: Ibbb811e988cc048e6c39a0c2336c716a1a0ddb30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a5ae9f4aa0e18787b4bb676b3e725c225deae96d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a5ae9f4aa0e18787b4bb676b3e725c225deae96d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
