)]}'
{"id":"openstack%2Foslo.messaging~742609","triplet_id":"openstack%2Foslo.messaging~stable%2Fussuri~Ibbcf21a57ab1e3f90c21901296e5c088b645127c","project":"openstack/oslo.messaging","branch":"stable/ussuri","topic":"test_setup.sh-stable/ussuri","hashtags":[],"change_id":"Ibbcf21a57ab1e3f90c21901296e5c088b645127c","subject":"tests: Resolves issues with kombu \u003e 4.6.8","status":"MERGED","created":"2020-07-23 08:09:47.000000000","updated":"2020-07-24 15:08:19.000000000","submitted":"2020-07-24 12:46:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742609-1595594799768-dfe663bb","meta_rev_id":"162245fb5ba42ccdb5549937e524219c288623bb","_number":742609,"virtual_id_number":742609,"owner":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2020-07-24 11:50:28.000000000","_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 12:46:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-07-23 15:17:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-07-24 11:12:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-07-24 11:12:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-23 15:17:39.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-07-24 11:12:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-24 12:46:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99dc13f87d9bb29de01443be2ede99e790c9e092","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-23 08:09:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a983e29d91969fc0a678b250be01c7f4baf5fa0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 10:03:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70f83e53b65240e3840cde71ba7ec36e : SUCCESS in 1h 21m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efb4fd7369344d4aacc2e8b060a7fb02 : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc93956160f345efb46f16396b9ab498 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4829df3cf1bb4cf1bc1923781fd88cab : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96e7f4bbf3a74203bf8ae0878d2cb7f6 : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54dd2b5dc9154db48bd9b83a5795cedd : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ef48b874fc246619d98eaa58f55fc24 : SUCCESS in 5m 01s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/79b8f586cdd7443d868a44252439fc50 : SUCCESS in 5m 20s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/bd5da83ac03f47029412aca4d4a7295b : SUCCESS in 7m 01s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/a7436439b9794c4e8e7b3706cb7a4f83 : SUCCESS in 6m 00s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/a6ae9ea9e0934512acd0ca2bbc9e5f85 : FAILURE in 6m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1bab8d72808b492d829620a0ecd3287e : SUCCESS in 1h 18m 15s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/96a1dc4d61314defa6d593e05ddaac53 : SUCCESS in 1h 47m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9d82b7e8b58f40c18fc1f8f650ad3785 : SUCCESS in 1h 39m 31s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/a653c7190b0649acaa453c371e8bf666 : SUCCESS in 1h 13m 56s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/601c8e1e44ac4b839f3f7e7eb7e9aaac : FAILURE in 28m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51c86b86fde01d70a3155f80f65df29b2cd9ad89","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-23 15:17:39.000000000","message":"Patch Set 1: Code-Review+2\n\nClean backport. LGTM","accounts_in_message":[],"_revision_number":1},{"id":"de76c4afc45b2d9566706b606a201de609a869f0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 11:12:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM\n\nI guess it should be backported to following too:\n- train (kombu!\u003d4.0.2,\u003e\u003d4.6.1)\n- stein (kombu!\u003d4.0.2,\u003e\u003d4.0.0)\n- rocky (kombu!\u003d4.0.2,\u003e\u003d4.0.0)\n- queens (kombu!\u003d4.0.2,\u003e\u003d4.0.0)","accounts_in_message":[],"_revision_number":1},{"id":"4c9113017e56d7a7bfef394b33e05a493885cbe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 11:12:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7e907f607d5d2823ef0d96bdd2812abe3750ff0","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 11:50:28.000000000","message":"Patch Set 1:\n\nok, also did the other cherry-picks. pike also has kombu\u003e\u003d4.0.0 .","accounts_in_message":[],"_revision_number":1},{"id":"ace75bbf3af4f2b60d7e6e97c700a162069188c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 12:46:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aaa957353647cb321eab5c40048b242ed9b13ed7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 12:46:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4531d763a364ae3aa4470d36d3f1c49 : SUCCESS in 1h 30m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0fccd3e80874edaadd2e7424d0f9cb7 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02692b32afbf480aa67db76a1241162f : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9ee191d26a04d7c93f8f3c00f663f2a : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d03f70bec3194a0da3078aa70be55090 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edba76eb7cc74bd3a152309c4f331f30 : SUCCESS in 4m 29s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/caf40f8059944c35b230f50963dccc51 : SUCCESS in 6m 24s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6f149edffd7641adbc9863407e016494 : SUCCESS in 1h 33m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5fd7ce9b8a929ef76f16a61d45a942f87675f44e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 12:48:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74890e084a324b60ae7861dc7032f34d : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":1},{"id":"957b8fbb62f356900b09c18de7be0284b8739426","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 13:01:00.000000000","message":"Patch Set 1:\n\nThis should probably be reverted. Max version of kombu supported in stable/ussuri is 4.6.8, so we shouldn\u0027t need to have compatibility code for newer versions.","accounts_in_message":[],"_revision_number":1},{"id":"d791f5e69b9c3ee7c3fa9f350e361ea9db83522f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-24 15:07:54.000000000","message":"Patch Set 1:\n\n\u003e This should probably be reverted. Max version of kombu supported in\n \u003e stable/ussuri is 4.6.8, so we shouldn\u0027t need to have compatibility\n \u003e code for newer versions.\n\nMan, how did I miss that /o\\ Apologies.","accounts_in_message":[],"_revision_number":1},{"id":"f84cf366153d53ade925153032e5013631c63ab1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-24 15:08:19.000000000","message":"Created a revert of this change as I50226ed476280f9e61c6b302a1f71a27ae88fbca","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6727be7890f64446f1658d79fac196d30b6f1a18","revisions":{"6727be7890f64446f1658d79fac196d30b6f1a18":{"kind":"REWORK","_number":1,"created":"2020-07-23 08:09:47.000000000","uploader":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"ref":"refs/changes/09/742609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/09/742609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/09/742609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/09/742609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/09/742609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/09/742609/1"}}},"commit":{"parents":[{"commit":"dfcc4b0010d39d1f16be413bb1c3a4808718cd19","subject":"Simplify tools/test-setup.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dfcc4b0010d39d1f16be413bb1c3a4808718cd19"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-01 14:01:00.000000000","tz":60},"committer":{"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","date":"2020-07-23 08:09:47.000000000","tz":0},"subject":"tests: Resolves issues with kombu \u003e 4.6.8","message":"tests: Resolves issues with kombu \u003e 4.6.8\n\nThe \u0027kombu.connection.Connection.ensure_connection\u0027 method has changed\nfrom calling \u0027retry_over_time\u0027 on \u0027self.connect\u0027 to calling it on\n\u0027self._connection_factory\u0027 [1], meaning our mocks are outdated. Address\nthis change.\n\n[1] https://github.com/celery/kombu/pull/1193/commits/398aa5b8cd1fe1fc\n\nChange-Id: Ibbcf21a57ab1e3f90c21901296e5c088b645127c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCloses-Bug: #1885923\n(cherry picked from commit afb035d97185fa9c622bdb9b4c31c71f01160370)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6727be7890f64446f1658d79fac196d30b6f1a18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6727be7890f64446f1658d79fac196d30b6f1a18"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
