)]}'
{"id":"openstack%2Foslo.messaging~742859","triplet_id":"openstack%2Foslo.messaging~stable%2Frocky~Ibbcf21a57ab1e3f90c21901296e5c088b645127c","project":"openstack/oslo.messaging","branch":"stable/rocky","topic":"test_setup.sh-stable/rocky","hashtags":[],"change_id":"Ibbcf21a57ab1e3f90c21901296e5c088b645127c","subject":"tests: Resolves issues with kombu \u003e 4.6.8","status":"ABANDONED","created":"2020-07-24 11:45:22.000000000","updated":"2020-07-24 15:31:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c0c5b2dbd86a7f5c790fd7a1234c7a3f16c69d81","_number":742859,"virtual_id_number":742859,"owner":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-24 13:40:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-1,"date":"2020-07-24 13:00:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-24 11:45:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 11:45:22.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-07-24 13:00:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-24 13:40:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0e0a61ed775e774adac77e530c7b2f79fe0102f","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 11:45:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f51e87c0f589da6e28066852b55792931289d37c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 13:00:09.000000000","message":"Patch Set 1: Backport-Candidate-1\n\nThe max version of kombu supported in stable/train is 4.6.6, so this should not be needed.","accounts_in_message":[],"_revision_number":1},{"id":"8cde3e5efbf1d970a30b0688335d07d39e9b5452","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 13:40:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f504990553d4fc8a9e6323dbacbc738 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/806375fb5a6744e984d45cead17679c6 : SUCCESS in 5m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/390b6d301ac14c1ca023a5c69667a2e7 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b4c09277434867b1aef31c5a467666 : SUCCESS in 4m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0b70125987a2448795a8b8ff20b49ee9 : SUCCESS in 59m 50s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/7dd521d40db14d5e960a55cd85b4500d : SUCCESS in 5m 11s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/7f5adcd762ca411c9e81c57bace47a3d : SUCCESS in 6m 48s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/a3171aceb3914abfa5a92ec1748e887b : FAILURE in 5m 51s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/7d40abd9222943bfaea741694a2d75a1 : SUCCESS in 8m 02s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/19ce32074ee849968cadc403aedcf078 : SUCCESS in 7m 40s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/b70e1e2dc2f94eecb9bf54af0098ec14 : FAILURE in 7m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/45bfe2aded5b4fcdbf139b8dd89dde3e : FAILURE in 1h 52m 58s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/61e66f7f00524646be8b7ee3598e3520 : FAILURE in 52m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/9786594dd71945fb9c61accd24b7a27e : FAILURE in 14m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2f914e648c314bbbaddbbd88690c82b1 : FAILURE in 28m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/4af7982884dd4e348eb47458c87e9375 : FAILURE in 25m 13s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d1b30123b0a4444b89a9e18aa7a9da88 : SUCCESS in 1h 18m 28s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/ee486734a68f43a8997409fcfbdf0232 : SUCCESS in 55m 25s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/0dd08a26379a4b71b2e6bcb1a22e56ef : FAILURE in 1h 03m 28s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/e016a98cd8ee4cc99f3a21b2ba8d1eab : FAILURE in 45m 22s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/10ca17b3addc492b92f4e93f4af5c233 : FAILURE in 1h 02m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16e5181595894c3f848d4d988f5c9184 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"7f962fb897175d0b81b9f8481d48d651e221e487","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 15:31:43.000000000","message":"Abandoned\n\nAbandoning because of resistance. I only need it in Ussuri. feel free to reopen.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4ffe7ab22c01d2abac92746dfe4ff380d370902","revisions":{"b4ffe7ab22c01d2abac92746dfe4ff380d370902":{"kind":"REWORK","_number":1,"created":"2020-07-24 11:45:22.000000000","uploader":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"ref":"refs/changes/59/742859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/59/742859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/59/742859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/59/742859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/59/742859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/59/742859/1"}}},"commit":{"parents":[{"commit":"171132ff69bb0edcc0ebe1044ae59d1a6d572fd3","subject":"Print warning message when connection running out","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/171132ff69bb0edcc0ebe1044ae59d1a6d572fd3"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-01 14:01:00.000000000","tz":60},"committer":{"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","date":"2020-07-24 11:45:22.000000000","tz":0},"subject":"tests: Resolves issues with kombu \u003e 4.6.8","message":"tests: Resolves issues with kombu \u003e 4.6.8\n\nThe \u0027kombu.connection.Connection.ensure_connection\u0027 method has changed\nfrom calling \u0027retry_over_time\u0027 on \u0027self.connect\u0027 to calling it on\n\u0027self._connection_factory\u0027 [1], meaning our mocks are outdated. Address\nthis change.\n\n[1] https://github.com/celery/kombu/pull/1193/commits/398aa5b8cd1fe1fc\n\nChange-Id: Ibbcf21a57ab1e3f90c21901296e5c088b645127c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCloses-Bug: #1885923\n(cherry picked from commit afb035d97185fa9c622bdb9b4c31c71f01160370)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b4ffe7ab22c01d2abac92746dfe4ff380d370902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b4ffe7ab22c01d2abac92746dfe4ff380d370902"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
