)]}'
{"id":"openstack%2Foslo.messaging~742860","triplet_id":"openstack%2Foslo.messaging~stable%2Fqueens~Ibbcf21a57ab1e3f90c21901296e5c088b645127c","project":"openstack/oslo.messaging","branch":"stable/queens","topic":"test_setup.sh-stable/queens","hashtags":[],"change_id":"Ibbcf21a57ab1e3f90c21901296e5c088b645127c","subject":"tests: Resolves issues with kombu \u003e 4.6.8","status":"ABANDONED","created":"2020-07-24 11:46:08.000000000","updated":"2020-07-24 15:31:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"498f666eb38969fd61118a54698b42deaca0bc21","_number":742860,"virtual_id_number":742860,"owner":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-24 13:38:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-1,"date":"2020-07-24 13:01:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-24 11:46:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 11:46:08.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-07-24 13:01:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-24 13:38:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"481471e4389a8d7ed81bc63f928d7670256ce78b","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 11:46:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"281f2b6fd0b5f89f2709b0ff5f0e5f473ecfb0a5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 13:01:29.000000000","message":"Patch Set 1: Backport-Candidate-1\n\n4.6.8 is not supported in this branch.","accounts_in_message":[],"_revision_number":1},{"id":"500fb93d2d50a709255f42250d5b4723f255cf08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 13:38:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95ccb82aa4f443f8a1681c152418ef55 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c0ffb45076b94e68bcb4737f9818eede : SUCCESS in 6m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9849dfdd129c41f7b1d8ee01f84386a0 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea6231de71f4e6f94f61b35557e3b7d : SUCCESS in 5m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/906218ba0a9147f989d5c20e2aee31cc : SUCCESS in 1h 39m 19s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/f7c339615e0741f1a52fec34ff211d63 : FAILURE in 6m 14s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/f75489c8dab7409090a3d25becf2dd95 : FAILURE in 6m 10s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/9e510946a3e640bcbcc269e04d531a3c : SUCCESS in 6m 08s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/7ec2e2e6286247bab7f360a1a6c73d17 : SUCCESS in 5m 27s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/b625f07a6497428e8f09b68f5bd075f0 : SUCCESS in 6m 15s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/0823b0326f6e493e8abbdeeb43db7b7b : SUCCESS in 9m 10s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/7e10c21f48b24a54adc6d1491ec95f7c : SUCCESS in 6m 46s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/a677beadea83434ead38db9650c9fc9e : SUCCESS in 8m 29s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/3d348b2e388b44c2ac0428816dcf651f : SUCCESS in 6m 32s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/f9d69b2fd4b44a8b89fa488f8b70df78 : SUCCESS in 6m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/cfc8357e964d4fa1af936531c0288ffb : SUCCESS in 1h 25m 22s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a1d69add1ced423c95894b18f2438030 : FAILURE in 37m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/01bc74aa9809417d95d95d4a14480559 : FAILURE in 17m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/bca021d9168d43ad9120f369515718bb : FAILURE in 12m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/ea9291adbea54c7bb32a6f73dffda313 : FAILURE in 22m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-pika-default https://zuul.opendev.org/t/openstack/build/22453c0d4abc448a8ff34f061d6a2695 : FAILURE in 1h 30m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-zmq-default https://zuul.opendev.org/t/openstack/build/5d23feac943f48018c77612607004e38 : FAILURE in 17m 56s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/2bf824559f6d4d74b0403a1c7770124b : SUCCESS in 1h 24m 21s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/206a43dfad23496b81c0f5bdf58b569e : SUCCESS in 1h 00m 19s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/5fff99e8cc18474d943a1bfadf32f1b3 : FAILURE in 42m 50s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/03a3d860d5454d01b69bdfe6f6e67dcd : FAILURE in 45m 20s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/884a3f01b265456286559c05172ce499 : FAILURE in 36m 20s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-pika https://zuul.opendev.org/t/openstack/build/47274c1ccc864fdfa96e79c4edd416cc : FAILURE in 54m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-zmq https://zuul.opendev.org/t/openstack/build/11a7191ecb1d4b18b6c012b83afbc73d : FAILURE in 28m 07s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/20a7c757cf314ee98a2125ebed308c27 : SUCCESS in 1h 43m 42s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7d37d060cdd942ad88fb03e8a07f9ec6 : FAILURE in 34m 55s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/de3fd6661c29432cbcb51d8b6e25af10 : FAILURE in 25m 46s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-pika-default https://zuul.opendev.org/t/openstack/build/208851ae12cc436a8cc946348773b58f : FAILURE in 1h 29m 07s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-zmq-default https://zuul.opendev.org/t/openstack/build/759aa165158c4feab7e40a9b3e75ee7b : FAILURE in 18m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f4fb3f452532121c941331340cb22bb0a22ad08","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 15:31:52.000000000","message":"Abandoned\n\nAbandoning because of resistance. I only need it in Ussuri. feel free to reopen.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d6f3edf94173ff096246e074d496cb4f2cbc352","revisions":{"8d6f3edf94173ff096246e074d496cb4f2cbc352":{"kind":"REWORK","_number":1,"created":"2020-07-24 11:46:08.000000000","uploader":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"ref":"refs/changes/60/742860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/60/742860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/60/742860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/60/742860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/60/742860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/60/742860/1"}}},"commit":{"parents":[{"commit":"c46367dad2904cfd3ff1b9cd722a08820dd60218","subject":"Simplify tools/test-setup.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c46367dad2904cfd3ff1b9cd722a08820dd60218"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-01 14:01:00.000000000","tz":60},"committer":{"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","date":"2020-07-24 11:46:08.000000000","tz":0},"subject":"tests: Resolves issues with kombu \u003e 4.6.8","message":"tests: Resolves issues with kombu \u003e 4.6.8\n\nThe \u0027kombu.connection.Connection.ensure_connection\u0027 method has changed\nfrom calling \u0027retry_over_time\u0027 on \u0027self.connect\u0027 to calling it on\n\u0027self._connection_factory\u0027 [1], meaning our mocks are outdated. Address\nthis change.\n\n[1] https://github.com/celery/kombu/pull/1193/commits/398aa5b8cd1fe1fc\n\nChange-Id: Ibbcf21a57ab1e3f90c21901296e5c088b645127c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCloses-Bug: #1885923\n(cherry picked from commit afb035d97185fa9c622bdb9b4c31c71f01160370)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8d6f3edf94173ff096246e074d496cb4f2cbc352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8d6f3edf94173ff096246e074d496cb4f2cbc352"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
