)]}'
{"id":"openstack%2Foslo.messaging~742907","triplet_id":"openstack%2Foslo.messaging~stable%2Fussuri~I50226ed476280f9e61c6b302a1f71a27ae88fbca","project":"openstack/oslo.messaging","branch":"stable/ussuri","topic":"test_setup.sh-stable/ussuri","hashtags":[],"change_id":"I50226ed476280f9e61c6b302a1f71a27ae88fbca","subject":"Revert \"tests: Resolves issues with kombu \u003e 4.6.8\"","status":"ABANDONED","created":"2020-07-24 15:08:19.000000000","updated":"2020-08-10 15:43:41.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f23e1bfd919761c15c841cab2818e697d2445472","_number":742907,"virtual_id_number":742907,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-07-24 17:28:40.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-24 17:03:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":-1,"date":"2020-07-24 15:33:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-24 15:21:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 15:21:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-24 15:33:45.000000000","updated_by":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"reviewer":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"state":"REVIEWER"},{"updated":"2020-07-24 17:03:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-24 17:28:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"0a967610df57c8a6cabc7873546bef928a375086","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-24 15:08:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94356e9bff08c248f34109a8d3b10984240d5e23","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 15:21:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c7d9f8e0cffca8edf35ba0fa3af5d83633758bd","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 15:30:12.000000000","message":"Patch Set 1:\n\nIn https://review.opendev.org/#/c/742609/ :\n\u003e This should probably be reverted. Max version of kombu supported in stable/ussuri is 4.6.8, so we shouldn\u0027t need to have compatibility code for newer versions.\n\nThat is not listed as a requirement in\nhttps://docs.openstack.org/project-team-guide/stable-branches.html#appropriate-fixes\nThe original patch is a small test only change, thus it matches the stable-branches requirements.\n\nWhile upper constraints shows what is supported and tested, there is no requirement to forbid changes to make newer versions work.\n\nThis should not be reverted as otherwise the latest OpenStack release won\u0027t work on any of Debian testing, Fedora Rawhide, openSUSE Tumbleweed. Also the compatibility is for a minor version change.","accounts_in_message":[],"_revision_number":1},{"id":"fdde9281d7a7210edee6f0d247d19a70aa7ac083","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 15:30:18.000000000","message":"Patch Set 1: Backport-Candidate-1","accounts_in_message":[],"_revision_number":1},{"id":"db699316ffa15da4827de57aae0531e0e0327f64","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 15:33:45.000000000","message":"Patch Set 1: -Backport-Candidate Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"6b5f5e90155c92dd740a31111f8dcdb3f978581a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 15:38:29.000000000","message":"Patch Set 1:\n\n\u003e In https://review.opendev.org/#/c/742609/ :\n \u003e \u003e This should probably be reverted. Max version of kombu supported\n \u003e in stable/ussuri is 4.6.8, so we shouldn\u0027t need to have\n \u003e compatibility code for newer versions.\n \u003e \n \u003e That is not listed as a requirement in\n \u003e https://docs.openstack.org/project-team-guide/stable-branches.html#appropriate-fixes\n \u003e The original patch is a small test only change, thus it matches the\n \u003e stable-branches requirements.\n \u003e \n \u003e While upper constraints shows what is supported and tested, there\n \u003e is no requirement to forbid changes to make newer versions work.\n \u003e \n \u003e This should not be reverted as otherwise the latest OpenStack\n \u003e release won\u0027t work on any of Debian testing, Fedora Rawhide,\n \u003e openSUSE Tumbleweed. Also the compatibility is for a minor version\n \u003e change.\n\nIs this testing not actually using upper-constraints? It shouldn\u0027t be possible to even hit this if constraints are being handled correctly.","accounts_in_message":[],"_revision_number":1},{"id":"cfcafff8dcb6c107c1b8a50423ee43f7dbcac188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 17:03:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e627274656f447d78832b5b7f09143c5 : SUCCESS in 1h 14m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a89ad0b0af2492ea3dea559884cd11f : SUCCESS in 9m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3368f8baf41e4248a782e4093fca1c00 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92afe501c0fc4a8cba9309d762e78789 : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57f747f36d39434f98bd9541df016393 : SUCCESS in 7m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f5015ea84954505bfea753000395826 : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd36aa021e5f4b47b613feacce20a121 : SUCCESS in 5m 39s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/b1b9643034ef405fabd00a89bac7a864 : SUCCESS in 6m 21s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/a3afee0e349b4f4a9c3a5cc56c8a83dc : SUCCESS in 7m 31s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/00eb8bd9c1a04afbad261281dce87be1 : SUCCESS in 6m 17s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/2b2efc180c4c400cacae406e3cfacafe : SUCCESS in 9m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/db3c05de42c2416daff4bf7e6b865d00 : SUCCESS in 1h 35m 09s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/52ab839799bd4948ae2295fa3180d9fe : SUCCESS in 1h 29m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2d00350e4d2543429ebff6cf16c10221 : SUCCESS in 1h 50m 06s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/886f0d7be69d4084aa3872fa5325f6be : SUCCESS in 1h 09m 45s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/d33a1d845e254044a84c3f5299fbccf9 : FAILURE in 30m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb4482e0b981e7620318f1ea02b951a4f8e8883e","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-24 17:23:34.000000000","message":"Patch Set 1:\n\nYou are thinking about installing via pip with not distribution provided packages, I\u0027m talking about deb/rpm packages. Those distributions all ship a kombu that is newer than upper-constraints.","accounts_in_message":[],"_revision_number":1},{"id":"efad8bd83d0cd721ce30db53e8082a7c0d9d7498","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 17:28:40.000000000","message":"Patch Set 1:\n\n\u003e You are thinking about installing via pip with not distribution\n \u003e provided packages, I\u0027m talking about deb/rpm packages. Those\n \u003e distributions all ship a kombu that is newer than upper-constraints.\n\nNo, that\u0027s also what I mean. The upper-constraints used in our testing is supposed to be used as the definitive lists for downstream packagers to know what versions are supported to be coinstallable. If a distro is packaging a version higher than what is defined in upper-constraints, then that is an error in packaging.\n\nBut fair enough, this should be harmless, so even though it should not be required, I guess it doesn\u0027t hurt much.","accounts_in_message":[],"_revision_number":1},{"id":"50c90bed0abf18453721d92424345c173399343f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-10 15:43:41.000000000","message":"Abandoned\n\nSean\u0027s happy to keep this as-is so let\u0027s do that","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3101c0436cb3a59ad0a6a7a0d2a5cc27b19c753","revisions":{"b3101c0436cb3a59ad0a6a7a0d2a5cc27b19c753":{"kind":"REWORK","_number":1,"created":"2020-07-24 15:08:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/07/742907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/07/742907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/07/742907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/07/742907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/07/742907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/07/742907/1"}}},"commit":{"parents":[{"commit":"6727be7890f64446f1658d79fac196d30b6f1a18","subject":"tests: Resolves issues with kombu \u003e 4.6.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6727be7890f64446f1658d79fac196d30b6f1a18"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-24 15:08:19.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-24 15:08:19.000000000","tz":0},"subject":"Revert \"tests: Resolves issues with kombu \u003e 4.6.8\"","message":"Revert \"tests: Resolves issues with kombu \u003e 4.6.8\"\n\nThis reverts commit 6727be7890f64446f1658d79fac196d30b6f1a18.\n\nChange-Id: I50226ed476280f9e61c6b302a1f71a27ae88fbca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b3101c0436cb3a59ad0a6a7a0d2a5cc27b19c753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b3101c0436cb3a59ad0a6a7a0d2a5cc27b19c753"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
