)]}'
{"id":"openstack%2Foslo.messaging~742917","triplet_id":"openstack%2Foslo.messaging~master~I9235560667f664643007b8ca0be1707eab4126ad","project":"openstack/oslo.messaging","branch":"master","topic":"bandit","hashtags":[],"change_id":"I9235560667f664643007b8ca0be1707eab4126ad","subject":"Bump bandit version","status":"MERGED","created":"2020-07-24 15:34:38.000000000","updated":"2020-07-28 16:34:29.000000000","submitted":"2020-07-28 16:25:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742917-1595953534448-0128338b","meta_rev_id":"95f97db72f8fae7b20b4d92fb4b29207894f1c52","_number":742917,"virtual_id_number":742917,"owner":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 16:25:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2020-07-28 14:36:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2020-07-27 07:54:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2020-07-26 13:03:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":1,"date":"2020-07-28 14:36:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-26 13:03:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-27 07:54:32.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-07-28 14:36:21.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-07-28 16:25:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"495d8334e6fbae8656cdc44174ed827491637697","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-07-24 15:34:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cee5cf0483aa0c3dae7247897a63a6169818ce02","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 16:44:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d0a5643c9d4c16d9e5ecc3bb9254421a5fe7f2db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 17:26:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ed9e30b4e54c4048b63357a4beeabacd : SUCCESS in 5m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29913b24750047189bae41572c4ffb13 : FAILURE in 1h 07m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13cf52f99e90474db7b5b1d477cd8510 : SUCCESS in 6m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/166e7768581f40c8bd6bf638ee72ff35 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02a5e735ce49438bb9d25292f5f0f06a : FAILURE in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76b0a833e8cb4ff389b4e39c4e072c3a : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43f02ba18e1746b28c81ff4d8903a749 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccbf286ead6446a586101d3326a9dd37 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/735e18b54ea54163a995b6f956ad40ff : SUCCESS in 5m 31s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/10605fd4291944058adcc71083f72054 : SUCCESS in 9m 49s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/a87ff3e2286947438d7d7cd239a8a307 : FAILURE in 9m 39s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/a1d1d11857a94fd1abf6db0b8dc3b2b9 : SUCCESS in 8m 56s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/29f412844c934180afb184394639f606 : SUCCESS in 7m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b7383e72e52c46fa990189656cbb1f7d : SUCCESS in 1h 34m 52s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e2eb7ebffb484572b79aa43f53725773 : SUCCESS in 1h 04m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ad7d3df73e49451eb76cbf6bf9445fbf : SUCCESS in 1h 02m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7b656b816137341a3c09c830067958c43329b70","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-07-25 09:48:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e7ec2769320a68531650947f8ebc5594ae94e398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 11:19:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/687dc352660b446eaa87ad4a0744a77a : SUCCESS in 3m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fe20679d8614290a9f82d4e90ee8c35 : SUCCESS in 1h 19m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c2086e4af014d7896ed9d0dcd7a26e7 : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35fb792772014457a9bb24b0deba9346 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cd6accf9cfc4968a210aa9342d715c0 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dee5e51241834c7b811c086983851b00 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd695a9131964f8badf337859eac07bc : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8534d8d7f53b40878523819cd894e545 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad45ad4597064e0da42c31067eb6cc89 : SUCCESS in 3m 44s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/9f73141a48e646048ea5365d0fff6e2f : SUCCESS in 5m 24s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/8482751a815547c7bf4365f1a9d7baa9 : SUCCESS in 5m 38s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/2b562aa6f46847b3af224d8828ef9112 : SUCCESS in 5m 40s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/4d61307b4bdc4707902af5cf95101e77 : SUCCESS in 5m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/27c17cd1c74e445ea5fb4f457677cd51 : SUCCESS in 1h 26m 08s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e7656c17631a4c13a32f3682f0784d72 : SUCCESS in 1h 17m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7835f4c4d77b43fd9fc516b6dc74a622 : SUCCESS in 1h 21m 15s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/12aa204787904c3d8a2465275be658ce : SUCCESS in 59m 33s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e649d235877b477398610c941c4448ea : SUCCESS in 1h 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d5c2798ee0a6df4fef96b65f31f7d5a6095933c1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-26 13:03:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2eec76314b39679cb6a687190cd4b27eb88c9274","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-07-27 07:54:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"78c080717c15c2290a628c9cc4268385209784a7","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-07-28 14:36:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8e5876516b155ba5795a57f9c268582eb3c77698","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 14:36:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"08122e1ae76eb1cd45b48296248c9e7da7315349","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:25:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d39adfc87044f0b3253042e0ecb3714d70263a04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:25:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/88112461ce4e4763976c4bb36b4c4e45 : SUCCESS in 3m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81e19baed14d481581b8810c9484305b : SUCCESS in 1h 09m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d67972698134a82826c2535e5a0c10d : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b7d7d3ffbbc451587c2ac091e4adb2e : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/696c07f027574991b50cb3a42210983e : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7887013f1b504181998fb98818c53ada : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80ea7bbb2b5243e19ba6c2ae39232f79 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71c38da0945941a48600d10ebb280737 : SUCCESS in 4m 33s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/1f46d9e5bec547448dff034df48870f4 : SUCCESS in 5m 42s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/817d02dbb3e0415d88ebce745e13008d : SUCCESS in 1h 21m 03s","accounts_in_message":[],"_revision_number":2},{"id":"4d6dc1e6f38db061f11328db0f30f4585252be9a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:34:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b79103bf23b042c88aaef00c7fc8ca45 : SUCCESS in 2m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d50a242a641646a19e1e84cbe6f46bc5 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6659cb8d192d127f314dcf8e6036b2f59b3c3540","revisions":{"09f098c519fb8446c5612db3f34174078f245ec9":{"kind":"REWORK","_number":1,"created":"2020-07-24 15:34:38.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/17/742917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/17/742917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/742917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/742917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/742917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/17/742917/1"}}},"commit":{"parents":[{"commit":"a5ae9f4aa0e18787b4bb676b3e725c225deae96d","subject":"Move legacy grenade jobs to Zuul v3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a5ae9f4aa0e18787b4bb676b3e725c225deae96d"}]}],"author":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-07-24 15:34:30.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-07-24 15:34:30.000000000","tz":120},"subject":"Bump bandit version","message":"Bump bandit version\n\nThis patch bumps bandit allowed version to \u003e\u003d1.6.0,\u003c1.7.0 in order to\navoid the errors detailed here https://github.com/PyCQA/bandit/pull/393\n\nChange-Id: I9235560667f664643007b8ca0be1707eab4126ad\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/09f098c519fb8446c5612db3f34174078f245ec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/09f098c519fb8446c5612db3f34174078f245ec9"}]},"branch":"refs/heads/master"},"6659cb8d192d127f314dcf8e6036b2f59b3c3540":{"kind":"REWORK","_number":2,"created":"2020-07-25 09:48:02.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/17/742917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/17/742917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/742917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/742917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/742917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/17/742917/2"}}},"commit":{"parents":[{"commit":"a5ae9f4aa0e18787b4bb676b3e725c225deae96d","subject":"Move legacy grenade jobs to Zuul v3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a5ae9f4aa0e18787b4bb676b3e725c225deae96d"}]}],"author":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-07-24 15:34:30.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-07-25 09:47:53.000000000","tz":120},"subject":"Bump bandit version","message":"Bump bandit version\n\nThis patch bumps bandit allowed version to \u003e\u003d1.6.0,\u003c1.7.0 in order to\navoid the errors detailed here https://github.com/PyCQA/bandit/pull/393\n\nChange-Id: I9235560667f664643007b8ca0be1707eab4126ad\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6659cb8d192d127f314dcf8e6036b2f59b3c3540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6659cb8d192d127f314dcf8e6036b2f59b3c3540"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
