)]}'
{"id":"openstack%2Foslo.messaging~747395","triplet_id":"openstack%2Foslo.messaging~master~I32a6c4ad0a456282ec02b5e4c8309489b3c17553","project":"openstack/oslo.messaging","branch":"master","topic":"rabbit-pthread","hashtags":[],"change_id":"I32a6c4ad0a456282ec02b5e4c8309489b3c17553","subject":"Run rabbitmq heartbeat in python thread by default","status":"MERGED","created":"2020-08-21 12:14:39.000000000","updated":"2020-10-15 17:54:54.000000000","submitted":"2020-10-15 17:53:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747395-1602784403765-349dc48e","meta_rev_id":"eebbedc10054509335656efb84f4916c43ec5b7f","_number":747395,"virtual_id_number":747395,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-15 17:53:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2020-10-15 16:21:12.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-09-03 14:54:35.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-15 08:50:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":1,"date":"2020-09-24 10:25:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2020-10-15 16:21:12.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2020-08-27 10:23:30.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":-1,"date":"2020-10-15 16:21:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2020-09-04 13:42:02.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":1,"date":"2020-09-24 10:25:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-27 09:40:42.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-09-03 14:54:35.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-09-04 13:42:02.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-09-24 10:25:55.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2020-10-15 08:50:17.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2020-10-15 16:21:12.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-10-15 17:53:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e8a47e8bb8761fffed2790282097bf0eabf32f8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-21 12:14:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a5f7715b5b957ee0a7f22bee8e30e8ffb65c598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 14:01:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2aaac9fe7fd43d2abe93949a4ca7d46 : SUCCESS in 1h 30m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06df3d3162b14aecbc471ca49a50afc3 : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28aa1ef76eca4fd39c65a24e0c55e71c : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be7684fdf64c4690870f9f96fa2cfc8d : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39450aadbbd04a8eaf3be3d0e1566c16 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/041ad7a332ce45508f443fca2b09bf55 : SUCCESS in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/587980cfa4c5452f872b3e4428318f65 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44979f927f70455c9675f807cf496132 : SUCCESS in 4m 22s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/d72a66c7a14d468facd2dd43919befb7 : SUCCESS in 6m 05s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/70f44a7daecc4a1e86509bb8ec2e4030 : FAILURE in 8m 24s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/964735225c664e4fb84818813edbc825 : SUCCESS in 6m 31s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/ce383b7088c5449b90e71a3a9e55748e : FAILURE in 6m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/fcaf0a60452d4f868c8a83fcff5c6cd2 : SUCCESS in 1h 26m 23s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/16d9caf587f24b29a2a18166523634c1 : FAILURE in 1h 42m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c9b6300abbcf4f7b9f29f7337ba12866 : FAILURE in 1h 38m 38s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/46334783fa0c4e5cb7714f05ebdd834e : SUCCESS in 1h 06m 10s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/180620ef94ba4d3792c524f26c38e46d : SUCCESS in 1h 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9e6c01a7eb86f5aa7834ffd602d55f170c81997","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-08-24 15:19:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"089f0926e2d70c9cf8cd48c5632c1189d5e45bb5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-27 09:40:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)\n\nreno needs some work, I think","accounts_in_message":[],"_revision_number":1},{"id":"de6a6899c3aceb47dd7723f75c3c1b910c5442a3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-27 09:43:01.000000000","message":"Patch Set 1: Workflow-1\n\nwe want to wait a bit before merge these changes","accounts_in_message":[],"_revision_number":1},{"id":"1666b7d4c059fdf5a33663219fc32f68b5e556c9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-27 10:18:12.000000000","message":"Patch Set 1:\n\n\u003e we want to wait a bit before merge these changes\n\nfor further details please take a look to http://eavesdrop.openstack.org/meetings/oslo/2020/oslo.2020-08-24-15.00.log.txt","accounts_in_message":[],"_revision_number":1},{"id":"192f1c5f155e0dbf6943d632386aebeefb1a02e0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-27 10:22:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"26e90a70edb50e846df7af114300198dd8ddf3a6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-27 10:23:30.000000000","message":"Patch Set 2: Workflow-1\n\nWe want to wait a bit before merge these changes\n\nfor further details please take a look to http://eavesdrop.openstack.org/meetings/oslo/2020/oslo.2020-08-24-15.00.log.txt","accounts_in_message":[],"_revision_number":2},{"id":"9ff583d86d40d34e459a50506760bc92ae832c37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 12:15:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/029f67aaa62e4f07bf4e1faa8b55a60a : SUCCESS in 1h 17m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b53df636ad844dc9bfdb73779aaff22 : SUCCESS in 7m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0068b6c891c459ebf9d6ec118039116 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ea716d3a6c48fd86b25f51b201f40f : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac5e73dc877e45f086c142f0f62f7f02 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/362a7de1a3cc4fe6a20a49a6af127b7d : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8937bbd56a0649c4aaca3b7fc7bdd15a : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa7b2d27bd8749bab6ca1b00c4b7fe70 : SUCCESS in 6m 00s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/515b6d47579147448218297a06964021 : SUCCESS in 6m 16s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/4014f9bda6b84608a6ca5cd7f9cfc81c : FAILURE in 8m 10s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/82808a765889461aab4002301c1df9ba : SUCCESS in 5m 52s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/c5160a4892bf4494a8d736f4771139ad : FAILURE in 6m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/708249b792fd4646b45a644e8100e2bc : SUCCESS in 1h 22m 50s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ac966288664748c0867e1632163f7cab : SUCCESS in 1h 25m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/4540f892b3734af4975728bc8100890d : SUCCESS in 1h 25m 53s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/a947c840909b4b49911f0f4f6553d17b : SUCCESS in 1h 10m 39s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b40fedb884c431eb8bc1e7d29596992 : SUCCESS in 1h 09m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f06b8f54beeb774362a461ba41631263ada07544","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-09-03 14:54:35.000000000","message":"Patch Set 2: Code-Review+2\n\nLet\u0027s land this now unless anyone objects.  I\u0027m good with it.","accounts_in_message":[],"_revision_number":2},{"id":"9295d2300677f871b7015c8df34effda0478dd66","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 07:32:42.000000000","message":"Removed Workflow-1 by Hervé Beraud \u003chberaud@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"5b58f1a1752f2cae2d4a097e6ff0c29f44fe30e9","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-09-24 10:25:42.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"e83eefe32401c37a48dbd129024447989b5d5fd6","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-09-24 10:25:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e33693ddd40a8827d05da439c8675439cfd8d904","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-09-24 10:26:22.000000000","message":"Patch Set 2:\n\nbackground http://lists.openstack.org/pipermail/openstack-discuss/2020-July/016128.html","accounts_in_message":[],"_revision_number":2},{"id":"cc15bdd69de171d9aed9a5815f41e6e2f92514ef","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-15 08:50:17.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cc90eaccc45b1871b110f971023cd380412e803b","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-10-15 16:21:12.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2 Workflow+1\n\nLet\u0027s get this in early in the cycle so we have time to tease out any issues.\n\nThis should not be backported as it changes the default behavior, and any users who want that can simply override the opt value.","accounts_in_message":[],"_revision_number":2},{"id":"e6e8caccc8224c2b872ac61a8d8a157b6b106fbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 16:21:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ba22d40b0288edc231961aa549f63e07b46718e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 17:53:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4ed44ed4926a4394b17d5f79268339715a928c8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 17:53:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fa41ffe06994aa1a65f9fccde3e73fb : SUCCESS in 1h 24m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d993d0e892343eb86c1e664afd8632c : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e82d8fb683ba4149aa54411135b85dc2 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2d0d139ce14447a951b0a05d72707ca : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c66c21a7c2c3448c9da593bc1314b98c : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa8b4ec8f58d4388bd6e92c33de30cd6 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df87dae534d2480e9456682a4f9e4c53 : SUCCESS in 4m 20s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/104d15cf57b3401ebbb02005b33f9741 : SUCCESS in 8m 45s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7115cb477b9746daadcd031264721425 : SUCCESS in 1h 23m 26s","accounts_in_message":[],"_revision_number":2},{"id":"097efa35463381cb961b4a1b89cce461865e6bff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 17:54:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0893fa90cd04313ad184deaf3a32f15 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1cc20a7d9494b519143b577a569572c : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"add5ab4ecec090efdb9864bc9385f871f2dd082a","revisions":{"264fc2780cca45d3e4ad17d973ff2d525174e627":{"kind":"REWORK","_number":1,"created":"2020-08-21 12:14:39.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/95/747395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/95/747395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/95/747395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/95/747395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/95/747395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/95/747395/1"}}},"commit":{"parents":[{"commit":"8d78ab2339d5798c7d8f5c49eaf6d2ef18859a07","subject":"Merge \"Cancel consumer if queue down\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8d78ab2339d5798c7d8f5c49eaf6d2ef18859a07"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-08-21 12:08:15.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-08-21 12:13:14.000000000","tz":120},"subject":"Run rabbitmq heartbeat in python thread by default","message":"Run rabbitmq heartbeat in python thread by default\n\nRemoving the experimental nature of this feature and activating it by default.\nNow to run heartbeat in a green thread users should set this option to False.\n\nAlso deprecating this option to prepare future removal and force to always run\nheartbeat in a python thread whatever the context.\n\nChange-Id: I32a6c4ad0a456282ec02b5e4c8309489b3c17553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/264fc2780cca45d3e4ad17d973ff2d525174e627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/264fc2780cca45d3e4ad17d973ff2d525174e627"}]},"branch":"refs/heads/master"},"add5ab4ecec090efdb9864bc9385f871f2dd082a":{"kind":"REWORK","_number":2,"created":"2020-08-27 10:22:55.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/95/747395/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/95/747395/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/95/747395/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/95/747395/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/95/747395/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/95/747395/2"}}},"commit":{"parents":[{"commit":"8d78ab2339d5798c7d8f5c49eaf6d2ef18859a07","subject":"Merge \"Cancel consumer if queue down\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8d78ab2339d5798c7d8f5c49eaf6d2ef18859a07"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-08-21 12:08:15.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-08-27 10:22:41.000000000","tz":120},"subject":"Run rabbitmq heartbeat in python thread by default","message":"Run rabbitmq heartbeat in python thread by default\n\nRemoving the experimental nature of this feature and activating it by default.\nNow to run heartbeat in a green thread users should set this option to False.\n\nAlso deprecating this option to prepare future removal and force to always run\nheartbeat in a python thread whatever the context.\n\nChange-Id: I32a6c4ad0a456282ec02b5e4c8309489b3c17553\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/add5ab4ecec090efdb9864bc9385f871f2dd082a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/add5ab4ecec090efdb9864bc9385f871f2dd082a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
