)]}'
{"id":"openstack%2Foslo.messaging~749193","triplet_id":"openstack%2Foslo.messaging~stable%2Fstein~I2ae53340783e4044dab58035bc0992dc08145b53","project":"openstack/oslo.messaging","branch":"stable/stein","topic":"stein-cancel-consumer","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-08 10:28:54.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-08 13:49:10.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-07-08 13:49:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2ae53340783e4044dab58035bc0992dc08145b53","subject":"Cancel consumer if queue down","status":"MERGED","created":"2020-09-01 07:29:52.000000000","updated":"2022-07-08 13:50:15.000000000","submitted":"2022-07-08 13:49:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749193-stein-cancel-consumer","meta_rev_id":"78121a02e3f6d85092b60ce2ece5f0f78a73b1bc","_number":749193,"virtual_id_number":749193,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-08 13:49:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2022-07-08 13:50:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2022-07-08 13:50:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2021-04-15 14:00:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-28 23:14:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2021-04-15 14:00:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 07:29:52.000000000","updated_by":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"reviewer":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"state":"REVIEWER"},{"updated":"2020-09-01 09:34:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-03 14:31:10.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-09-09 16:58:53.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2021-02-09 13:26:26.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"},{"updated":"2021-03-20 18:34:06.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2021-04-15 14:00:10.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-04-28 23:14:51.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2022-06-28 23:29:23.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"4ca9234a9f828916dfad00e7079c5532b1257ea0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-01 07:29:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e31a23e8e59fa2fbbee68f1f6ec1eed316d32d15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 09:34:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/7145f95abf6b458ea4b038ae71360155 : SUCCESS in 1h 53m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2785aa7c4d142b38561c1f6bfa60e7d : FAILURE in 21m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab2e7d3ef6b241d8811c46abd324699a : SUCCESS in 7m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/449a40153936422f859870f8152c2835 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b68ff4410b6c499b91548d956ed39b68 : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6d093a3dc07477287b82f3e29b625f9 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/827acbb9cdde4b9594fe1ef22167f7c1 : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65ef455be899428f9d6096e3f8a4c993 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4156c843a23a46f1b08be59edf87aabe : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46043b10ae274827a3b7b11443fa98a0 : SUCCESS in 4m 50s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/8d4189971ad5400fba5e36c5959d6c3c : FAILURE in 7m 06s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/c7335b386afd46eba7453aab30449505 : SUCCESS in 5m 26s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/cafe6dd1e378431e81ce2a0ca3af5644 : FAILURE in 5m 40s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/29b2ae96d88a4a0180f6981a67948ea0 : FAILURE in 7m 27s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/bef8c6b99dfa439fba7a30134cf6cc05 : SUCCESS in 6m 47s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/4c2154ee15ce438fa674dc37537e96a5 : FAILURE in 5m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/aa086cb5a5f84b018baaf471bb178f69 : SUCCESS in 1h 23m 01s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/dc63067dc2b449edb0ea4b5e0ebf0fe1 : SUCCESS in 1h 35m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/45ae0df71bd746b88ba5a500a6d712b4 : FAILURE in 13m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3c6902e6be8f47f79cc16448029da3fb : SUCCESS in 1h 22m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/51e80d66d1f24d979d07f1728009e693 : FAILURE in 11m 52s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/c528e721fae74f49b30c8e22fdffec13 : SUCCESS in 50m 57s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/6e4810b2285a46059da6ba6abdd2da96 : FAILURE in 10m 55s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/a39cf942405e4a7897f99edfee31cd09 : FAILURE in 32m 13s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/eaca0c72f420468590387280fc46e96a : FAILURE in 30m 35s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/6bc6c31e65c24948b5584d2257f8c63e : FAILURE in 22m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae36a91d7a4a224ef9477d1e99e47fa3b165fe7f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-03 10:43:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"333a9c513316de4205b5dd3ecb699471f247e5ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 12:19:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/949a8f01d9ec47dfa91eee644ab4ece6 : SUCCESS in 1h 22m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/596e4c96ce1e402a9c0db63a1a9b3fdc : SUCCESS in 1h 19m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c2d308b873b48b59e9a0050749ae6c4 : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17145d06033e431d9383551822798289 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ab930ed34d948508fc4a2beaccb2f39 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61db3aa885ce4d2988d30729de1514ea : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dda9a5bd156f4a44b1e9a8812e2cea7e : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2502cecc44874445a2ddce7c5d92891c : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9822782a174b4c6996d70d399571a621 : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbaeca9fa8334a5a9af00a732477c460 : SUCCESS in 4m 05s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/b136219e1e0144219cb9c1e513854f3e : FAILURE in 5m 29s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/6467ad26dad14528a5f34243eb08eada : SUCCESS in 4m 40s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/95d3127ce5c646d6b6a33ee2cc72fb60 : FAILURE in 4m 08s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/e1fe1698826e4e958c90965a50001e85 : FAILURE in 5m 51s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/8c7c36bea4cb41e98992d1c3cdffe98b : SUCCESS in 5m 05s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/1b42abc2336e48a3b7ded0361c29e260 : FAILURE in 4m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ba15c00f9e764605b6e4f3f0e74209b2 : SUCCESS in 1h 24m 27s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/9080cfd9844c46f8a13af85aa124040e : SUCCESS in 1h 22m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/f9ea34ab17ec4decbddd8a1c88728feb : FAILURE in 17m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/efc6f07a50df411082eaf0ca8927b4e7 : SUCCESS in 1h 26m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/13cd8efbd43e45689d6a6b795195a370 : FAILURE in 19m 21s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/74cd57766662447bacc305c735f1f236 : SUCCESS in 56m 01s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/7f5555dcf7254f4e852f193cf1eb8ae6 : FAILURE in 1h 00m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/329b35ba9a5646889b1adfa300ece5b1 : FAILURE in 29m 46s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/177c52dd0b164561ba774fbfe99f8ad9 : FAILURE in 27m 50s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/db98228c88ae48abae6a27f85999be6e : FAILURE in 28m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3375ee1015e25ebe2ad9e7394866b295d5381c48","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-09-03 14:31:10.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"81e992548f946c9eb538e9efbe21d4d5d2df2ee2","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-09-03 14:58:02.000000000","message":"Patch Set 1:\n\nHmm, for some reason the failover functional test is failing. It doesn\u0027t _seem_ related, but any idea what\u0027s going on there?","accounts_in_message":[],"_revision_number":1},{"id":"12191b5263f31df3f5e2cf3ae1bc0067cd1fb2ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 16:17:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8954fda797fb4285ad9582486434bb8d : SUCCESS in 1h 23m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab820820e74841d8af9658d4523db2fc : SUCCESS in 1h 34m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/539b02743e28429caed554001aba2104 : SUCCESS in 12m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9d99db4e76845909485b4fe5ca7adce : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd67a0279ef4dcfa984cce9fbdbdb23 : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45c4aa3a7e454229bc271de967754ea4 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b3af5830c7a4fcc8fb9e207c950ed9f : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa6e0d237e8d4f24954ce1b567995aaa : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5882050fed2f425a8f334a1a65fd1a90 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4809b2a38ed146eb9033b0618caebd3c : SUCCESS in 5m 57s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/ce567ab04ee249a69ba0ae4f0dc6215d : FAILURE in 5m 40s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/444e4be532cd4c4682bd4cf42f10762e : SUCCESS in 6m 12s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/f8fb8fbf4662435f881d5fcd35b598bd : FAILURE in 6m 32s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/4bc9eee355f841d28d1a713cc477179f : POST_FAILURE in 36m 38s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/56998a6f396f44f19c46d69ccec261a0 : SUCCESS in 5m 10s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/2bd7ab89b321467d87bffd11302cf911 : FAILURE in 6m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/835d6a7b1aeb4841a2dd8a20c5e8b532 : SUCCESS in 1h 37m 04s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6655b2ed561544ae949cd7c4d16c4fbe : SUCCESS in 1h 19m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/941ca7564bfc42ec90c7db845d79d0f4 : FAILURE in 18m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/0e7859aeff644c7d82d859d9437fb1ea : SUCCESS in 1h 23m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/719222996689480fb649705040f728ba : FAILURE in 20m 22s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/8e99cc89bb5b4b42957d6043bb5854c4 : SUCCESS in 56m 50s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/869612d2ffee4961a6cf0e73f6d212f4 : FAILURE in 1h 00m 31s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/849335058cda47b4a7a6813cae0566ea : FAILURE in 50m 54s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/a2f817da30f54261b483c0a1e9fbeedd : FAILURE in 46m 21s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/9eabd14147e54322be9b4513152d07fd : FAILURE in 40m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a91054c506bec5025ec6a927ef8db6e6858fe0fc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 07:25:34.000000000","message":"Patch Set 1:\n\n\u003e Hmm, for some reason the failover functional test is failing. It\n \u003e doesn\u0027t _seem_ related, but any idea what\u0027s going on there?\n\nit looks like to pifpaf issues, I don\u0027t think this is related to these changes:\n\n```\n2020-09-03 14:46:17.495410 | ubuntu-bionic |     Traceback (most recent call last):\n2020-09-03 14:46:17.495442 | ubuntu-bionic |       File \"/home/zuul/src/opendev.org/openstack/oslo.messaging/.tox/py27-func-rabbit/lib/python2.7/site-packages/fixtures/fixture.py\", line 197, in setUp\n2020-09-03 14:46:17.495473 | ubuntu-bionic |         self._setUp()\n2020-09-03 14:46:17.495493 | ubuntu-bionic |       File \"/home/zuul/src/opendev.org/openstack/oslo.messaging/.tox/py27-func-rabbit/lib/python2.7/site-packages/pifpaf/drivers/rabbitmq.py\", line 136, in _setUp\n2020-09-03 14:46:17.495514 | ubuntu-bionic |         self.start_node(n1)\n2020-09-03 14:46:17.495533 | ubuntu-bionic |       File \"/home/zuul/src/opendev.org/openstack/oslo.messaging/.tox/py27-func-rabbit/lib/python2.7/site-packages/pifpaf/drivers/rabbitmq.py\", line 86, in start_node\n2020-09-03 14:46:17.495553 | ubuntu-bionic |         wait_for_line\u003d(\"completed with .* plugins\"))\n2020-09-03 14:46:17.495573 | ubuntu-bionic |       File \"/home/zuul/src/opendev.org/openstack/oslo.messaging/.tox/py27-func-rabbit/lib/python2.7/site-packages/pifpaf/drivers/__init__.py\", line 259, in _exec\n2020-09-03 14:46:17.495592 | ubuntu-bionic |         % (wait_for_line, b\"\".join(lines)))\n2020-09-03 14:46:17.495611 | ubuntu-bionic |     RuntimeError: Program did not print: `completed with .* plugins\u0027\n2020-09-03 14:46:17.495648 | ubuntu-bionic |     Output: ERROR: node with name \"pifpaf-1\" already running on \"localhost\"\n```","accounts_in_message":[],"_revision_number":1},{"id":"998cabbabd18ab6a944c49210c86bda70b7ca49e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 07:25:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6bf5ab3f3d18bfc8f1578bd335215eebda339c4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 09:15:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/626e206f560a4ce99970bc386706071f : SUCCESS in 1h 43m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/474bb5fba8284f11a2e158ee820925ef : SUCCESS in 1h 27m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f144e3eb9f64fd6a32130d086fc6550 : SUCCESS in 6m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/132df05512f043e8a14f3207202c24bc : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f3d797c30c04ec1bb279452f84e24ff : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/684551e37d744146bbc1ea3e4fbc867f : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de6e42f9287746caa3c0488864ec91ec : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/748a970f65ed4e6cb0870c9e796bfc52 : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23f35b8ec68f48dd844657cb64a6a000 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1933d789cbd54798a7b0e20ed352f778 : SUCCESS in 6m 33s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/8b4fad0e9b8e4e80af601b126bf3f5e9 : FAILURE in 8m 00s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/c63621cc2c0c4f1da542b23e958aefff : SUCCESS in 5m 52s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/f7ca052f966f457aa1b0b75eb67af52c : FAILURE in 6m 49s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/4c0b6354dba147319af0977063a5a6a2 : FAILURE in 7m 25s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/3501a97892d34fb9ab4376fc03745d2b : SUCCESS in 7m 58s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/55963e1ba35648b09beabffea255aa1a : FAILURE in 5m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e0c4c585d34b4b068851afabe7c0203a : SUCCESS in 1h 22m 49s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/fefe105a1e69454a9a374e5a15ffd08f : SUCCESS in 1h 34m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/9b216b4101a94fdbbabc016897dac504 : FAILURE in 17m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2fd46f5d9f4947e185c8b480acb14fdf : SUCCESS in 1h 03m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/9c0d933208ee453a8bcbbc41cbf4ee2f : FAILURE in 11m 22s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/fd3d4dfd4aad45088b1a643e5136655a : SUCCESS in 1h 00m 30s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/927693c3eed443b3a1567da209014359 : FAILURE in 1h 02m 45s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/6c7949c8ded146a19bb165149744fa30 : FAILURE in 29m 03s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/3d8abd0f740142b8bd44674509fc393d : FAILURE in 31m 44s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/8d51ef23d3954e96b4bdf3977672c3b9 : FAILURE in 39m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa2baaaa156ec13e079231422fabd4b36d5459c2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-08 09:35:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0bf9c24df92955d5cefe39b61aed9c290c5b60e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 11:20:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/05ff305ca48741659feafa58e3ea42ce : SUCCESS in 1h 19m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0fd469af2df402c8cacf1900d9541fb : SUCCESS in 1h 09m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da3fc946334a4987865628bff0f69de2 : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29064a24e23847a9b1e73267b93c051a : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a5155f5117c481ab792b3c995d92fd4 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97d0aa6fb0e84b5ba134043e18d974a5 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2484cae8c4c74828a27d0f3744e4c1d9 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b22b403d12614df78c065b50b468c1af : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f71e0d85d064c2a8774dda167b45983 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/097eb98ddeef4a5e9a9829ad1ed019b9 : SUCCESS in 4m 56s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/df5c0a0c311146d085dc894579572514 : FAILURE in 9m 04s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/95cc8a6bdcf74732aba8e6479ed69265 : SUCCESS in 6m 23s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/6c7375f1401c4d8ab26eb548f72813a3 : FAILURE in 9m 56s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/078e6748c5c940faa0e7a47ba45f0d7c : FAILURE in 6m 50s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/4bd77e1ae5dc4246a220741558f48c80 : SUCCESS in 7m 02s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/bc12b9e47bf7447fb3aab03cf3a727c7 : FAILURE in 5m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bb2084758a2e478095806cbb1f928283 : FAILURE in 1h 39m 20s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3ce1cbac4a3a427992a5550061e04fab : SUCCESS in 1h 12m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/037b6e9a1d7a4341b826486c9d7a6066 : FAILURE in 13m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ae97e9f52b0d4e1780abe23dcafa805f : SUCCESS in 1h 17m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/65ddd23bd0334b5599d916af5873c059 : FAILURE in 13m 32s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/2a783129eecb4be3b3e00d8b3b5ac590 : SUCCESS in 1h 09m 04s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/2db83a0d8c1548be914f9e9363f7baef : FAILURE in 1h 08m 40s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/1a626b986eaa4be7b89ae6849d7e853f : FAILURE in 37m 14s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/ae2519de22bd4fafb722e3fc18612ee9 : FAILURE in 40m 16s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/801a9e4e0f454899b6fc68d036842eeb : FAILURE in 39m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5322b767c22ae8ba66c8893d7743ec36cdbf9bdf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-09 11:41:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bd4ff47700116f8f69fa0d23f85a27a8a40e2293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 13:46:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/a13ea6c4779a4d4589c4bb6cac8edcd9 : SUCCESS in 1h 36m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1813f2b8e9a4f8c91302777c8326ddc : SUCCESS in 1h 15m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df70ff1e3f9042e08be923a49ca9559f : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/541c0b0587cd45d6b27042d8b8d892d3 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9636583b47254b89970d9d5aab527a79 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5af7fc29724b4f5fbdfd74b97c4cf540 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce69a612a1274593bdd81b4c7d13dd4c : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4dc2fb408be143c09b149f1713626b5d : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a62d4cf7d64438aa3705822dbfc827 : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8ae4010937a401881481174abfc1480 : SUCCESS in 5m 25s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/45e1f0b5ef914ee3a13378077bdaa2f9 : FAILURE in 6m 01s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/43c17ef8a170480ab4fe1192b4baf533 : SUCCESS in 4m 43s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/90180c19e9ff41efb6899facce704bc4 : FAILURE in 5m 25s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/de6f149ce7c742bea3aadb620de5cc57 : SUCCESS in 6m 45s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/07763bc5289f48b394409c23c9f151d6 : SUCCESS in 6m 15s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/5b8c5fc114384bd1bc4a62ac0735ec69 : FAILURE in 8m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/408fca8684cc4d80a9ddc7e3668edbbc : SUCCESS in 1h 42m 42s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/5c88e97bc8294046be5ea5ba3fe10d49 : SUCCESS in 1h 45m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/3725186aa33443f7a3e1dacc0d621d92 : FAILURE in 17m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/65f9521d8320466f89f2b28378ed059a : SUCCESS in 1h 35m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/62fa59f7f38245d3bb81a3fa2f395fcb : FAILURE in 18m 39s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/8a73a93c4b974d00997188ccfaf634e4 : SUCCESS in 58m 34s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/e60291684dd1446692e42c150b3a52a7 : FAILURE in 1h 04m 07s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/e6e4502f8f4d4648a2937eea22f81228 : FAILURE in 32m 06s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/65b83729815648b2ac154764e0cda7ed : FAILURE in 34m 07s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/6f5ba752179c48348cf8b1bfda6b33ec : FAILURE in 35m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77a1cc2214aee7ad82cae038f37f0da11d628d62","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-09-09 16:58:53.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1662955fd6222d99084beda2618d02eeb03893f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 19:09:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d3feae0cdf0647969aec7f4021bee607 : SUCCESS in 1h 41m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b0d7a6f48b449f8a58c432dc75a67e3 : SUCCESS in 1h 54m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d03ddadce234d91b8227042dbab498e : SUCCESS in 6m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ceff51aa14243318a46626e37736ecc : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a4bc3cfb9a846609f188ff8432969b6 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5606030f030a44dfb86323aba19fd427 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40938ae2114547f3b7c2075076b0bf96 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8967e8fca6e546f18638a390aa8c64dd : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b77d1f6b634e4bea980d131a8d03cf15 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88778b40fa20470f8303d584b63a89fd : SUCCESS in 3m 52s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/c35c975c835d462ca951d6ace2f9382e : FAILURE in 5m 02s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/2368159331dc4c41b361f2800bbef7a3 : SUCCESS in 4m 42s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/9fd20fafcf074fd9ba9f79551ca8daed : FAILURE in 4m 08s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/b6cb5efb21454de7bc85a000739168a3 : SUCCESS in 5m 26s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/147c385971384b5a8f941716c4f640ee : SUCCESS in 5m 02s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/c063eaff368e463cb110571a4525c3e1 : FAILURE in 5m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3ac13bf2cd8e4664bca66a0bb00718fa : SUCCESS in 1h 30m 48s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d36dd293510d4a3c92383eb026d3815f : SUCCESS in 1h 43m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/a4b4e2b0153d474ab9868f6f399a7ab9 : FAILURE in 15m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9e83e4be92364bdcafc58f4dfaa8804f : SUCCESS in 1h 43m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/5b2505a49fac4f64b1af01e19d62b2ae : FAILURE in 9m 09s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/05ad388d0831414581914f75e5e6ec95 : SUCCESS in 46m 45s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/f0a063d463e44c88b39cdba9df5bced8 : FAILURE in 49m 58s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/20819bd4970d4fb0b5068c2b0c24c3ba : FAILURE in 25m 08s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/c756e9c723354b5a97e80f862ffaefdf : FAILURE in 25m 34s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/e17d7075360946afaaa3f3a8361b06b7 : FAILURE in 25m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef4a44d751e8bbcbed8e1c8ea8b8e33c9e045943","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-15 09:23:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bd5b80f4519d171b5ce731be11b4c1520cd156ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 10:52:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/be461fbbadf3468292e671fcf2f1d204 : SUCCESS in 1h 10m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/617352dc2f874a03a3f1d3c5bdcfadf6 : POST_FAILURE in 29m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/328c8247c35d4c2194d6cae386989744 : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09ba2fb40a9340ef82734f806e9981ba : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/525163f5d5a14a6b8069ed52d2ba304b : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af438ecd12344e3daf92840e7d32e481 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35b649f370b0495cb44b1447d7824040 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/adeb513d1ab54cd3ad178a01b33f310c : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4adff121495c40fa8cd8bd8becb13952 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b31620d337f64a3f8b5761de6922297f : POST_FAILURE in 4m 00s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/28abe9cdcdfe410fbd71a65e6a0bc7df : FAILURE in 6m 08s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/9350db81497f41a1ba18f50dff481563 : POST_FAILURE in 5m 26s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/5ba281d623164d408050ec09a20c3b78 : FAILURE in 4m 04s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/9671506cc760405694e79a3bea3a56b4 : FAILURE in 5m 06s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/5420393928664ee6a1d417d0087e5fad : SUCCESS in 5m 14s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/5880c22a5f3245caad2891d05e923779 : FAILURE in 5m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1ac5a81cc6d842bb83552984b8711eb4 : FAILURE in 15m 56s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/987f87f3aaea448cb7bc84252d4b91c9 : SUCCESS in 1h 23m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/e0b3858e110d4042a21ee614afaf3f0c : FAILURE in 8m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f9d4d11a16714528982b068a4a3d959e : SUCCESS in 1h 13m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/5a1d9ee4783541ef891375682088bf85 : FAILURE in 11m 29s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/2b33ee4df314491b89e6368979e25834 : SUCCESS in 1h 02m 40s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/e59b86f2dd2f4949abff51ed8cd6331d : FAILURE in 58m 55s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/97307faaa4aa4b9f93f978fb237f34dd : FAILURE in 23m 58s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/2ec392b6a7f243cebe7f260d601e734a : FAILURE in 25m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/7da65c9bcfca4b10843bbcd93d7b37f9 : FAILURE in 18m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d765556d3667a4d5f5254c06b1ddfd5f7ff9f325","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-15 15:59:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b18f1bb658eba3c8dbd16de41e85657b9249b684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 17:45:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/bf41da77006241d7a608883726684e7e : SUCCESS in 1h 22m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d1a9c193b13481a8374dde5d88056b3 : SUCCESS in 1h 19m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c22edf355014c609162becbfc45aa42 : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15b042a832ea4fb7af5a0b46ee79a949 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bede20c5fef04123a57822b3e30a0fff : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/33abee91bf4b4247b860fbe5e22c4441 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/317ee2264d2b496b988a52b87db6b669 : SUCCESS in 6m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1d4a72527db49339bde792ab3593383 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8292fb35cf054746af746b3e47c5be2d : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9d3d5be12234e7b921c4bf3b1317f61 : SUCCESS in 4m 02s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/e05adfdb14ff4ed8bb03bb37c21eca2e : FAILURE in 5m 38s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/02bc9e7e02d44e91835cc70d26cb090d : SUCCESS in 6m 41s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/a2396753608d440f94cf1b7cdd9cdd80 : FAILURE in 6m 10s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/0cedb46d23bc45f79664416371355d5c : SUCCESS in 6m 41s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/9ac761259ef6446e908ebd1fb1d286c4 : SUCCESS in 6m 25s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/0a1e90766e204bf5a45f03baa784dac9 : FAILURE in 6m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8322d78a5a36449bb757cb32a5f3745d : FAILURE in 27m 44s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6d99a90df4784784838749e43d8eb1e0 : FAILURE in 32m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/8f120a2627f2434491fb487074b5f1c9 : FAILURE in 11m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/18a90ce7677749ed88c26d2e398ffc5a : SUCCESS in 1h 37m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/873839aecac341be98fca4da357ac66d : FAILURE in 16m 42s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/1ecb1f6fadc045379571815706e9fe47 : SUCCESS in 1h 04m 55s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/c829470fc7034ce5a091d0d918a9df1d : FAILURE in 6m 04s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/74f858a3b55f45ffbfdddc89ad731b27 : FAILURE in 25m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/643f8ae8abf0436cb15887d0eabd047d : FAILURE in 24m 46s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/8ef2d94e596b48ad8bf18aec99939f91 : FAILURE in 34m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcbacd9d6ef3522928a8e9a530898276ba30d9aa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-16 10:05:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5087aec50e50e5f13e19d3e03b0aff23a838f430","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 11:52:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6467e5ba079b40a68b80493729b5786b : SUCCESS in 1h 32m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/626d57391db44a2d8f0ff216c24a63d3 : SUCCESS in 1h 40m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9395d3b80824fcf83be905031072746 : SUCCESS in 4m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbb4d5fa77204ad7b4ac4139156465f6 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b464fe36a879423da570d58874198bc4 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d13acc9f88484e9e8587382785846ca0 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c22a05f96c674f4faf2364566d907971 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31fa7f27fd0d400b8ea7be8a24525cdf : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad84a2c0397b45108b4814131554d866 : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a797e81fa8454240a95018e3d2ea0f54 : SUCCESS in 4m 04s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/a7329bfcfff94ad182d75988c114d93e : FAILURE in 5m 42s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/e9037e0b14aa4064a9c3ffe3dddde24b : SUCCESS in 4m 12s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/d1037205a1ca4fa69ff4780c3f7f1a16 : FAILURE in 4m 37s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/720d740068f5434d871c1c0c2e94dc2b : FAILURE in 5m 42s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/7f4a71bbefff49fb8c69fb0235391f76 : SUCCESS in 4m 58s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/57dad75dc32f46b1a0543e125160a72a : FAILURE in 6m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6f96f0e881794e2ba3968af6d219d06e : SUCCESS in 1h 23m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a33fde019b1b4f0d9607363b0c59ebd8 : SUCCESS in 1h 39m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/904c1ffbc83b4109a86d071d366733e1 : FAILURE in 17m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ce4c7c1ff59b4106bc0765ca8fb9b46f : SUCCESS in 1h 20m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/87d380238a784506ba6427df9548700a : FAILURE in 16m 02s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/6a2890d62bcf4a2aa45595c5b1c55720 : SUCCESS in 52m 36s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/60109fc4aaab4c2898b53a85190a2a1b : FAILURE in 1h 02m 02s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/5906f762d2d54e63ad0aba901e109a5a : FAILURE in 24m 34s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/8e97a9d4de2a4c4d916488490bc49cdc : FAILURE in 28m 40s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/1eb78554afdf458490e56234e59c11ee : FAILURE in 31m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7badac9dde0da8681551f7fdf92baeb04b2fdb6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-16 13:01:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9210d110ff1303055a159388305bb000d35c4384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 14:42:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6c9cf729242d481ba31485691519d049 : SUCCESS in 1h 30m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9bd5850961064e428f896b9986027213 : SUCCESS in 1h 32m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/131ba8ee89f9475d81170c4395b0cd5d : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9a71256b5be49239fa0565d503ed895 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a0b5143ea7b421b8f7c6a05ad5b70b9 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/699564c4901f4cbcae120b4569cad2e8 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adad46cf4215413da55a5c240edeb190 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/971c0cc8ca144ed89fe6dd003f360102 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f91b96a25c1471c8ec10812a1929aba : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9196168e1743456e82d6c354c259bb06 : SUCCESS in 4m 46s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/db208b5279334ab49cc1bd05cc5c3150 : FAILURE in 8m 27s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/547f7217b2f5495cb9b4ba77bff0269f : SUCCESS in 6m 10s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/e0268514b19c48b5b748e4f66d3bf272 : FAILURE in 4m 51s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/b17c07d351984d1b918d370a073278d8 : FAILURE in 5m 45s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/edd29d59c99a42a4950ea0e17d53daf3 : SUCCESS in 6m 12s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/74ba7e1b62c84e24b4d2e21667277e5f : FAILURE in 7m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9e99b16014e0418daec7fc7c1db21b6d : SUCCESS in 1h 29m 02s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2c2f190ebf2e48ef9c7347d75227085f : SUCCESS in 1h 26m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/297728ad135d4f38b9ccf86d2438ab6e : FAILURE in 13m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9db281b9f1624dc4bab0d50a059559f0 : SUCCESS in 1h 21m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/ed90b4d062824104abfa7ccb994c0ee3 : FAILURE in 12m 25s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/58be4a993c6d432db250cc689c29d88e : SUCCESS in 52m 25s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/7710b252ddb34534a5d691fea4fc399a : FAILURE in 1h 07m 53s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/a94819915951422d842a11345d585bb0 : FAILURE in 31m 00s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/08f6537a38a149fb8954672518a09f17 : FAILURE in 30m 22s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/43030b2c01e849ad97ee7e66d63fef02 : FAILURE in 31m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73997d9ede5b704ec969adb6ba0545dd22745b21","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-17 08:52:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2f054814123afb2a29f310d8c0a7ce50025c06f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 10:26:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/416082ccdaed4b9abee6a7fcd42dbeac : SUCCESS in 1h 25m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3cc25c4e49f34a818d660c60584a0ff0 : SUCCESS in 1h 16m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e661fca945d4eeeb2c54879743bf58f : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3659a12c69ed47ba8d7233ac594e0356 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85778f8b89a44eb88e738308999c9862 : SUCCESS in 8m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fa293c8de794ca682174859fc3dfd8f : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ca657d399af4681b27119486e662232 : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7e1ae9c548745ed9c6e46d43290a04c : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/105c9f4cc8d042aeb24005a7f440c55a : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea1b626eb5d542fc864cec999c22cfa3 : SUCCESS in 5m 05s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/5adc0e071e2046e6b801db34c24c9b1f : FAILURE in 6m 17s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/7beaac394ecf4a3b9c58c5086a4496cd : FAILURE in 6m 56s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/b1937b27d5064731bfadd45df2066147 : FAILURE in 5m 40s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/59f44301ed9f4cefa0364c0e75d3a47d : FAILURE in 7m 05s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/30a23aaf33014f00b70208a0875cf4c9 : SUCCESS in 5m 46s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/36cc6f2f04144310bf0825e0726412bc : FAILURE in 5m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/779d21683b034ec9ad372ab64a399f33 : SUCCESS in 1h 24m 18s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d3fe41e11d6a4d82a5943ad89d145d17 : SUCCESS in 1h 16m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/476326a7b97042819344c8ada3765dac : FAILURE in 13m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/0e74f49c85324d82974c0fc1eb08168a : SUCCESS in 1h 21m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/d7d1a602be884b60ae2c66a7117d6a8d : FAILURE in 16m 17s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/09b3372dddf64365b16b5418f04ed1ce : SUCCESS in 1h 08m 36s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/fea58b4982fc4176ac5d413765efbd7f : FAILURE in 1h 01m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/4ab4c25c36e14f06b5366deadb0477c6 : FAILURE in 36m 02s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/cdb10fed89b64af0a716a77986667577 : FAILURE in 36m 36s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/0892007a1e9a434aa3120b399f851346 : FAILURE in 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83be6a5cb58e7214bd7a05a3f4aad8aa6276974b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-18 09:59:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"296c87b6426f482430ee17ba6f98390c3fc81519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 11:52:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/210a17db5c374e2f96dc012106766148 : SUCCESS in 1h 26m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/951d0e39e1c2445b936c64cc77d0e13c : SUCCESS in 1h 27m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f25269072dd4e629e00c696be6d5ce8 : SUCCESS in 7m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74145a392bb24113a5e49d9a82b67b14 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fac09bb23734cd5a20a9b44fdf393f3 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/602d6e30360b4a5e8084b9d3659f734f : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b97dfe3861c1449ba78f6941022f97ec : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6404a766dba46cfa7e0f319dc412225 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db7412339bdc47b09c24e8bbdc9cf919 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e6301a8d96e4249b2c4a2845e86afc7 : SUCCESS in 5m 06s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/dfb8c565bf5949cbb606faf6730529fc : FAILURE in 5m 58s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/ea887c5116ff476fbfe32320b4a97934 : SUCCESS in 4m 46s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/820a2b23df61457ab8d40e3e18ed0055 : FAILURE in 5m 16s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/5a8bdf74139743b0ae014c2a117285df : FAILURE in 6m 30s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/e23e32754fd744b398f5086a016fcd74 : SUCCESS in 4m 46s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/9014c999868d4a118d642802d56bafe0 : FAILURE in 4m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/fa44958c0b6d42d48a7dc3a66f0393ed : SUCCESS in 1h 42m 44s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ab80e81746cf4508931f62fac491beb4 : SUCCESS in 1h 24m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/63120bdd439141288feb4383b4c9d06e : FAILURE in 12m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3ef486b1fb2f4076ba1ad28d2d7b7e74 : SUCCESS in 1h 20m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/b16bcd48b354467ba6ac37c245a2deba : FAILURE in 16m 22s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/ba04430bd35f4db5aaaea22afcc395cc : SUCCESS in 46m 37s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/caddb553d2ed46df9196cd0c77eeafeb : FAILURE in 50m 44s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/58dad3a886de4a89b5d40f996d4826f9 : FAILURE in 26m 07s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/50f1545e81294f9abf6706bbc9d38b6a : FAILURE in 27m 31s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/399b94d3dfff470fbc8117b184e9993e : FAILURE in 26m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"368729ca122507a90885c111510f5e40cf935ae4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-18 12:50:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c286177bca41e0c9bfc6e77d153aa9fb54de968a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 14:35:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/30ecb616a1c34f4e9db8843721bc393c : SUCCESS in 1h 11m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3dfaab8304b140b3be045d3bc514ca44 : SUCCESS in 1h 10m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6f6881ebe934ff2a9d8de86a14ea628 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba00743475ad4d20b265f80366779c07 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e452eb2fa7b14766993ec3f1bc12f937 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/949592821f624400b0f100a990e2af9b : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6783f40d0d3748be9073307ee6cfdd9a : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9dcddb1e69af4edebb7acc4fe90db948 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bd99ed98a30400685c3487a708f221c : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/233b36140559435abaa25ec4397f698f : SUCCESS in 5m 37s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/fa45bb433e25492e89a63fe076571612 : FAILURE in 6m 18s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/ec27b4d8a90e4fc7ba02325a216f3bc8 : SUCCESS in 5m 43s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/7660f4c2688f45209c30513e43d6af40 : FAILURE in 5m 12s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/9faf79c2d13a4abfa87686a0b36f9951 : FAILURE in 5m 32s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/c4fdd032fc4b4b799287995696c1d923 : SUCCESS in 5m 01s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/5db101fa092447cc9183a84ef16ed755 : FAILURE in 4m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0accb881899f49f4b334ff84fae4d23f : SUCCESS in 1h 24m 45s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/b5717bb5d5d84bdb9dc33fc686e6135a : SUCCESS in 1h 27m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/47620b1c56074bd285be49dc49f9241b : FAILURE in 21m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/13d173fa4d474b14b91e9c1b33c35229 : SUCCESS in 1h 30m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/0547ea3ac2914b969240949700e4d7a6 : FAILURE in 11m 18s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/c81301c2668643e7a95b54db381eba1c : SUCCESS in 1h 02m 50s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/5d3994f6796648ce8b29d01f84e1e80b : FAILURE in 1h 05m 01s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/abb16784fb3743ce9101607bdc87cd3b : FAILURE in 23m 34s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/75b76af89fd24ab9bdb951316bbca238 : FAILURE in 23m 03s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/11207679cf8645adb231a9ca45f1a7f7 : FAILURE in 35m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44b4361aaaf039ce873ec0c2ed94b82dac296b3e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-18 14:49:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d05a7f155cbca07aa4435382e1af357bf29f8a3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 16:27:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4bbd3dad40214843b85ef7023e96f536 : SUCCESS in 1h 31m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3e473db4e41460797784f1d6259d081 : SUCCESS in 1h 29m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf32b6a2458e4d329b208d29e732b146 : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4314b2037dde400abe5dabf621f7ff69 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f07498765a1745748bdc7eed5b866583 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a10e9e4b54ec4c059927b76e631875c4 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a16070510a304c2391ef47857954e248 : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/366e6091c9324fda8df6d8d509d631f3 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd1b064dbb94a83b6f1fd9f6b01b1a4 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8baee17da5e34acda132aad854b7e760 : SUCCESS in 4m 14s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/8691e16fea2e44d7abb220d77d09bf35 : FAILURE in 4m 58s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/67b5658210d84ac9822ed4f449ad98bb : SUCCESS in 5m 38s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/187ec24833a542a197bd82bbda4daf1a : FAILURE in 4m 12s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/b798d49a29604c788af3d68974abd6c7 : FAILURE in 5m 12s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/37b2ae3511a54571bbf6679325e2538b : SUCCESS in 5m 42s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/9f4bd26387ad4dc7aa370c2b59719cca : FAILURE in 5m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/05d969b2fe844e9bb50abffa5c76c0f4 : SUCCESS in 1h 16m 08s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/61fa5aea8ba54a33bdc5d5f9ae4c7c31 : SUCCESS in 1h 14m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/249dcc7e73ca4225888f12bc73d756ca : FAILURE in 11m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/32024ea5acd9444db0ac5112835fcf12 : SUCCESS in 1h 27m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/3d1350aa7760448e8127a4b5ee577a68 : FAILURE in 10m 59s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f868e5cadb5b42b9819d238d3c550207 : SUCCESS in 56m 38s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/ed9d2171eefd4312bd1dceb386319476 : FAILURE in 1h 14m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/8e3c2714126a4d97b1f4e322bf0e2853 : FAILURE in 23m 28s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/3fcf2e0de36744db81f63fc8965c237a : FAILURE in 24m 06s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/1ff27f5a93d94e28b6f3d822c565bbe3 : FAILURE in 25m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9755e9fa09ebafe746bcda8faea99e925b64a6c9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-21 15:01:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d95d08a6b8a709e643b9793c8aee0ca758109665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 17:05:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/181aee6aea5a492889b0bf7975cac14f : SUCCESS in 1h 19m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8ea014aaeb04944819687ffb5949e04 : SUCCESS in 1h 13m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94a2f74440144166a8d564bf781bfd99 : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a1e843cc77c49c0b9f3f5614718127a : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb6b4c7283b34466bb8b59db627ad1b3 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/969d9f7b6ba642e1838b9c87b34a4472 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47ca809191614df69a3e53446759d656 : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ccbc15c1b9d640c78d99f2d824e8a253 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d600aeb46974452f8d4163db69034def : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0e253959fe04a54887682f479106967 : SUCCESS in 5m 22s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/9c3f3ab520d040d9982cdbd6a91dbfd6 : FAILURE in 5m 39s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/4aea59e1a51e47b6b8cc4ac3d3c83661 : SUCCESS in 4m 58s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/b0275b08f5b14c83a17a096fd13550d9 : FAILURE in 5m 30s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/99f416ac5a44434da8f69dfd194ac025 : FAILURE in 5m 37s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/d6933792ef54423d9a5aaf0f6f4168fb : SUCCESS in 5m 34s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/a38fdb242f454de4ba8c3174f2718a29 : FAILURE in 7m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/aca50e159d4449ecb6fce087599a1aa5 : SUCCESS in 1h 29m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/adb332b077074183890bc09cf89ea262 : SUCCESS in 1h 25m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/b6d78e7539074bcda5566c4eb4bb3544 : FAILURE in 17m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/92d1020f4d364fd89f681f712d35b84a : SUCCESS in 1h 28m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/97c8732e693941c384f2c2492b362254 : FAILURE in 18m 58s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/92318da6423f4fb8bf8f0384e4ee8fbc : SUCCESS in 57m 26s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/df4ea56feb0c4ac78f68d371147d7ce3 : FAILURE in 1h 05m 22s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/cda15b6d33ed41cea06c6c9f0f2c5832 : FAILURE in 37m 33s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/9615651e31b04b7f89ace720e4ad08ae : FAILURE in 40m 35s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/3629d36e888f42c8be2b9cd1d25363f2 : FAILURE in 33m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6461f7e624bfb25c03d6aa44e8ed2f87096fcdee","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-23 13:35:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"07bf49307e45cbb60a41b309b52e10514e2ba38c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 15:45:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ef4ab861dc7546118e2f2634765ea3c2 : SUCCESS in 1h 20m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/799f2dbb842046a686feb38e5a36320b : SUCCESS in 1h 19m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/512960e702e540a4b65dc89a5df7e3f4 : SUCCESS in 9m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73737feda20c41c59f5dd4769160d785 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e691044347804e26833fa62e55d7d0b5 : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6f5dad5816424dfcba184208cf79cc76 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8da6e79dfb84cea92d41a67801f0643 : SUCCESS in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88f9db7598d046b5afd4271ed5635a37 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9268805bd21e4a4cbc016856468ad5ed : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1560785b1f494c259c92683674c7c917 : SUCCESS in 4m 13s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/23352f68d3244b7cbe717374654926c8 : FAILURE in 6m 55s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/f9780f392f7442b781556e43ca0f24ae : FAILURE in 6m 19s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/7f9bde1408e04e87a716e7f850f432a0 : FAILURE in 7m 51s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/f77364daf4ad4ef586957d79252d6302 : FAILURE in 6m 49s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/e4d0db039dd341b4a5b210dfe4b439be : SUCCESS in 5m 48s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/e7ddf2f03fc24132b08446f44701fc3d : FAILURE in 5m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/55b34ab1c5204326a8e94cd4afc5f3e9 : FAILURE in 1h 18m 24s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/62cd86a1b3b8491db31454b2f3f95bae : SUCCESS in 1h 22m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/2f26f3d916ec4425a83ff88072de3513 : FAILURE in 10m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/0bb63d016d9b476aaeb5e1fea2720798 : SUCCESS in 1h 40m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/330e598c8256460e95c20409b632930c : FAILURE in 12m 16s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/48f351e37464472481460687998f2e79 : SUCCESS in 53m 54s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/8409fb151304424ca74e778a0296476e : FAILURE in 1h 01m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/0ae934f546914b38a6f57e4446013915 : FAILURE in 30m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/d380579573504eeeb2d199e31c18aff8 : FAILURE in 32m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/ea5fc3b0eab741c19546ed1a00ed9a75 : FAILURE in 35m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcb6ae6d37058cff2e65b7e811e168aff8df7211","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-09 11:27:23.000000000","message":"Patch Set 1:\n\nrecheck until stars are not fully aligned...\n\nlocally functional tests pass...","accounts_in_message":[],"_revision_number":1},{"id":"44283af96ff47a5e4f66a602d43132c5e4346034","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 13:14:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/18401150e9324775966b2c93558d79cd : SUCCESS in 1h 40m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d70cb7bb71c458080f48df38e49738c : SUCCESS in 1h 17m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c37420bd51041bfb2c54b5ed6eb47b4 : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89bbe34709d74dfebc3ec745cf3ee9e6 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b484086fa9744a9c91d92fa44d845d26 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e660dc4f9ea4058937300d15e2d6e3e : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de254a8ec5ac42a68cb7b1960cc88506 : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c29011c4d924a90976393a6ddefed5a : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c6c9dd683f4e3280f41e7f8e769b3b : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d8b88eb76f541bf82d7989f6bbcee19 : SUCCESS in 4m 30s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/0817078b55d5451483bc8ab5e75c1076 : FAILURE in 5m 03s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/085b0df1663541bfadc722cb708e9100 : SUCCESS in 4m 42s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/256f3f49285f43388f0788574f83aa78 : FAILURE in 4m 35s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/dccd8a0d950e41919940d666964789cd : SUCCESS in 5m 54s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/7070f8aba06f44e3a07efa0ee7c50207 : SUCCESS in 5m 04s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/b211ac59b46c484a96acadfd86b0e7b4 : FAILURE in 10m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2fd193c40e1140d5aa8bfa40f6b37878 : SUCCESS in 1h 24m 15s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/254b2e4859d04d99b65fbd4fe145b0df : SUCCESS in 1h 11m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/55e221dc89b041d99090d17181d6b8b0 : FAILURE in 10m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ce654cfff9794f899d4fc8a19872d9cb : SUCCESS in 1h 24m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/60f7efcfac354022b14c7f436900ead2 : FAILURE in 15m 12s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/5dc62d3c852d408088152948cd712d0c : SUCCESS in 47m 51s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/38b8753d49f74cb5af12b5b34482eec2 : FAILURE in 1h 01m 50s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/c82ac061d6ff4dba9bf829eec26803e7 : FAILURE in 32m 18s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/acd26a56c30144dfb83c9da4b96bfcb7 : FAILURE in 34m 18s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/8a17ceedb53c4e8796061b868450de3e : FAILURE in 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f98a1fcb1bf55944ad4b4683ec1ad7aaff482190","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-19 15:08:54.000000000","message":"Patch Set 1:\n\nCould be rebased on the top of https://review.opendev.org/#/c/763343/ when it will be merged","accounts_in_message":[],"_revision_number":1},{"id":"6bea57946ec2e7cafa82e992a3dc9c8f0bdf29a0","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-02-09 13:26:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e2f948ac4c38d4b1e0e30d226bc83bf67ef9f60b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 15:17:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b9763472ec964450986c53f4afe1f018 : SUCCESS in 1h 21m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc20a4a94e7a48838d89354e9733b867 : SUCCESS in 1h 24m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4920a44c5f514f0c98e175b5ff5a5259 : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af5992056b60450cae70720226ce6594 : FAILURE in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbb3530fad8546ccaf216f7d89054822 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/815881291791431597dbddc94f1ba686 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76160a4599c14629b77052e23e623be6 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a22f5930a4fa4587b12d289957585894 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f813f30e9c491ab2afb17d5ae9ed8c : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a194d281122148bea4506b0c2a4b4588 : SUCCESS in 3m 36s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/55058a9a54394c05a3f89643f6bd7e83 : FAILURE in 5m 18s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/68000d1b8ce84b909dd5d775e635783c : SUCCESS in 4m 25s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/f37c7d6518444d1f91486dbae9548358 : FAILURE in 5m 04s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/e72d94b7d14f463cb476e18db513f543 : FAILURE in 7m 30s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/3ec8f3ff4d8a46ad95b909bb34882a06 : SUCCESS in 5m 36s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/d8c1f2b4c9c6409d8bdaa611df5af68f : FAILURE in 5m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0d9d43f3a9314a2faf3d6d6a6ebfca50 : SUCCESS in 1h 21m 44s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/398d8a1bfbb6417f9633e1a7a8ae20bf : SUCCESS in 1h 19m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/e827dfb408cc41bdb9111ca0e86eafff : FAILURE in 13m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a5086782da80422f95ba6124cdeda48d : SUCCESS in 1h 32m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/6a5f193e903c4c04992bafdd30e3dfbc : FAILURE in 14m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/8efe2f85370f4611b1604dacad54dfa5 : SUCCESS in 41m 50s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/9bc37ffc4cf34b78b9acd63ce643eff8 : FAILURE in 1h 12m 14s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/e3f04c9113914c0f8a3501164226c0dc : FAILURE in 30m 39s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/c902a2c6798b4570b19bfb809350c044 : FAILURE in 38m 50s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/58b89ba3855c46f89f34fd7e44738f33 : FAILURE in 29m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9790f28c5add53ff8fb044061586567e23977ff","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-02-17 19:41:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4f9389b3d593727898be511f8f5fbd94c2d6036e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 21:10:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e15e5dadf45d4dca8aaae6aa46971932 : SUCCESS in 1h 16m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f38b182457b4e83a62d3955f4f5243c : SUCCESS in 59m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b52a7c96fa754f98ba754ff0f9c8d8e0 : SUCCESS in 6m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c928fdf0afba4dd5b840597755ab2afd : FAILURE in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b6aef9ef55e4e2cb0916e188915f8ef : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a92d349bf6644393a594f5e1207d74a8 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/752c868c5be8455aab1eb22a88d43894 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ba3fc150ddf483e9bbad3bd49180e8b : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cfd5ca4b13348999f83e035cf31ef9a : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/106bb1e08e9344e2a0b92566e80cc34a : SUCCESS in 5m 01s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/b2ff2040632d4844a0bcbb55ad853e36 : FAILURE in 6m 48s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/6b61b3e2295349fa8ea00728a44a9172 : SUCCESS in 4m 35s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/16532fde99a848e4b30507202b5541b5 : FAILURE in 6m 16s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/69bc9c5f65f94708b618d83e879bb3fb : FAILURE in 8m 24s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/d18ea1ceaaee43328a71a4fc07350493 : SUCCESS in 5m 28s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/a546a78b192b4151adf1d535bf80c560 : FAILURE in 6m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ce8a20abc1584022920b8aa278d3a573 : SUCCESS in 1h 07m 58s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/978464b509b54ca09712f9aa67afd2fb : SUCCESS in 1h 22m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/b7186e3efc0943eca31bd9a83e28424f : FAILURE in 10m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f4285beef0f24311986616081cba620f : SUCCESS in 1h 25m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/769dbf78344b4be6b2f515741a6fa4b7 : FAILURE in 10m 27s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/c19479eee6d546b8b25df75cae8e9d0c : SUCCESS in 58m 44s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/b09056570ff3496fa46439412b618004 : FAILURE in 53m 43s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/c96a1f0b47ee42f38fe2a72ee35a3441 : FAILURE in 26m 39s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/7f7056daf2784aa2b40dab8f8ae33b1f : FAILURE in 34m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/0c7eb94e50b34ed993749605d43b4a31 : FAILURE in 30m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8eec82c8b58da25f2a765bf2ca104acad0e2e808","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-02-23 23:38:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"760ee201ecec1b85c3a0121e58fe9aef627eb8c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 01:35:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b67194c9952c466c8ef3f00e716944ee : SUCCESS in 1h 43m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/962bd772a91d491ca81fb9c96b435bbd : SUCCESS in 1h 33m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/211f87ec0c9f4044a606c6a603071c47 : SUCCESS in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab509704a6ef41adab79ee76edccb7c1 : FAILURE in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c17caede1c748e6a45ab8b9cfc15eed : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf55f751c9454f70bad74fa1ae8f79bb : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c68b5b8761524f1598fb41a411cfcf7d : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac18190c071e4541a1548ef65a628356 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26192ef5a46f4bb38ba25d6b76432f24 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d54103eb789425db48a7d704b86ec8c : SUCCESS in 4m 20s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/fa6e5649b5fd4f918b457b7a8a60e190 : FAILURE in 5m 19s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/146ad615799f41228ed944aa019138f0 : SUCCESS in 4m 44s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/5f1911f902e74208a49bd26a61c1d7c6 : FAILURE in 4m 20s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/8164f9920d48440fa512f67e1376237a : FAILURE in 7m 47s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/d0d2672eaeaf420ca501576cc3b11cf8 : SUCCESS in 7m 34s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/558f081177444ef98c04d05cc3017f4b : FAILURE in 5m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8a04abcae7f84e8fa59fe0f7f63bf40a : SUCCESS in 1h 19m 55s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6dee091828634b3b92449cddbe492d5d : SUCCESS in 1h 25m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/80a6a2d6753a4605af1c8aebf92d257b : FAILURE in 11m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1af70f96b1314ef1a97e455cc2db8b70 : SUCCESS in 1h 49m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/37db2b5482354074acc18af0744bee50 : FAILURE in 11m 58s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/ec884a058ba841fea57d4b896f6c9d76 : SUCCESS in 51m 30s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/a8e257736f4d4045897d72142dc4ccd3 : FAILURE in 1h 05m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/74a05f3fd51d4b36b186ca8751d718e9 : FAILURE in 26m 40s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/0473df962550487fb20208fe3d121741 : FAILURE in 29m 16s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/667e075426664d51bfdb100b5ee6599f : FAILURE in 31m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e8ba2b21aba138b003a5b90ff57a7252def78b9","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-03-02 17:21:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6150b19c7b13b7a25a0468991bee25db4fa1fef1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 17:21:32.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"3941feaf9bfa2e66e1147bef7dcb023b074d4da1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-20 18:34:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2a665fdb19fc4ae38361afc8736ba7e6f4281a2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 20:11:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3955104a052948f09a8dc083637abd27 : SUCCESS in 1h 32m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f16ef5ffb1274871a5c794db75c81845 : SUCCESS in 1h 10m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6abf091c26043c59f11b208f856fc8f : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e78bacb5dc448c39ebe015f6874322b : FAILURE in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a770e9470414110a2ac7f835187e124 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b14aa32eb32e4cb887a1c2e505ba7a46 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4911f4bb371c45d78f980fffcdb8017e : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5e796e71f114868abee10f67215ff92 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9825c4f0bd347a3beb6f1b61047d1f9 : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/594d60c20af04f8b910f4a5f6132e3cd : SUCCESS in 4m 25s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/726e081e9de74556b3866a0133e040e8 : FAILURE in 4m 50s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/09cbce36424b45beb9e47b0b10e82d00 : SUCCESS in 5m 08s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/404dc58e52d9486a951cc504153e65d2 : FAILURE in 5m 02s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/612f6010fad44ad4897825bb393073d2 : FAILURE in 9m 06s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/9aa29eead1fa42e682354aeee98f71d6 : SUCCESS in 5m 27s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/cd55f82931b6414297d8ae06c598ed29 : FAILURE in 4m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d3983c8538bb4bad96b40d810bf8d631 : SUCCESS in 1h 13m 37s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/93b8eb1ecb96451e87c46fd705d3ff31 : SUCCESS in 1h 19m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/060cd24408c2466fad4e6ac0698e7ba2 : FAILURE in 12m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/acbfd87288d54e49848cf72508193b35 : SUCCESS in 1h 05m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/86da463cede64f0a9ec02787cf96b12f : FAILURE in 9m 41s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/6e3eee72556945f4b4a7c9dbfe4dc0c6 : SUCCESS in 49m 52s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/4ec6cede29ff47db832005bf4d2f473b : FAILURE in 46m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/5316bee2f75d429ebef686255cdfa42c : FAILURE in 28m 04s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/7682276b6bbb4de088782c370e6d6ace : FAILURE in 26m 01s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/c055ef587542413bb50d9d58de4d4283 : FAILURE in 21m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6e4d64aebbc48917d4eeeb9a916714daea6f901","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-22 12:57:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"77e71d93978e5623276d8e6052051c0524b72ae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 16:27:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/aaa66c3702f84e21bb1b93ef3d2f6784 : SUCCESS in 1h 32m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a7848399b224a808ca1249317ebeeb9 : SUCCESS in 1h 30m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddb92ace3c2c4617b3cacc79a5214954 : SUCCESS in 6m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/889fc24cc8924834bc2804df1230e5c3 : FAILURE in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a9e3287197846f1bd1a2d28c22e1267 : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0291e578eff040de921437d7e7c09555 : SUCCESS in 9m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64400ed2192840ae9f45e17f5eb6da65 : SUCCESS in 7m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8c9bf6f5f6d405891f3ac24c6deb265 : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62871aed98284202aa11147a0fb55573 : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5060b06bc2dc49afb8ba5bc99c5bb20e : SUCCESS in 6m 33s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/2401e7665df14be6bbc40e70c3814e38 : FAILURE in 7m 51s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/05ff02758a66429e99508a05998afe0e : SUCCESS in 6m 39s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/5b5f0080b8d14dea910625aac01b2129 : FAILURE in 6m 32s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/749936c7ece946cf949113b3a45c5e4d : FAILURE in 7m 42s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/f8c24846e6b349f59760559ef998f480 : FAILURE in 7m 46s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/82394873ec9f4486a79ae5b0dda91788 : FAILURE in 7m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/74b01a0d39924ab38cf44dfa50db0c62 : SUCCESS in 1h 27m 03s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2c585b4991314d30a0b0ab146a0718c1 : SUCCESS in 1h 47m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/28781aab6a474fb88de60eb1237d8b27 : FAILURE in 11m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c8102b3feee54692bb4e1a01dd637815 : SUCCESS in 1h 31m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/d8be82a1c2394ac297dca9cc6dfdf73c : FAILURE in 12m 48s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/b8cd82d8776945b783161bb6a5251297 : SUCCESS in 1h 01m 32s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/510ba548eef34eafb88c48e380360c2b : FAILURE in 56m 11s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/4e39b32d27c54fe38165febd7aa79795 : FAILURE in 27m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/a17d3fd9bd8742b8ad1c6c6834aa2de1 : FAILURE in 37m 02s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/da3425b88ac346578f37d3c7f3537b2f : FAILURE in 29m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"afe12bf745dc47e5c15e95ad0288bfdf1610ee0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-15 13:59:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ef9665e2c4fb9a5772d5d452f543b3b23533b0ba","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-15 14:00:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"06fa77ddf0bbd072354ba2a563c73ef25c07f575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 15:27:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/07dfd9398ce649dc967360ec9fc838ea : SUCCESS in 1h 10m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/894724ed06f2447496e105a73b55d0dc : SUCCESS in 1h 22m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed94609fdfa4436eb1f917b61dc78ca9 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/309ae2b092f24ee0955d0e21743cb6d2 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aca5d5bc16fa42e0b465a1221a154543 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0aee141d1ef549adad645fc227d3b837 : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1624a5f342bc404692e48d87e40c35e4 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baf91fbe4a354d3ca3c6f7d0769d66e4 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22b7db5d030a474d93302df7dae1335c : SUCCESS in 4m 44s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/7575a806d5c24ce0b9bf89b5c040f19a : SUCCESS in 6m 55s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/5f912cb83fde4275911a3377497b7692 : SUCCESS in 7m 00s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/8b4d53e30b2c4d9686ecf8f57283cf02 : FAILURE in 5m 25s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/b7ffaaac78cd4babaadb881d1cfed3a7 : FAILURE in 4m 16s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/39b0477daff74ea0b4115e1a9b923341 : FAILURE in 4m 43s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/98101db0216442b4a2ad264c7d352106 : FAILURE in 5m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/62857f6362114a3f80fc2e8a579b97fc : SUCCESS in 1h 21m 06s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/5dea0b2339b44b5bb3d5e818e8a45411 : SUCCESS in 1h 20m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/13e619b20ad34d8484176546f9fa229c : FAILURE in 11m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/838c7775c07141868eff4eb123697b61 : SUCCESS in 1h 18m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/0625b2b8a0ed43a6a875ef8438de77c3 : FAILURE in 13m 04s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/6744c95f579748769832a68c8141d443 : SUCCESS in 48m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/84d87e948ddd47aaaaf0f067101946d9 : FAILURE in 56m 20s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/2d02b1e3d5e048adb45f78ee8f7d03e6 : FAILURE in 27m 00s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/6a58c5a4f71742d6bdfdf48dad51accf : FAILURE in 27m 14s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/0f1c3f8617704bf88341e886d5ea7d81 : FAILURE in 27m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"905e73088a814c6a84bd548f4d8d4e44fcfce239","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:28:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c1b8f370bb418b972a05bbddd9419485bacb99c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:27:20.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/408feccfeeb2456f8e448c550fef37e0 : SUCCESS in 1h 07m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/673adfc6ad094f5b8c9eb892233a9fae : SUCCESS in 1h 05m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78b5676471544c2c82ec5806fffe5aba : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bfbb2b946b5a432088129ec097db1d1a : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b00c1474f0944804b59fc3833fde491f : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a572712b40194dd080a1cf7668a39060 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0487cdc23914c81938d111fe775dc48 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e528aac746334a349159f797d8d386b4 : SUCCESS in 5m 12s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/993385130f294a0d9acb845d6927626c : FAILURE in 6m 51s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/218d6b1cef7f47c8b659ad025a4770c2 : SUCCESS in 1h 56m 52s","accounts_in_message":[],"_revision_number":2},{"id":"04fb9d23da2d9ef35ddfd313d3019378e4089183","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 13:44:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b80bc5c91d76e9c3b9f8a8f3e5d5a6034128c594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 15:48:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3b096f5a82974c8394148ac8bdc50cec : SUCCESS in 1h 46m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a85e566b97c1498f8911579a538e21ce : SUCCESS in 1h 30m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33af5049bbd046249f639e01bac76b89 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c742c03022e6440f939091538f277e4f : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/11d43e284a4447c08b911e6a33733445 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9009c691263740ba9217a5f37c932fe6 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49755f3cf237413fa79cba66954a34bf : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d980816d4c7b40c68c6563c928f1e413 : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f70ee11a734c403fbba6953ca48a1e1a : SUCCESS in 4m 26s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/41206d750d8e461db1322c56ce898df6 : FAILURE in 6m 01s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/0336fc0cccb044d691cd99fd8ea0c379 : SUCCESS in 4m 55s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/a3cbc63fac8846d3bad5095a3876d81c : FAILURE in 4m 48s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/6f4bc4b39080405591021ef0a8f116a6 : FAILURE in 3m 53s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/ddc774f57f7c4c44a0447b9e2031b508 : SUCCESS in 7m 55s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/290569d3a4c54fe5941628ec94b9ee43 : FAILURE in 3m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/29e0051e65524aa6823739822ac6077a : SUCCESS in 1h 31m 49s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/957c1710f948481fbff3aad6a74d368a : SUCCESS in 1h 36m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/1828d6efca7e430f83e47c4478a29578 : FAILURE in 14m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2fb2b889050d4e3d87d692778e80d789 : SUCCESS in 1h 37m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/6e12acabfca54a638afd20164b924774 : FAILURE in 29m 28s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/53f6df3262b44f9aa2049764956c89ad : SUCCESS in 52m 00s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/1dd87177c70c454f8feb2a0fab79a1f3 : FAILURE in 16m 04s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/193df08615514b8987e9f4668a12f365 : FAILURE in 23m 28s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/606f5f0a214c4cdeb600e56258ce7887 : FAILURE in 27m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/55463352da98422e95f197b22f74ec2a : FAILURE in 23m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f76f6d5e10168e6a70cf191e7de200416cd84105","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 15:51:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5558b491395e720d1034933bef92db1ffebd176a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 17:29:25.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d775f6e3428947b5b95030120e9249d2 : SUCCESS in 1h 35m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d7844fc726c400ba1f7cea04795ed94 : SUCCESS in 1h 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f6bdb2b5234d8ba5e8a65af4529a90 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45c960c7d3874ea3a0fc41da56c878ec : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57792e01d6644a4f98037cf49fd421f8 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83e3d9120d9345fd9e3e10b84d3ec755 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b137ce9548b24bba865ebf47f5fb99c6 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0c7647a45ec4c169fe9aa0f60fbf419 : SUCCESS in 5m 09s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/3657d2bab6ff478cb3a9682b4ce4cf28 : FAILURE in 7m 06s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/67c26c8a3bc848878a9e231f9d1ca0d5 : SUCCESS in 1h 14m 56s","accounts_in_message":[],"_revision_number":2},{"id":"40e2a5051677a5b9cd59ae89180d8f7bbe8eaa2d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-20 14:11:26.000000000","message":"Patch Set 2:\n\nrecheck\n\ntox-py27-func-rabbit:\n* test_failover_scenario - Output: ERROR: node with name \"pifpaf-1\" already running on \"localhost\"\n* test_failover_scenario_enable_cancel_on_failover - oslo_messaging.exceptions.MessagingTimeout: Timed out waiting for a reply to message ID f8abd0fef89c41dfa63e91c7dc0e95a8","accounts_in_message":[],"_revision_number":2},{"id":"3581fd01187ae1f232c930bea2fb0b63d65ad83a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 15:47:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f21422d3d9a84fb698e5729c674c1c11 : SUCCESS in 1h 32m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9deab80922a744f1a08bc46d388a14ca : SUCCESS in 1h 21m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb9ca578d9774ea4a7c421026dc083ce : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa0a90a4e41247f49f8a46cf916577c1 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5342fbcca32740e8b7ca6bd9e0d8d9d5 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb638445b2a74a68b0cdbd9da82c2d7a : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e0bef51ff194c9283b2f7eb116aa795 : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f08d9238133a4de68b8ed96abaa58218 : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4b73b0c27fc4f1c85d20021b7942423 : SUCCESS in 5m 03s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/08f395ef529f449bae6e818143bc9614 : SUCCESS in 6m 21s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/d5fbc13068654312b6b78cf37a12ba1e : SUCCESS in 6m 58s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/cec7e714bbb64784a4c575de9686ab3b : FAILURE in 6m 11s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/9e9cec8660b04c8bac96c05970138490 : FAILURE in 4m 12s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/56c6639353ed4c81a3c62204f15efbff : SUCCESS in 8m 35s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/2881d53be2294629b7e6bfa2c1e4f0a2 : FAILURE in 7m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b7dc67bc6e0e42d982ff8e23f1437209 : SUCCESS in 1h 20m 19s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6445f9f531e74b55b7d12219b8f755d6 : SUCCESS in 1h 10m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/dccf8966eb744c1eb6f634ee1b48a05b : FAILURE in 11m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/0a7b6a329f5048cf901706c95078e3d9 : SUCCESS in 1h 28m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/6d6ca93190a64db08c6e11db6d456d48 : FAILURE in 9m 57s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/453cc8c0544947549b6a00ae382ee183 : SUCCESS in 46m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/f62b4e0b09fc46419bf816b510dafa95 : FAILURE in 53m 28s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/b542ca45f3e741a7a1ecc59a2c696e52 : FAILURE in 26m 39s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/e9fbd9302b9f49228b3db5562ea67d39 : FAILURE in 27m 46s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/1ae7a1723596457e91adb54c4e6f0f53 : FAILURE in 32m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5fd7a4c1c2720bd712a9dd76239ad7efc94a5cce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 15:52:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b81cd251c171ef9df15aea2b0deb967d0f050671","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 17:31:21.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3f71f761d8bf4f3aae87b2a8d4e3ff61 : SUCCESS in 1h 35m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a46973b06b434d32a792141172734be1 : FAILURE in 1h 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f240af0dcb894fb790f482376f83e7d8 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2282bdbde3494963a75f212580f05e2d : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8047199f44934ee5a9b776a03c627bb9 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ae650f2e47a4bad9c99aaa301025729 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87392feba1c94e11a73832bc98f36ac1 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1568aa8fa5f64095a8a3c78f7c67383c : SUCCESS in 4m 31s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/3c0fa1746be7493db8ee97c54ae0ff63 : FAILURE in 6m 08s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/dd5842ec56bd418b99cbe7b4889abd45 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":2},{"id":"44833206854c7307b31d6e9ce669f9d8a49becf1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-05 12:15:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"31e11703ca4c6f83a5f4eaf9d66bb97e51bd55d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 13:56:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4264a79ca976400fa31bfc9f439d9ada : SUCCESS in 1h 15m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28629b353db3492bb19f58fcc2430884 : SUCCESS in 1h 27m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f2ac98818654639b73e409a7fe0272f : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e330142d5b4e418ecc599c39020527 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57577f5183be44db980c3e7ac7ec284c : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff50d81fc5d94367ad8e72b5efa826bb : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13c48db5f07b458f901db13cfddbb3b8 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc25292acbf45fa8fbb34f0bac7b878 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8edf4f6c0b71427cbb575cc0c1598418 : SUCCESS in 6m 57s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/aeaf5a7cf51747c6a23d181e4ace5762 : FAILURE in 6m 33s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/9a0e5d9746b2474b8a9b5a87119f6d9b : SUCCESS in 5m 35s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/23e7061ed134473e8f1664da400fa79a : FAILURE in 4m 48s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/0fb321aa154a499fb072ee0f2ca2ba9e : FAILURE in 4m 08s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/da56ae07a0d64b6083e1e9a0a1e31dd1 : FAILURE in 4m 07s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/351c2ca50bf948c19f0f1c00a92872c2 : FAILURE in 4m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f9d95ccd832d496a81a4d33bfb2cbe20 : SUCCESS in 1h 22m 32s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ad39d3ab37da4007a301305010a52d92 : SUCCESS in 1h 34m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/b461aeb60ee64efd8fe1721ad31aad38 : FAILURE in 17m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/711190702872446bb3a4182e0505d992 : SUCCESS in 1h 19m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/b1e17804aad74c678fc4e51a4b8262a7 : FAILURE in 17m 27s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/096f42d4b38d4832b0e409a0fe709649 : SUCCESS in 51m 07s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/a50d68cc99344ef8b5021c4731d79dde : FAILURE in 13m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/32a649ffd3b943888ea1fad1796d5b72 : FAILURE in 28m 28s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/afa7e7f59218445baeb0b358756c926a : FAILURE in 27m 36s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/3e7d110b36494b84a70a7f539b87ef97 : FAILURE in 27m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74d259abfa8be1ebd5866941f34d3c39d8d366e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 13:56:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"44acd4f354c8257d256eface23fab20fd74a3370","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 15:39:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/93700a4800e24a409fd408984613eac4 : SUCCESS in 1h 41m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1723f2c939f435ea13bf2ae7574a237 : SUCCESS in 1h 38m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87b80b4a1b2c468c9b584255220715e9 : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08108a1831414b8f856337e1328da8fe : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a49da8c5b8f840d29606311e8cbf4b80 : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8536b76e1a149d285bf090719549a52 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fb2a53612b34efeaa9eba6316fdcab8 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05bb10b3e1c642aaa79f9304ccdc340b : SUCCESS in 5m 53s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/f3546cf8125c45adbefeacdf4b939653 : FAILURE in 6m 13s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8a1c0daf66de4caa9af7601260efd3bd : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":2},{"id":"8318dd611ec344cdaef0a2ed39969a31e3739111","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-05-05 15:58:01.000000000","message":"Patch Set 2:\n\noslo.messaging-tox-py27-func-rabbit is non-voting in the check pipeline, but remained voting in the gate pipeline. That\u0027s why this patch passes the check and fails on gate.","accounts_in_message":[],"_revision_number":2},{"id":"e023ea8a3ad1e3a5fedddf346efada898213725e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-06-01 10:55:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0ffa336a8ea398eabbc61ed92b8ec389e6551071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 12:15:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ed77306b794f4d709a380e1f4f91b52d : SUCCESS in 1h 15m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/431c72459e9642cd8a0ad0fe09d6d99c : SUCCESS in 1h 13m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5af28a0d0824be0aa5ec5a95410434b : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b061936e90f24849bb44f67652504add : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b7913590f380493d8187421bf0ccdb99 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84e774d86f1a4e659d739f4844489b60 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f63f199734694354a02ae76afd2053aa : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bed5418d3f364c7d8a60c0ba8372238a : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74628346b0ba4b708c2e2e9e33e9af2e : SUCCESS in 4m 52s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/2f332784dfd148ea90620fa118baf3b7 : FAILURE in 6m 01s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/05ba165b462e46dd98974540006442d6 : SUCCESS in 5m 04s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/af9daf8a9ec84ebb8a8ed31dbf2acf17 : FAILURE in 5m 14s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/88645e23e8ce4035ab430223b522cab3 : FAILURE in 4m 43s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/488d5b11341f449f8b53901b9fdbc5a2 : FAILURE in 4m 05s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/b0b153fe61dc43398aaf9a1544e43fa6 : FAILURE in 5m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/55c2e46e5c7741b084d7cbe044f4d94f : SUCCESS in 1h 15m 28s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/9d449096dbc74149b2b9f41af5186c92 : SUCCESS in 1h 18m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/5e0e7559ada34da7a1a9776fb54b8427 : FAILURE in 10m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/aaaf52b5c69a4ea0adc05514f31f0734 : SUCCESS in 1h 18m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/8f01ef99f9ed48448148da04cf19853e : FAILURE in 9m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/3a42b63f25ff48caa822e88a9a5091cd : SUCCESS in 58m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/8213e15a24e047a7b6a6147e34bf2687 : FAILURE in 20m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/33eb011b14f647d38e460b663637f309 : FAILURE in 36m 42s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/a398eff38922416d8dacc25014cf3708 : FAILURE in 34m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/816ef98ad7c740198155327317e48d04 : FAILURE in 34m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"49001fad63dd10cca56b8bb137b5f34de6aac1b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 12:15:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"765b8995e35acbde8d05a0ffdb11868898eb44df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 14:09:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/50bf5b95b1b241eeac636e931ddf8dea : SUCCESS in 1h 14m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53742994316c4c44b2273689439fb999 : SUCCESS in 1h 07m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/069182832fdc47e68c619888b69f1cbe : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a035829e5ee4673b8b9886dc932cf7e : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2cc411457d6422e96891535bec5e2a7 : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc74308b83aa4ef28e6abdd056c00f9b : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b2b0302804d47abb201f692a69e055d : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2baa4b458e5b42e1818adfd8e534397c : SUCCESS in 4m 35s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/8fb73bc3356242d1803e8ea276cf2962 : FAILURE in 5m 23s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/200bca57e79741d68d643df81c22119d : SUCCESS in 1h 47m 06s","accounts_in_message":[],"_revision_number":2},{"id":"9f87974c7fc45ae0883a24886ca7b3da931dbefb","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-02-09 15:56:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a64a6ec31c177a68b1802e16f3667bfc43b558be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 17:51:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4041687988d64f32a036d3bb1437a859 : SUCCESS in 1h 28m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67e8f243ea2b4e3c8b0bdae19c1ffee1 : FAILURE in 19m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b25247a277454e2a907bf568e8143b31 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25544d46afd3427789bb61167ee76346 : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76b3128c110d46aa9778f83291cadf41 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18612679fccd48618df905b1de6d3306 : SUCCESS in 7m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c2e402a297a4f6ca1762a22cbe9ca5a : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40b9ec944f4b4d6da62080118555800f : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bda54c259374433cbfbc7d54deb1f81b : SUCCESS in 5m 47s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/1cf606632bb9447694b11240088386f0 : FAILURE in 7m 33s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/d0e7e0cf711f44ed8424a4f12e3c38c7 : SUCCESS in 5m 45s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/812ae01efd3f431c86a642c280bab895 : FAILURE in 5m 45s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/57f2850147244c66b82958fa8c76c15d : FAILURE in 5m 00s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/462955db4e3d466fadf7547d4df2ac6f : FAILURE in 5m 48s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/18e34a25db67420483689aac0711b472 : FAILURE in 12m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/99bc1137bf28434caee96a0445b48598 : SUCCESS in 1h 19m 38s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6a4767ab24884f729c49052201f870ec : SUCCESS in 1h 20m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/440bcbfe92f648a78cd25c19e0a01dba : FAILURE in 13m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8289502cc64744f0a99d4a963d1d0961 : SUCCESS in 1h 20m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/f56550ebe5a74049bc8d968e2ee41918 : FAILURE in 20m 09s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/c70b76e232414a4eaab093aa5f57ffd8 : SUCCESS in 1h 03m 41s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/721b23eeaa864c8f88b9ccb00af481aa : FAILURE in 17m 39s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/6bd716c8b3df4ba3841276b88c661023 : FAILURE in 33m 54s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/a609fe815c7545a2bcfd81483dbd8dff : FAILURE in 33m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/85f47f06b4594f2bbdf03b763b839be3 : FAILURE in 36m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b2a3935601ff8eaa3d8c4d23f135ee0033f00dd4","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-04-27 13:35:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e400457e26a3dfa3a545129459ff04d9a6983e56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:06:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f763fbb33148473f9acd278344b94221 : SUCCESS in 1h 18m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65ee2b6f3d334eb48dd6e3d873838431 : SUCCESS in 1h 23m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bce6d482bf564e69b728224ed40ae37a : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a78c7963bf7b47a29541a2895f71ca53 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6336658e5c6404e9d7a96207c2fdaaa : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c722a893184f4fccbba405dffdb797ce : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea5b8ff7d6fd43189cff236e74aa5281 : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c05abe4a626f41df8733e9a865246eea : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf7010860bc64629beb2925225578971 : SUCCESS in 5m 10s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/9cf4c8e4b2174d7aa4787eeeb78aa60d : SUCCESS in 7m 31s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/dcbf79345a5d485898bf6302dca5e5b4 : SUCCESS in 7m 53s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/055e03f6d13243de8c20b7999c9cc361 : FAILURE in 6m 26s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/c514fdc65e5c4239bf714979742e6406 : FAILURE in 7m 18s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/94f0b0c94cfa4157b850229898337d09 : FAILURE in 7m 19s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/1126c23ab132401e9941e1a04c38161e : FAILURE in 8m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d2a0d54b4e5a4d19a2f9865557ca3b00 : SUCCESS in 1h 17m 26s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/dd1c83ea891f434dafe54bd40dfb39e7 : SUCCESS in 1h 19m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/63117bce65fd41018ff6b160b2f38b68 : FAILURE in 11m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f5bf4b50f5764803bc17be918eb042f5 : SUCCESS in 1h 17m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/d18fdb2a14334b569044fc02d0aeacec : FAILURE in 12m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/0101474e2751449a9190d56fdb66991e : SUCCESS in 52m 19s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/1c14f888504c47ff8532850ca4ac2f14 : FAILURE in 16m 07s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/af1dcd77529240eaa8967b81b82ee01e : FAILURE in 28m 55s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/1ded486d5eb2456681e92910bb0a0069 : FAILURE in 28m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/2d360e0a8f0547dc85ecb6e288990260 : FAILURE in 28m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b68394a906008565bd6a131fbd9c0b8afd9d037","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:06:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c3d72b66da3831a62a05b9ad11f43eee542c7076","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:31:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f0247fa98f084746ac083bf0fd858fca : SUCCESS in 1h 20m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a291a3df0a57475d926525f3be2ddb87 : SUCCESS in 1h 16m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b40369d61f64829a3d145ca01f2df77 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d4a84a9998c4a909567b5bf031d8ee1 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f3072c840a144b785ef074b5371a8de : SUCCESS in 7m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58400360bea14167b56a197ec4113fe4 : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53531cc44dbe4ef5b3364937b4ff3727 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/762aa0bb0fed460ab15ee5b11a26d8c4 : SUCCESS in 4m 25s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/7f7cfb4044604f899cf757168d767f92 : FAILURE in 7m 35s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/da1c14fa9d1e4f388fdb776e04a81ca4 : SUCCESS in 1h 16m 24s","accounts_in_message":[],"_revision_number":2},{"id":"78d09455b6563bcc9fdf6c752221eb48a269943e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-04-28 23:14:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8d83957e136bf6046e6060a585a3f5a59db2e738","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-04-28 23:14:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8ffbfcf0ace3f4dc6146bbc22896669e3dfbe683","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 00:47:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b366594b2b93464686aac30bc1444e58 : SUCCESS in 1h 31m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8164d4fe47443879f8b9d0748715b7f : SUCCESS in 1h 17m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7da68bfbe1e4f6388507d93b4cce31b : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8035fe6e71d54a208347308661da513f : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66afdc44f8c14290bb3987c4b00b7418 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42ba42130b3344579f23dfd8a5ed9336 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e06ada96f0b4eb9b8ce9618ff31c897 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a60869acf4ef4e16beb998e75e8ae35a : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43fc7632e2b342cb9dd33bf2f2a62a05 : SUCCESS in 3m 56s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/84c2ea9f1ff245fd81502da93cc20c7f : FAILURE in 5m 22s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/f1f1b97949df4e87b9c678ccdbca4858 : SUCCESS in 5m 01s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/6bba38d31d45492bbb7def08c0e680e1 : FAILURE in 5m 23s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/f0b2fd9e8b494d4a88d145abbe15c4f6 : FAILURE in 3m 47s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/0b45393c431b44539c3c6f0ee68471a8 : FAILURE in 4m 01s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/79db51ef81eb44baad8c6d49c993f64b : FAILURE in 4m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b2eb855eea4c407eb55700185d91fc06 : SUCCESS in 1h 20m 27s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/43b238fbdabb4fe7862f2118ddbd0708 : SUCCESS in 1h 17m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/026328403481465b95fb682b165ab174 : FAILURE in 10m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b12fbcf430f64ac9b9ecc7d6e03430c6 : SUCCESS in 1h 15m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/02da33ca03bf4139b4552ae5e6b9d91d : FAILURE in 10m 18s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/00a87db41abc4bb88b43aed81e8a7989 : SUCCESS in 50m 08s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/df04973b65b847d6bc1d5515b2782018 : FAILURE in 22m 08s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/b8eeb45ec75642b189640c7dfebfbbe3 : FAILURE in 29m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/e7e74039110642db991aaa717f050800 : FAILURE in 28m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/6279a59b65e547658e77e28bbce0dc96 : FAILURE in 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa4fc51f5a51c24f32c674b9161d673cec47da9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 00:48:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"974de278ed25c0aea39628d82b1edfd366e9a835","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 01:48:17.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1305803076e44c209a7fa49137b33fad : SUCCESS in 53m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be6653175ca0469ea0dfa93cf1709fc2 : SUCCESS in 59m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ce14880d6db4435801fc0459fe1f005 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e055de8bf3c4429c832f8747673af005 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc2c7a8fce4d481eaf5cffde386aad25 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d916253aa934b51a8af7cc5523220ad : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/334e9c3d9bcf439185a535a8185ebe0c : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a037ed97019a4fa3b3c72dc3be5dafc4 : SUCCESS in 3m 37s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/52decf3ca4974d4c8c4012b320ad1ce8 : FAILURE in 4m 36s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5c9491d5c4ab4646a7cc381e4eab8786 : SUCCESS in 58m 19s","accounts_in_message":[],"_revision_number":2},{"id":"0c06e30a8cebb24dce1c192c9a06a996df5c812d","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-06-13 15:23:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"56dbb5d6b1e1af06b20673039751d294ba2bd417","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 17:15:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/cbcc2395852048009c49ee524c8c187c : SUCCESS in 1h 47m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2056011d8a264923a8620949f4f5226d : SUCCESS in 1h 23m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b21d582cae4426896904876f4dd73a3 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31391681082648d0bca4c2326b996848 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/890acc64c844447b9fe6a87c31cbb2ea : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aff1a0b76fad42f3a0d6f2164ec0e1fb : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e726b4b618b143b38f0d36fd3aea5b08 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/649a26f2c28d4ad89602f91fa9892ace : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27f1ef624c89417ea03c042a1b4a719c : SUCCESS in 3m 59s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/25e01b1e89a84b45b7e11e35497b3622 : FAILURE in 5m 03s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/c0dda37e132f4584adef4c4adb10d5a1 : SUCCESS in 4m 57s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/88c8810fc78f481f94cfc18eecb93203 : FAILURE in 5m 44s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/64bb88b9819045bfbf0410e9ea22c63a : FAILURE in 3m 20s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/cc6d27ee4b3b4e2b81ad20a9de25bb00 : FAILURE in 3m 13s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/24aa583214124b3394fa08ada528d655 : FAILURE in 4m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5ad7b3efd4db4bdf86ca298fbf1f71b4 : SUCCESS in 1h 49m 14s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/67f0879c12a84cfaafe9e0ee81575e8a : SUCCESS in 1h 13m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/9b105074513b4a019af8ec5afa2b8b5b : FAILURE in 10m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b575b658876b49aa9a72db6eebb1f0f7 : SUCCESS in 1h 36m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/30d56e16a47441839866c0aadb8a6a0e : FAILURE in 9m 14s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/9e151ed46daf4e77af2724369639c013 : SUCCESS in 51m 38s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/eccf26511ea54aeea1ce39d6f3ffa114 : FAILURE in 14m 06s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/086356f0b4284921aea24b90ffe1b944 : FAILURE in 26m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/d473ef17d3764bc2804de2c023847b79 : FAILURE in 24m 43s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/d4ba6b9278914df5bbf8c2d6a35e40c6 : FAILURE in 26m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b519cdbebf40b13cb0bf1ae595b74cc1691c5377","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 17:15:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"96c3deb1ab59057c50c52be601edf7f75188d639","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 18:53:46.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e2ff1bafcd1b428aa7053851b63b0151 : SUCCESS in 1h 17m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e86276c5b1f8472c92bd6aebee04b243 : SUCCESS in 1h 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/748e0167b1144ee79ff9969397eb0d92 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b21a1e40102c4609b5bde04525d45d9a : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efa29fc287664eb29167398a12005fdb : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8bf75d276954233a20ed325561b1b2e : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf624b939c8d4f55b54ef73287eeb2ed : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/108f751d32454e70b13c3bcbe37d3781 : SUCCESS in 3m 02s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/0903729a17f046e59f1d1695f8ff09c3 : FAILURE in 4m 47s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9705d6ac04694dfd8a709376273b3e68 : SUCCESS in 1h 31m 54s","accounts_in_message":[],"_revision_number":2},{"id":"4a052bdaf148be4108d8a149bd9d4f92a7574eea","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-28 23:29:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f81b862e9d76fa33357298d2f49192e3547c9a8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-08 10:28:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2564b74f440a82119562f247977f9b33739a90a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 12:07:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e39ca436af84c61933954185ac6c07a\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d20d3cc5a12a49f882e58638a77744b5 : SUCCESS in 1h 23m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08272914ceb647178e4decccc10b24d6 : SUCCESS in 1h 30m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/640128a177a6467583423612dab45d1b : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d6f6df2a5e9486a9df7d1bac17d8a0a : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21124fd840b74345a3d208a42aa713f5 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f1247f205494b609e238408e026fa7e : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d855c0be38204279bf43e81bca04921e : SUCCESS in 11m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b535b875fdb43c58de10ddf93e5023c : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1ff2720ef334fccb15b04b2cacdc979 : SUCCESS in 3m 07s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/c19debca57f94e0eaa5f379125dfd263 : FAILURE in 4m 21s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/44448bd512204ef191fbf3d1c1ddac25 : SUCCESS in 7m 02s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/7da67069045343e9add7ef16d13c1217 : FAILURE in 3m 35s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/0db5dcd2ba85428d9b6324bb7f4043c6 : FAILURE in 2m 49s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/711afeddf56346859062a89036c2c3cc : FAILURE in 2m 40s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/e71945226ecd4a16948a4a14beacecc1 : FAILURE in 3m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3bdfc0cf3348493dba3ffb3f38851976 : SUCCESS in 1h 30m 55s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/db87bd3bd8904f6abdbe9b00e50d1ed6 : SUCCESS in 1h 29m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/71aa2f2050314230821871f73ead3223 : FAILURE in 19m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ba5fae10961b48bba07131cfde8cae26 : SUCCESS in 1h 29m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/027d3f0964804a28b75db1bc96cf8b4d : FAILURE in 7m 39s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/fd43c8e2a11b40b6a89f56c74d06b3a8 : SUCCESS in 1h 01m 09s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/c1741ce3645a4a77b2a8abe41251392b : FAILURE in 12m 45s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/b7374becf386462f92be17a4a4c1a168 : FAILURE in 26m 53s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/c5fb071c19df4e7eab745388ff735e5d : FAILURE in 34m 00s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/5dc13ac8aa694c5d853f211074d69d6e : FAILURE in 34m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"21f56d0fb82a3334abef51eb46b9175859d0c704","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 12:08:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f34fe78704e658c1c48b59a0a29f719b7dd9b427","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 13:49:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af14c71270e945e0abf1acd8ec6d052c\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/fa8df22a17a34dbc9c09ae8bf568af5d : SUCCESS in 1h 35m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2fd9d3fb537747f491113a60146564cb : SUCCESS in 1h 25m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d10e6652a21d409aa30508e8a949b620 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c022625fadbd44b2abdad120d695709b : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90cae0f29f794adb97259803d3f65cf1 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/303f5455bb444232bc958ee2885c6451 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/863617c14bf9472b93d263eeff16f625 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d114db8d19b446aa7d3009c5e480393 : SUCCESS in 3m 25s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/ce7c5f1aaeef4582ae0ac1ed4703a7e1 : SUCCESS in 4m 59s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b37bca464f154874b0ef3401e1cd4863 : SUCCESS in 1h 32m 52s","accounts_in_message":[],"_revision_number":2},{"id":"7dd270e81bdcd5442f05de03cbd8485a8f441c7a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 13:49:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"74bb13d836b1219f8537965a9228710229c89919","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 13:50:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/216e967ba20f4ac6916cc34e9dad96d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/716bba743246429c9d441f84ecdac71a : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad89be6297b846cca6db5de52eda3f89 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b2acc6663f6c3f60e07cdeb1eae97fd1210a4d81","revisions":{"56bab1deab167e621d0a263f3fcf95b75c7d1bad":{"kind":"REWORK","_number":1,"created":"2020-09-01 07:29:52.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/93/749193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/93/749193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/749193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/749193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/749193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/93/749193/1"}}},"commit":{"parents":[{"commit":"fd795bbb8f5aaa740fa810f2248646f6ec3a211e","subject":"Print warning message when connection running out","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fd795bbb8f5aaa740fa810f2248646f6ec3a211e"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-07-03 07:51:21.000000000","tz":480},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-01 07:29:11.000000000","tz":120},"subject":"Cancel consumer if queue down","message":"Cancel consumer if queue down\n\nPreviously, we have switched to use default exchanges\nto avoid excessive amounts of exchange not found messages.\nBut it does not actually solve the problem because\nreply_* queue is already gone and agent will not receive callbacks.\n\nafter some debugging, I found under some circumstances\nseems rabbitmq consumer does not receive basic cancel\nsignal when queue is already gone. This might due to\nrabbitmq try to restart consumer when queue is down\n(for example when split brain). In such cases,\nit might be better to fail early.\n\nby reading the code, seems like x-cancel-on-ha-failover\nis not dedicated to mirror queues only, https://github.com/rabbitmq/rabbitmq-server/blob/master/src/rabbit_channel.erl#L1894,\nhttps://github.com/rabbitmq/rabbitmq-server/blob/master/src/rabbit_channel.erl#L1926.\n\nBy failing early, in my own test setup,\nI could solve a certain case of exchange not found problem.\n\nChange-Id: I2ae53340783e4044dab58035bc0992dc08145b53\nRelated-bug: #1789177\nDepends-On: https://review.opendev.org/#/c/747892/\n(cherry picked from commit 196fa877a90d7eb0f82ec9e1c194eef3f98fc0b1)\n(cherry picked from commit 0a432c7fb107d04f7a41199fe9a8c4fbd344d009)\n(cherry picked from commit 5de11fa752ab8e37b95b1785f4c71210bf473f0c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/56bab1deab167e621d0a263f3fcf95b75c7d1bad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/56bab1deab167e621d0a263f3fcf95b75c7d1bad"}]},"branch":"refs/heads/stable/stein"},"b2acc6663f6c3f60e07cdeb1eae97fd1210a4d81":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-15 13:59:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/93/749193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/93/749193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/749193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/749193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/749193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/93/749193/2"}}},"commit":{"parents":[{"commit":"0c47ed4e1b03dd6999a3ae464eb311297a8ac4cd","subject":"Dropping lower constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0c47ed4e1b03dd6999a3ae464eb311297a8ac4cd"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-07-03 07:51:21.000000000","tz":480},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-15 13:59:07.000000000","tz":0},"subject":"Cancel consumer if queue down","message":"Cancel consumer if queue down\n\nPreviously, we have switched to use default exchanges\nto avoid excessive amounts of exchange not found messages.\nBut it does not actually solve the problem because\nreply_* queue is already gone and agent will not receive callbacks.\n\nafter some debugging, I found under some circumstances\nseems rabbitmq consumer does not receive basic cancel\nsignal when queue is already gone. This might due to\nrabbitmq try to restart consumer when queue is down\n(for example when split brain). In such cases,\nit might be better to fail early.\n\nby reading the code, seems like x-cancel-on-ha-failover\nis not dedicated to mirror queues only, https://github.com/rabbitmq/rabbitmq-server/blob/master/src/rabbit_channel.erl#L1894,\nhttps://github.com/rabbitmq/rabbitmq-server/blob/master/src/rabbit_channel.erl#L1926.\n\nBy failing early, in my own test setup,\nI could solve a certain case of exchange not found problem.\n\nChange-Id: I2ae53340783e4044dab58035bc0992dc08145b53\nRelated-bug: #1789177\nDepends-On: https://review.opendev.org/#/c/747892/\n(cherry picked from commit 196fa877a90d7eb0f82ec9e1c194eef3f98fc0b1)\n(cherry picked from commit 0a432c7fb107d04f7a41199fe9a8c4fbd344d009)\n(cherry picked from commit 5de11fa752ab8e37b95b1785f4c71210bf473f0c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b2acc6663f6c3f60e07cdeb1eae97fd1210a4d81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b2acc6663f6c3f60e07cdeb1eae97fd1210a4d81"}]},"branch":"refs/heads/stable/stein","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
