)]}'
{"id":"openstack%2Foslo.messaging~749196","triplet_id":"openstack%2Foslo.messaging~stable%2Fqueens~I2ae53340783e4044dab58035bc0992dc08145b53","project":"openstack/oslo.messaging","branch":"stable/queens","topic":"queens-cancel-consumer","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-11-30 16:24:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2ae53340783e4044dab58035bc0992dc08145b53","subject":"Cancel consumer if queue down","status":"ABANDONED","created":"2020-09-01 07:39:41.000000000","updated":"2022-11-30 16:24:48.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6174db9d17d906b6654f3f138bc37ef3c5f71b31","_number":749196,"virtual_id_number":749196,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"date":"2020-09-16 13:01:28.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-05 16:08:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"recommended":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":2,"date":"2020-09-03 14:30:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-04 11:16:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"date":"2020-09-01 07:39:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 07:39:41.000000000","updated_by":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"reviewer":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"state":"REVIEWER"},{"updated":"2020-09-03 14:30:46.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-09-04 11:16:45.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2020-09-16 14:44:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-28 23:28:10.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"b9cd2a1aee035298879528be20109e8712a26825","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-01 07:39:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eaa900ea5a81eb520e132a993f61a0cc1684a840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 09:12:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a7e1f5c698421295932761e4c2d601 : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbc9e2345f874319a4ada9271f9712d6 : SUCCESS in 6m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bce7f1196765462589cf0276186bcbc5 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c006889b054c908d3e695eabd36235 : SUCCESS in 5m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3d64cb6ab8a84f8b92aa9d26fc2fb46d : SUCCESS in 1h 20m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8307d2ed254f46138017d6776fb34196 : SUCCESS in 4m 37s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/f639cad65404431eb1d25a70dd46cc1a : FAILURE in 5m 15s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/d11bfbb48360491c941fdc5e1871c131 : FAILURE in 5m 42s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/2c1039175d534dd0a10f2071ca1ba7c9 : SUCCESS in 5m 25s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/03d80807c6a34fd0b05afefa42ce0c4e : FAILURE in 7m 23s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/27654e1aea39434da06c93150e0acd89 : SUCCESS in 6m 54s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/df9584afff5d405ca0f4c51202b66a18 : SUCCESS in 6m 42s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/f73fa7d0006948c09af2a1a7817a1fb8 : SUCCESS in 6m 21s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/bed7f132efa14d58bf0b05914b707105 : SUCCESS in 5m 14s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/d8bd85257edd4bb79d34e2d450f6087a : FAILURE in 7m 06s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/8aeea782636746fbb95bf3f463f5ad36 : SUCCESS in 5m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/3e37ae612df7407486e7d98971656aa7 : FAILURE in 6m 50s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/72ab87cdf2354f72b3e502c0e88e65d5 : FAILURE in 6m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/ce2bd07ca4bf47ddba0a0bac48b9c95c : FAILURE in 12m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/dabb173881fa42e18df9dab7773030c6 : FAILURE in 13m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/c9389b9af2004931806a94f5807ab3f5 : FAILURE in 8m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-pika-default https://zuul.opendev.org/t/openstack/build/681c099573f44ae69619e74ccb5efb14 : FAILURE in 6m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-zmq-default https://zuul.opendev.org/t/openstack/build/732a37de15bc49c88de6b2ca92f6f8f5 : FAILURE in 6m 30s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/75cf306625d94b61a225283cc3818f11 : FAILURE in 14m 56s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/745147dccd064a9c9cdbf6e6c81317de : FAILURE in 9m 01s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/28a1404536ae4b00a6cd6f7ad0ff315d : FAILURE in 8m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/b4e9b584424343a1934150ed4e8aa44d : FAILURE in 7m 54s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/71c0fc18ee7344d1ab385de2c5ad3b40 : FAILURE in 7m 24s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-pika https://zuul.opendev.org/t/openstack/build/775fc287a1324107b94cd33b24d2aab2 : FAILURE in 9m 12s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-zmq https://zuul.opendev.org/t/openstack/build/6a070bb9e46f489c813c6997b111ec75 : FAILURE in 8m 18s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/98e1dc6a6fc246d69754feb8f2e7f673 : FAILURE in 7m 38s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/37efd6eba2ee4c119e79b77e22c14a47 : FAILURE in 7m 17s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/c83e67715c2547429595321942f6dc70 : FAILURE in 8m 04s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-pika-default https://zuul.opendev.org/t/openstack/build/e9f43140df3d4b2d9b065f55656d4441 : FAILURE in 7m 26s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-zmq-default https://zuul.opendev.org/t/openstack/build/80fc78c45d0f4ff0baf8aa431565e88e : FAILURE in 7m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a09f250db9b489af6c088a18be922a663ded0e2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-03 10:44:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b95e7424d9150684cc2107e3e941f2b245dbb888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 12:24:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3861b35599243c1a9c260511626710a : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/36cb64483b244139b002db30b9aed7b3 : SUCCESS in 6m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/10c36f54f2b6498ca7c1eecec1916727 : FAILURE in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d80298a30df0471cb1be9d711b609465 : SUCCESS in 5m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/888b0fd120ac415e9be7f8e51cb356e8 : SUCCESS in 1h 18m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d2c0df583f84adf80bb6b2def4ed33b : SUCCESS in 5m 27s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/754a8581de36413ba6b5e38f974621ce : SUCCESS in 5m 09s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/d0de464904da4ed7b92dd62f01cc5764 : FAILURE in 6m 35s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/6bf867040b7d454c97ce176f35cf0dd2 : SUCCESS in 5m 46s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/96044354317a4eaa9e8a82f428c9ee36 : FAILURE in 5m 36s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/71494b581e4c45cb9a025f89a21e54aa : SUCCESS in 5m 57s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/5bfc9a5048b549d7b3aaf3ec9da01dff : FAILURE in 5m 57s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/580e165805814c7f9d02747669c2c9ae : SUCCESS in 5m 43s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/4f8407315e024fa182b74febe82857cd : SUCCESS in 5m 26s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/7f655b8a556e4be199f57c8b2726b2c5 : FAILURE in 4m 44s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/2acea4e8e1c24cb28fce51a388c024a3 : TIMED_OUT in 40m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/ee2e3d96045f4fae8cc3f41489b88e8b : SUCCESS in 1h 19m 28s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/67332fd88d6d49df8210d4db2f0a339a : FAILURE in 30m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/f64815d5a5fd44fc94c5e640cfbefa72 : FAILURE in 16m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/087ba6f22b9f46378c1318b9e7a67719 : FAILURE in 18m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/a9fb9c74fefd4c42ab149fa83ef2ae06 : FAILURE in 24m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-pika-default https://zuul.opendev.org/t/openstack/build/a8f09b9fdac24b80ad45148bca3d85af : FAILURE in 1h 10m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-zmq-default https://zuul.opendev.org/t/openstack/build/38477092af3540f98716757616982142 : FAILURE in 26m 48s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/3fb2c0fdd5294df0b968dac09996c78b : SUCCESS in 1h 08m 02s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/83d9bc3b38ad420b9209531d4605c97b : SUCCESS in 1h 06m 22s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/2e8f6fa9539d40dd9923c12e881dd32d : FAILURE in 44m 13s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/3573b21473e049898d558c548ca3c8db : FAILURE in 42m 14s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/b5ab6e28bd9942f09b3c74e67591d3a3 : FAILURE in 25m 32s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-pika https://zuul.opendev.org/t/openstack/build/d42917f8e42d4b5cb627154501ded8c3 : FAILURE in 54m 04s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-zmq https://zuul.opendev.org/t/openstack/build/990e6752f4634f719ebb3268251dcb5d : FAILURE in 23m 18s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/68806dca5a5d42bb8cd04e161f2f5d58 : SUCCESS in 1h 11m 00s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/27e4108208514dc39a2191c8beaaa420 : FAILURE in 29m 11s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/00ca78de05e543c08a2c5ebed80179ba : FAILURE in 37m 54s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-pika-default https://zuul.opendev.org/t/openstack/build/993b90e2d18b4e97a7846038cab8315e : FAILURE in 1h 25m 01s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-zmq-default https://zuul.opendev.org/t/openstack/build/76e2c9fc24e34750a03dce39dba4f77a : FAILURE in 23m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e9f1325c595949fb20865d9fa07bd9c733ed03b","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-09-03 14:30:46.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a6f8142c93912bf27cae9e4b164a32b4ba24195f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 16:14:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5072ac63e0946dfba0981934a15d43d : SUCCESS in 6m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08e028b89b034e85b9a20cc56cd05791 : SUCCESS in 6m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/99997a1d18594529b42aa19af576862f : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f5484d313d407c9a8a03b3daf056fb : SUCCESS in 5m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7196eb0246884c28b8ee2296751e6156 : SUCCESS in 1h 20m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81a9c567ae7f4e74a0f560b9a385e87d : SUCCESS in 3m 50s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/a8b4139758524282a9a5366f75f0f98e : SUCCESS in 6m 32s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/954490d62a1b4d32bb698e2712c56dc9 : FAILURE in 10m 07s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/9a4522d2444249e9acdf6ec0f00d4d01 : SUCCESS in 6m 15s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/672d3b61ee0a4f05972dea2d62d78e75 : FAILURE in 5m 28s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/dfd41fba31794e7eb7eb344f7dee5cb4 : SUCCESS in 5m 40s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/0406b8300f904e1c911f5830dfe609e6 : FAILURE in 5m 45s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/2723bfb8aa774a1f9e05efbacfe0122d : SUCCESS in 5m 45s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/2e89a574777e462abf4eb10c7e6867dd : SUCCESS in 5m 52s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/893341e1ec714405adde3d61ab7a948f : FAILURE in 6m 56s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/c035b34773ba4201a42da181b38e5d87 : SUCCESS in 5m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/88b9739ee61b4d529819685d84431b0a : SUCCESS in 1h 12m 48s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c503f2d338874ba4b96dfd691da741e7 : FAILURE in 29m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/90aade1877eb4d65908cac656371c8bf : FAILURE in 26m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/dadeee0b17864af292f904c35cb80b78 : FAILURE in 19m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/d12fd96e209e494682a16bd5d493d567 : FAILURE in 19m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-pika-default https://zuul.opendev.org/t/openstack/build/24139be941dc4714bd897c9342a20c39 : FAILURE in 1h 18m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-zmq-default https://zuul.opendev.org/t/openstack/build/4a7f2ba8c0a441628072b7db09c67d4c : FAILURE in 18m 01s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/a8841203900f4965b163b06b03c52fa2 : SUCCESS in 1h 18m 31s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/ec8e6b6262f84287a6a5c3575366d4d9 : SUCCESS in 1h 35m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/81d8db60ac9d43239f0177b714668e04 : FAILURE in 39m 29s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/205b9990a96a4f4e95b9927973f30dab : FAILURE in 38m 04s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/11f4cf7efb0e43f08549d3240217d833 : FAILURE in 32m 02s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-pika https://zuul.opendev.org/t/openstack/build/636a1a6ea16e4388ab733237a494d574 : FAILURE in 42m 34s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-zmq https://zuul.opendev.org/t/openstack/build/3af2ea70294e407ea962070ec94a231d : FAILURE in 26m 24s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/8a76680b5c8b4543a0d7e28b9ec87edb : SUCCESS in 48m 49s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e0948e40fa7949148f78eae2118cbf13 : FAILURE in 25m 22s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/dfb940eadb3e41319af141f9557ae37d : FAILURE in 33m 24s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-pika-default https://zuul.opendev.org/t/openstack/build/6544bd2985ba4d50b887bdc92a4288a7 : FAILURE in 1h 17m 24s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-zmq-default https://zuul.opendev.org/t/openstack/build/b91341d2ac86458aaa1377834950e8d2 : FAILURE in 20m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d61acc8de8863638a563406e118ad4236aa258f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 07:32:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2292dae17ca501f80fea6af92ced569b071b771a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 09:28:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07f5c9c15cd74c59af30a60de6a50bb0 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4492a0f658c74208b99072b51a905225 : SUCCESS in 6m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/807f2aa542c740c4b160bfeb2664206d : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa5caed437444e9afe27cb1eae460af : SUCCESS in 5m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2b013b382b7c4cf5be61dce834238807 : SUCCESS in 1h 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/985123abbb5e498c8ec728a4bb99dc0b : SUCCESS in 5m 32s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/ef60d9089f0c4fedb3fac4801926bfb1 : SUCCESS in 7m 12s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/facdbd527588432fbec7aa9b56364bb0 : FAILURE in 8m 28s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/661e29f545cc41f492a84aa821a43a33 : SUCCESS in 6m 23s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/92d6d5a719ad40b29e41244fe855fbae : FAILURE in 7m 00s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/79b85eba30da46f7b51d6ba637642955 : SUCCESS in 6m 53s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/3ee159a3623a424fab4106242a3cd852 : FAILURE in 9m 02s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/54d5c09d2c30485f9e7926f21e2e448a : TIMED_OUT in 41m 03s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/771c3cfd945b4bbabcef4aca48756ede : SUCCESS in 6m 40s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/275e07c9f877415c963e105256d7494b : SUCCESS in 6m 49s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/d0361f64eb1c4c05b394bb837f3deb9a : SUCCESS in 6m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/1bd93c7e5e6f49878dda5edb81d094d7 : SUCCESS in 1h 07m 42s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/49bdf8bbef3c4a3a8417442ba64666f2 : FAILURE in 28m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/193f755b56ab4fc98c8e443857cc64b1 : FAILURE in 13m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/af75e962a7aa4640b3fdf4c504546d0e : FAILURE in 11m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/5d3297fe2a1d4a089ae66a866dbba371 : FAILURE in 19m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-pika-default https://zuul.opendev.org/t/openstack/build/a2c16565f8fe495dbd48b3e023979d7e : FAILURE in 1h 13m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-zmq-default https://zuul.opendev.org/t/openstack/build/80d45a116dde4ead8a2f3fd4a1164a8c : FAILURE in 21m 29s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/be0de945def944faa943159df5bab81c : SUCCESS in 59m 36s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/f97f5eb9bfaf45f2b35db200095e6ac7 : SUCCESS in 1h 03m 31s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/8b6e865ec2c84576b0d43404fdc12410 : FAILURE in 40m 05s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/771637b4ad0141f796931610dd73e7c4 : FAILURE in 40m 03s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/743c2679aea2402ca312d5ab89e403ee : FAILURE in 24m 31s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-pika https://zuul.opendev.org/t/openstack/build/0556047a62f146cca22fc3c5a2081e03 : FAILURE in 30m 55s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-zmq https://zuul.opendev.org/t/openstack/build/58923e3c72fe4e8dbad121cb02bbb846 : FAILURE in 14m 13s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/e39b1501a7544b02b8f352b9f0aa1400 : SUCCESS in 1h 20m 33s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2caab6b00ccb4e44982e5e13a7106b0d : FAILURE in 36m 38s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/1b979d44843e4ff7b5bd5cd9fcb7a64b : FAILURE in 22m 46s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-pika-default https://zuul.opendev.org/t/openstack/build/47f99ee9ce32408cbbe0a004f816d21c : FAILURE in 1h 43m 25s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-zmq-default https://zuul.opendev.org/t/openstack/build/3e5571b10ad04c15bdb4aec760245246 : FAILURE in 20m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6fd7c94d579a7c4cf05dbb9bcef984be182579e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-09-04 11:16:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"83fd6cbbf879ac33b2d72e9451c0095fc1970e0f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-08 09:36:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"75b33bd42801117f2b2129b72023311a14198eb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 11:37:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74923d3acc3f42548dc5782660fc1fab : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cde949e2aac4401b8f26e9aa9dd2e345 : SUCCESS in 5m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/11846aa16619493b8b56d46e76f51bde : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e485c9fc09345d18e78a9f3c95ef56f : SUCCESS in 4m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3bb823ad62964716a37dcdce0bd0f6c2 : SUCCESS in 1h 22m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bacb50a9021c41ec8e377662acbd522e : SUCCESS in 4m 22s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/60941298501148fc90fe466cdb661a50 : SUCCESS in 5m 16s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/f2c76ea8afdf4e259ead75d65da054a2 : FAILURE in 4m 55s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/dd0a245ff700453687665374fa512383 : SUCCESS in 5m 01s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/b3655495736c4dcd92cd29389838f70c : FAILURE in 5m 23s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/84eb75a6bdc74daa87f94ffa67cac1a7 : SUCCESS in 5m 41s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/cb4b0a24672747cbb658436a29a9a557 : FAILURE in 5m 26s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/6d164b0da300453d81a429ecb73c4bb8 : SUCCESS in 5m 45s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/87cb5942f141460b996072b80e3b4a16 : SUCCESS in 5m 36s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/a7722f54adf84528b6ecdfa73fceca93 : FAILURE in 5m 24s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/4f1ef7e144474c8984eb48f0cc1966ec : TIMED_OUT in 41m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/f5ea43fc948744649e0f05490be6ad2d : SUCCESS in 1h 27m 51s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4e9cc78d9456436fb1ffb0f10ca74953 : FAILURE in 31m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/17ba3381cbdc4ce99f70908f3c75c83b : FAILURE in 17m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/1bd4be92857b4e4a9642a80b5b876cc7 : FAILURE in 17m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/e77552aa331541f7a137e181b5c5af88 : FAILURE in 23m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-pika-default https://zuul.opendev.org/t/openstack/build/d3508bea08564bba8fff92083995a0bf : FAILURE in 1h 18m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-zmq-default https://zuul.opendev.org/t/openstack/build/fc97b46bac3f4323bf3284bb9b503813 : FAILURE in 20m 36s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/456528808f36437a9f3975e5bb3d4fd1 : SUCCESS in 55m 27s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/747dd8525a86412bb9bdffdf07bd1dd9 : SUCCESS in 1h 04m 32s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/a63b9d6beeb846d391be5d8e4e0a841e : FAILURE in 40m 48s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/399288c1ff614b24af217a12ede712e1 : FAILURE in 50m 42s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/2e8abd4b189f4a0a809621ab8d8d36ea : FAILURE in 17m 06s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-pika https://zuul.opendev.org/t/openstack/build/588cb632c6554692a81278061d408f4d : FAILURE in 1h 02m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-zmq https://zuul.opendev.org/t/openstack/build/752fa96fea824a72a3493ee030941cae : FAILURE in 14m 15s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/a49798eba7c341f6a809f7aeecee60f2 : SUCCESS in 1h 54m 49s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6e977badb2ce4f769464043c7e214f85 : FAILURE in 23m 44s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/63b9d8273521480a85318f53ceb56436 : FAILURE in 15m 20s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-pika-default https://zuul.opendev.org/t/openstack/build/cab0da6a411940ee87b3d4405ed6d702 : FAILURE in 1h 04m 37s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-zmq-default https://zuul.opendev.org/t/openstack/build/8302a83841c74bf18a5aaf9ed0dab5d2 : FAILURE in 14m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"168e51d6ba7c504202c6ee456ed17bb03f48be58","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-09 11:41:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c4810204396cbf438a87542878d3c0ded73babee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 13:22:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3a9f0cfb6cf41c3a3e49d05bbe94d77 : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f86e207419fd41a291912d9330721e12 : SUCCESS in 6m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1438b478347a4cf29ffc2b58753c6dbc : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ba88d1b6da1467baa93bd19ca2ae97c : SUCCESS in 5m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c0405d81b841483ca0e0f941b3b6381a : SUCCESS in 1h 19m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ca1de842d7140749301287be24516b0 : SUCCESS in 6m 03s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/e896529a63444ad2adde2ec8090ddeba : SUCCESS in 7m 00s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/88a3a8b3c7984471b298cfda5485c6f7 : FAILURE in 6m 01s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/d3cad65776b74995bbcf994538ba818c : SUCCESS in 5m 59s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/5fa9ffca30404b4ca27a488300a62b5a : FAILURE in 5m 55s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/f8416a528b914025a1db91fb07c1eb79 : SUCCESS in 7m 20s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/3657e8bd1cef430296e1c3a870316591 : FAILURE in 6m 42s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/c695195ad0094c1b862facecd54899ff : TIMED_OUT in 41m 09s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/b532beb9292e471f967676ce2c8008b4 : SUCCESS in 6m 01s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/93ad1105a52a45b6a96c87b70d6742b2 : SUCCESS in 7m 44s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/aeb7e74c336c44178f83234839574da5 : SUCCESS in 7m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/962f99bffd34480980e76d5a7187bb66 : SUCCESS in 1h 15m 51s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d98398c1a5c548c2906e8c50e57d8c8e : FAILURE in 31m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/e13aa3d1acbe4000a6de5d31ed629321 : FAILURE in 12m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/3ddf0c23df62471a9edbfb6887bcde3c : FAILURE in 12m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/254f3b17de7b48daa31bbd874160a746 : FAILURE in 23m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-pika-default https://zuul.opendev.org/t/openstack/build/691552f24c4f4ac78bcec66b4091a07f : FAILURE in 1h 10m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-zmq-default https://zuul.opendev.org/t/openstack/build/f3422b2a7a71476db29701a3f603171c : FAILURE in 23m 14s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/ac8102fd2d0444fb86de3c95a7af1c12 : SUCCESS in 1h 04m 37s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/037482cf0047402ab6bb44796e63f783 : SUCCESS in 56m 25s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/06736e02f19844a19a50045ba0de26b9 : FAILURE in 37m 40s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/8f4dff910b454a88b9516b1d075e8dfa : FAILURE in 39m 20s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/60686b832cc143419429620f9328b7c3 : FAILURE in 23m 27s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-pika https://zuul.opendev.org/t/openstack/build/b841b6527edd447889bb113b8b6d379e : FAILURE in 45m 30s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-zmq https://zuul.opendev.org/t/openstack/build/3e0d110f42d3463c93874826578bdb78 : FAILURE in 18m 36s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/dc2987d78ee343919e8ae8ccb5533a70 : SUCCESS in 1h 26m 15s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/5dae6aa342d242f8a00bf2ccd56e5c0f : FAILURE in 36m 02s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/d4fb9dc52692483981247cec5d7e2ff6 : FAILURE in 23m 49s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-pika-default https://zuul.opendev.org/t/openstack/build/a858df087e874f8f915ba0ab747594f6 : FAILURE in 51m 18s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-zmq-default https://zuul.opendev.org/t/openstack/build/d5f313ba48744964974c5991036ec1ed : FAILURE in 18m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0bd02d71b5d5c7fd4fa462f2489849d03aca740e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-15 09:23:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7ed8e2aec7f3b0f71b5941c19718cdabf0c4e067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 10:41:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffc8dea75eb64082acf1e5def10c31b6 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/12b40ed1f7d0467eac2dc265948b1c8c : SUCCESS in 5m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7345c5494f8e4a858799bef9b4e5843f : FAILURE in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b4d48a6cd9e4840830f6d7a888f6143 : SUCCESS in 3m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7a4a4c3e88be42a1a7fb5c8b9a3ddc0b : SUCCESS in 1h 14m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99bac9e9fdfc41c0a1da13e7060c19d8 : SUCCESS in 4m 41s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/a524c5a2fb5b42e2878ec559e71ebb95 : SUCCESS in 5m 29s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/e852fa8e251b492199c7415f5b608b72 : FAILURE in 4m 53s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/f34eff343809401c830effe69dff0f6e : SUCCESS in 4m 56s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/b11c31df9525405e938fca1d8ecb6cb3 : FAILURE in 5m 06s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/2aff2ff3304645cca403fb78e94b3611 : SUCCESS in 5m 37s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/cda151b18f5d4590a47d4ec142c20a36 : POST_FAILURE in 7m 21s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/7778c3a826b74ff3b687080a6285b05a : SUCCESS in 6m 35s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/089082fc88e24fc78197fd64d11255b8 : SUCCESS in 5m 36s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/3509c7357c164be68ad80b61761ca88b : SUCCESS in 5m 17s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/3f3195741b7d4812b3b76d0aaef59245 : SUCCESS in 5m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/3891040b6c6541cdae0de8161b016c21 : SUCCESS in 1h 13m 03s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/b67e1c299d0a432bad874c43d19c3e7e : FAILURE in 33m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/571944c88564407c803ff792800f7fc0 : FAILURE in 11m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/bfb4679397124c6d95e39c6b2bb07784 : POST_FAILURE in 11m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/0b281f3720d1420caf8be1aebb57259e : FAILURE in 17m 18s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/7b445a11d04f4289bc7de466a5436164 : POST_FAILURE in 4m 54s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/16a6d5a1dbe24035a2480aa08089cad4 : SUCCESS in 1h 13m 20s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/b6293da8f16e4a2a817b417bed81a7da : POST_FAILURE in 40m 31s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/84bbac9857cd4bd8b72a9b02794dc18e : FAILURE in 54m 14s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/f0d0a7922f9b48dbb237197571b65107 : FAILURE in 22m 38s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/9a1a9b81ce62435485b824ae1df14598 : SUCCESS in 1h 12m 42s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2102b125f4ba4489998defd963858a99 : FAILURE in 43m 50s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/ac35020a60664c7187a5cc2ca14e78cd : FAILURE in 17m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7eaf5581825a9e7a7f2b369d13c4455de7b0b35a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-15 15:59:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"773ecd3160b9822adaffd9432569e220f573483e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 17:26:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6c9c29009b1495b941cff508c681067 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89b49b5141fa4e07a29e68b667488605 : SUCCESS in 6m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/804823046a19408998c9f5dd63f911ed : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/716f54cf4c1e4fbb87a6e38329664c23 : SUCCESS in 5m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f1e9e65d551542ee8ed85813cd78fdc2 : SUCCESS in 1h 17m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f3cfe4eb72744679248fcf3c344255d : SUCCESS in 3m 58s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/922548a11e6745139369d0ffc73d9972 : SUCCESS in 7m 09s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/d28a40ae2ef7411da3e1c9f43c7210e6 : FAILURE in 5m 42s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/909ff15ce3454615ae1d7370da9a924c : SUCCESS in 6m 53s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/495831c022024796bd25c47208d87ee4 : SUCCESS in 7m 01s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/dd1226defcf140398e0c6b8948902dd6 : SUCCESS in 7m 58s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/1875a9a8bd7c4fb5bf36c5d3c8a5fd6c : FAILURE in 6m 28s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/c1211947a4964e6694428c241a0495c7 : SUCCESS in 6m 41s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/412b25e8548c4a14af69224d5ae89ec1 : SUCCESS in 6m 25s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/3774356d45e8419b86465bf392cc009d : FAILURE in 6m 15s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/e442f3cb5c834ef884d3e940f632d0d8 : SUCCESS in 7m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/400382ea0afb4adb97f2cb8f72b96ba5 : SUCCESS in 1h 21m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d37fa75d2a554874837f0b4f2ae15c9b : FAILURE in 44m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/ed9f4b5e63e34674a299a0dc702e3ef2 : FAILURE in 19m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/80dc155e79b94deda8e8430850de070d : FAILURE in 19m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/fcb07d70a23f4362bbb3ccca4990e9c4 : FAILURE in 22m 30s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/e02e1008c84444d981e042431ed7307f : FAILURE in 4m 51s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/3b26850cb0fb4bb39baca834dc2d531d : SUCCESS in 1h 12m 28s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/51764020bf4d4f4d80904119b94fd542 : FAILURE in 47m 12s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/28488356e031444d8afc7adc5dbe1df8 : FAILURE in 6m 28s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/a7832734df02420ba4c99c93cce026e0 : FAILURE in 35m 10s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/776045a7da8847cf8f639c404d855b2f : FAILURE in 5m 17s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/658daf0be3f14986a445add9171fff5f : FAILURE in 30m 09s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/72279e6775da4f53b2769e8dc8f12d51 : FAILURE in 34m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"212ea808f083c052657c611b0d64854ce3f8216f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-16 10:04:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"188ba5bf4e536454f3a117dc645e4cb9d2e80769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 11:27:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ef803b3a84b4a46acddeb337e6bce68 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5332234cc93e4cf2bbbd5bddb99920ad : SUCCESS in 5m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3ff14c81087a4e02998b9e2aaa8d5f83 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2dbf1c4986445f5b3a965ae47613d4b : SUCCESS in 4m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0c8fd5bb58b14137bc44927678a16fde : SUCCESS in 1h 17m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8034cdc14c546da8eb109d987fa6f8a : SUCCESS in 4m 38s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/c0b6cbc076f541a193fadfa9970334c8 : SUCCESS in 5m 09s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/c9d1e26d1b014f9f9b3d2e78ef46d332 : FAILURE in 6m 59s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/04c287bfefca48c797e17cafac198a90 : SUCCESS in 6m 03s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/798a600a64474e3cb7e7719852c8a0ba : FAILURE in 6m 17s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/e04fd976032941928165d0931235eb52 : SUCCESS in 6m 46s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/2038c3313658476db68378b1fd629189 : SUCCESS in 5m 29s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/8f77f0234fc549468b898c55f5b0d5c9 : SUCCESS in 5m 55s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/6d172343536241bda29d47bf7188f9e0 : SUCCESS in 5m 37s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/f788fc61c6104e6dbc958d2006ed56f8 : FAILURE in 8m 39s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/5d4b49debf594493a7a1d772b0690b15 : TIMED_OUT in 41m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/35ff561d93994cbab8151fba31157e97 : SUCCESS in 1h 15m 17s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7835823b4b6d49e79d3912f4886859f4 : FAILURE in 31m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/d2aea5810d634007973819c830efebcd : FAILURE in 22m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/0eb8d203c0b246338d15440b967d8e43 : FAILURE in 17m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/bc621f7c75f94acb88025e93e2c8d897 : FAILURE in 18m 24s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/b27ee3ac2661495386e3690e1508971c : SUCCESS in 1h 17m 33s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/fbdb61dd8a164fb6b168dc9a70f8a032 : SUCCESS in 1h 04m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/21822c2c05d34373a279aa1e30becd97 : FAILURE in 38m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/94f47acc5b88453084ef76bbbc09480e : FAILURE in 58m 06s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/51e9df15b9e0492c9286886ca0383b8c : FAILURE in 26m 15s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/2403bcf1847a488f9f04be11e9bb9635 : SUCCESS in 1h 05m 59s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/85883caa5db84bff8c197a079e3915a4 : FAILURE in 24m 43s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/01698476f4894cb0bf6bcbc7448abbce : FAILURE in 36m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f1fc698b0ca8df85187d3a6c8d8fc1d3eb072fb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-16 13:01:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cb3af13e69eeb6a438231cbb7c588026ebf225c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 14:44:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71becce5d7264e1ab604596344c21f6b : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/87c557f8f86e484a906f3c291a01cab6 : SUCCESS in 6m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/08f4bdc5563447db86d8fc1d041375a6 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/366f2b5efb1241a7811d75d6d6d5483f : SUCCESS in 6m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b20cf41934f34538bb347920b633f28d : SUCCESS in 1h 18m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c27192b9bec44f0fb36c6d6390635aac : SUCCESS in 4m 28s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/51ff6468b24b47eaa8dbe9a39315d417 : SUCCESS in 7m 33s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/8aa83ad2aa8242439918eed5ce6b5967 : FAILURE in 7m 41s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/3791476ee6ec454d8e883da993d4f50b : SUCCESS in 8m 00s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/9c83957716e546569bccc25b64f0645b : SUCCESS in 6m 45s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/2a700a3dae644ebda92e6b82794969c2 : SUCCESS in 5m 45s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/c574cc98e1ba4df69ea5517370f79bc4 : FAILURE in 9m 11s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/ca74e0d2bc804049a6e8502598becd13 : SUCCESS in 6m 18s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/655e24a3f5694e1097f1db68e9835f20 : SUCCESS in 7m 32s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/c8f4bf18f4d84449b00e71f07ce7ff2f : SUCCESS in 5m 54s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/5ccdbfca050e4a77a8671a4b684aaf53 : SUCCESS in 7m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/8b48cf4e2aa944e1b7510d9f581095f1 : SUCCESS in 1h 12m 43s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a3cd3145d3584beba064511fb2599ef0 : FAILURE in 35m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/a4f412a1790b4321adcc0bf53d8f185b : FAILURE in 22m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/dd69939b7b11407aa0d4076d0a51657c : FAILURE in 11m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/f67e5018a3e34b2faf346fed093d5deb : FAILURE in 20m 52s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/333645ef8d44439aa34e4ef77ee60c19 : SUCCESS in 55m 06s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/4a5e1cc3a01248f792288e08f2a11094 : SUCCESS in 1h 04m 19s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/18bd64a0cb104ec398f84a1785f6b04f : FAILURE in 39m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/e0541d4c38fc48df84974ee3a2d330d0 : FAILURE in 1h 08m 18s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/ea455fca192c4f9e8257673598120d28 : FAILURE in 21m 41s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/b9b85c0e8c9548f2a48cc44638aededa : SUCCESS in 1h 17m 18s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/9d4f41c0761346c5916535c49ac500b5 : FAILURE in 30m 08s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/8da314c29d8b405582b7ee5995ca0bce : FAILURE in 17m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ac98af051a41180d7af554f42080d0cc7670cfc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 14:26:43.000000000","message":"Abandoned\n\nQueens is no longer maintained (https://releases.openstack.org/).","accounts_in_message":[],"_revision_number":1},{"id":"90c90f82322d60ec852406394ec54cd7aa76f60d","tag":"autogenerated:gerrit:restore","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 14:28:00.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"c77e66ad8ebce3b48ef19846e0654f16fe2e82e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 16:08:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53abdcd741ba48f581bcafbbab660f16 : FAILURE in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af04abb02e5b4d32ae86a73933eab0ac : FAILURE in 4m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1f414570b3244ddb9d52bc18834d6e23 : FAILURE in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/187214f4b92b4ca981e40d91551158b0 : FAILURE in 4m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4e169166f2fc4235872d3e1948af7c89 : SUCCESS in 1h 15m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07bd4968c6464fefa6a0db1e87eebdfe : SUCCESS in 5m 08s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/42debb1a15574da288b9aa653c77d203 : FAILURE in 3m 57s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/677f5eb7f08a494f8c026470aead21eb : FAILURE in 5m 06s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/c40c0b13c55e4a73ab95e7aaa2114098 : FAILURE in 4m 43s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/bea677e99d0b4a328201dfe02e4a82f1 : FAILURE in 4m 11s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/8481e8dd374248a881c8b5ea6d808fc6 : FAILURE in 4m 49s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/115a131edff847f68c7bfdfae67f44b0 : FAILURE in 4m 55s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/c3bcb5eb383044de8549c12ed8228763 : FAILURE in 5m 01s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/d428ba3d2bde42778e560f1e5cba65fc : FAILURE in 4m 11s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/294136c4c163428296e7def4400b73aa : FAILURE in 4m 49s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/571cc07d46e74b9fa6cfeb249df20bdf : FAILURE in 5m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/0a54807647f14491bbfe304d460224b6 : SUCCESS in 1h 08m 54s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4e40cab3572648b2878f27ae737f0dbb : SUCCESS in 1h 17m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/b4e5bdee65dc45fd826d427f042873e8 : SUCCESS in 1h 37m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/43f5fd34efd94c699a9fc940d330b554 : FAILURE in 21m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/52618a1231d24a61aa2761bbfba4dabc : FAILURE in 18m 42s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/47e034bc28cb46389eaa76ac04b365a7 : SUCCESS in 51m 15s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/d99dd9a2d1544082b2c822e8a625b0dd : SUCCESS in 1h 01m 29s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/ec23742b37414a0e84cf08dc9017a24e : FAILURE in 27m 20s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/b54240d7f4ea4785903e1e0497460fa5 : FAILURE in 26m 48s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/36a9fda84ecb4579922d7f648f6b2031 : FAILURE in 21m 16s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/a316342b87814f059f586f4043f2f265 : SUCCESS in 1h 17m 05s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/18f8875e565140028ded43443e7878ae : SUCCESS in 1h 18m 57s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/56dbbd21bb864c3998b965b6b7ce591b : FAILURE in 17m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6894096120c6630213a65882293eb481e8b7c328","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-02-08 13:49:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da439b882c90983391f55e22165614f48789083a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 15:31:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f5b618e407143b68884935a0caa9bec : FAILURE in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0649de149f624e669eabb3f7a4781dca : FAILURE in 5m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7b523893c8a242eea8e6efe701c3686b : FAILURE in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bfa34fc88e54fa9b411641ea465e5a6 : FAILURE in 4m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fca3c04b38bc4606a4df2df2341b6d1b : SUCCESS in 1h 17m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35259ff089464c82acb9ffa7f9bccef1 : SUCCESS in 6m 52s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/e8dbeb4f75fa4ee9a81edd23e29cd753 : FAILURE in 4m 40s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/63239af21490404895a2a6802097d2fd : FAILURE in 5m 12s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/e9d37ec9152644cca541055117b33b1c : FAILURE in 5m 12s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/19125c79101e4aaaa1e719810fec46fb : FAILURE in 4m 51s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/699cdb8ebebc4975a69c0f601dde85ff : FAILURE in 4m 40s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/ff31adbb858b470e8084750ae84200a9 : FAILURE in 5m 16s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/3ac1566a8b924bd08b4ef47ab40dc624 : FAILURE in 6m 12s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/10d81f3d9bb4458eaea15bc3faeeb805 : FAILURE in 4m 51s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/4bf9fe90d8cf42bf9f58fad4e2e70d31 : FAILURE in 4m 52s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/bc45d0c8bd6c49b786ec71df77c048ee : FAILURE in 5m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/7cb99212ea28476fafe03eb779ddaa1c : SUCCESS in 1h 16m 27s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/49cf4b5787804bad8d770f039e39c476 : SUCCESS in 1h 16m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/194f9396fcbe453ba45df3139dc93d08 : FAILURE in 22m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/2f0559500e624f1699e28606e17c8f4a : FAILURE in 21m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/d4dd16a6552e43438f1841243fb30083 : FAILURE in 17m 29s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/b2eab7241f0c4823900ca312f252f300 : SUCCESS in 54m 13s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/c686e954989b45cc973b6f3aedb74a0b : SUCCESS in 1h 05m 04s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/c3c29c953b3b4994b3d0aa98a4f00813 : FAILURE in 31m 12s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/63d353c0006a465ba86eba546d99526b : FAILURE in 29m 46s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/cbb22d7f9ce74aa7b261c96dd91f9c29 : FAILURE in 24m 19s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/1c886e2f5a904e40a3a366a1e49fe93d : SUCCESS in 1h 12m 44s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/38e5c3f3ca814557ab2f3130ef111398 : SUCCESS in 1h 13m 59s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/e4312bfee2294433a6f5c38ecbbb3207 : FAILURE in 26m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"538d66597c6c239cde15fb81d410b9504b242f53","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-02-09 15:52:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68434949a552220a316653500f284ab8f9ca31f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 17:43:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d45c9e234b55439e82239b4c96b183ee : FAILURE in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c0d8a1b18434472c8930c7b59b8876db : FAILURE in 10m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/27bce2147b59484fb2a3072339c2282a : FAILURE in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c592e6c9865f41ecb171801c796ecf48 : FAILURE in 7m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c6b6d69ebaf242b98407acafd534ff22 : SUCCESS in 1h 17m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22d35878dcaf415d995613dd6ef29e3c : SUCCESS in 5m 43s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/0ce29c2402a84fa2b9d737560077dc8c : FAILURE in 7m 41s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/1ff3a0c955f44fee91e8214c47a8928a : FAILURE in 6m 37s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/373cd47236034104a6380e45fd51a550 : FAILURE in 6m 22s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/d63fc5452eaf42eb96dc1552b6ca9510 : FAILURE in 5m 32s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/84ed7bbf63784eeeaeebee12dbbef009 : FAILURE in 8m 17s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/6d4cdcd283f24669aa89b2dce7ecc1bc : FAILURE in 7m 56s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/5757d750d0254cc1b5366d48babab78f : FAILURE in 7m 47s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/69f003c10bb24bc0a230077573deb669 : FAILURE in 9m 15s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/e93a3bbf988e43ff8a6ee0385924e4f6 : FAILURE in 8m 07s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/a08cabc5f4db43b09cb884c47cce4c71 : FAILURE in 7m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/642e3b3dd861429680c100e66ca50fd6 : SUCCESS in 1h 19m 34s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/37f65a4fc99c463bacc0ceb9d659775a : SUCCESS in 1h 14m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/1985c48021404bb39c7ff9690862200c : FAILURE in 28m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/544aa86053964a799ab957048d8eaf5c : FAILURE in 27m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/023ecede5c2948eb86cd25662981e5bd : FAILURE in 26m 21s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/fd3faec0f29d437f81ec090184eb51e1 : SUCCESS in 53m 04s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/f2397a1f5d5d466e977364d93f0cc658 : SUCCESS in 1h 03m 48s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/60362809f5124134ac35c48b7133d703 : FAILURE in 26m 55s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/c1278951830e4a948ffe146f222bdfb2 : FAILURE in 28m 01s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/310f67829e864cc5a6bdc04ac79e5df1 : FAILURE in 28m 20s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/2956215429d74fad820d466e422fc08b : SUCCESS in 1h 04m 16s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/9fc23b85b56545749698857cacb7dcef : SUCCESS in 1h 22m 39s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/b8722981b6a44249b20923ea23c7ea58 : FAILURE in 26m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b131a5f632164a06ac267d5a11de08e633f92b7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-28 23:28:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6174db9d17d906b6654f3f138bc37ef3c5f71b31","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-30 16:24:48.000000000","message":"Abandoned\n\nThis branch of this project has transitioned to End of Life, open patches need to be abandoned to be able to delete the branch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a74872ac890461a0abbc79ab995231bffe252722","revisions":{"a74872ac890461a0abbc79ab995231bffe252722":{"kind":"REWORK","_number":1,"created":"2020-09-01 07:39:41.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/96/749196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/96/749196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/749196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/749196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/749196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/96/749196/1"}}},"commit":{"parents":[{"commit":"c46367dad2904cfd3ff1b9cd722a08820dd60218","subject":"Simplify tools/test-setup.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c46367dad2904cfd3ff1b9cd722a08820dd60218"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-07-03 07:51:21.000000000","tz":480},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-01 07:39:05.000000000","tz":120},"subject":"Cancel consumer if queue down","message":"Cancel consumer if queue down\n\nPreviously, we have switched to use default exchanges\nto avoid excessive amounts of exchange not found messages.\nBut it does not actually solve the problem because\nreply_* queue is already gone and agent will not receive callbacks.\n\nafter some debugging, I found under some circumstances\nseems rabbitmq consumer does not receive basic cancel\nsignal when queue is already gone. This might due to\nrabbitmq try to restart consumer when queue is down\n(for example when split brain). In such cases,\nit might be better to fail early.\n\nby reading the code, seems like x-cancel-on-ha-failover\nis not dedicated to mirror queues only, https://github.com/rabbitmq/rabbitmq-server/blob/master/src/rabbit_channel.erl#L1894,\nhttps://github.com/rabbitmq/rabbitmq-server/blob/master/src/rabbit_channel.erl#L1926.\n\nBy failing early, in my own test setup,\nI could solve a certain case of exchange not found problem.\n\nChange-Id: I2ae53340783e4044dab58035bc0992dc08145b53\nRelated-bug: #1789177\nDepends-On: https://review.opendev.org/#/c/749194/\n(cherry picked from commit 196fa877a90d7eb0f82ec9e1c194eef3f98fc0b1)\n(cherry picked from commit 0a432c7fb107d04f7a41199fe9a8c4fbd344d009)\n(cherry picked from commit 5de11fa752ab8e37b95b1785f4c71210bf473f0c)\n(cherry picked from commit 56bab1deab167e621d0a263f3fcf95b75c7d1bad)\n(cherry picked from commit c49667d54ab792bd017989b245ec37e711fc81e3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a74872ac890461a0abbc79ab995231bffe252722"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a74872ac890461a0abbc79ab995231bffe252722"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
