)]}'
{"id":"openstack%2Foslo.messaging~753526","triplet_id":"openstack%2Foslo.messaging~stable%2Frocky~I8f6b4a669ec11721f76cf03abcd7d802c3c11eb5","project":"openstack/oslo.messaging","branch":"stable/rocky","topic":"bug/1857319-stable/rocky","hashtags":[],"change_id":"I8f6b4a669ec11721f76cf03abcd7d802c3c11eb5","subject":"Don\u0027t log NoSuchMethod for special non-existing methods","status":"ABANDONED","created":"2020-09-23 07:03:55.000000000","updated":"2020-09-24 07:30:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5b7258355a8934bdfbb28ddbaba748a64db90742","_number":753526,"virtual_id_number":753526,"owner":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-23 08:41:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2020-09-23 09:26:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 08:41:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-23 09:26:12.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"66c54d2f3b0eeb0a065e538792d080f29650e2d2","author":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"date":"2020-09-23 07:03:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2da8422cf01b20ab5de89039d5d4ae2a27de1cda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 08:41:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9191d07642ed408db570e863dd8fde2b : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/037100cb83e34f68a8c69ae72facf128 : SUCCESS in 6m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6935b022c3234509b6dc1892f81a1656 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49b289fcfde64341ac18fe0abd224d30 : SUCCESS in 4m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/63661cd222a54937a57dc077e52ac5dc : SUCCESS in 1h 31m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae66d0a68c6347988ce7d8893087de87 : SUCCESS in 4m 07s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/18950206bc25483697a808f2bd6f90ff : SUCCESS in 5m 45s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/815df52f224c46129d8238dd17be816e : SUCCESS in 6m 21s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/655b18980f8148a0bf4696e6f4a59cb6 : FAILURE in 5m 58s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/1d776612cc59429e98af0213dec025cc : SUCCESS in 5m 48s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/812e91c999ca40b5bcf9f78d2f14542f : SUCCESS in 5m 33s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/86de4abde7ec4912885a812323d70fa3 : FAILURE in 6m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f655d47dc0834dc58190e8ad4defffca : SUCCESS in 1h 18m 27s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4d1c12600cbc47199301928b39fd210c : FAILURE in 34m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/6383290b7c2a409bbdb245eb5cf8b3ca : FAILURE in 21m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f3945b2c62644cef8e7775f07cabf4e1 : FAILURE in 31m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/132ab8b8ca7c47148477614b93e6838e : FAILURE in 17m 28s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f851c14bee6a4e8c9a75e0ec75e2d2be : SUCCESS in 57m 31s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/ca2214ab9d0d4720ad5c9ffc74381b17 : SUCCESS in 57m 04s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/2a077d749e324554a4ad28cdc000bd87 : FAILURE in 41m 10s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/634f95f9bd9742d490e41b67a92e40db : FAILURE in 39m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/4ef363f3b3d647c9950389c073b60e5d : FAILURE in 40m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccf9d2e3a7444933839e6af08bc36101 : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"bd0b92b921c24e1081fedf29247cb32da9c2f547","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-23 09:26:09.000000000","message":"Patch Set 1: Code-Review-2\n\nThis needs to be backported in order, so stable/ussuri -\u003e stable/train -\u003e ...","accounts_in_message":[],"_revision_number":1},{"id":"a29b3a34ce630bfda1001a71b6e6cb141d634365","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-23 09:26:12.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"ec16ee780a763962ad1e15833922524a692a3312","author":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"date":"2020-09-24 07:30:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c2a66e8c78d7c4325734b55186f6aead52029e6","revisions":{"3c2a66e8c78d7c4325734b55186f6aead52029e6":{"kind":"REWORK","_number":1,"created":"2020-09-23 07:03:55.000000000","uploader":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"ref":"refs/changes/26/753526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/26/753526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/26/753526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/26/753526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/26/753526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/26/753526/1"}}},"commit":{"parents":[{"commit":"171132ff69bb0edcc0ebe1044ae59d1a6d572fd3","subject":"Print warning message when connection running out","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/171132ff69bb0edcc0ebe1044ae59d1a6d572fd3"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-12-23 07:32:28.000000000","tz":240},"committer":{"name":"zhen","email":"zhen00fa@gmail.com","date":"2020-09-23 07:03:55.000000000","tz":0},"subject":"Don\u0027t log NoSuchMethod for special non-existing methods","message":"Don\u0027t log NoSuchMethod for special non-existing methods\n\nSuch non-existing methods are used as health probes.\nPlease see bug for details.\n\nChange-Id: I8f6b4a669ec11721f76cf03abcd7d802c3c11eb5\nCloses-Bug: #1857319\n(cherry picked from commit d7eb89eac8f76c46849f5e0d18a7ea00dc2d07bb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3c2a66e8c78d7c4325734b55186f6aead52029e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3c2a66e8c78d7c4325734b55186f6aead52029e6"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
