)]}'
{"id":"openstack%2Foslo.messaging~757242","triplet_id":"openstack%2Foslo.messaging~stable%2Ftrain~I97287e81b1ea6a910e2543eea79c0d78e51743d6","project":"openstack/oslo.messaging","branch":"stable/train","topic":"bug/1883038","hashtags":[],"change_id":"I97287e81b1ea6a910e2543eea79c0d78e51743d6","subject":"Catch ConnectionForced Exception","status":"ABANDONED","created":"2020-10-10 03:23:45.000000000","updated":"2021-09-02 06:23:29.000000000","total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f0fe61401515aa059eac39f69b349a3a384e6809","_number":757242,"virtual_id_number":757242,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-10 05:05:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"date":"2021-08-30 15:11:56.000000000","_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":-1,"date":"2021-08-11 09:12:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2021-08-11 09:05:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-10-14 09:33:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-10 05:05:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 09:33:48.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2021-05-19 09:50:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-08-11 07:06:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-11 09:12:47.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"08742bdd5b911c8538cd5a90048bf4744c05fa94","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-10-10 03:23:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d72770da78d817c71bf40e52b6036a1267225551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 05:05:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d2df664962454ac58d489f0393f61e2e : SUCCESS in 1h 31m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba604d3f95744bc1943109fbdad726c3 : SUCCESS in 1h 26m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05859c2a105c4bbe95599e77668a4ff5 : SUCCESS in 11m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6cb9e87487647bbbc32f25804e3081f : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b7725d3725a4fa898612cad1c3669f3 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b76cb61c91e847c0898ce5a7a337e369 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdf18ee668ce4f29822363829158e334 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ec282bf01c7482ea1901ae011ef57db : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e97c7a57c0484213ae0cc743e684c745 : SUCCESS in 4m 09s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/eaddd23394504c6a86ac05b6f9132f35 : SUCCESS in 4m 52s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/1914394e9e734292967c5a1d3b119355 : FAILURE in 11m 28s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/64f03ed9c8e94e079e9fd344d67c46f0 : SUCCESS in 6m 03s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/5e6096a53d974f0d94031e29eb5e73a7 : SUCCESS in 5m 09s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/4d42673766df4ff0b5b4fb0b282bcc90 : FAILURE in 11m 34s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/f3aaaefbcafe45a3b752fcd310d5a1cb : SUCCESS in 5m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f2e895ad5b804f5b8bec01e9832145fd : SUCCESS in 1h 32m 49s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2684525dcfbe4a1f848e4583902a3391 : SUCCESS in 1h 29m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/e7447e54812f4c0bbf6e594c21823b81 : FAILURE in 16m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/34b11229ed5140a3afb927a56b2e2659 : SUCCESS in 1h 26m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/02fa73ce5cf6401282f11387c9880357 : FAILURE in 11m 16s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/a23df406db0b4a268713f1a066cdf6ba : SUCCESS in 1h 00m 45s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/5b79e9aefd3b42928d759ca7dd48ccc2 : FAILURE in 25m 03s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/59a7f4f19c634d8ba70b08b654f3d979 : FAILURE in 1h 22m 30s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/fe81db8c3a3740d9b81b86d828b66141 : FAILURE in 48m 12s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/1a169d7e8b6d40d3a69cb5f4e4a57891 : FAILURE in 41m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aae2b14433b39d0034ce4d4dc2fd55a9ac037c78","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-14 09:33:48.000000000","message":"Patch Set 1: Code-Review-1\n\nSounds good but can you explain why you don\u0027t make a cherry-pick please?","accounts_in_message":[],"_revision_number":1},{"id":"554161741009123a9fd3ecc31477aba9813c3160","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-10-15 12:23:24.000000000","message":"Patch Set 1:\n\n\u003e Sounds good but can you explain why you don\u0027t make a cherry-pick\n \u003e please?\n\nthank you for reply. The reason is the original patch does not consider python2.7 and thus will fail every python2.7 test cases. This patch in my mind will be compatible with both 2.7 and 3.x. But it looks not very elegant to myself. Hopes this explains my idea...","accounts_in_message":[],"_revision_number":1},{"id":"3e6a23b7193cfea5adc902a727b148c2a2d55ce7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-19 09:50:33.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Sounds good but can you explain why you don\u0027t make a cherry-pick\n\u003e  \u003e please?\n\u003e \n\u003e thank you for reply. The reason is the original patch does not consider python2.7 and thus will fail every python2.7 test cases. This patch in my mind will be compatible with both 2.7 and 3.x. But it looks not very elegant to myself. Hopes this explains my idea...\n\nHello,\n\nFirst sorry for our late reply.\n\nI\u0027m not sure to see what\u0027s not compatible with py2.7 in the original patch?\n\nhttps://opendev.org/openstack/oslo.messaging/commit/26e8bb4cc9097fddf3c2e34599b131e3f5d73867\n\nAlso we should notice that train will switch to extended maintenance soon so it could be interesting to see these changes released before EM.\n\nhttps://review.opendev.org/c/openstack/releases/+/790765","accounts_in_message":[],"_revision_number":1},{"id":"24427fdd77e8ede5954489d8e85e66c9e3044c48","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-03 15:01:44.000000000","message":"Patch Set 1:\n\nrecheck\n\nI have digged this topic a bit more, and, indeed, this patch seems a good approach.\nI couldn\u0027t ran successfully the py2.7 with your other patch even by tweaking it... https://review.opendev.org/c/openstack/oslo.messaging/+/744262\n\nLet\u0027s give another try to this solution, if the test works then I think that we could continue with this patch. I just recheck to refresh the logs and observe them again.","accounts_in_message":[],"_revision_number":1},{"id":"086995db8ffaba3021203ad00facaf9e6001723a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 16:53:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ef60ac96cb734972bfdf23b9cf0935f0 : SUCCESS in 1h 24m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6af78ebf99454dababbce4dab489c6c9 : SUCCESS in 1h 14m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9ecbd3aefc5418e8cba6a6a51147924 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11ad547f5ca140ada4c772ea6a117a31 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95325bf8226a4485a1b75317f0a551a9 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87737d4282c2414ca5f4d31ca86efb91 : SUCCESS in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3e1e35f34e4450e8caf04b7f0125f67 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a6e379d8dcd4f3182b8050b45738408 : SUCCESS in 4m 56s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/8ac829de7c7a45c6996aa5683fda4f72 : FAILURE in 9m 05s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/8890f0c4a9914439a83ec16897c94166 : FAILURE in 13m 07s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/897761f2705646d5b2629f927622486d : SUCCESS in 6m 19s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/eefd072cfc6f4da1a7d56070e553ce06 : SUCCESS in 6m 16s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/600619942a4f488b94828cff7db35bd5 : FAILURE in 13m 22s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/c6f22480fafb407093dc7f2dc438c2bc : SUCCESS in 6m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/764ce5922df34ade8df44bbbc2a86cec : SUCCESS in 1h 24m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/79a6628b1b16492f94da9aa3fa139388 : SUCCESS in 1h 45m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/5a9c4675572d4148b388f8daddcfa163 : FAILURE in 15m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ad2adf74c723460f99c75b59c0099360 : SUCCESS in 1h 31m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/c71f14bd8978493fa9f052299ef4d65e : FAILURE in 17m 16s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/9dd7b42ef6db4e6da96e79bd8f8986d0 : FAILURE in 11m 29s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/12018559af5442198bec2cc23c1447b3 : FAILURE in 19m 39s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/e4bc0f20c260406c80b6bc759571382a : FAILURE in 14m 28s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/0d276e78829b49cd95b02b9a13eb9ce0 : FAILURE in 14m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/4db3156bcbe54be0b7a70451fcdaa8a0 : FAILURE in 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e8b2bc88d4395a355d19174183d1b2cfac317a2","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2021-08-04 07:16:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b1a8ea19e329486e453911705b763459d52e9a6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 07:56:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7fe2e1b01259a60726d0fc88b6714c6b677dad5c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 14:07:18.000000000","message":"Abandoned\n\nrocky is now unmaintained","accounts_in_message":[],"_revision_number":1},{"id":"d9a62b87aa044ee228181398248ee56d35afb7d1","tag":"autogenerated:gerrit:restore","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 14:07:32.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"c5c41f4776be72853b4e959335dc506fb5761da2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 14:08:11.000000000","message":"Patch Set 1:\n\nSorry ignore my previous comment, wrong patch on wrong window :(","accounts_in_message":[],"_revision_number":1},{"id":"f13d4687ade1578966ed0eb5abd94e3be7343024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 15:54:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/83fc641ff849458d938447bb8e402bee : SUCCESS in 1h 38m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79490b47b346492496dbd5762559ebd5 : SUCCESS in 1h 15m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/491e0c490bf449c6b784022da3154d86 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e75545c5b5614bad9e159054b5785b9d : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc842f7b021341de950ca52b35cd8bfe : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97622f5a5de9401194207a016c892881 : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed1337c149b243229112532135baee58 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92109b216c9a4dc4a04aa13aa1ddbf4b : SUCCESS in 4m 51s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/5b06180cc0794ed388763a39a3b3aa48 : SUCCESS in 5m 45s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/175d9391c4e0419cb93214ef11204e51 : FAILURE in 14m 19s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/de6f9dc2b89b4a93a528ea67d66bd11e : FAILURE in 7m 06s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/5f403be419f648a2a853710cdab0dc86 : SUCCESS in 6m 01s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/48d3bf8ae15a478db9b020397872b506 : FAILURE in 13m 41s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/46576d0310dd4f84be866582e9e83ad6 : SUCCESS in 7m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ce2c1bdba77140cb9cdceea0205ed9ad : SUCCESS in 1h 24m 42s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d6af5b05864a4f6287bb644ca4d9f3dd : SUCCESS in 1h 24m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/b9b088926d1f4735874c2fceee2caff1 : FAILURE in 11m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/38b98af5fdc448609290245514320afe : SUCCESS in 1h 44m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/46cc75ab546d4c16983c56d663b297e0 : FAILURE in 11m 04s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/57179b8b4e0a4ccebc8a33f2e3cb0d6c : SUCCESS in 56m 01s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/23d55d8a6aa74328a747fbe52a66602b : FAILURE in 14m 03s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/a7cf1a6966dc49738846bad1760f1032 : FAILURE in 41m 46s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/12ade5abef2c4383b42dd2728dc21e02 : FAILURE in 45m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/8ebc827173424bafa59c5b89081d3c41 : FAILURE in 46m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64c57d0f8f89c4f9e0cd3aabe4af32de714304cd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-10 10:30:25.000000000","message":"Patch Set 1:\n\nAs I said in my previous messages, these changes still behave weirdly with py27 (at least locally). Did you experienced a similar behavior?\n\nPlease have a look to the stacktrace below\n\n```\nCaptured traceback:\n~~~~~~~~~~~~~~~~~~~\n    Traceback (most recent call last):\n      File \"oslo_messaging/tests/drivers/test_impl_rabbit.py\", line 89, in test_test_heartbeat_sent_connection_fail\n        info\u003d\u0027A recoverable connection/channel error occurred, \u0027\n      File \"/home/hberaud/dev/redhat/upstream/openstack/oslo/oslo.messaging/.tox/py27/lib/python2.7/site-packages/mock/mock.py\", line 1330, in patched\n        return func(*args, **keywargs)\n      File \"oslo_messaging/tests/drivers/test_impl_rabbit.py\", line 77, in _do_test_heartbeat_sent\n        self.assertEqual(2, fake_ensure_connection.call_count)\n      File \"/home/hberaud/dev/redhat/upstream/openstack/oslo/oslo.messaging/.tox/py27/lib/python2.7/site-packages/testtools/testcase.py\", line 411, in assertEqual\n        self.assertThat(observed, matcher, message)\n      File \"/home/hberaud/dev/redhat/upstream/openstack/oslo/oslo.messaging/.tox/py27/lib/python2.7/site-packages/testtools/testcase.py\", line 498, in assertThat\n        raise mismatch_error\n    testtools.matchers._impl.MismatchError: 2 !\u003d 1\n```","accounts_in_message":[],"_revision_number":1},{"id":"5d95968665f8424bab4914507ce0a9be3e0ef42b","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2021-08-10 11:33:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c5a404e9feb494d6247735ecb9ae23d9cb1646cf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-10 12:08:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f935d40cf028c0e341c32f748fa45fa2a2223c5","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2021-08-10 13:27:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a83ad6bb5990f4cdb8668fdecea6613faa64f9c","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2021-08-11 06:01:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"38a2ff93c8c2115b87441a894d36818b3e35282d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 07:04:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83190bd6dbc679ef0c2baabb20e7ae3b0e58d66c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 07:06:41.000000000","message":"Patch Set 1: Code-Review+2\n\nAs the local failures aren\u0027t there all the time and as our gates seems happy let\u0027s give a chance to this patch.","accounts_in_message":[],"_revision_number":1},{"id":"de32f27197c68c6df008a4ab58bf2beb506574ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 08:09:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ca95b3274585428b88c41846c1d7c8ca : SUCCESS in 1h 32m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3da233e45f0c44cfb4685a1db5efe812 : SUCCESS in 1h 21m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/045623e4d1a84ff5b01f5e3e03824870 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09daa2aec0f34168a0db8be6c0caf039 : SUCCESS in 7m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82cb9c63b32b40b188a7ea12888bf0b3 : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21555f56be1a410e9d7c3d244273b61d : SUCCESS in 7m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/910448ba773548718b3d80cca6458575 : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c69c541244044dfa147c4bc6e2451d4 : SUCCESS in 5m 30s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/977e3a8f5da74a09bf7898da0d8ecb94 : SUCCESS in 7m 23s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/b7c9baf41d084ad1a8a1e80a36f3ca08 : FAILURE in 18m 49s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/911b226736ff42848fd7bc88022989c8 : SUCCESS in 9m 41s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/36c36523693743f8b60ab223a9a3a2f6 : SUCCESS in 6m 47s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/4db12f687e3c4338bfbef002d07bc0ed : FAILURE in 14m 53s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/ca1889b632594cf782269fc972d64d63 : SUCCESS in 13m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/074db79e893e47918f700e1bd981bf32 : SUCCESS in 1h 50m 05s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/5f98769f2b9748a1b3f244fef67fc636 : SUCCESS in 1h 30m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/dad88e9c6f984d4a995b67da8a20a045 : FAILURE in 18m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ca978c02f32d4cd8a83f8e3026927c87 : SUCCESS in 1h 32m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/33004c5dba0649c99b326b443b7a2cce : FAILURE in 19m 00s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/27a6d663b19d4749813442a9251d7463 : SUCCESS in 1h 01m 24s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/50fec8d2bf18446aaebf72f2e575c7b9 : FAILURE in 25m 41s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/bbe88d9c081d40cd8e96700b4d2f881b : FAILURE in 48m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/4842179cdb5d4b4d95bfbf93b01e2339 : FAILURE in 51m 20s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/5c0f71236dae419a9c76e554786d4d0e : FAILURE in 49m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"833d1d72f25050e729f3b3e22fa07a330396c75d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 08:50:18.000000000","message":"Patch Set 1:\n\nPlease can you update the commit message by adding the cherry-pick information and adding a note concerning your rewriting of the import part to indicate why you changed the original commit. It will help us to keep a tracks of the history of these changes.\n\nThank you by advance.","accounts_in_message":[],"_revision_number":1},{"id":"cecd0f71dbbc6fb1526612d909b36f2c82e82abc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 08:51:32.000000000","message":"Patch Set 1:\n\ncherry-picked from https://review.opendev.org/c/openstack/oslo.messaging/+/743647","accounts_in_message":[],"_revision_number":1},{"id":"01d354d79a8dd67aff389dae56338fe332dc9fde","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 09:05:26.000000000","message":"Patch Set 1: Code-Review+1\n\nI\u0027m waiting for your feedback concerning the outcomes of your local tests","accounts_in_message":[],"_revision_number":1},{"id":"f6e94289bfebc2dd899036a1e6692f517e3bb9b7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-11 09:12:47.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks for the patch. Unfortunately, I\u0027m finding it rather confusing. Is there a reason this is proposed against stable/train rather than master? The standard practice is to fix on master and then backport as far as you need. Perhaps this is a thing that only affects Python 2.7? If so, you need to provide this additional context in the commit message. You should also tag the commit message with StableOnly or similar to show why you have done things the way you have. If not, this needs to be moved to master and merged before we start backporting it the various stable branches.","accounts_in_message":[],"_revision_number":1},{"id":"318ab445687b45d5ff295088cfb2e880e29b88eb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-11 09:16:31.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Thanks for the patch. Unfortunately, I\u0027m finding it rather confusing. Is there a reason this is proposed against stable/train rather than master? The standard practice is to fix on master and then backport as far as you need. Perhaps this is a thing that only affects Python 2.7? If so, you need to provide this additional context in the commit message. You should also tag the commit message with StableOnly or similar to show why you have done things the way you have. If not, this needs to be moved to master and merged before we start backporting it the various stable branches.\n\nAh, okay, I read the comments and see the issue now. I don\u0027t think this is the correct approach to take. What you should do is cherry-pick the original change from stable/ussuri and modify it to fix the Python 2.7 issues. You can then call out the modifications you made in the commit message like so:\n\n  NOTE(normanshen): Moved the exception to ... because ...","accounts_in_message":[],"_revision_number":1},{"id":"05fd13b20aa968514d336ee58e931cab595aa08c","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2021-08-11 11:46:55.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \n\u003e \u003e Thanks for the patch. Unfortunately, I\u0027m finding it rather confusing. Is there a reason this is proposed against stable/train rather than master? The standard practice is to fix on master and then backport as far as you need. Perhaps this is a thing that only affects Python 2.7? If so, you need to provide this additional context in the commit message. You should also tag the commit message with StableOnly or similar to show why you have done things the way you have. If not, this needs to be moved to master and merged before we start backporting it the various stable branches.\n\u003e \n\u003e Ah, okay, I read the comments and see the issue now. I don\u0027t think this is the correct approach to take. What you should do is cherry-pick the original change from stable/ussuri and modify it to fix the Python 2.7 issues. You can then call out the modifications you made in the commit message like so:\n\u003e \n\u003e   NOTE(normanshen): Moved the exception to ... because ...\n\nOK, I follow this approach..","accounts_in_message":[],"_revision_number":1},{"id":"aadbda97ff4f603a4d27d84fc4d84091328772d9","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2021-08-11 11:48:23.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+1\n\u003e \n\u003e I\u0027m waiting for your feedback concerning the outcomes of your local tests\n\nI\u0027ve tested it locally and got a different exception (which I cannot reproduce...)\nI\u0027ll follow stephen\u0027s suggestion and do a cherry-pick and fix.. How do you think?","accounts_in_message":[],"_revision_number":1},{"id":"8315b5eb6dfaf66cff28a2dc9e29641226dae49a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-16 11:45:20.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review+1\n\u003e \u003e \n\u003e \u003e I\u0027m waiting for your feedback concerning the outcomes of your local tests\n\u003e \n\u003e I\u0027ve tested it locally and got a different exception (which I cannot reproduce...)\n\u003e I\u0027ll follow stephen\u0027s suggestion and do a cherry-pick and fix.. How do you think?\n\nWork for me.\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"e7d649ab31132dc766fed46da06b56885d0e062f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-30 09:32:45.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review+1\n\u003e \u003e \n\u003e \u003e I\u0027m waiting for your feedback concerning the outcomes of your local tests\n\u003e \n\u003e I\u0027ve tested it locally and got a different exception (which I cannot reproduce...)\n\u003e I\u0027ll follow stephen\u0027s suggestion and do a cherry-pick and fix.. How do you think?\n\nHello,\n\nDo you have news to share with us?","accounts_in_message":[],"_revision_number":1},{"id":"29979af0bda7587ae2677d3b225641819fc51fc4","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2021-08-30 14:50:14.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1: Code-Review+1\n\u003e \u003e \u003e \n\u003e \u003e \u003e I\u0027m waiting for your feedback concerning the outcomes of your local tests\n\u003e \u003e \n\u003e \u003e I\u0027ve tested it locally and got a different exception (which I cannot reproduce...)\n\u003e \u003e I\u0027ll follow stephen\u0027s suggestion and do a cherry-pick and fix.. How do you think?\n\u003e \n\u003e Hello,\n\u003e \n\u003e Do you have news to share with us?\n\nsorry.. I forgot this issue .. Let me work on it right away... sorry..","accounts_in_message":[],"_revision_number":1},{"id":"baaf84d3a3976ac71ae5ad9f471e978e2e1eda71","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2021-08-30 15:11:56.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1: Code-Review+1\n\u003e \u003e \u003e \n\u003e \u003e \u003e I\u0027m waiting for your feedback concerning the outcomes of your local tests\n\u003e \u003e \n\u003e \u003e I\u0027ve tested it locally and got a different exception (which I cannot reproduce...)\n\u003e \u003e I\u0027ll follow stephen\u0027s suggestion and do a cherry-pick and fix.. How do you think?\n\u003e \n\u003e Hello,\n\u003e \n\u003e Do you have news to share with us?\n\ncould you have a look at this commit https://review.opendev.org/c/openstack/oslo.messaging/+/744262? I restore it and amend commit instead...","accounts_in_message":[],"_revision_number":1},{"id":"f0fe61401515aa059eac39f69b349a3a384e6809","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2021-09-02 06:23:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efa71ef78adbaea61532499b1dc6dcc8f82c8102","revisions":{"efa71ef78adbaea61532499b1dc6dcc8f82c8102":{"kind":"REWORK","_number":1,"created":"2020-10-10 03:23:45.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/42/757242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/42/757242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/42/757242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/42/757242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/42/757242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/42/757242/1"}}},"commit":{"parents":[{"commit":"1613b7a968f33ef7dc683d05fde05be0c458c8fc","subject":"Print warning message when connection running out","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1613b7a968f33ef7dc683d05fde05be0c458c8fc"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-10-10 03:16:38.000000000","tz":480},"committer":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-10-10 03:18:19.000000000","tz":480},"subject":"Catch ConnectionForced Exception","message":"Catch ConnectionForced Exception\n\nOccasionally I saw large number of connectionforced exceptions.\nsuch exceptions cannot be healed automatically and requires a reboot.\ncatch it and ensure connection established.\n\nChange-Id: I97287e81b1ea6a910e2543eea79c0d78e51743d6\nCloses-Bug: #1883038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/efa71ef78adbaea61532499b1dc6dcc8f82c8102"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/efa71ef78adbaea61532499b1dc6dcc8f82c8102"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
