)]}'
{"id":"openstack%2Foslo.messaging~75815","triplet_id":"openstack%2Foslo.messaging~master~I871703e4cdc04cee3e6c214e911c9df464ede2ed","project":"openstack/oslo.messaging","branch":"master","topic":"bp/amqp10-driver-implementation","hashtags":[],"change_id":"I871703e4cdc04cee3e6c214e911c9df464ede2ed","subject":"An initial implementation of an AMQP 1.0 based messaging driver","status":"MERGED","created":"2014-02-24 09:26:22.000000000","updated":"2014-09-06 08:56:41.000000000","submitted":"2014-09-06 08:56:41.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":230,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7b67aade4c0c525e37e191fa385bd8a6b892deb0","_number":75815,"virtual_id_number":75815,"owner":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2014-09-06 08:56:41.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":7818,"name":"Ted Ross","email":"tross@redhat.com","username":"tross"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"value":0,"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},{"value":0,"_account_id":7536,"name":"Ilya Pekelny","email":"ipekelny@mirantis.com","username":"I159"},{"value":0,"date":"2014-09-04 22:04:19.000000000","_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":7763,"name":"Alexei Kornienko","email":"akornienko@mirantis.com","username":"akornienko"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"value":0,"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},{"value":0,"_account_id":8415,"name":"Denis M.","email":"lildee1991@gmail.com","username":"dmakogon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":2,"date":"2014-09-05 13:24:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":7818,"name":"Ted Ross","email":"tross@redhat.com","username":"tross"},{"value":2,"date":"2014-09-05 12:54:27.000000000","_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"value":1,"date":"2014-09-05 09:23:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},{"value":1,"date":"2014-09-05 09:18:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7536,"name":"Ilya Pekelny","email":"ipekelny@mirantis.com","username":"I159"},{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":7763,"name":"Alexei Kornienko","email":"akornienko@mirantis.com","username":"akornienko"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":2,"date":"2014-09-05 12:48:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"value":0,"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},{"value":0,"_account_id":8415,"name":"Denis M.","email":"lildee1991@gmail.com","username":"dmakogon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":1,"date":"2014-09-05 13:24:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":7818,"name":"Ted Ross","email":"tross@redhat.com","username":"tross"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"value":0,"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},{"value":0,"_account_id":7536,"name":"Ilya Pekelny","email":"ipekelny@mirantis.com","username":"I159"},{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":7763,"name":"Alexei Kornienko","email":"akornienko@mirantis.com","username":"akornienko"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"value":0,"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},{"value":0,"_account_id":8415,"name":"Denis M.","email":"lildee1991@gmail.com","username":"dmakogon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":7818,"name":"Ted Ross","email":"tross@redhat.com","username":"tross"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"value":0,"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},{"value":0,"_account_id":7536,"name":"Ilya Pekelny","email":"ipekelny@mirantis.com","username":"I159"},{"value":0,"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"_account_id":7763,"name":"Alexei Kornienko","email":"akornienko@mirantis.com","username":"akornienko"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"value":0,"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},{"value":0,"_account_id":8415,"name":"Denis M.","email":"lildee1991@gmail.com","username":"dmakogon"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"_account_id":7536,"name":"Ilya Pekelny","email":"ipekelny@mirantis.com","username":"I159"},{"_account_id":7763,"name":"Alexei Kornienko","email":"akornienko@mirantis.com","username":"akornienko"},{"_account_id":7818,"name":"Ted Ross","email":"tross@redhat.com","username":"tross"},{"_account_id":8415,"name":"Denis M.","email":"lildee1991@gmail.com","username":"dmakogon"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-02-24 09:33:59.000000000","updated_by":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"reviewer":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"state":"REVIEWER"},{"updated":"2014-02-24 10:56:40.000000000","updated_by":{"_account_id":8415,"name":"Denis M.","email":"lildee1991@gmail.com","username":"dmakogon"},"reviewer":{"_account_id":8415,"name":"Denis M.","email":"lildee1991@gmail.com","username":"dmakogon"},"state":"REVIEWER"},{"updated":"2014-07-02 08:27:31.000000000","updated_by":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"reviewer":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"state":"REVIEWER"},{"updated":"2014-07-25 15:43:41.000000000","updated_by":{"_account_id":7763,"name":"Alexei Kornienko","email":"akornienko@mirantis.com","username":"akornienko"},"reviewer":{"_account_id":7763,"name":"Alexei Kornienko","email":"akornienko@mirantis.com","username":"akornienko"},"state":"REVIEWER"},{"updated":"2014-07-31 19:21:07.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2014-09-03 15:40:33.000000000","updated_by":{"_account_id":7818,"name":"Ted Ross","email":"tross@redhat.com","username":"tross"},"reviewer":{"_account_id":7818,"name":"Ted Ross","email":"tross@redhat.com","username":"tross"},"state":"REVIEWER"},{"updated":"2014-09-04 18:39:12.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2014-09-04 22:04:19.000000000","updated_by":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"reviewer":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"state":"REVIEWER"},{"updated":"2014-09-05 09:18:34.000000000","updated_by":{"_account_id":7536,"name":"Ilya Pekelny","email":"ipekelny@mirantis.com","username":"I159"},"reviewer":{"_account_id":7536,"name":"Ilya Pekelny","email":"ipekelny@mirantis.com","username":"I159"},"state":"REVIEWER"},{"updated":"2014-09-05 12:48:25.000000000","updated_by":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"reviewer":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"state":"REVIEWER"},{"updated":"2014-09-05 12:54:27.000000000","updated_by":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"reviewer":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"state":"REVIEWER"},{"updated":"2014-09-05 13:24:20.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2014-09-06 08:56:41.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"bc52ffd0824a10434f3483e7d1bb5ef954ad392e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-24 09:26:23.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"9582efbca4f9afa392429454f0d46f6a0156556d","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-02-24 09:32:37.000000000","message":"Patch Set 1: Work In Progress\n\nThis is an initial cut of the code to implement the driver. The code evolved in https://github.com/FlaPer87/oslo.messaging/tree/gordon mostly by Ken Giusti and myself.\n\nThis version implements intermediated communication only, though I do have a branch I will soon publish that shows non-intermediated communication as an option also.\n\nI\u0027m keen to get any early feedback that I can and will push up improvements to the patch as progress is made.","accounts_in_message":[],"_revision_number":1},{"id":"8195dca5c32cf4252ab03ae9ee6aade6b4fc236c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-24 10:21:44.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/1/check/gate-oslo.messaging-pep8/2a067ac : FAILURE in 51s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/1/check/gate-oslo.messaging-docs/e575680/doc/build/html/ : SUCCESS in 1m 04s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/1/check/gate-oslo.messaging-python26/f2e4d73 : SUCCESS in 1m 16s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/1/check/gate-oslo.messaging-python27/1c2a39d : SUCCESS in 1m 11s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/1/check/check-devstack-dsvm-cells/f75c1b8 : FAILURE in 8m 58s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/1/check/check-tempest-dsvm-full/441d55b : SUCCESS in 52m 09s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/1/check/check-tempest-dsvm-neutron/dc77f27 : SUCCESS in 33m 23s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/15/75815/1/check/check-tempest-dsvm-neutron-full/9052b59 : SUCCESS in 39m 48s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/1/check/gate-tempest-dsvm-large-ops/56c56bb : SUCCESS in 15m 30s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/1/check/gate-tempest-dsvm-neutron-large-ops/fe09d09 : SUCCESS in 17m 54s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/1/check/check-grenade-dsvm/24d6779 : SUCCESS in 34m 41s\n","accounts_in_message":[],"_revision_number":1},{"id":"3b0923683b4c31ee342097a737b8304bb28a31d2","author":{"_account_id":8415,"name":"Denis M.","email":"lildee1991@gmail.com","username":"dmakogon"},"date":"2014-02-24 10:56:40.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n(37 inline comments)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"dab02a1abdcb6cef123f102d9534ae2b3016e90c","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-02-24 15:08:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a80a554b3064c4d3d044b3ca95123d14d07ad134","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-24 15:08:55.000000000","message":"Patch Set 2:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"8a5069d53f2fc60f61daf62e83b3ee658f5c6ec5","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-02-24 15:17:49.000000000","message":"Patch Set 1: (34 inline comments)\n\nThanks very much for the review Denis! I\u0027ve uploaded a new set with most of the changes and will tackle the remaining couple shortly.","accounts_in_message":[],"_revision_number":1},{"id":"1854231e2c29b2199baedcf26a69551980609b59","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-24 16:10:38.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/2/check/gate-oslo.messaging-pep8/6e20c50 : SUCCESS in 1m 33s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/2/check/gate-oslo.messaging-docs/71d3f17/doc/build/html/ : SUCCESS in 1m 19s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/2/check/gate-oslo.messaging-python26/bc4d4e9 : SUCCESS in 49s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/2/check/gate-oslo.messaging-python27/bbcbf55 : SUCCESS in 1m 32s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/2/check/check-devstack-dsvm-cells/fc3328f : SUCCESS in 13m 44s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/2/check/check-tempest-dsvm-full/0ff4de4 : SUCCESS in 53m 23s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/2/check/check-tempest-dsvm-neutron/4c4f7c5 : SUCCESS in 31m 34s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/15/75815/2/check/check-tempest-dsvm-neutron-full/c5c6e60 : SUCCESS in 51m 02s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/2/check/gate-tempest-dsvm-large-ops/5f9791e : SUCCESS in 16m 22s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/2/check/gate-tempest-dsvm-neutron-large-ops/b758059 : SUCCESS in 13m 54s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/2/check/check-grenade-dsvm/4360206 : SUCCESS in 30m 04s\n","accounts_in_message":[],"_revision_number":2},{"id":"13485c45e4fc461b5f4db2f45d415b1b3698949c","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-02-25 14:47:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3a329f4873eaa9370e9407fb1532c4d0dcfa080f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-25 14:47:52.000000000","message":"Patch Set 3:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"da5e6055d08c80298b94a367397d933ef08f7ead","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-02-25 14:50:16.000000000","message":"Patch Set 1: (3 inline comments)\n\nAddressed remaining issues in the review from Denis. Thanks again!","accounts_in_message":[],"_revision_number":1},{"id":"374da39bf540b5e62cba8cb89cd2412ff520d87e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-25 15:47:48.000000000","message":"Patch Set 3: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/3/check/gate-oslo.messaging-pep8/dd8a4f7 : SUCCESS in 1m 23s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/3/check/gate-oslo.messaging-docs/6b0b190/doc/build/html/ : SUCCESS in 1m 34s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/3/check/gate-oslo.messaging-python26/1d36e1c : SUCCESS in 1m 06s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/3/check/gate-oslo.messaging-python27/63d3a42 : SUCCESS in 1m 20s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/3/check/check-devstack-dsvm-cells/3fef446 : SUCCESS in 10m 48s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/3/check/check-tempest-dsvm-full/5eea8af : SUCCESS in 52m 50s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/3/check/check-tempest-dsvm-neutron/4704c2f : FAILURE in 7m 43s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/15/75815/3/check/check-tempest-dsvm-neutron-full/4c1ce6a : SUCCESS in 39m 27s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/3/check/gate-tempest-dsvm-large-ops/a16454c : SUCCESS in 13m 52s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/3/check/gate-tempest-dsvm-neutron-large-ops/22d201f : SUCCESS in 12m 45s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/3/check/check-grenade-dsvm/9d34146 : SUCCESS in 33m 30s\n","accounts_in_message":[],"_revision_number":3},{"id":"178218bf7ce9afe37d11ef8263c3f2eccf237ae6","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2014-03-05 06:03:36.000000000","message":"Patch Set 3:\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":3},{"id":"724416760251fa9bf5c6f0c3234e1b6c53503b3a","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2014-03-05 06:03:36.000000000","message":"Patch Set 3: Abandoned\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":3},{"id":"e52bd5dd24f78ca4b8c657feb81d541eb025be77","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-03-05 10:50:16.000000000","message":"Patch Set 3: Restored\n\nI\u0027m still keen to get feedback on this. I believe I have addressed all the points raised in the first review.","accounts_in_message":[],"_revision_number":3},{"id":"f199d652516ae513a89c5a324a712cc510a76dff","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-05 10:51:32.000000000","message":"Patch Set 3: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"9f5922919b4e6d7ac316ee3899410f51211920b0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-05 11:40:50.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/3/check/gate-oslo.messaging-pep8/3e82582 : SUCCESS in 40s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/3/check/gate-oslo.messaging-docs/222046d/doc/build/html/ : SUCCESS in 1m 23s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/3/check/gate-oslo.messaging-python26/0c5ad6a : SUCCESS in 1m 01s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/3/check/gate-oslo.messaging-python27/fbab91a : SUCCESS in 1m 04s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/3/check/check-devstack-dsvm-cells/24bd38a : SUCCESS in 12m 12s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/3/check/check-tempest-dsvm-full/87cb8a9 : SUCCESS in 39m 20s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/3/check/check-tempest-dsvm-neutron/0a98a7d : SUCCESS in 32m 26s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/15/75815/3/check/check-tempest-dsvm-neutron-full/247ce70 : SUCCESS in 42m 42s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/3/check/gate-tempest-dsvm-large-ops/3b928aa : SUCCESS in 24m 54s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/3/check/gate-tempest-dsvm-neutron-large-ops/d61a42d : SUCCESS in 15m 44s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/3/check/check-grenade-dsvm/920f0a4 : SUCCESS in 33m 43s\n","accounts_in_message":[],"_revision_number":3},{"id":"12e42839640089a3bbc3ace286100ae7228391a3","author":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"date":"2014-03-18 12:03:52.000000000","message":"Patch Set 3: Do not merge\n\nGordon - I\u0027m sorry I and the rest of oslo-core didn\u0027t get around to reviewing this earlier. We\u0027re a couple of weeks into Icehouse feature freeze now, so I think this has to wait until Juno. Temporarily marking with -2 to reflect that.\n\nPlease do make sure to restore this when Juno development starts.\n\nApologies again. I know missing feature freeze is a pain, I\u0027ve been there.","accounts_in_message":[],"_revision_number":3},{"id":"02bb033ab2691e6766c473359404d46bb5d49a7e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-18 12:03:53.000000000","message":"Patch Set 3: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"e9b881462ca264321e3f9413dd93aa41b7b012f9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-18 13:11:11.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/3/check/gate-oslo.messaging-pep8/84d9fb2 : SUCCESS in 53s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/3/check/gate-oslo.messaging-docs/8010459/doc/build/html/ : SUCCESS in 1m 23s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/3/check/gate-oslo.messaging-python26/8c22d0e : SUCCESS in 1m 35s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/3/check/gate-oslo.messaging-python27/14ef128 : SUCCESS in 1m 00s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/3/check/check-grenade-dsvm/ccfb077 : SUCCESS in 30m 11s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/3/check/check-tempest-dsvm-full/86da12d : SUCCESS in 52m 36s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/3/check/check-tempest-dsvm-neutron/4f42888 : SUCCESS in 39m 24s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/3/check/check-tempest-dsvm-neutron-heat-slow/43440c3 : SUCCESS in 23m 38s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/3/check/check-devstack-dsvm-cells/0f48675 : SUCCESS in 10m 15s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/3/check/gate-tempest-dsvm-large-ops/f6e866a : SUCCESS in 14m 01s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/3/check/gate-tempest-dsvm-neutron-large-ops/da7d2c2 : SUCCESS in 14m 53s\n- check-oslo.test-dsvm-oslo.messaging http://logs.openstack.org/15/75815/3/check/check-oslo.test-dsvm-oslo.messaging/20af198 : SUCCESS in 3m 30s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"eaa1deb519b155c2eeb40822c271a2bf6ab84c0a","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2014-03-26 06:04:10.000000000","message":"Patch Set 3:\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":3},{"id":"fc51329063df6513c0fca8acaf2bd98caecc9f2e","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2014-03-26 06:04:10.000000000","message":"Patch Set 3: Abandoned\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":3},{"id":"6ede0aac163cba38d96ff2bcfcc0b8d03458a64c","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-03-26 09:09:59.000000000","message":"Patch Set 3: Restored\n\nI understand this will now be targeted for Juno at the earliest. I\u0027m restoring now to get any early comments/feedback. I have some tests to add to the patch also, which I will update shortly.","accounts_in_message":[],"_revision_number":3},{"id":"6150bb270766ff3ca70588dfa598b176650b2bfd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-26 09:10:00.000000000","message":"Patch Set 3: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"a77122b1831c22cb428af10f034c15dd09e91aca","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-26 11:53:42.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/3/check/gate-oslo.messaging-pep8/dc441ff : SUCCESS in 2m 48s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/3/check/gate-oslo.messaging-docs/d60ba9e/doc/build/html/ : SUCCESS in 51s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/3/check/gate-oslo.messaging-python26/891e590 : SUCCESS in 1m 12s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/3/check/gate-oslo.messaging-python27/7d1869f : SUCCESS in 2m 07s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/3/check/check-grenade-dsvm/3e2573c : SUCCESS in 34m 01s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/3/check/check-tempest-dsvm-full/7d4ddfd : SUCCESS in 52m 12s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/3/check/check-tempest-dsvm-neutron/556a079 : SUCCESS in 50m 54s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/3/check/check-tempest-dsvm-neutron-heat-slow/9aa1bc0 : SUCCESS in 42m 36s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/3/check/check-devstack-dsvm-cells/680ca41 : SUCCESS in 12m 42s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/3/check/gate-tempest-dsvm-large-ops/7161db0 : SUCCESS in 15m 37s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/3/check/gate-tempest-dsvm-neutron-large-ops/e336aed : SUCCESS in 17m 34s\n- check-oslo.test-dsvm-oslo.messaging http://logs.openstack.org/15/75815/3/check/check-oslo.test-dsvm-oslo.messaging/dcaaa91 : SUCCESS in 5m 09s (non-voting)\n\n","accounts_in_message":[],"_revision_number":3},{"id":"62255275f43d10971d5c8d3d171afbce9a6a39df","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-31 17:08:41.000000000","message":"Patch Set 4:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"2f658e6ca89234c46bf1784f5670b3f67e0cbdfd","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-03-31 17:08:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"18415adad9a33931bf4018efff584977ba208dc9","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-03-31 17:11:44.000000000","message":"Patch Set 4:\n\nSome fixes made in getting the functional tests - https://review.openstack.org/#/c/84164/ - to pass.","accounts_in_message":[],"_revision_number":4},{"id":"c0c3e38cd90de76408087d9d917002e5766c90de","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-31 18:00:51.000000000","message":"Patch Set 4: Works for me\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/4/check/gate-oslo.messaging-pep8/4b53b4a : SUCCESS in 1m 15s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/4/check/gate-oslo.messaging-docs/fcdbe10/doc/build/html/ : SUCCESS in 1m 35s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/4/check/gate-oslo.messaging-python26/0a6d5e8 : SUCCESS in 1m 50s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/4/check/gate-oslo.messaging-python27/2e9aabd : SUCCESS in 1m 39s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/4/check/check-grenade-dsvm/ca1be60 : SUCCESS in 36m 56s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/4/check/check-tempest-dsvm-full/95a89ca : SUCCESS in 51m 23s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/4/check/check-tempest-dsvm-neutron/adeaf13 : SUCCESS in 44m 51s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/4/check/check-tempest-dsvm-neutron-heat-slow/b24695f : SUCCESS in 26m 45s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/4/check/check-devstack-dsvm-cells/7682118 : SUCCESS in 12m 08s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/4/check/gate-tempest-dsvm-large-ops/50abf1f : SUCCESS in 16m 13s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/4/check/gate-tempest-dsvm-neutron-large-ops/3ede7a4 : SUCCESS in 16m 53s\n- check-oslo.test-dsvm-oslo.messaging http://logs.openstack.org/15/75815/4/check/check-oslo.test-dsvm-oslo.messaging/d25a181 : SUCCESS in 5m 00s\n\n","accounts_in_message":[],"_revision_number":4},{"id":"35724e65de40d4a42b52e43b7485dad10c1203b6","author":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"date":"2014-04-02 06:30:02.000000000","message":"Patch Set 4: No score\n\nmaster is open for Juno development","accounts_in_message":[],"_revision_number":4},{"id":"d9ba98e10611f874504599996eafa9ddc89aefea","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-04-17 10:31:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1665c2a01723a0d2d9f714654edd896e17d2f877","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-17 10:32:24.000000000","message":"Patch Set 5:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":5},{"id":"0d605f2763c3c7a71573e1876b4031d7ae70a654","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-17 11:45:10.000000000","message":"Patch Set 5: Works for me\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/5/check/gate-oslo.messaging-pep8/d957a77 : SUCCESS in 1m 16s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/5/check/gate-oslo.messaging-docs/e55e500/doc/build/html/ : SUCCESS in 1m 21s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/5/check/gate-oslo.messaging-python26/29d55dd : SUCCESS in 1m 02s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/5/check/gate-oslo.messaging-python27/9fb0a0e : SUCCESS in 1m 42s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/5/check/check-tempest-dsvm-full/c243149 : SUCCESS in 1h 00m 47s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/5/check/check-tempest-dsvm-neutron/4eaa14f : SUCCESS in 52m 38s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/5/check/check-tempest-dsvm-neutron-heat-slow/f56ecba : SUCCESS in 17m 28s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/15/75815/5/check/check-grenade-dsvm/84f0744 : SUCCESS in 36m 05s\n- check-grenade-dsvm-neutron http://logs.openstack.org/15/75815/5/check/check-grenade-dsvm-neutron/3fda115 : FAILURE in 27m 02s (non-voting)\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/5/check/check-devstack-dsvm-cells/3588042 : SUCCESS in 11m 06s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/5/check/gate-tempest-dsvm-large-ops/082f0f2 : SUCCESS in 14m 39s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/5/check/gate-tempest-dsvm-neutron-large-ops/29c03c8 : SUCCESS in 16m 42s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/5/check/check-oslotest-dsvm-oslo.messaging/94a5ba3 : SUCCESS in 4m 49s\n","accounts_in_message":[],"_revision_number":5},{"id":"25289f3a8f89d658f82c20c8cf1f0cfc38838ee6","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2014-05-19 16:05:03.000000000","message":"Patch Set 5: Code-Review-1\n\nBased on the feedback from the summit session. This patch needs to be cleaned up. In more detail, the engine code should be removed in favor of pyngus.\n\nLooking forward to the new patch-set.","accounts_in_message":[],"_revision_number":5},{"id":"f47e72f38e21ba7ed544759260ff8e60772b9269","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-19 16:05:04.000000000","message":"Patch Set 5: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":5},{"id":"049ad53506d64d7712b0568ffe1ea04acd2bbf80","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-19 17:15:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/5/check/gate-oslo.messaging-pep8/ac72b5a : SUCCESS in 1m 08s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/5/check/gate-oslo.messaging-docs/4001c0b/doc/build/html/ : SUCCESS in 1m 22s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/5/check/gate-oslo.messaging-python26/75ef3a8 : SUCCESS in 1m 31s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/5/check/gate-oslo.messaging-python27/a5e08ec : SUCCESS in 1m 41s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/5/check/check-tempest-dsvm-full/eed4009 : SUCCESS in 1h 06m 11s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/5/check/check-tempest-dsvm-neutron/c4d94d1 : SUCCESS in 57m 18s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/5/check/check-tempest-dsvm-neutron-heat-slow/a947ade : FAILURE in 19m 58s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/15/75815/5/check/check-grenade-dsvm/588d636 : SUCCESS in 47m 20s\n- check-grenade-dsvm-neutron http://logs.openstack.org/15/75815/5/check/check-grenade-dsvm-neutron/98a4300 : FAILURE in 33m 12s (non-voting)\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/5/check/check-devstack-dsvm-cells/9f673d2 : SUCCESS in 11m 52s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/5/check/gate-tempest-dsvm-large-ops/abaf080 : SUCCESS in 22m 05s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/5/check/gate-tempest-dsvm-neutron-large-ops/b2f2dfe : SUCCESS in 19m 44s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/5/check/check-oslotest-dsvm-oslo.messaging/9dfde37 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":5},{"id":"bc8c20ce76fae599aa87b7ec5baef1f582b85fe8","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-06-03 13:03:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8a65bc74854946ea59b6e17ac18efd5e56c08311","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-03 13:03:41.000000000","message":"Patch Set 6:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"a5fe45a0d07818aa2175adb563b40695ee770b20","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-03 22:59:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/6/check/gate-oslo.messaging-pep8/50074e1 : FAILURE in 26s\n- gate-oslo.messaging-docs http://logs.openstack.org/15/75815/6/check/gate-oslo.messaging-docs/4c5f0b4 : FAILURE in 24s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/6/check/gate-oslo.messaging-python26/0b79c46 : FAILURE in 1m 08s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/6/check/gate-oslo.messaging-python27/40bd44a : FAILURE in 22s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/6/check/gate-oslo.messaging-requirements/9b86558 : Incompatible requirement found; see https://wiki.openstack.org/wiki/Requirements in 8s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/6/check/check-tempest-dsvm-full/6f7d67b : FAILURE in 6m 15s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/6/check/check-tempest-dsvm-postgres-full/b51adbd : FAILURE in 4m 47s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/6/check/check-tempest-dsvm-neutron/6bc522f : FAILURE in 4m 28s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/6/check/check-tempest-dsvm-neutron-heat-slow/6d0dfa3 : FAILURE in 5m 51s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/15/75815/6/check/check-grenade-dsvm/785e52c : SUCCESS in 58m 43s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/6/check/check-devstack-dsvm-cells/9c92f1d : FAILURE in 4m 59s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/6/check/gate-tempest-dsvm-large-ops/180580a : FAILURE in 4m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/6/check/gate-tempest-dsvm-neutron-large-ops/d08afd2 : FAILURE in 5m 18s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/6/check/check-oslotest-dsvm-oslo.messaging/778b7c8 : FAILURE in 3m 53s","accounts_in_message":[],"_revision_number":6},{"id":"1e2f153e985c089c424393933888a0da936312a4","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-06-04 20:02:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f56e5769cdcb59b65b886f462c275f3b0e6adefc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-04 20:02:24.000000000","message":"Patch Set 7:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":7},{"id":"db3ea22ecfb7016d1285c91294003920e01acd8c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-04 21:08:02.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/7/check/gate-oslo.messaging-pep8/530072c : SUCCESS in 52s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/7/check/gate-oslo.messaging-docs/435bfc7/doc/build/html/ : SUCCESS in 54s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/7/check/gate-oslo.messaging-python26/04e0f1e : SUCCESS in 2m 20s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/7/check/gate-oslo.messaging-python27/89b52bd : SUCCESS in 1m 28s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/7/check/gate-oslo.messaging-requirements/499038e : SUCCESS in 24s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/7/check/check-tempest-dsvm-full/af76168 : SUCCESS in 56m 23s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/7/check/check-tempest-dsvm-postgres-full/98f3b65 : SUCCESS in 51m 27s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/7/check/check-tempest-dsvm-neutron/6cdf0f8 : SUCCESS in 53m 51s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/7/check/check-tempest-dsvm-neutron-heat-slow/09bbfcb : SUCCESS in 17m 01s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/15/75815/7/check/check-grenade-dsvm/eba51ed : SUCCESS in 42m 21s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/7/check/check-devstack-dsvm-cells/d221985 : SUCCESS in 12m 18s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/7/check/gate-tempest-dsvm-large-ops/57c2d8f : SUCCESS in 26m 30s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/7/check/gate-tempest-dsvm-neutron-large-ops/9a17fa1 : SUCCESS in 19m 53s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/7/check/check-oslotest-dsvm-oslo.messaging/af357ab : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":7},{"id":"e35b566fcdc7e02a9b9bab8c9b30b102d9f7659b","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2014-06-18 10:10:26.000000000","message":"Patch Set 7: Code-Review-1\n\n(18 comments)\n\nI added some comments.\n\nSince the patch is still a bit big, I started reviewing the driver\u0027s code and things related to it. I\u0027ll now start reviewing the controller and other parts of the patch.\n\nThe review I did has some of python-style nitpicks and some logic/driver/structure related questions.","accounts_in_message":[],"_revision_number":7},{"id":"7bf4506047ab3d468c0ef3ef5ebecd56d8055c01","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-18 10:10:27.000000000","message":"Patch Set 7: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":7},{"id":"9e572a8bdfb8ff2a6635e2b665f77bef0a9e1bc8","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-06-18 10:54:03.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a84ab01b99a43838a5306696978457bca5c7d58f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-18 11:06:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/7/check/gate-oslo.messaging-pep8/c7b1e3a : SUCCESS in 53s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/7/check/gate-oslo.messaging-docs/75f0bf4/doc/build/html/ : SUCCESS in 42s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/7/check/gate-oslo.messaging-python26/585cbc3 : SUCCESS in 1m 22s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/7/check/gate-oslo.messaging-python27/f347cf0 : SUCCESS in 1m 17s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/7/check/gate-oslo.messaging-requirements/3566544 : SUCCESS in 14s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/7/check/check-tempest-dsvm-full/4f94694 : SUCCESS in 39m 59s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/7/check/check-tempest-dsvm-postgres-full/46439c8 : SUCCESS in 46m 15s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/7/check/check-tempest-dsvm-neutron/38dfb23 : SUCCESS in 56m 13s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/7/check/check-tempest-dsvm-neutron-heat-slow/76c57c4 : SUCCESS in 19m 37s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/15/75815/7/check/check-grenade-dsvm/b6e4279 : SUCCESS in 37m 59s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/7/check/check-devstack-dsvm-cells/6706cb3 : SUCCESS in 13m 14s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/7/check/gate-tempest-dsvm-large-ops/9b0d284 : SUCCESS in 25m 20s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/7/check/gate-tempest-dsvm-neutron-large-ops/257d251 : SUCCESS in 24m 05s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/7/check/check-oslotest-dsvm-oslo.messaging/dc2a45b : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":7},{"id":"21c991d5e9d57637a5d1d511070873efc0c7ea82","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-06-24 01:58:52.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8b94cf501b27d565989fd2e6af56d941d1cc1681","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-24 01:59:06.000000000","message":"Patch Set 8:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":8},{"id":"4d1f1c57d99bc861031a0a5215339083d204b06e","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-06-24 02:00:52.000000000","message":"Patch Set 8:\n\nPatch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"cb8b0ad0527c0bb2fa4e88d77c8e2f8614de40cc","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-06-24 02:07:43.000000000","message":"Patch Set 8:\n\nThis patch is a work-in-progress.  I\u0027ve added a set of unit tests that exercise the driver.\n\nThe unit tests introduce a dependency on the Proton AMQP libraries. Proton packages are currently available for Centos 6, and we are working to make a PPA available for Debian.\nI also need to address the comments left by Flavio and Gordon - this patch does not yet include that.","accounts_in_message":[],"_revision_number":8},{"id":"804574b8233e74bac2cc6a368c7b7bc6a3428d6a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-24 02:36:14.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/8/check/gate-oslo.messaging-pep8/6fe1693 : FAILURE in 38s\n- gate-oslo.messaging-docs http://logs.openstack.org/15/75815/8/check/gate-oslo.messaging-docs/9cafbbf : FAILURE in 38s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/8/check/gate-oslo.messaging-python26/3b8d0cc : FAILURE in 19s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/8/check/gate-oslo.messaging-python27/d16b60b : FAILURE in 42s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/8/check/gate-oslo.messaging-requirements/2a33b8d : Incompatible requirement found; see https://wiki.openstack.org/wiki/Requirements in 12s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/8/check/check-tempest-dsvm-full/7ba22b5 : FAILURE in 6m 28s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/8/check/check-tempest-dsvm-postgres-full/b2b134d : FAILURE in 5m 48s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/8/check/check-tempest-dsvm-neutron/ab3ac86 : FAILURE in 6m 33s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/8/check/check-tempest-dsvm-neutron-heat-slow/e92df1e : FAILURE in 6m 26s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/15/75815/8/check/check-grenade-dsvm/cd22d29 : SUCCESS in 36m 59s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/8/check/check-devstack-dsvm-cells/ea80e35 : FAILURE in 3m 56s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/8/check/gate-tempest-dsvm-large-ops/18bf55f : FAILURE in 5m 32s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/8/check/gate-tempest-dsvm-neutron-large-ops/5467c09 : FAILURE in 5m 40s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/8/check/check-oslotest-dsvm-oslo.messaging/1c3d0bc : FAILURE in 4m 25s","accounts_in_message":[],"_revision_number":8},{"id":"7cb6901c6843b150e394147b177f1b238ee6abef","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-06-24 02:41:47.000000000","message":"Patch Set 8:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-full: unrecognized error\n- check-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1286818\n- check-tempest-dsvm-postgres-full: unrecognized error\n- gate-tempest-dsvm-large-ops: unrecognized error\n- gate-tempest-dsvm-neutron-large-ops: unrecognized error\n- check-tempest-dsvm-neutron: unrecognized error\n\nYou have some unrecognized errors.\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":8},{"id":"15d7d052ca5a1d7bf0883522a3bd2c8160858d8b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-26 20:00:09.000000000","message":"Patch Set 9:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":9},{"id":"5fcc227d43d22590b8b88fdc553beb87c44a3ce0","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-06-26 20:00:09.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"86ae1edddb4180384f8ba8cc4aa61f30fb0d4802","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-06-26 20:00:36.000000000","message":"Patch Set 7:\n\n(21 comments)","accounts_in_message":[],"_revision_number":7},{"id":"d9ee999562af82faa8c4df550e0f883750631c83","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-06-26 20:02:59.000000000","message":"Patch Set 9:\n\nWorkflow-1\n\nThis patch should address all current review comments.\n\nThe unit tests will not pass due to dependencies on pyngus and proton packages.\n\npyngus - is available at pypi.python.org, and I\u0027ve added it to test-requirements.txt, yet it does not appear in pypi.openstack.org\n\nproton - these packages are available on EPEL for centos/rhel.  The qpid team is currently creating a PPA for debian - it should be available soon.","accounts_in_message":[],"_revision_number":9},{"id":"522f4b66779b2be00308182dbd438426d81c93b8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-26 20:46:25.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/9/check/gate-oslo.messaging-pep8/5c43b08 : FAILURE in 1m 09s\n- gate-oslo.messaging-docs http://logs.openstack.org/15/75815/9/check/gate-oslo.messaging-docs/a0ce405 : FAILURE in 40s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/9/check/gate-oslo.messaging-python26/1458bbe : FAILURE in 18s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/9/check/gate-oslo.messaging-python27/f25af4c : FAILURE in 58s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/9/check/gate-oslo.messaging-requirements/25021da : Incompatible requirement found; see https://wiki.openstack.org/wiki/Requirements in 35s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/9/check/check-tempest-dsvm-full/cf1728c : FAILURE in 6m 33s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/9/check/check-tempest-dsvm-postgres-full/fccf514 : FAILURE in 5m 28s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/9/check/check-tempest-dsvm-neutron/b1c9776 : FAILURE in 4m 30s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/9/check/check-tempest-dsvm-neutron-heat-slow/2a28db0 : FAILURE in 5m 33s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/9/check/check-grenade-dsvm/6573c8d : SUCCESS in 39m 03s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/9/check/check-devstack-dsvm-cells/26ef7df : FAILURE in 4m 39s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/9/check/gate-tempest-dsvm-large-ops/264ebfe : FAILURE in 5m 56s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/9/check/gate-tempest-dsvm-neutron-large-ops/f1d1054 : FAILURE in 6m 20s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/9/check/check-oslotest-dsvm-oslo.messaging/fb88f97 : FAILURE in 3m 48s","accounts_in_message":[],"_revision_number":9},{"id":"26e5c10b39c02ecd3137063447215b4fa3bcdc9f","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-06-30 01:51:54.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a4f4721e370a1f7db83cf27269bf3265ab2edb2d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-30 01:51:55.000000000","message":"Patch Set 10:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":10},{"id":"884ddf0fd3aa981073d44376255e51e84bea6411","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-30 02:34:07.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/10/check/gate-oslo.messaging-pep8/4d38abc : FAILURE in 25s\n- gate-oslo.messaging-docs http://logs.openstack.org/15/75815/10/check/gate-oslo.messaging-docs/f894f4d : FAILURE in 34s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/10/check/gate-oslo.messaging-python26/c6c2e64 : FAILURE in 18s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/10/check/gate-oslo.messaging-python27/ccef79c : FAILURE in 16s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/10/check/gate-oslo.messaging-requirements/d53bd33 : Incompatible requirement found; see https://wiki.openstack.org/wiki/Requirements in 13s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/10/check/check-tempest-dsvm-full/5367a59 : FAILURE in 6m 23s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/10/check/check-tempest-dsvm-postgres-full/a9389ea : FAILURE in 4m 23s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/10/check/check-tempest-dsvm-neutron/b49ea2e : FAILURE in 5m 17s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/10/check/check-tempest-dsvm-neutron-heat-slow/8f61950 : FAILURE in 7m 26s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/10/check/check-grenade-dsvm/6a4172a : SUCCESS in 42m 06s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/10/check/check-devstack-dsvm-cells/0246142 : FAILURE in 5m 24s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/10/check/gate-tempest-dsvm-large-ops/b77e7e8 : FAILURE in 5m 49s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/10/check/gate-tempest-dsvm-neutron-large-ops/aa54409 : FAILURE in 6m 57s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/10/check/check-oslotest-dsvm-oslo.messaging/4e3e978 : FAILURE in 4m 06s","accounts_in_message":[],"_revision_number":10},{"id":"8e444e655e613234b7d8af7c4fc16233106373bc","author":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"date":"2014-07-02 08:27:31.000000000","message":"Patch Set 10: Workflow-1\n\nMarking as WIP because pyngus isn\u0027t available - see Iff0684de2e72d39d065d549447e6d59464b914d5","accounts_in_message":[],"_revision_number":10},{"id":"4c3653beb05c5b3b8741cee5f351675d437cc54a","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2014-07-02 11:17:58.000000000","message":"Patch Set 10: Code-Review-1\n\n(14 comments)\n\nHey,\n\nI just left some comments in both driver and controller modules. It\u0027s getting better, however, I think there\u0027s a bit of mixed up logic in some parts.\n\nFor example, it\u0027s not clear to me what parts actually belong to the driver, what parts belong to the eventloop and what other pats belong to pyngus. To be more precise, there are some callbacks like `connection_failed` that are being called from the _SocketConncetion instance and I wonder whether that logic should go into pyngus or not.\n\nOne more thing, the driver, especially the controller, seems to ask a lot for permission instead of forgiveness. I\u0027ve nothing against that but I\u0027m really concerned about unforeseen failures/exceptions that should be used as a landing pad to call callbacks etc.","accounts_in_message":[],"_revision_number":10},{"id":"595455f3c7b2720a92aeb674d1ad8ef7a7911287","author":{"_account_id":1247,"name":"Mark McLoughlin","email":"markmc@redhat.com","username":"markmc"},"date":"2014-07-02 11:52:57.000000000","message":"Patch Set 10:\n\n(1 comment)\n\nMinor suggestion inline","accounts_in_message":[],"_revision_number":10},{"id":"293579aadc7d3d902ac3d93d49536e1f814fd4f2","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-07-02 20:57:20.000000000","message":"Patch Set 10:\n\n(15 comments)","accounts_in_message":[],"_revision_number":10},{"id":"6c193244f106598b5306cabaa08924a1065c562a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-02 20:58:05.000000000","message":"Patch Set 11:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":11},{"id":"8a194a7e625d73f7949edadade322a01b641d314","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-07-02 20:58:05.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"4297c315a0158ac7ceda70460d18bdf82a65c6cb","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-07-02 21:02:32.000000000","message":"Patch Set 11:\n\nHi Flavio,\n\nThanks for the feedback - I\u0027ve pushed up another patchset that I think addresses most of the points you raised.\n\nI\u0027ve tried to simplify the callback logic a bit.  I agree with your last comment regarding the use of the \u0027connection-failed\" callback.  This is a callback that is used by Pyngus to signal a protocol-level failure of the connection.  The eventloop.py/_SocketConnection code was also leveraging that callback for handling socket failures, since in the end the same actions needed to be performed.\n\nI\u0027ve moved the socket failures into their own callback, separate from the protocol error callback used by pyngus in the latest patch.","accounts_in_message":[],"_revision_number":11},{"id":"32eb35e710497c0e2d138adbc2f26ef3627bafea","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-02 21:48:14.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/11/check/gate-oslo.messaging-pep8/09fb0ab : SUCCESS in 1m 16s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/11/check/gate-oslo.messaging-docs/d27d570/doc/build/html/ : SUCCESS in 53s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/11/check/gate-oslo.messaging-python26/36e934b : FAILURE in 17s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/11/check/gate-oslo.messaging-python27/be3011d : FAILURE in 1m 08s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/11/check/gate-oslo.messaging-requirements/a9d43b1 : Incompatible requirement found; see https://wiki.openstack.org/wiki/Requirements in 32s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/11/check/check-tempest-dsvm-full/295b85d : FAILURE in 9m 00s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/11/check/check-tempest-dsvm-postgres-full/b41e155 : FAILURE in 6m 37s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/11/check/check-tempest-dsvm-neutron/fb3df2e : FAILURE in 6m 04s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/11/check/check-tempest-dsvm-neutron-heat-slow/716d382 : FAILURE in 6m 25s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/11/check/check-grenade-dsvm/030fa30 : SUCCESS in 42m 22s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/11/check/check-devstack-dsvm-cells/638ff3b : FAILURE in 7m 59s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/11/check/gate-tempest-dsvm-large-ops/d0a03a0 : FAILURE in 6m 33s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/11/check/gate-tempest-dsvm-neutron-large-ops/cd35bd9 : FAILURE in 6m 45s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/11/check/check-oslotest-dsvm-oslo.messaging/cec5c9d : FAILURE in 5m 36s","accounts_in_message":[],"_revision_number":11},{"id":"732f7cc1580555188802ef80f4404896c285295c","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-07-03 11:26:29.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"327401234b0455f90435bb14d1fdf8d2e4ee21bf","author":{"_account_id":7763,"name":"Alexei Kornienko","email":"akornienko@mirantis.com","username":"akornienko"},"date":"2014-07-09 19:26:32.000000000","message":"Patch Set 11: Code-Review-1\n\nThis driver doesn\u0027t respect flow defined by MessagingServer. Cause of this I think that it should implement it\u0027s own messaging server (since it\u0027s using it\u0027s own controller and main loop).","accounts_in_message":[],"_revision_number":11},{"id":"f07fcf0cd92c9a7a1d771aed4f55e20a754069af","author":{"_account_id":7763,"name":"Alexei Kornienko","email":"akornienko@mirantis.com","username":"akornienko"},"date":"2014-07-23 21:24:25.000000000","message":"Patch Set 11:\n\n(2 comments)\n\nsome minor questions inline","accounts_in_message":[],"_revision_number":11},{"id":"4d25bdb36a3f548c7fe693996fd5a4012502ae1f","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-07-25 14:50:54.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"aeded5005c33d1f1a624d87b7a5ba1e8fa2301c0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-25 14:50:56.000000000","message":"Patch Set 12:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":12},{"id":"5d26d8cac6b3932726ba18e1a6eca07c116286df","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-07-25 14:58:01.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"76624f747e58f2153c9baf5e4bbfb1bcea2855de","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-07-25 15:03:50.000000000","message":"Patch Set 12:\n\nThis patch is an update against the current master branch.  It includes a few fixes for issues Gordon and I have found while doing some perf testing.\n\nNote well: Pypi now includes the proton python wrappers.  This new package includes a C extension that is built when the package is installed.  In order to build\u0026install this package, you\u0027ll need to have the Proton C development libraries installed locally.   These can be installed from EPEL (qpid-proton-c-devel) for rhel-based systems.  Packages have also been proposed for debian, but have yet to be included in that distro.  There\u0027s a PPA available for Ubuntu 14 LTS otherwise (ppa:qpid/released).","accounts_in_message":[],"_revision_number":12},{"id":"3f7881f4a014149746c676fb9e141bf47fff7c56","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-07-25 15:22:30.000000000","message":"Patch Set 12:\n\nIn reply to Alexei\u0027s feedback:\n\n\u003e Alexei Kornienko\t\tJul 9 3:26 PM\n\u003e \n\u003e Patch Set 11: Code-Review-1\n\u003e \n\u003e This driver doesn\u0027t respect flow defined by MessagingServer. Cause of this I think that it should implement it\u0027s own messaging server (since it\u0027s using it\u0027s own controller and main loop).\n\nThanks for the feedback Alexei.  You\u0027re correct, currently the driver\u0027s ability to back pressure senders in inadequate.  This especially affects the \u0027cast\u0027 pattern, which lacks the explicit backpressure of an rpc that waits for a reply.\n\nGordon and I have been working on changes to the way the driver grants credit on behalf of subscribers, as well as changes that will backpressure senders based on available credit.\n\nI haven\u0027t made much progress in the area to date - I\u0027ve been focusing on trying to get CI/devstack/gate support for this driver in place as that seems to be a major block to getting this driver included upstream.  My plan was to open a bug for the credit work after I put something in place for CI/devstack/gate.","accounts_in_message":[],"_revision_number":12},{"id":"3799c19e25e07038ba2d28d4d4103e5b2895709f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-25 15:32:44.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/12/check/gate-oslo.messaging-pep8/7e770ec : FAILURE in 1m 04s\n- gate-oslo.messaging-docs http://logs.openstack.org/15/75815/12/check/gate-oslo.messaging-docs/527322d : FAILURE in 1m 20s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/12/check/gate-oslo.messaging-python26/2438e73 : FAILURE in 57s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/12/check/gate-oslo.messaging-python27/e817d31 : FAILURE in 1m 16s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/12/check/gate-oslo.messaging-requirements/4a08b6b : Incompatible requirement found; see https://wiki.openstack.org/wiki/Requirements in 44s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/12/check/check-tempest-dsvm-full/e5d446c : FAILURE in 9m 30s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/12/check/check-tempest-dsvm-postgres-full/81e0f1c : FAILURE in 6m 26s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/12/check/check-tempest-dsvm-neutron/3a9fcdc : FAILURE in 7m 42s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/12/check/check-tempest-dsvm-neutron-heat-slow/ff23069 : FAILURE in 6m 34s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/12/check/check-grenade-dsvm/271eacb : SUCCESS in 40m 31s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/12/check/check-devstack-dsvm-cells/f6a585f : FAILURE in 8m 19s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/12/check/gate-tempest-dsvm-large-ops/09b22e0 : FAILURE in 5m 23s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/12/check/gate-tempest-dsvm-neutron-large-ops/70c8832 : FAILURE in 8m 15s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/12/check/check-oslotest-dsvm-oslo.messaging/30bb786 : FAILURE in 4m 12s","accounts_in_message":[],"_revision_number":12},{"id":"779b8761bbe72da593f63731bdf519788b0c71b1","author":{"_account_id":7763,"name":"Alexei Kornienko","email":"akornienko@mirantis.com","username":"akornienko"},"date":"2014-07-25 15:43:41.000000000","message":"Patch Set 12:\n\nNote to my comment about bad server flow:\n\nI\u0027m actually working on zeromq driver as a replacement for existing one.\n\nYou can check existing code here - https://github.com/Alexei-Kornienko/oslo.messaging/commits/zeromq\n\nTake a look on how I implement a custom Server with a main loop. It can make your controller stuff simpler since you won\u0027t have to pass messages between your main loop and Listener poll method.","accounts_in_message":[],"_revision_number":12},{"id":"b2d8735df0765d2ef6bd177a39290ecc1c730caa","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-07-31 17:39:50.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"326d0d5621303b3204b1a2ed5b828f8bc51c9455","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-31 17:39:51.000000000","message":"Patch Set 13:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":13},{"id":"a819882610e369859e6250ad1214c9c0f50c1505","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-31 19:02:01.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-pep8/b25c48f : SUCCESS in 1m 52s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/13/check/gate-oslo.messaging-docs/44a9de1/doc/build/html/ : SUCCESS in 1m 34s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-python26/5fea37f : SUCCESS in 1m 25s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-python27/9d41718 : SUCCESS in 2m 28s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-requirements/c494eed : SUCCESS in 29s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-full/613e17b : FAILURE in 33m 23s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-postgres-full/bd64bcc : SUCCESS in 35m 58s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-neutron/34a92e3 : SUCCESS in 59m 20s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-neutron-heat-slow/d3f3d6f : SUCCESS in 28m 07s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/13/check/check-grenade-dsvm/17db084 : SUCCESS in 33m 23s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/13/check/check-devstack-dsvm-cells/b6f817e : SUCCESS in 14m 21s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/13/check/gate-tempest-dsvm-large-ops/024fe25 : SUCCESS in 27m 44s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/13/check/gate-tempest-dsvm-neutron-large-ops/6dfce0f : SUCCESS in 21m 47s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/13/check/check-oslotest-dsvm-oslo.messaging/6774d89 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":13},{"id":"4649705461be7778d1738b066a7451a688553ed0","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-07-31 19:21:07.000000000","message":"Patch Set 13:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1324793\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1324793\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":13},{"id":"2d2004ac238b2d755da508fa8acd3cc0a9b0c406","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-07-31 19:30:15.000000000","message":"Patch Set 13:\n\nrecheck bug 1324793","accounts_in_message":[],"_revision_number":13},{"id":"5828bb13541d6c1af29356540dd88cc4de52a179","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-31 19:30:17.000000000","message":"Patch Set 13: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":13},{"id":"7655da721eab6632270f80a08b4783ad1bf7fc50","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-31 21:10:00.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-pep8/ed2ccdd : SUCCESS in 1m 24s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/13/check/gate-oslo.messaging-docs/8538d40/doc/build/html/ : SUCCESS in 1m 26s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-python26/403c26e : SUCCESS in 2m 05s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-python27/966e196 : SUCCESS in 1m 40s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-requirements/a382481 : SUCCESS in 49s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-full/088b41a : SUCCESS in 37m 27s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-postgres-full/b6906f4 : FAILURE in 42m 12s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-neutron/97d40fc : SUCCESS in 1h 02m 29s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-neutron-heat-slow/bed506d : SUCCESS in 28m 43s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/13/check/check-grenade-dsvm/3521cfa : SUCCESS in 36m 26s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/13/check/check-devstack-dsvm-cells/68cb0cd : SUCCESS in 18m 37s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/13/check/gate-tempest-dsvm-large-ops/369a411 : SUCCESS in 19m 01s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/13/check/gate-tempest-dsvm-neutron-large-ops/80edc3d : SUCCESS in 18m 02s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/13/check/check-oslotest-dsvm-oslo.messaging/d7b17c5 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":13},{"id":"45e03258ce32d40d6573629e06f6d76b275c2e53","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2014-08-01 13:50:09.000000000","message":"Patch Set 13:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":13},{"id":"7ff2bf11cd91155823b84763fab404a6d8260732","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-01 13:50:10.000000000","message":"Patch Set 13: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":13},{"id":"f940e5db4c9b33d6fbb108095463811eafcb6dfc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-01 15:00:03.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-pep8/eadd756 : SUCCESS in 1m 13s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/13/check/gate-oslo.messaging-docs/abab9ea/doc/build/html/ : SUCCESS in 1m 16s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-python26/45a6017 : SUCCESS in 2m 40s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-python27/209ee7b : SUCCESS in 1m 43s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-requirements/05f732e : SUCCESS in 16s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-full/ae348ec : SUCCESS in 37m 27s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-postgres-full/68f9512 : FAILURE in 50m 34s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-neutron/1866c72 : SUCCESS in 49m 22s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-neutron-heat-slow/15f9aa8 : SUCCESS in 20m 19s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/13/check/check-grenade-dsvm/77cd75d : SUCCESS in 36m 42s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/13/check/check-devstack-dsvm-cells/8b35592 : SUCCESS in 15m 37s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/13/check/gate-tempest-dsvm-large-ops/2b4af51 : SUCCESS in 19m 55s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/13/check/gate-tempest-dsvm-neutron-large-ops/13257ee : SUCCESS in 22m 24s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/13/check/check-oslotest-dsvm-oslo.messaging/27305d7 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":13},{"id":"0ebd1857c45b41f33550ad0c664ac3f922174b4d","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2014-08-05 14:28:09.000000000","message":"Patch Set 13:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":13},{"id":"9beb2b72d9629b2c722111e33d233f53d5b27a33","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-05 14:28:10.000000000","message":"Patch Set 13: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":13},{"id":"f99bffaa4abadddae16af93431afa3d6d864e009","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-05 15:53:33.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-pep8/883c1ca : SUCCESS in 1m 20s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/13/check/gate-oslo.messaging-docs/1f3da03/doc/build/html/ : SUCCESS in 1m 46s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-python26/3f4af37 : SUCCESS in 2m 28s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-python27/2adce04 : SUCCESS in 2m 08s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/13/check/gate-oslo.messaging-requirements/a02e833 : SUCCESS in 53s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-full/4f09f5a : SUCCESS in 45m 20s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-postgres-full/e815866 : SUCCESS in 43m 46s\n- check-tempest-dsvm-neutron http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-neutron/4d27a7d : SUCCESS in 50m 13s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/13/check/check-tempest-dsvm-neutron-heat-slow/b2ee86f : SUCCESS in 20m 35s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/13/check/check-grenade-dsvm/eebfa2c : SUCCESS in 43m 20s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/13/check/gate-tempest-dsvm-large-ops/65a0163 : SUCCESS in 18m 18s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/13/check/gate-tempest-dsvm-neutron-large-ops/3b3e3cc : SUCCESS in 18m 15s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/13/check/check-devstack-dsvm-cells/ac3bea8 : SUCCESS in 15m 14s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/13/check/check-oslotest-dsvm-oslo.messaging/7f9b937 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":13},{"id":"876e5a8fd68e7d6222a29b512a4b133517d4e31b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-19 00:20:45.000000000","message":"Patch Set 13: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":13},{"id":"7c24e789026336aaf0d1ebf7e4e64a6d1593ea54","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-08-19 19:42:57.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"a4df1983133276da7d0157f1ab55ea882d0839df","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-19 19:44:24.000000000","message":"Patch Set 14:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":14},{"id":"113f08d52063b636474bb043367d46d384438144","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-19 21:32:25.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/14/check/gate-oslo.messaging-pep8/c6497c8 : SUCCESS in 1m 44s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/14/check/gate-oslo.messaging-docs/cf8295d/doc/build/html/ : SUCCESS in 1m 05s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/14/check/gate-oslo.messaging-python26/6e91bf9 : SUCCESS in 2m 33s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/14/check/gate-oslo.messaging-python27/ba46d91 : SUCCESS in 1m 50s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/14/check/gate-oslo.messaging-requirements/92b2988 : SUCCESS in 23s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/14/check/check-tempest-dsvm-full/5f9bffd : SUCCESS in 48m 01s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/14/check/check-tempest-dsvm-postgres-full/cc3bb8e : SUCCESS in 54m 09s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/15/75815/14/check/check-tempest-dsvm-neutron-full/1da1c59 : SUCCESS in 47m 10s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/14/check/check-tempest-dsvm-neutron-heat-slow/5ee9008 : SUCCESS in 25m 18s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/14/check/check-grenade-dsvm/f1e6eb2 : SUCCESS in 37m 34s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/14/check/gate-tempest-dsvm-large-ops/93a33f7 : SUCCESS in 19m 08s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/14/check/gate-tempest-dsvm-neutron-large-ops/28bd363 : SUCCESS in 21m 20s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/14/check/check-devstack-dsvm-cells/9d2bdb6 : SUCCESS in 15m 09s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/14/check/check-oslotest-dsvm-oslo.messaging/b07a550 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":14},{"id":"7078389abcc0d8e08942f71061cd4971b86247c6","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-08-22 14:06:13.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"33625bc992cd2c511ce8c44638da1eca312f2a33","author":{"_account_id":7818,"name":"Ted Ross","email":"tross@redhat.com","username":"tross"},"date":"2014-08-22 14:48:57.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"783238715314ba5132ad28b09b4765e03e275997","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2014-08-26 09:16:06.000000000","message":"Patch Set 14: Code-Review-1\n\n(2 comments)\n\nCouple of nits.\n\nI think we need to agree on what to do with this driver.\n\nI\u0027m not really happy about this driver landing without a proper gate but I know the patches are up for review and there\u0027s commitment on making that happen. The driver won\u0027t reach a perfect state until we merge it and let people give it a try. Furthermore, at the Juno summit, we had a session on this driver and some benchmarks and live-tests results were presented.\n\nWhat I\u0027m getting at is that I\u0027d be happy to let this driver land as-is if we get full commitment from Ken on making the gate thing happen asap.\n\nThis patch has been around since Icehouse. We either let it land as-is or block it until all the gates are setup, which probably means L (not even Kilo). Lets agree on something before the feature freeze.\n\nKen, it\u0027d be nice if you could extend the commit message with some info about the tests you\u0027ve done.","accounts_in_message":[],"_revision_number":14},{"id":"d5a51c5a4f484553a79d3776ab92ea2cabc9a8a9","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-08-27 12:46:45.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"75d29afcc5061abeff2dfd5b948913abaf31802c","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-08-27 12:52:56.000000000","message":"Patch Set 14:\n\n(2 comments)\n\nI\u0027ve updated the commit message to more clearly explain the included functional tests and the platform dependencies.  Let me know what you think, thanks.","accounts_in_message":[],"_revision_number":14},{"id":"85d21dd033d616532b90d7073a5a9678d1ab4112","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-27 13:55:43.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/15/check/gate-oslo.messaging-pep8/77d6cd9 : SUCCESS in 1m 15s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/15/check/gate-oslo.messaging-docs/6d37be5/doc/build/html/ : SUCCESS in 1m 41s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/15/check/gate-oslo.messaging-python26/6c6269e : SUCCESS in 1m 14s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/15/check/gate-oslo.messaging-python27/2b5c0f7 : SUCCESS in 2m 10s\n- gate-oslo.messaging-python33 http://logs.openstack.org/15/75815/15/check/gate-oslo.messaging-python33/50c87c8 : SUCCESS in 3m 21s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/15/check/gate-oslo.messaging-requirements/dbec4e6 : SUCCESS in 24s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/15/check/check-tempest-dsvm-full/5cde948 : SUCCESS in 56m 46s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/15/check/check-tempest-dsvm-postgres-full/8b7086d : SUCCESS in 1h 08m 53s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/15/75815/15/check/check-tempest-dsvm-neutron-full/5b853b7 : SUCCESS in 47m 39s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/15/check/check-tempest-dsvm-neutron-heat-slow/3b900a5 : SUCCESS in 27m 44s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/15/check/check-grenade-dsvm/d5d554b : SUCCESS in 42m 16s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/15/check/gate-tempest-dsvm-large-ops/f2d95df : SUCCESS in 22m 32s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/15/check/gate-tempest-dsvm-neutron-large-ops/d2e8f4b : SUCCESS in 25m 12s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/15/check/check-devstack-dsvm-cells/8ace832 : SUCCESS in 17m 06s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/15/check/check-oslotest-dsvm-oslo.messaging/4d8f062 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":15},{"id":"8b41d94858367b1dcec30be5ffbe497682f18fd5","author":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"date":"2014-08-28 08:41:19.000000000","message":"Patch Set 15: Code-Review-1\n\n(7 comments)\n\n@Ken: some nits\n\nThe patch have now a good state, and it will be easier to review, small change ant 0bug fix than reread everything of thoses patchsets.\nI a bit sad to see it\u0027s blocked because of packaging issue :(","accounts_in_message":[],"_revision_number":15},{"id":"5b78b0fa11fa0c648c18ede3020c867590dea957","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-08-28 20:42:43.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"3bb45a950586972b83b9a01764aa2f48fb9dd457","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-08-28 20:43:38.000000000","message":"Patch Set 15:\n\n(7 comments)\n\nThank you for the feedback Mehdi!","accounts_in_message":[],"_revision_number":15},{"id":"e07a736392ae35fedf4d030bf31bfa9c4bf18b59","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-29 01:42:29.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/16/check/gate-oslo.messaging-pep8/a18674a : SUCCESS in 1m 07s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/16/check/gate-oslo.messaging-docs/7f5d2e1/doc/build/html/ : SUCCESS in 41s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/16/check/gate-oslo.messaging-python26/1fc8e1d : FAILURE in 2m 13s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/16/check/gate-oslo.messaging-python27/b72b837 : SUCCESS in 1m 39s\n- gate-oslo.messaging-python33 http://logs.openstack.org/15/75815/16/check/gate-oslo.messaging-python33/0ed63d2 : FAILURE in 3m 03s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/16/check/gate-oslo.messaging-requirements/7a28bf6 : SUCCESS in 10s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/16/check/check-tempest-dsvm-full/b9b950b : SUCCESS in 1h 03m 33s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/16/check/check-tempest-dsvm-postgres-full/6e6ca49 : SUCCESS in 1h 13m 50s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/15/75815/16/check/check-tempest-dsvm-neutron-full/849aaf9 : SUCCESS in 52m 54s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/16/check/check-tempest-dsvm-neutron-heat-slow/44b5ccb : SUCCESS in 26m 31s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/16/check/check-grenade-dsvm/511c3b8 : SUCCESS in 40m 29s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/16/check/gate-tempest-dsvm-large-ops/46ef994 : SUCCESS in 21m 43s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/16/check/gate-tempest-dsvm-neutron-large-ops/0c76cd6 : SUCCESS in 27m 44s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/16/check/check-devstack-dsvm-cells/b9c8af2 : SUCCESS in 18m 05s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/16/check/check-oslotest-dsvm-oslo.messaging/e8320e2 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":16},{"id":"6f1b8f6ca854329696c68d0db99da77bd8bb9d8f","author":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"date":"2014-08-29 07:05:00.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"0867f857d7524188fa403e46365c7b00ca00154a","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2014-08-29 07:16:04.000000000","message":"Patch Set 16:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":16},{"id":"6722f019b6d1101b3750927668bbf86d1e158309","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-29 08:33:27.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/16/check/gate-oslo.messaging-pep8/f507f85 : SUCCESS in 1m 01s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/16/check/gate-oslo.messaging-docs/d545f02/doc/build/html/ : SUCCESS in 1m 51s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/16/check/gate-oslo.messaging-python26/8f315a9 : FAILURE in 1m 21s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/16/check/gate-oslo.messaging-python27/823b2b8 : FAILURE in 1m 05s\n- gate-oslo.messaging-python33 http://logs.openstack.org/15/75815/16/check/gate-oslo.messaging-python33/51b39d5 : FAILURE in 2m 33s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/16/check/gate-oslo.messaging-requirements/8bfb385 : SUCCESS in 20s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/16/check/check-tempest-dsvm-full/c396bf2 : SUCCESS in 1h 01m 38s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/16/check/check-tempest-dsvm-postgres-full/fee9905 : SUCCESS in 1h 05m 25s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/15/75815/16/check/check-tempest-dsvm-neutron-full/24b45e6 : SUCCESS in 1h 07m 45s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/16/check/check-tempest-dsvm-neutron-heat-slow/79e30f7 : SUCCESS in 24m 38s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/16/check/check-grenade-dsvm/5b0a2f1 : SUCCESS in 47m 51s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/16/check/gate-tempest-dsvm-large-ops/a47da2a : SUCCESS in 27m 37s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/16/check/gate-tempest-dsvm-neutron-large-ops/15f6409 : SUCCESS in 41m 55s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/16/check/check-devstack-dsvm-cells/5ea1388 : SUCCESS in 15m 31s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/16/check/check-oslotest-dsvm-oslo.messaging/55b1245 : FAILURE in 4m 36s","accounts_in_message":[],"_revision_number":16},{"id":"1f12b1ddf113f62377f8031f59e9dbbba0d533fb","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-08-29 21:49:24.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"d3dd760bf9e525501043552b1d930d3be5429fd3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-29 22:59:52.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/17/check/gate-oslo.messaging-pep8/cb58bab : SUCCESS in 1m 16s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/17/check/gate-oslo.messaging-docs/0594983/doc/build/html/ : SUCCESS in 1m 16s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/17/check/gate-oslo.messaging-python26/3154156 : SUCCESS in 2m 01s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/17/check/gate-oslo.messaging-python27/b4621f3 : SUCCESS in 1m 04s\n- gate-oslo.messaging-python33 http://logs.openstack.org/15/75815/17/check/gate-oslo.messaging-python33/b97fe16 : SUCCESS in 4m 06s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/17/check/gate-oslo.messaging-requirements/b73a22b : SUCCESS in 4s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/17/check/check-tempest-dsvm-full/6cc6f46 : SUCCESS in 51m 17s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/17/check/check-tempest-dsvm-postgres-full/b508a79 : SUCCESS in 1h 01m 36s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/15/75815/17/check/check-tempest-dsvm-neutron-full/850507a : SUCCESS in 52m 41s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/17/check/check-tempest-dsvm-neutron-heat-slow/2d7f044 : SUCCESS in 18m 22s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/17/check/check-grenade-dsvm/4efe285 : SUCCESS in 46m 21s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/17/check/gate-tempest-dsvm-large-ops/7b0dcfb : SUCCESS in 22m 31s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/17/check/gate-tempest-dsvm-neutron-large-ops/d41d34a : SUCCESS in 24m 46s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/17/check/check-devstack-dsvm-cells/13a29bc : SUCCESS in 17m 40s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/17/check/check-oslotest-dsvm-oslo.messaging/e108553 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":17},{"id":"d036395d6516fa05954d6d5546adc7ce3796dcbc","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2014-08-30 19:24:12.000000000","message":"Patch Set 17:\n\nJust out of curiosity why is this being put into oslo.messaging?\n\nKombu now has a qpid driver, why not put it in kombu so it can be used in a more general purpose manner by entities outside openstack?\n\nIf the question was already raised let me know, but it seems like we should (imho) be contributing to libraries that are general purpose, not contributing to libraries that are only useful in openstack (and really are only useful in openstack applications...)\n\nThoughts?\n\nMaybe I should raise this on the ML instead since it might involve others that want to have some input; something just doesn\u0027t seem right when we are duplicating (and thus having to maintain) things that are better served by other open solutions...","accounts_in_message":[],"_revision_number":17},{"id":"16ba55fc1f7a410954c651c1ddc39d7a5d41ed7d","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-09-01 09:22:42.000000000","message":"Patch Set 17:\n\n\u003e Just out of curiosity why is this being put into\n\u003e oslo.messaging?\n\u003e\n\u003e Kombu now has a qpid driver, why not put it in kombu so\n\u003e it can be used in a more general purpose manner by\n\u003e entities outside openstack?\n\nThis is intended to be an AMQP 1.0 (ISO/IEC 19464) driver, with a clean and well defined mapping to the protocol.\n\nIt is not intended as a qpid specific driver. There already is one of those for oslo.messaging. It uses an older version of AMQP superceded by 1.0 (as does the qpid support in kombu, I believe). Though anyone wanting to use qpidd could do so using this new driver (which I believe will perform and scale much better than the existing qpid specific one), this driver is intended to be of much wider use.\n\nThe wire protocol becomes an interface, that when honoured allows different solutions at deployment. (The pre-1.0 iterations of AMQP are really entirely different protocols).\n\n\u003e If the question was already raised let me know, but it\n\u003e seems like we should (imho) be contributing to libraries\n\u003e that are general purpose, not contributing to libraries\n\u003e that are only useful in openstack (and really are only\n\u003e useful in openstack applications...)\n\nI\u0027m not sure I understand your point here. The oslo.messaging library itself is only used in openstack. As I understand it, it is intended as a pluggable RPC+notify model that is actually implemented over other messaging technologies, through different drivers. This is such a driver, whose aim is to provide AMQP 1.0 protocol support to users of oslo.messaging. \n\nAre you saying you don\u0027t think oslo.messaging should be pluggable in this way? That it should just use kombu and whatever that offers?\n\nIf so, the problem I see with that is that the kombu model and API is really driven by a specific older version of AMQP (0-9-1). Using it to drive other protocols obscures the mapping, may prevent access to useful features and  impact the efficiency of use. The oslo.messaging library itself offers a slightly higher level abstraction (rpc \u0026 notify) and I believe adapting that to different solutions gives more flexibility and is more likely to offer real choice by allowing different technologies supported to be fully exploited to the best of their capabilities.\n\n\u003e Thoughts?\n\u003e \n\u003e Maybe I should raise this on the ML instead since it\n\u003e might involve others that want to have some input;\n\u003e something just doesn\u0027t seem right when we are duplicating\n\u003e (and thus having to maintain) things that are better \n\u003e served by other open solutions...\n\nI don\u0027t think anything is being duplicated by this driver. As above the qpid support in kombu and the existing qpid driver could be considered to be overlapping. This driver however is offering a completely different protocol.","accounts_in_message":[],"_revision_number":17},{"id":"fecc1a1f4ba4fa4e196b18010e703018d2b908a4","author":{"_account_id":7536,"name":"Ilya Pekelny","email":"ipekelny@mirantis.com","username":"I159"},"date":"2014-09-01 15:36:47.000000000","message":"Patch Set 17: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":17},{"id":"f98b95282fa8406a1cc15ba8adf8921534a680f4","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2014-09-01 20:51:18.000000000","message":"Patch Set 17:\n\nI\u0027m not saying much either way.\n\nI just wanted to hear the opinions and see the feedback so that at least someone asked the obvious (at least I think it\u0027s obvious) question about helping kombu become more capable instead of just helping oslo.messaging (which is a big user of kombu) become better (even if kombu may not be the perfect API match at the current moment).\n\nIt just feels like we have taken a lot from kombu and how much have we given back? That\u0027s all...","accounts_in_message":[],"_revision_number":17},{"id":"5abadf17ff218099dc100935b0b39fbad2b4e06e","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-09-02 16:21:15.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"755d3da65f8a117d64b990cae34a8337c4033bfa","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-09-02 16:21:28.000000000","message":"Patch Set 17:\n\n(10 comments)","accounts_in_message":[],"_revision_number":17},{"id":"1bd4fae8c6695e4ad5609b364a92e4af823e5f72","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-02 18:44:53.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/18/check/gate-oslo.messaging-pep8/08c42e3 : SUCCESS in 47s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/18/check/gate-oslo.messaging-docs/76a3800/doc/build/html/ : SUCCESS in 1m 33s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/18/check/gate-oslo.messaging-python26/5baff54 : SUCCESS in 2m 38s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/18/check/gate-oslo.messaging-python27/d9954df : SUCCESS in 2m 02s\n- gate-oslo.messaging-python33 http://logs.openstack.org/15/75815/18/check/gate-oslo.messaging-python33/4c8783b : SUCCESS in 4m 49s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/18/check/gate-oslo.messaging-requirements/029ad5c : SUCCESS in 26s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/18/check/check-tempest-dsvm-full/2ede21c : SUCCESS in 1h 01m 03s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/18/check/check-tempest-dsvm-postgres-full/8e3ce4a : SUCCESS in 1h 00m 55s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/15/75815/18/check/check-tempest-dsvm-neutron-full/107deff : SUCCESS in 1h 08m 30s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/18/check/check-tempest-dsvm-neutron-heat-slow/dd79bd2 : SUCCESS in 24m 12s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/18/check/check-grenade-dsvm/ec78a72 : SUCCESS in 43m 47s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/18/check/gate-tempest-dsvm-large-ops/1ab515c : SUCCESS in 25m 05s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/18/check/gate-tempest-dsvm-neutron-large-ops/d86d648 : SUCCESS in 24m 01s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/18/check/check-devstack-dsvm-cells/be21165 : SUCCESS in 16m 50s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/18/check/check-oslotest-dsvm-oslo.messaging/cec38c0 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":18},{"id":"d7fa275e395c9adca6981ac47cad2785c384355c","author":{"_account_id":7536,"name":"Ilya Pekelny","email":"ipekelny@mirantis.com","username":"I159"},"date":"2014-09-03 10:06:53.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"cfb7943a40ecb36937b789d0d0358cb0b2faabbe","author":{"_account_id":7536,"name":"Ilya Pekelny","email":"ipekelny@mirantis.com","username":"I159"},"date":"2014-09-03 10:07:39.000000000","message":"Patch Set 18: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"c2cd5545a6003d08cc794d4d2811bef35e5c1365","author":{"_account_id":7536,"name":"Ilya Pekelny","email":"ipekelny@mirantis.com","username":"I159"},"date":"2014-09-03 10:09:10.000000000","message":"Patch Set 18:\n\nThanks for positive attitude!","accounts_in_message":[],"_revision_number":18},{"id":"b59b640db4c59732e59c6933a39ee84b10134204","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-09-03 12:40:23.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"47f28a30c7c49ddfa5f777fdc8b6e8c16da2f871","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-09-03 12:41:26.000000000","message":"Patch Set 17:\n\n(1 comment)\n\nThanks for your advice, I appreciate the help.","accounts_in_message":[],"_revision_number":17},{"id":"0cb2355f66e76ed7eb93c8384d2fa3a3b9c99a65","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-09-03 12:42:00.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"5447de7f047e47311e62ee5029d2b12709c6f6ce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-03 13:57:32.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/19/check/gate-oslo.messaging-pep8/89203ab : SUCCESS in 52s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/19/check/gate-oslo.messaging-docs/335273d/doc/build/html/ : SUCCESS in 41s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/19/check/gate-oslo.messaging-python26/f06eb59 : SUCCESS in 1m 47s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/19/check/gate-oslo.messaging-python27/3565765 : SUCCESS in 1m 04s\n- gate-oslo.messaging-python33 http://logs.openstack.org/15/75815/19/check/gate-oslo.messaging-python33/7fcb4ab : SUCCESS in 4m 08s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/19/check/gate-oslo.messaging-requirements/0cea788 : SUCCESS in 14s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/19/check/check-tempest-dsvm-full/aa72152 : SUCCESS in 59m 04s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/19/check/check-tempest-dsvm-postgres-full/b223824 : SUCCESS in 1h 13m 32s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/15/75815/19/check/check-tempest-dsvm-neutron-full/d68119f : SUCCESS in 1h 07m 18s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/19/check/check-tempest-dsvm-neutron-heat-slow/0987b34 : SUCCESS in 23m 31s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/19/check/check-grenade-dsvm/6de02a8 : SUCCESS in 46m 33s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/19/check/gate-tempest-dsvm-large-ops/12e673e : SUCCESS in 27m 02s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/19/check/gate-tempest-dsvm-neutron-large-ops/e64719e : SUCCESS in 24m 19s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/19/check/check-devstack-dsvm-cells/f3e0bf1 : SUCCESS in 16m 30s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/19/check/check-oslotest-dsvm-oslo.messaging/7ce6093 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":19},{"id":"a1252143376fb2486645bdafc5e9950fb8e3263f","author":{"_account_id":7536,"name":"Ilya Pekelny","email":"ipekelny@mirantis.com","username":"I159"},"date":"2014-09-03 14:12:03.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"78c1aadbce0a1cc382578b9880c82f0e475108a5","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-09-03 14:37:31.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"c3d4e02bc6071384e8aea4ad726f98d9e4ae4304","author":{"_account_id":7818,"name":"Ted Ross","email":"tross@redhat.com","username":"tross"},"date":"2014-09-03 15:40:33.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"1e5cc3f7bc25472cdf9f0d5cc293b1e388b4b6e2","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2014-09-03 20:42:02.000000000","message":"Patch Set 19: Code-Review-1\n\n(5 comments)\n\nI have a couple of comments inline. I think we\u0027ll review the fixes faster if they are submitted as separate dependent patches after this one. The only one I would want to block this on is the one related to the config option registration. The others can land any time before RC1, but we don\u0027t want to release a version of the library with the options in a position we know we\u0027re going to deprecate later.","accounts_in_message":[],"_revision_number":19},{"id":"3a6257f42825148234dbe8a8978d788b6ec9a3da","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2014-09-03 23:26:34.000000000","message":"Patch Set 19:\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"022e7eadbd493b1dfaba6e8b71404d3b24ee53ae","author":{"_account_id":7536,"name":"Ilya Pekelny","email":"ipekelny@mirantis.com","username":"I159"},"date":"2014-09-04 13:14:57.000000000","message":"Patch Set 19: Code-Review-1\n\n(4 comments)\n\nI see a sense in Doug\u0027s and Joshua\u0027s comments. Let\u0027s continue to improve this patch.","accounts_in_message":[],"_revision_number":19},{"id":"b57212b854966f10ef158fac74a2eaad03900513","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-09-04 18:39:12.000000000","message":"Patch Set 19:\n\n(9 comments)","accounts_in_message":[],"_revision_number":19},{"id":"f88f05f577f8b9923dabd745f0fd3afe08619b25","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-09-04 19:15:40.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"7d2b6ea997b9705a86d219291e88a772d5eec254","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2014-09-04 22:04:34.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"d7a8bdaacfaf35372ac9cc1b33fe9fbe41743f62","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2014-09-04 22:13:12.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"9c6600c8591dbc515a89de61118010323906640f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-05 00:28:48.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/20/check/gate-oslo.messaging-pep8/5cdd7ce : SUCCESS in 1m 05s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/20/check/gate-oslo.messaging-docs/ae1e6f9/doc/build/html/ : SUCCESS in 1m 17s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/20/check/gate-oslo.messaging-python26/5450d9e : SUCCESS in 2m 57s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/20/check/gate-oslo.messaging-python27/7f440e3 : SUCCESS in 1m 40s\n- gate-oslo.messaging-python33 http://logs.openstack.org/15/75815/20/check/gate-oslo.messaging-python33/1fee51e : SUCCESS in 4m 24s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/20/check/gate-oslo.messaging-requirements/b4189f1 : SUCCESS in 10s\n- check-tempest-dsvm-full http://logs.openstack.org/15/75815/20/check/check-tempest-dsvm-full/ae018f8 : SUCCESS in 1h 01m 46s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/20/check/check-tempest-dsvm-postgres-full/4341e0a : SUCCESS in 1h 16m 51s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/15/75815/20/check/check-tempest-dsvm-neutron-full/61b3909 : SUCCESS in 53m 19s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/20/check/check-tempest-dsvm-neutron-heat-slow/a8bed2e : SUCCESS in 20m 54s\n- check-grenade-dsvm http://logs.openstack.org/15/75815/20/check/check-grenade-dsvm/9474b8a : SUCCESS in 43m 40s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/20/check/gate-tempest-dsvm-large-ops/48a0b78 : SUCCESS in 26m 28s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/20/check/gate-tempest-dsvm-neutron-large-ops/c80472c : SUCCESS in 19m 04s\n- check-devstack-dsvm-cells http://logs.openstack.org/15/75815/20/check/check-devstack-dsvm-cells/ccafd5b : SUCCESS in 18m 55s\n- check-oslotest-dsvm-oslo.messaging http://logs.openstack.org/15/75815/20/check/check-oslotest-dsvm-oslo.messaging/a70d4a4 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":20},{"id":"876b68d5cf5bd0fb75046e77b1dfb6e82aa59569","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2014-09-05 07:14:35.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"ae82bec15ad62573fdec58e163fd511e758772aa","author":{"_account_id":7536,"name":"Ilya Pekelny","email":"ipekelny@mirantis.com","username":"I159"},"date":"2014-09-05 09:18:34.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"ea972e7d47108157347f57ec8b13a1c4c5278963","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-09-05 09:23:49.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"346093f9029267d05169133c3f9090618bd1ab90","author":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"date":"2014-09-05 09:29:24.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"a1b5910c8e67b7befcda052298ac0bf96f0434d9","author":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"date":"2014-09-05 12:48:25.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"127f5b98d65e4bdc771900ddb530ce400a975fdc","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2014-09-05 12:54:27.000000000","message":"Patch Set 20: Code-Review+2\n\nI\u0027ll let Doug approve this.","accounts_in_message":[],"_revision_number":20},{"id":"f100104f7a162485601473eee4fff86fbf74c89d","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2014-09-05 13:24:20.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1\n\nFFE approved","accounts_in_message":[],"_revision_number":20},{"id":"8d3076848a2cf415de9a3f0f69f8d7e989ac0394","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-05 13:24:21.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":20},{"id":"90f55aec9625008eafbabf8f857b79b678162542","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2014-09-05 15:34:05.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"611ed12f4724800b388ad743cb7a4559ba9b3504","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2014-09-05 18:20:23.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"517e7a79c10f54635075267bb2bf9948c45b7f3d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-06 08:56:41.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":20},{"id":"6071e63ab0e51cceca4d0b434812cc63ae8700f2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-06 08:56:41.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded.\n\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/15/75815/20/gate/gate-oslo.messaging-docs/4c444c0/doc/build/html/ : SUCCESS in 1m 07s\n- gate-oslo.messaging-pep8 http://logs.openstack.org/15/75815/20/gate/gate-oslo.messaging-pep8/2dc9342 : SUCCESS in 48s\n- gate-oslo.messaging-python26 http://logs.openstack.org/15/75815/20/gate/gate-oslo.messaging-python26/00c21ff : SUCCESS in 1m 31s\n- gate-oslo.messaging-python27 http://logs.openstack.org/15/75815/20/gate/gate-oslo.messaging-python27/e8c28b3 : SUCCESS in 1m 17s\n- gate-oslo.messaging-python33 http://logs.openstack.org/15/75815/20/gate/gate-oslo.messaging-python33/47c85dc : SUCCESS in 2m 59s\n- gate-oslo.messaging-requirements http://logs.openstack.org/15/75815/20/gate/gate-oslo.messaging-requirements/e3d16a4 : SUCCESS in 12s\n- gate-tempest-dsvm-full http://logs.openstack.org/15/75815/20/gate/gate-tempest-dsvm-full/e073035 : SUCCESS in 48m 41s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/15/75815/20/gate/gate-tempest-dsvm-postgres-full/d8fb0d1 : SUCCESS in 1h 03m 45s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/15/75815/20/gate/gate-tempest-dsvm-neutron-full/4ee6b29 : SUCCESS in 51m 34s\n- gate-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/15/75815/20/gate/gate-tempest-dsvm-neutron-heat-slow/45ea834 : SUCCESS in 19m 07s\n- gate-grenade-dsvm http://logs.openstack.org/15/75815/20/gate/gate-grenade-dsvm/6b92533 : SUCCESS in 37m 34s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/15/75815/20/gate/gate-tempest-dsvm-large-ops/c84fd69 : SUCCESS in 18m 53s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/15/75815/20/gate/gate-tempest-dsvm-neutron-large-ops/ba9e550 : SUCCESS in 19m 03s\n- gate-devstack-dsvm-cells http://logs.openstack.org/15/75815/20/gate/gate-devstack-dsvm-cells/32c957e : SUCCESS in 14m 13s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"fbee94170f918c165ac4e7eb3bacd0dc193acc50","revisions":{"8021997e0e1df12e9b708d0093efde27cc27a2c7":{"kind":"REWORK","_number":1,"created":"2014-02-24 09:26:22.000000000","uploader":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"ref":"refs/changes/15/75815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/1"}}},"commit":{"parents":[{"commit":"8a3744fe055843ecf5a7208ea32fb42e6c4224b1","subject":"Pass the session to QpidMessage constructor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8a3744fe055843ecf5a7208ea32fb42e6c4224b1"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-24 09:26:29.000000000","tz":0},"subject":"An initial implementation of a 1.0 based messaging driver","message":"An initial implementation of a 1.0 based messaging driver\n\nThis uses the proton protocol engine library from the Apache Qpid project. The\ndriver has its own IO thread and event loop. The files under engine are really\ngeneric and could at some point be separated into their own library or folded\ninto the engine library itself somehow. The key driver interface is implemented\nin driver.py, tmost of the logic of mapping to AMQP 1.0 is in controller.py and\nthe eventloop.py is also fairly generic.\n\nImplements: blueprint Provide a AMQP 1.0 implementation for oslo.messaging\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8021997e0e1df12e9b708d0093efde27cc27a2c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8021997e0e1df12e9b708d0093efde27cc27a2c7"}]},"branch":"refs/heads/master"},"0b1a4a8a48a7ec1db2f7061d8538b951f4d2bac4":{"kind":"REWORK","_number":2,"created":"2014-02-24 15:08:19.000000000","uploader":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"ref":"refs/changes/15/75815/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/2"}}},"commit":{"parents":[{"commit":"8a3744fe055843ecf5a7208ea32fb42e6c4224b1","subject":"Pass the session to QpidMessage constructor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8a3744fe055843ecf5a7208ea32fb42e6c4224b1"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-24 15:09:59.000000000","tz":0},"subject":"An initial implementation of a 1.0 based messaging driver","message":"An initial implementation of a 1.0 based messaging driver\n\nThis uses the proton protocol engine library from the Apache Qpid project. The\ndriver has its own IO thread and event loop. The files under engine are really\ngeneric and could at some point be separated into their own library or folded\ninto the engine library itself somehow. The key driver interface is implemented\nin driver.py, tmost of the logic of mapping to AMQP 1.0 is in controller.py and\nthe eventloop.py is also fairly generic.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0b1a4a8a48a7ec1db2f7061d8538b951f4d2bac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0b1a4a8a48a7ec1db2f7061d8538b951f4d2bac4"}]},"branch":"refs/heads/master"},"5a5e5f1046fa35f2b34a2784a946f5191c0ea7e2":{"kind":"REWORK","_number":3,"created":"2014-02-25 14:47:51.000000000","uploader":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"ref":"refs/changes/15/75815/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/3"}}},"commit":{"parents":[{"commit":"e9821fe602344c6e755a1510ddd60f7ed0f09853","subject":"Merge \"Improve help strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e9821fe602344c6e755a1510ddd60f7ed0f09853"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-25 14:49:50.000000000","tz":0},"subject":"An initial implementation of a 1.0 based messaging driver","message":"An initial implementation of a 1.0 based messaging driver\n\nThis uses the proton protocol engine library from the Apache Qpid project. The\ndriver has its own IO thread and event loop. The files under engine are really\ngeneric and could at some point be separated into their own library or folded\ninto the engine library itself somehow. The key driver interface is implemented\nin driver.py, tmost of the logic of mapping to AMQP 1.0 is in controller.py and\nthe eventloop.py is also fairly generic.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5a5e5f1046fa35f2b34a2784a946f5191c0ea7e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5a5e5f1046fa35f2b34a2784a946f5191c0ea7e2"}]},"branch":"refs/heads/master"},"60c63861c66421bb0787e98a4167411d9f89753e":{"kind":"REWORK","_number":4,"created":"2014-03-31 17:08:41.000000000","uploader":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"ref":"refs/changes/15/75815/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/4"}}},"commit":{"parents":[{"commit":"19053b6e652d54444af44cff2c7db640b5b5b207","subject":"Merge \"Use messaging_conf fixture configuration by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/19053b6e652d54444af44cff2c7db640b5b5b207"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-03-31 17:10:32.000000000","tz":60},"subject":"An initial implementation of a 1.0 based messaging driver","message":"An initial implementation of a 1.0 based messaging driver\n\nThis uses the proton protocol engine library from the Apache Qpid project. The\ndriver has its own IO thread and event loop. The files under engine are really\ngeneric and could at some point be separated into their own library or folded\ninto the engine library itself somehow. The key driver interface is implemented\nin driver.py, tmost of the logic of mapping to AMQP 1.0 is in controller.py and\nthe eventloop.py is also fairly generic.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/60c63861c66421bb0787e98a4167411d9f89753e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/60c63861c66421bb0787e98a4167411d9f89753e"}]},"branch":"refs/heads/master"},"aae932bd5512927771870a7b8a0a44953be8c37e":{"kind":"REWORK","_number":5,"created":"2014-04-17 10:31:15.000000000","uploader":{"_account_id":8784,"name":"Gordon Sim","email":"gsim@redhat.com","username":"gordonsim"},"ref":"refs/changes/15/75815/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/5"}}},"commit":{"parents":[{"commit":"dc5cb4a19271c29efe520d0ecaa2f72feeafef6c","subject":"Merge \"Ensure routing key is specified in the address for a direct producer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dc5cb4a19271c29efe520d0ecaa2f72feeafef6c"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-04-17 10:30:53.000000000","tz":60},"subject":"An initial implementation of a 1.0 based messaging driver","message":"An initial implementation of a 1.0 based messaging driver\n\nThis uses the proton protocol engine library from the Apache Qpid project. The\ndriver has its own IO thread and event loop. The files under engine are really\ngeneric and could at some point be separated into their own library or folded\ninto the engine library itself somehow. The key driver interface is implemented\nin driver.py, tmost of the logic of mapping to AMQP 1.0 is in controller.py and\nthe eventloop.py is also fairly generic.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/aae932bd5512927771870a7b8a0a44953be8c37e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/aae932bd5512927771870a7b8a0a44953be8c37e"}]},"branch":"refs/heads/master"},"6bd4912f53d50a6ce167924a450888857a1b2633":{"kind":"REWORK","_number":6,"created":"2014-06-03 13:03:39.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/6"}}},"commit":{"parents":[{"commit":"dc5cb4a19271c29efe520d0ecaa2f72feeafef6c","subject":"Merge \"Ensure routing key is specified in the address for a direct producer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dc5cb4a19271c29efe520d0ecaa2f72feeafef6c"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-06-03 13:02:41.000000000","tz":-240},"subject":"An initial implementation of a 1.0 based messaging driver","message":"An initial implementation of a 1.0 based messaging driver\n\nThis uses the proton protocol engine library from the Apache Qpid project. The\ndriver has its own IO thread and event loop. The files under engine are really\ngeneric and could at some point be separated into their own library or folded\ninto the engine library itself somehow. The key driver interface is implemented\nin driver.py, tmost of the logic of mapping to AMQP 1.0 is in controller.py and\nthe eventloop.py is also fairly generic.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6bd4912f53d50a6ce167924a450888857a1b2633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6bd4912f53d50a6ce167924a450888857a1b2633"}]},"branch":"refs/heads/master"},"a9ac3a759b9818d138ca68ba14b830aa9f16ccb0":{"kind":"REWORK","_number":7,"created":"2014-06-04 20:02:23.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/7"}}},"commit":{"parents":[{"commit":"dc5cb4a19271c29efe520d0ecaa2f72feeafef6c","subject":"Merge \"Ensure routing key is specified in the address for a direct producer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dc5cb4a19271c29efe520d0ecaa2f72feeafef6c"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-06-04 20:01:20.000000000","tz":-240},"subject":"An initial implementation of a 1.0 based messaging driver","message":"An initial implementation of a 1.0 based messaging driver\n\nThis uses the proton protocol engine library from the Apache Qpid project. The\ndriver has its own IO thread and event loop. The files under engine are really\ngeneric and could at some point be separated into their own library or folded\ninto the engine library itself somehow. The key driver interface is implemented\nin driver.py, tmost of the logic of mapping to AMQP 1.0 is in controller.py and\nthe eventloop.py is also fairly generic.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9ac3a759b9818d138ca68ba14b830aa9f16ccb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9ac3a759b9818d138ca68ba14b830aa9f16ccb0"}]},"branch":"refs/heads/master"},"8317afce139d9fafdaf7cd7a0feae2461334cdb1":{"kind":"REWORK","_number":8,"created":"2014-06-24 01:58:52.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/8"}}},"commit":{"parents":[{"commit":"dc5cb4a19271c29efe520d0ecaa2f72feeafef6c","subject":"Merge \"Ensure routing key is specified in the address for a direct producer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dc5cb4a19271c29efe520d0ecaa2f72feeafef6c"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-06-24 01:57:57.000000000","tz":-240},"subject":"An initial implementation of a 1.0 based messaging driver","message":"An initial implementation of a 1.0 based messaging driver\n\nThis uses the proton protocol engine library from the Apache Qpid project. The\ndriver has its own IO thread and event loop. The files under engine are really\ngeneric and could at some point be separated into their own library or folded\ninto the engine library itself somehow. The key driver interface is implemented\nin driver.py, tmost of the logic of mapping to AMQP 1.0 is in controller.py and\nthe eventloop.py is also fairly generic.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8317afce139d9fafdaf7cd7a0feae2461334cdb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8317afce139d9fafdaf7cd7a0feae2461334cdb1"}]},"branch":"refs/heads/master"},"37e1c24925f5628fc78c9657b43a902d092cd500":{"kind":"REWORK","_number":9,"created":"2014-06-26 20:00:09.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/9"}}},"commit":{"parents":[{"commit":"dc5cb4a19271c29efe520d0ecaa2f72feeafef6c","subject":"Merge \"Ensure routing key is specified in the address for a direct producer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dc5cb4a19271c29efe520d0ecaa2f72feeafef6c"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-06-26 19:57:06.000000000","tz":-240},"subject":"An initial implementation of a 1.0 based messaging driver","message":"An initial implementation of a 1.0 based messaging driver\n\nThis uses the proton protocol engine library from the Apache Qpid project. The\ndriver has its own IO thread and event loop. The files under engine are really\ngeneric and could at some point be separated into their own library or folded\ninto the engine library itself somehow. The key driver interface is implemented\nin driver.py, tmost of the logic of mapping to AMQP 1.0 is in controller.py and\nthe eventloop.py is also fairly generic.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/37e1c24925f5628fc78c9657b43a902d092cd500"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/37e1c24925f5628fc78c9657b43a902d092cd500"}]},"branch":"refs/heads/master"},"e5b959de6a5743ce12c84489ccb97bcd9d0679c0":{"kind":"REWORK","_number":10,"created":"2014-06-30 01:51:54.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/10"}}},"commit":{"parents":[{"commit":"dc5cb4a19271c29efe520d0ecaa2f72feeafef6c","subject":"Merge \"Ensure routing key is specified in the address for a direct producer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dc5cb4a19271c29efe520d0ecaa2f72feeafef6c"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-06-30 01:51:23.000000000","tz":-240},"subject":"An initial implementation of a 1.0 based messaging driver","message":"An initial implementation of a 1.0 based messaging driver\n\nThis uses the proton protocol engine library from the Apache Qpid project. The\ndriver has its own IO thread and event loop. The files under engine are really\ngeneric and could at some point be separated into their own library or folded\ninto the engine library itself somehow. The key driver interface is implemented\nin driver.py, tmost of the logic of mapping to AMQP 1.0 is in controller.py and\nthe eventloop.py is also fairly generic.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e5b959de6a5743ce12c84489ccb97bcd9d0679c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e5b959de6a5743ce12c84489ccb97bcd9d0679c0"}]},"branch":"refs/heads/master"},"1a9c63c8181a7968d957b039c1c69ce7f2ee3a1a":{"kind":"REWORK","_number":11,"created":"2014-07-02 20:58:05.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/11"}}},"commit":{"parents":[{"commit":"270e9cb00c634d4e232ecb518c96395e0c115cf3","subject":"Merge \"Remove duplicate docs for MessageHandlingServer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/270e9cb00c634d4e232ecb518c96395e0c115cf3"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-07-02 20:56:44.000000000","tz":-240},"subject":"An initial implementation of a 1.0 based messaging driver","message":"An initial implementation of a 1.0 based messaging driver\n\nThis driver uses the Pyngus module, which is a client API built on the\nProton AMQP 1.0 protocol engine library from the Apache Qpid project.\n\nThe key driver interfaces are implemented in the ProtonDriver class in\ndriver.py.  The logic for interfacing with Pyngus in order to\nsend/receive messages, manage AMQP connections and links, and handle\nprotocol events is in controller.py.  eventloop.py is a fairly generic\nsocket connection and I/O processor which runs in its own thread.\ncontroller.py uses the eventloop.py thread to schedule subscription\nand message send requests from the driver, as well as handle all\nprotocol event callbacks coming from Pyngus.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1a9c63c8181a7968d957b039c1c69ce7f2ee3a1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1a9c63c8181a7968d957b039c1c69ce7f2ee3a1a"}]},"branch":"refs/heads/master"},"d12560b69018a27eb91a478f6b0956e51bfcdcf6":{"kind":"REWORK","_number":12,"created":"2014-07-25 14:50:54.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/12"}}},"commit":{"parents":[{"commit":"5be1b6a6a93e2be1fd6adef109a016f49ed98b0a","subject":"Enabled hacking checks H305 and H307","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5be1b6a6a93e2be1fd6adef109a016f49ed98b0a"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-07-25 14:34:23.000000000","tz":-240},"subject":"An initial implementation of a 1.0 based messaging driver","message":"An initial implementation of a 1.0 based messaging driver\n\nThis driver uses the Pyngus module, which is a client API built on the\nProton AMQP 1.0 protocol engine library from the Apache Qpid project.\n\nThe key driver interfaces are implemented in the ProtonDriver class in\ndriver.py.  The logic for interfacing with Pyngus in order to\nsend/receive messages, manage AMQP connections and links, and handle\nprotocol events is in controller.py.  eventloop.py is a fairly generic\nsocket connection and I/O processor which runs in its own thread.\ncontroller.py uses the eventloop.py thread to schedule subscription\nand message send requests from the driver, as well as handle all\nprotocol event callbacks coming from Pyngus.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d12560b69018a27eb91a478f6b0956e51bfcdcf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d12560b69018a27eb91a478f6b0956e51bfcdcf6"}]},"branch":"refs/heads/master"},"db6c38636429f149c7438e951adfff65b9eaef87":{"kind":"REWORK","_number":13,"created":"2014-07-31 17:39:50.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/13"}}},"commit":{"parents":[{"commit":"63a6d6229b3a58af3056477e631480b7dec96024","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/63a6d6229b3a58af3056477e631480b7dec96024"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-07-31 17:27:51.000000000","tz":-240},"subject":"An initial implementation of a 1.0 based messaging driver","message":"An initial implementation of a 1.0 based messaging driver\n\nThis driver uses the Pyngus module, which is a client API built on the\nProton AMQP 1.0 protocol engine library from the Apache Qpid project.\n\nThe key driver interfaces are implemented in the ProtonDriver class in\ndriver.py.  The logic for interfacing with Pyngus in order to\nsend/receive messages, manage AMQP connections and links, and handle\nprotocol events is in controller.py.  eventloop.py is a fairly generic\nsocket connection and I/O processor which runs in its own thread.\ncontroller.py uses the eventloop.py thread to schedule subscription\nand message send requests from the driver, as well as handle all\nprotocol event callbacks coming from Pyngus.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/db6c38636429f149c7438e951adfff65b9eaef87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/db6c38636429f149c7438e951adfff65b9eaef87"}]},"branch":"refs/heads/master"},"3885f6d44878c5568de60dc524e9fec998753da5":{"kind":"REWORK","_number":14,"created":"2014-08-19 19:42:57.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/14"}}},"commit":{"parents":[{"commit":"1ee33509e7a6cfa2c9d6ae4f80cf0c3abdd4599b","subject":"Merge \"Fix error in example of an RPC server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1ee33509e7a6cfa2c9d6ae4f80cf0c3abdd4599b"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-08-19 19:34:51.000000000","tz":-240},"subject":"An initial implementation of an AMQP 1.0 based messaging driver","message":"An initial implementation of an AMQP 1.0 based messaging driver\n\nThis driver uses the Pyngus module, which is a client API built on the\nProton AMQP 1.0 protocol engine library from the Apache Qpid project.\n\nThe key driver interfaces are implemented in the ProtonDriver class in\ndriver.py.  The logic for interfacing with Pyngus in order to\nsend/receive messages, manage AMQP connections and links, and handle\nprotocol events is in controller.py.  eventloop.py is a fairly generic\nsocket connection and I/O processor which runs in its own thread.\ncontroller.py uses the eventloop.py thread to schedule subscription\nand message send requests from the driver, as well as handle all\nprotocol event callbacks coming from Pyngus.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3885f6d44878c5568de60dc524e9fec998753da5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3885f6d44878c5568de60dc524e9fec998753da5"}]},"branch":"refs/heads/master"},"c9d3f3f5676615f548d3a025abf70fa0812f5546":{"kind":"REWORK","_number":15,"created":"2014-08-27 12:46:45.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/15"}}},"commit":{"parents":[{"commit":"48a9ba4e27bcf237f8b9032e70d79a3e22a429aa","subject":"Fix Python 3 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/48a9ba4e27bcf237f8b9032e70d79a3e22a429aa"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-08-27 12:37:18.000000000","tz":-240},"subject":"An initial implementation of an AMQP 1.0 based messaging driver","message":"An initial implementation of an AMQP 1.0 based messaging driver\n\nThe key driver interfaces are implemented in the ProtonDriver class in\ndriver.py.  The logic for interfacing with Pyngus in order to\nsend/receive messages, manage AMQP connections and links, and handle\nprotocol events is in controller.py.  eventloop.py is a fairly generic\nsocket connection and I/O processor which runs in its own thread.\ncontroller.py uses the eventloop.py thread to schedule subscription\nand message send requests from the driver, as well as handle all\nprotocol event callbacks coming from Pyngus.\n\nIncluded in this patch are a set of functional tests that can be run\nunder tox (tox -eamqp1).  These tests fully exercise the new driver,\nfrom the driver API down to the \u0027wire\u0027 - nothing in the driver is\nmocked out.  The functional tests implement a simple loopback test\nbroker, which allows the driver to send and receive messages via the\nlocal network.  All RPC call patterns, RPC timeouts, and even broker\nfailover are verified by the included functional tests.\n\nThis driver uses the Pyngus module, which is a pure-python client API\nbuilt on the Proton AMQP 1.0 protocol engine library from the Apache\nQpid project.  Pyngus is available via pypi.python.org.\n\nThis driver introduces a dependency on the Proton AMQP 1.0 protocol\nlibrary, which is a platform-dependent library that must be installed\nin order to use this driver and run the functional tests.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c9d3f3f5676615f548d3a025abf70fa0812f5546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c9d3f3f5676615f548d3a025abf70fa0812f5546"}]},"branch":"refs/heads/master"},"802cf2fe97a0b9f6c5a88e9cc50952ac86efca86":{"kind":"REWORK","_number":16,"created":"2014-08-28 20:42:43.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/16"}}},"commit":{"parents":[{"commit":"48a9ba4e27bcf237f8b9032e70d79a3e22a429aa","subject":"Fix Python 3 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/48a9ba4e27bcf237f8b9032e70d79a3e22a429aa"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-08-28 20:42:33.000000000","tz":-240},"subject":"An initial implementation of an AMQP 1.0 based messaging driver","message":"An initial implementation of an AMQP 1.0 based messaging driver\n\nThe key driver interfaces are implemented in the ProtonDriver class in\ndriver.py.  The logic for interfacing with Pyngus in order to\nsend/receive messages, manage AMQP connections and links, and handle\nprotocol events is in controller.py.  eventloop.py is a fairly generic\nsocket connection and I/O processor which runs in its own thread.\ncontroller.py uses the eventloop.py thread to schedule subscription\nand message send requests from the driver, as well as handle all\nprotocol event callbacks coming from Pyngus.\n\nIncluded in this patch are a set of functional tests that can be run\nunder tox (tox -eamqp1).  These tests fully exercise the new driver,\nfrom the driver API down to the \u0027wire\u0027 - nothing in the driver is\nmocked out.  The functional tests implement a simple loopback test\nbroker, which allows the driver to send and receive messages via the\nlocal network.  All RPC call patterns, RPC timeouts, and even broker\nfailover are verified by the included functional tests.\n\nThis driver uses the Pyngus module, which is a pure-python client API\nbuilt on the Proton AMQP 1.0 protocol engine library from the Apache\nQpid project.  Pyngus is available via pypi.python.org.\n\nThis driver introduces a dependency on the Proton AMQP 1.0 protocol\nlibrary, which is a platform-dependent library that must be installed\nin order to use this driver and run the functional tests.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/802cf2fe97a0b9f6c5a88e9cc50952ac86efca86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/802cf2fe97a0b9f6c5a88e9cc50952ac86efca86"}]},"branch":"refs/heads/master"},"e3b7bb52a0847505fe75b25f71c982ab43af44fe":{"kind":"REWORK","_number":17,"created":"2014-08-29 21:49:24.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/17"}}},"commit":{"parents":[{"commit":"48a9ba4e27bcf237f8b9032e70d79a3e22a429aa","subject":"Fix Python 3 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/48a9ba4e27bcf237f8b9032e70d79a3e22a429aa"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-08-29 21:49:05.000000000","tz":-240},"subject":"An initial implementation of an AMQP 1.0 based messaging driver","message":"An initial implementation of an AMQP 1.0 based messaging driver\n\nThe key driver interfaces are implemented in the ProtonDriver class in\ndriver.py.  The logic for interfacing with Pyngus in order to\nsend/receive messages, manage AMQP connections and links, and handle\nprotocol events is in controller.py.  eventloop.py is a fairly generic\nsocket connection and I/O processor which runs in its own thread.\ncontroller.py uses the eventloop.py thread to schedule subscription\nand message send requests from the driver, as well as handle all\nprotocol event callbacks coming from Pyngus.\n\nIncluded in this patch are a set of functional tests that can be run\nunder tox (tox -eamqp1).  These tests fully exercise the new driver,\nfrom the driver API down to the \u0027wire\u0027 - nothing in the driver is\nmocked out.  The functional tests implement a simple loopback test\nbroker, which allows the driver to send and receive messages via the\nlocal network.  All RPC call patterns, RPC timeouts, and even broker\nfailover are verified by the included functional tests.\n\nThis driver uses the Pyngus module, which is a pure-python client API\nbuilt on the Proton AMQP 1.0 protocol engine library from the Apache\nQpid project.  Pyngus is available via pypi.python.org.\n\nThis driver introduces a dependency on the Proton AMQP 1.0 protocol\nlibrary, which is a platform-dependent library that must be installed\nin order to use this driver and run the functional tests.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e3b7bb52a0847505fe75b25f71c982ab43af44fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e3b7bb52a0847505fe75b25f71c982ab43af44fe"}]},"branch":"refs/heads/master"},"fd29df5193ecd74cb6223c4699d6003cfaf2347c":{"kind":"REWORK","_number":18,"created":"2014-09-02 16:21:15.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/18"}}},"commit":{"parents":[{"commit":"48a9ba4e27bcf237f8b9032e70d79a3e22a429aa","subject":"Fix Python 3 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/48a9ba4e27bcf237f8b9032e70d79a3e22a429aa"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-09-02 16:21:03.000000000","tz":-240},"subject":"An initial implementation of an AMQP 1.0 based messaging driver","message":"An initial implementation of an AMQP 1.0 based messaging driver\n\nThe key driver interfaces are implemented in the ProtonDriver class in\ndriver.py.  The logic for interfacing with Pyngus in order to\nsend/receive messages, manage AMQP connections and links, and handle\nprotocol events is in controller.py.  eventloop.py is a fairly generic\nsocket connection and I/O processor which runs in its own thread.\ncontroller.py uses the eventloop.py thread to schedule subscription\nand message send requests from the driver, as well as handle all\nprotocol event callbacks coming from Pyngus.\n\nIncluded in this patch are a set of functional tests that can be run\nunder tox (tox -eamqp1).  These tests fully exercise the new driver,\nfrom the driver API down to the \u0027wire\u0027 - nothing in the driver is\nmocked out.  The functional tests implement a simple loopback test\nbroker, which allows the driver to send and receive messages via the\nlocal network.  All RPC call patterns, RPC timeouts, and even broker\nfailover are verified by the included functional tests.\n\nThis driver uses the Pyngus module, which is a pure-python client API\nbuilt on the Proton AMQP 1.0 protocol engine library from the Apache\nQpid project.  Pyngus is available via pypi.python.org.\n\nThis driver introduces a dependency on the Proton AMQP 1.0 protocol\nlibrary, which is a platform-dependent library that must be installed\nin order to use this driver and run the functional tests.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fd29df5193ecd74cb6223c4699d6003cfaf2347c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fd29df5193ecd74cb6223c4699d6003cfaf2347c"}]},"branch":"refs/heads/master"},"55b22631d90859f84e38288c472452a50d35b0d2":{"kind":"REWORK","_number":19,"created":"2014-09-03 12:40:23.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/19"}}},"commit":{"parents":[{"commit":"48a9ba4e27bcf237f8b9032e70d79a3e22a429aa","subject":"Fix Python 3 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/48a9ba4e27bcf237f8b9032e70d79a3e22a429aa"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-09-03 12:40:11.000000000","tz":-240},"subject":"An initial implementation of an AMQP 1.0 based messaging driver","message":"An initial implementation of an AMQP 1.0 based messaging driver\n\nThe key driver interfaces are implemented in the ProtonDriver class in\ndriver.py.  The logic for interfacing with Pyngus in order to\nsend/receive messages, manage AMQP connections and links, and handle\nprotocol events is in controller.py.  eventloop.py is a fairly generic\nsocket connection and I/O processor which runs in its own thread.\ncontroller.py uses the eventloop.py thread to schedule subscription\nand message send requests from the driver, as well as handle all\nprotocol event callbacks coming from Pyngus.\n\nIncluded in this patch are a set of functional tests that can be run\nunder tox (tox -eamqp1).  These tests fully exercise the new driver,\nfrom the driver API down to the \u0027wire\u0027 - nothing in the driver is\nmocked out.  The functional tests implement a simple loopback test\nbroker, which allows the driver to send and receive messages via the\nlocal network.  All RPC call patterns, RPC timeouts, and even broker\nfailover are verified by the included functional tests.\n\nThis driver uses the Pyngus module, which is a pure-python client API\nbuilt on the Proton AMQP 1.0 protocol engine library from the Apache\nQpid project.  Pyngus is available via pypi.python.org.\n\nThis driver introduces a dependency on the Proton AMQP 1.0 protocol\nlibrary, which is a platform-dependent library that must be installed\nin order to use this driver and run the functional tests.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/55b22631d90859f84e38288c472452a50d35b0d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/55b22631d90859f84e38288c472452a50d35b0d2"}]},"branch":"refs/heads/master"},"fbee94170f918c165ac4e7eb3bacd0dc193acc50":{"kind":"REWORK","_number":20,"created":"2014-09-04 19:15:40.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/15/75815/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/15/75815/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/15/75815/20"}}},"commit":{"parents":[{"commit":"a9ec73f38db804858c10267ca5da721232c4dd70","subject":"Switch to oslo.utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9ec73f38db804858c10267ca5da721232c4dd70"}]}],"author":{"name":"Gordon Sim","email":"gsim@redhat.com","date":"2014-02-21 21:13:17.000000000","tz":0},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2014-09-04 19:12:43.000000000","tz":-240},"subject":"An initial implementation of an AMQP 1.0 based messaging driver","message":"An initial implementation of an AMQP 1.0 based messaging driver\n\nThe key driver interfaces are implemented in the ProtonDriver class in\ndriver.py.  The logic for interfacing with Pyngus in order to\nsend/receive messages, manage AMQP connections and links, and handle\nprotocol events is in controller.py.  eventloop.py is a fairly generic\nsocket connection and I/O processor which runs in its own thread.\ncontroller.py uses the eventloop.py thread to schedule subscription\nand message send requests from the driver, as well as handle all\nprotocol event callbacks coming from Pyngus.\n\nIncluded in this patch are a set of functional tests that can be run\nunder tox (tox -eamqp1).  These tests fully exercise the new driver,\nfrom the driver API down to the \u0027wire\u0027 - nothing in the driver is\nmocked out.  The functional tests implement a simple loopback test\nbroker, which allows the driver to send and receive messages via the\nlocal network.  All RPC call patterns, RPC timeouts, and even broker\nfailover are verified by the included functional tests.\n\nThis driver uses the Pyngus module, which is a pure-python client API\nbuilt on the Proton AMQP 1.0 protocol engine library from the Apache\nQpid project.  Pyngus is available via pypi.python.org.\n\nThis driver introduces a dependency on the Proton AMQP 1.0 protocol\nlibrary, which is a platform-dependent library that must be installed\nin order to use this driver and run the functional tests.\n\nChange-Id: I871703e4cdc04cee3e6c214e911c9df464ede2ed\nImplements: blueprint amqp10-driver-implementation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fbee94170f918c165ac4e7eb3bacd0dc193acc50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fbee94170f918c165ac4e7eb3bacd0dc193acc50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
