)]}'
{"id":"openstack%2Foslo.messaging~758531","triplet_id":"openstack%2Foslo.messaging~master~I9d671abcd2cea9c0c726edaddcd65e1093d96731","project":"openstack/oslo.messaging","branch":"master","hashtags":[],"change_id":"I9d671abcd2cea9c0c726edaddcd65e1093d96731","subject":"Python 3.9: isAlive is removed","status":"MERGED","created":"2020-10-16 07:45:39.000000000","updated":"2020-10-19 15:00:42.000000000","submitted":"2020-10-19 14:58:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758531-1603119538518-1d1e633c","meta_rev_id":"2529c4c525e17322ad20f7ff6b2a855c6d068d65","_number":758531,"virtual_id_number":758531,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 14:58:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-16 08:29:37.000000000","_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-16 12:57:57.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-10-19 13:25:04.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2020-10-16 08:33:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-10-19 13:25:04.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 08:29:37.000000000","updated_by":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"reviewer":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"state":"REVIEWER"},{"updated":"2020-10-16 08:33:51.000000000","updated_by":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"reviewer":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"state":"REVIEWER"},{"updated":"2020-10-16 12:57:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-19 13:25:04.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-10-19 14:58:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91b898c4234d3e85179fb957e4a21b71e5a17f0d","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-10-16 07:45:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5996bfe0bc2bdebd506fbcd8239dc82eb6128d01","author":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"date":"2020-10-16 08:29:37.000000000","message":"Patch Set 1:\n\n+1","accounts_in_message":[],"_revision_number":1},{"id":"11889eb33c2c844e56d8328131f64b3e6af57220","author":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"date":"2020-10-16 08:33:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"021e25fff87345d5bba59327f60f7d704bdc160a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 09:23:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9a04abc61f64d95a734ab6d1e4402a9 : SUCCESS in 1h 15m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebd7defbd33e411882a2b5d570fd85d0 : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8da60068f0b45e48458ec20b77fc079 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7129165f8174dcf87af43d05ba0476f : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efe5b351a40847aa92abfbe8f27cbd12 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bff65656dad4b2496ece573493443ab : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0496be8ea5e74f6fa21629cc24697aac : SUCCESS in 6m 39s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/cc0d58deafc14f11833421c7125fed1c : SUCCESS in 7m 19s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/c7376a70a67a4dcda300858e7e6cfb6e : FAILURE in 9m 33s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/24c4414676594dada88d567c6bc24cb8 : SUCCESS in 6m 43s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/a668274078234138a2ac1fbbd68a95ea : FAILURE in 10m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/84f6096f784d4ab5822b8bfbf9a7b08d : SUCCESS in 1h 30m 18s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/5fc3ad935c6f48319027a2feba3eeae0 : FAILURE in 1h 17m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/18d2c143018d4ae1a0369f5d0a9d0346 : SUCCESS in 1h 30m 02s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/c64a70a853ff48e8bc389c518582e3b8 : SUCCESS in 59m 13s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8cc22105b594334985a1c6622d6195e : SUCCESS in 1h 06m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"018cb413ce04bfcd527cee4d323cde061d6bd3eb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-16 12:57:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"755b4f8a677f23a3761dd57a24ecde80a7a52a91","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-10-19 13:25:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"38a737e2131fab76bb565e7bb99c3d3ec322e9d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 13:25:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7bd7bfa5dd3ce02dbbc50d72d00d7335a0b63cbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 14:58:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ed4884826c46877c4197dbcd1b1659b9f579ac76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 14:58:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b94c1c89a5bd41a39df0623975ff06ab : SUCCESS in 1h 27m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/397daab41a3e440b9b4d1cc078083b55 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a6529cd1af84092a8b17bc2d640d12b : SUCCESS in 10m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4429b0c15d34d5ba1802bb8efa920aa : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4731587a1fc14c96b4b6ebe0cf61d646 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de84f717be6543e6b1cc6219f5f371ae : SUCCESS in 8m 31s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/80a7f8a8ab75455f9cf2fbeeebbad685 : SUCCESS in 8m 38s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b654a2660357462ebac679a54434cdd3 : SUCCESS in 1h 22m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2529c4c525e17322ad20f7ff6b2a855c6d068d65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 15:00:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab215209e8b4f62bfa867076cd65f07 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c4203bbdc463010d756a71bbd06720867122bcb","revisions":{"0c4203bbdc463010d756a71bbd06720867122bcb":{"kind":"REWORK","_number":1,"created":"2020-10-16 07:45:39.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/31/758531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/31/758531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/758531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/758531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/758531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/31/758531/1"}}},"commit":{"parents":[{"commit":"2cc35f6b1ac95f2c472658a18131dc292f89f30e","subject":"Merge \"Run rabbitmq heartbeat in python thread by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2cc35f6b1ac95f2c472658a18131dc292f89f30e"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-10-16 07:41:44.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-10-16 07:45:36.000000000","tz":120},"subject":"Python 3.9: isAlive is removed","message":"Python 3.9: isAlive is removed\n\nThe Threading method isAlive was deprecated a long time ago, in the favor\nof is_alive(). Though in Python 3.9, isAlive is removed. We must switch\nto it. Note that is_alive() is available at least in Python 3.5 (I tried)\nand probably even earlier, so switching to is_alive() is not a problem\nfor the Python interpreter versions currently supported by OpenStack.\n\nChange-Id: I9d671abcd2cea9c0c726edaddcd65e1093d96731\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0c4203bbdc463010d756a71bbd06720867122bcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0c4203bbdc463010d756a71bbd06720867122bcb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
