)]}'
{"id":"openstack%2Foslo.messaging~758793","triplet_id":"openstack%2Foslo.messaging~stable%2Fussuri~I2ad4b9f1d012c9f0586a932ac27d96da1bcc4e4c","project":"openstack/oslo.messaging","branch":"stable/ussuri","topic":"ussuri","hashtags":[],"change_id":"I2ad4b9f1d012c9f0586a932ac27d96da1bcc4e4c","subject":"add min of 1 to rpc_conn_pool_size","status":"MERGED","created":"2020-10-19 11:03:45.000000000","updated":"2020-10-22 12:16:57.000000000","submitted":"2020-10-22 12:15:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758793-1603368925505-2ee07cca","meta_rev_id":"d353d88da6d84c9ab8a36159e99e89d7d6dd0e64","_number":758793,"virtual_id_number":758793,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-22 12:15:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-21 07:51:11.000000000","_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2020-10-19 13:47:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2020-10-22 10:03:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"date":"2020-10-19 11:03:45.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2020-10-19 11:18:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-19 11:03:45.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-10-22 10:03:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-22 12:15:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89e2c684df62b91225e8e8ddc3b32e3857923f1e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-19 11:03:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8670ce62764d7574694ba51f9fead604a3fbe39","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-19 11:18:14.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+1\n\nClean backport, waiting for CI (only +1 for now), else LGTM for now.\n\nAs Sean noticed fstring should be updated for train as they was introduced with python 3.6 and train still support py35 and py27.","accounts_in_message":[],"_revision_number":1},{"id":"d926b6b87772dfbf3699965a1fee9bb5c2f3d0db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 12:47:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b0fb1483a1d4e9cbcc0fe347b987c93 : RETRY_LIMIT in 20m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ebd9bbbf06f45398a145f4181c0e5bc : SUCCESS in 6m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e34836c7ed7c4fc6ab6b4d5d04b74883 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7578ccbda38e45ce8cbe6ee8a6a55e37 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/103a502d8e334660b0bdf1119097e1ed : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b8e3e7e94fc482aab9cb5113b83fef8 : SUCCESS in 9m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce8aad4e9aa4862a36d802e2a9a5729 : SUCCESS in 7m 11s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/5615386903df49628048bdc710d088fb : SUCCESS in 8m 07s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/0ef07c2ba7ca4c33b76b8545a442f618 : SUCCESS in 9m 07s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/107d8c5b58344a50be1be3bfd78cd7af : SUCCESS in 7m 50s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/90a6e850737345c3a0227ba6e387991b : FAILURE in 11m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/70353091e18c461ead90dae71d19b7cc : SUCCESS in 1h 23m 19s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/9efaebeac96340beb690fc5d4458baf1 : SUCCESS in 1h 36m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2e704b2dc8cd46359c1726753b16152c : SUCCESS in 1h 24m 43s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/43d296090fe745d8baf5c8a43b3107f1 : SUCCESS in 1h 14m 47s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/f4c0e6b0b85f4bfc8c662f1610c85305 : FAILURE in 26m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ecc2c118f0d21e78c925933d187500231d4aa2d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-19 13:47:15.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"96c4b77356baeb882083cfcbbc363c0f33a2da02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 15:56:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eeb595cc9b34403184ddfd5f40f49f85 : FAILURE in 1h 55m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de172d90873a4a7d8b0cd4cf179d12ce : SUCCESS in 7m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa5ec97040a54f3f8b1f86695275f1dc : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/940e76a1076a40dc91827233878b6ce4 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc2b2a9999e34e44b008af3edc18f3e9 : SUCCESS in 7m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7af9025f30364174a8d2d867cdba068e : SUCCESS in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6293cb1ba7d044299fc46732b571848c : SUCCESS in 6m 58s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/79ef282655194dbe8a0a7b2b456dd6f9 : SUCCESS in 8m 34s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/e406f6c456984befad3913f50362c731 : SUCCESS in 9m 12s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/78aa48440c0f4befab4269678a783902 : SUCCESS in 13m 56s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/f3cebdb5854249f6aabed8222ac18b84 : SUCCESS in 8m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d30536357bd8429bac6134d044eb9255 : SUCCESS in 1h 35m 44s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7444bd3da8444749b1e18e54434b5909 : SUCCESS in 1h 28m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/62e09679e24f46efac342c69dafdd870 : SUCCESS in 1h 33m 35s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/41c5c2126f9a4dcfa2842203a808b63d : RETRY_LIMIT in 22m 19s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/67bb514ae5db4598a5c9328e770cf208 : FAILURE in 30m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4f625125158b9ee14b72104764b5a1343a7ba3a","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-21 07:51:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bfb7d15a72b030cc20cc3bbb866a4519aa749753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 09:46:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5819b3879f274c96956734d9652878bf : SUCCESS in 1h 18m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f46691c7da704586804e5c97ecdc65d5 : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/377286c05b6d462ca5634ce23ae9b302 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e78c9e3956d04384b6bc460088bb0030 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60e778ba05894088837e4f8ade567823 : SUCCESS in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40c5c9cea8e74d4c902608556fac0851 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e8f5737e6634facb9249eecbd09df3b : SUCCESS in 7m 11s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/70cf9502fa1a4122a4120d98982a9bd2 : SUCCESS in 6m 51s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/cce0988591564ae890e20a13ceb92d50 : SUCCESS in 9m 01s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/ce5d3f1321494c21aa8921bfc74d3c47 : SUCCESS in 8m 01s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/a747c39b732e4d4d931bd542120b89be : SUCCESS in 9m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/446e2c4dc6ef4632bc406e34f6e2d4d6 : SUCCESS in 1h 21m 38s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/64919dcd95dc4bf5bb360c2d6e17f8fe : SUCCESS in 1h 44m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/798a2aa2e99b41b1ba924d87e31b8ee8 : RETRY_LIMIT in 21m 22s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d5fbc92aa134436a8405c18bedc252a6 : SUCCESS in 1h 18m 12s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/75db92338c4143cd9acb8e70df935982 : FAILURE in 31m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c81f19c173d60ad94f6489daf1675e8dc3464de","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-22 10:03:02.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"010dcae58ca7c378e6207d714ffb9d1cccc7f52a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 10:03:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d146ddd8418d1cfbc96b9be2b5071eb1c3d612ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 12:15:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a97fb387c3df78f2d499bd3bcd05ffe0c6ec112e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 12:15:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a6961fba73e4ec7975bd8c6ccc4a1c2 : SUCCESS in 1h 26m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eed296bef4bb438f9644a2d21735ed4d : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e96a418a8e8a445ca2805e6107886709 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ddd024a85e849d3b8477134ed81f8c0 : SUCCESS in 8m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8790e01fb924d35a6190ca92fecd57c : SUCCESS in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f77f802a3cf49d882ba603422a304dc : SUCCESS in 8m 30s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/5a5609d2fb0d45f38e84aa782989096d : SUCCESS in 9m 22s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8731c031cd714fd3ae3e21200ed7f949 : SUCCESS in 1h 23m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c16106dfeda70c9a796c139b652ab2be3f5c99dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 12:16:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a84c981ec5949cf89a39ff2e92d651c : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c9b070f5bc07692c837f9915eb0d8525065d0f6","revisions":{"7c9b070f5bc07692c837f9915eb0d8525065d0f6":{"kind":"REWORK","_number":1,"created":"2020-10-19 11:03:45.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/93/758793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/93/758793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/758793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/758793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/758793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/93/758793/1"}}},"commit":{"parents":[{"commit":"0a432c7fb107d04f7a41199fe9a8c4fbd344d009","subject":"Cancel consumer if queue down","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0a432c7fb107d04f7a41199fe9a8c4fbd344d009"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-10-12 21:43:31.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-10-19 11:09:54.000000000","tz":120},"subject":"add min of 1 to rpc_conn_pool_size","message":"add min of 1 to rpc_conn_pool_size\n\nThis change add a min value of 1 to\n[oslo_messaging_rabbit]/rpc_conn_pool_size\nsuch that there is always at least 1 connection avaiable.\n\nThis change add a runtime check to ensure that\n[oslo_messaging_rabbit]/rpc_conn_pool_size is greater than\nor equal too [oslo_messaging_rabbit]/conn_pool_min_size\n\nChange-Id: I2ad4b9f1d012c9f0586a932ac27d96da1bcc4e4c\nCloses-Bug: #1899533\n(cherry picked from commit 00d15eaeaba0ded0330cdcec7b19eee3adbfb1e1)\n(cherry picked from commit 8f5847dcbe5144272ff7295fce2b50a9f544e5b0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7c9b070f5bc07692c837f9915eb0d8525065d0f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7c9b070f5bc07692c837f9915eb0d8525065d0f6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
