)]}'
{"id":"openstack%2Foslo.messaging~761307","triplet_id":"openstack%2Foslo.messaging~master~I2c517401aa1ae464e6563800ecb0a459076655b3","project":"openstack/oslo.messaging","branch":"master","topic":"tox-python38","hashtags":[],"change_id":"I2c517401aa1ae464e6563800ecb0a459076655b3","subject":"Use py3 as the default runtime for tox","status":"MERGED","created":"2020-11-04 09:02:53.000000000","updated":"2021-01-18 16:28:12.000000000","submitted":"2021-01-18 16:20:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761307-tox-python38","meta_rev_id":"beace4dfcebd065266449131ebeee705e36ff2b9","_number":761307,"virtual_id_number":761307,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-18 16:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-12-11 15:42:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-12-11 15:42:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-04 10:52:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 15:42:31.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"ff1b69df8615c94f269b234c54f967da264abbc7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-04 09:02:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29499dfec795236d7ace930c01967434af67e0d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 10:52:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d7127f661344474ba22748a2276c44a : SUCCESS in 1h 20m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22e67bb7e87c446b88e3a638c9748e14 : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96435b53ad604a89a9654e545df8f045 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f10a76a9fcc4aa3b9d28c3064514cac : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc6d957fae2140199e0f53f94cd79c48 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc9c466fa5024aefa6b00ba337e3da11 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/114237113f194086bc6db932361d3361 : FAILURE in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/626a3a59b3b44379ad090ef38b136091 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cd2f99183c04a0eae2071e2354a92bc : SUCCESS in 5m 12s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/1bccfdffbcb6440ab65bc27a00691529 : SUCCESS in 6m 05s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/0cba374d726947c499827db131f854c5 : FAILURE in 6m 32s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/f025579a1a954ad1b969f5837762e2d9 : SUCCESS in 6m 14s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/5aef8ca6489642f4be0e47df60a14325 : FAILURE in 5m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b14d5581a09f4d46b76c4d8affebbc52 : SUCCESS in 1h 37m 20s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4160d31655f44dfd8fd3acb2c735fa90 : SUCCESS in 1h 18m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b7d61bc638bb4e15b62dafd7e8ce7bed : SUCCESS in 1h 36m 40s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b7920ac65c6549ab8ed9595fd86e8ff4 : SUCCESS in 52m 32s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/fcc3d7ae89a841d6b9ad56dc93dc8ef5 : SUCCESS in 1h 08m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a2b5c8855031a4a7b64a6da50cd3bf213cca7f0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-11 15:42:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4819d6fa970a611fa6c9614659cc9d844d86eff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 15:42:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1cdcc8fb937743c1e9e6c3cf8d7a2a6754b31c74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 17:36:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/018318d6c537415598556af6b6f1ee7a : SUCCESS in 1h 32m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36633d5b799e4a31829a7dd2e07ef167 : FAILURE in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acb16d190c744f8eb2338f427447c8aa : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05bd44ac856948d781c16f68571748d6 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca14c2d9bc4e4a40bb490195925791a4 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c83089972c6f41f79306575be0d41a31 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6780e4e1c13a46a1ad0eb134f6ef85f7 : SUCCESS in 5m 58s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/75ab3012bac345418f2e475a9237f2ba : SUCCESS in 9m 52s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/51db9afb98d44fc7a164017fdd9f05cf : SUCCESS in 1h 47m 40s","accounts_in_message":[],"_revision_number":1},{"id":"bf207f51c173d6f4f0013bfe5acef8f21e0dac5a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-08 10:25:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"10f70233ec4d17de1b4b7f23d73107df23c065ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 13:00:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0366a1b5ced444c9b332951f8849d9e5 : TIMED_OUT in 2h 05m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57266ec424134d13912bf8ba4dce70cb : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5e7d4715c2a45329a9c08c86bb4993e : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d578a066c5cf43ee9b957fefb2bbdf9d : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d46e67c0e0044ac925169236b9aa35d : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4aeefa0fb36c4c7d9305d136f684f008 : FAILURE in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bac294bf68b4a98b3c27a76905c42bc : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/579f7746c94a434981a1e067800a01d0 : SUCCESS in 3m 36s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/8006bb58ff054edd9a6d1dabbc40ff15 : SUCCESS in 5m 07s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/0bc678b871074274b43d5b2d27fcd6b8 : FAILURE in 6m 42s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/13f3948ddaed4e9cb1cfab656dabae9b : SUCCESS in 5m 20s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/73ef70991cc64e5d8cd6196c0d0f9dd6 : FAILURE in 5m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/102a6dcf3d3e4215bb652202223f56bf : SUCCESS in 1h 27m 29s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/852de2ac43cc4e368920f4205f865576 : SUCCESS in 1h 42m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1897308fe7684a17b76b1363a4b434e2 : SUCCESS in 1h 32m 12s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0c883338ff0c48f7842be027a2a6a072 : SUCCESS in 1h 16m 10s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c3adfbf509040d9b3859c7d556cb971 : SUCCESS in 1h 23m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5fbfc485b007dab21a253521e779273a588e00c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-08 14:55:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c319f5948374f46cf2fdce1305446460e6b28d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 17:02:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2009f7de6d1f4e11b430dd50a264f1b2 : SUCCESS in 1h 24m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6297e374cee74ee2a05fb84268664497 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/228e4474f8e14a21baecf562348bc649 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86c393755dab49bf98c6ed4f8b95013c : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fb212cc10c24f849be35cafb9a2a60b : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10532b678cdd4914a500915786e07aa1 : FAILURE in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64d5aed1f16d4a17b9f46bfd463fa131 : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dd6a0d2fae5442482287eaf83a73980 : SUCCESS in 3m 17s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/31e31c900bc94fbb98929ad77b26234c : SUCCESS in 5m 24s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/30c4c5e27082427a9de71cd730808327 : FAILURE in 6m 32s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/e22d1293cc2c4135b99fde42bbebd092 : SUCCESS in 5m 17s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/76bf1f383e284d7da694bdb084c8342d : FAILURE in 6m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a7b955eb1d814e50b3cad9e3718df34e : SUCCESS in 1h 29m 01s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a10551241a294c42af7f008433410957 : SUCCESS in 1h 33m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b75f9380ea3c43b78c2937af52a79a69 : SUCCESS in 1h 42m 27s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/29f6f8c51d4540c8932428cc11f25cf6 : SUCCESS in 48m 09s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/855d8a6b5e5f47a2b2b8f2afd0764b86 : SUCCESS in 1h 08m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3b17ff9b17e73738cedcd7b2bbb95b85738df92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 17:02:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e31694dd6ec7fbf163487097ef44988b321d0726","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 18:35:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15efddd26a9c48feae1a72bf967d0472 : SUCCESS in 1h 21m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408480c44f344ae2a0c1574d8f4112fd : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8a819676284492c926922037967155d : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1af67b54c6b9436a9163770fb890b0f9 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d259dd1633574df9b8bbde6a79f3afd2 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/168285d991ab4c01bea1bb642eddb3f3 : SUCCESS in 4m 30s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/2d2dbb591bc446b8908b9edb6d6298a4 : SUCCESS in 5m 31s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/fa31fcfe83624f4ab57189eba294bb9f : FAILURE in 1h 30m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c496e49080c0692b583ae5862cb17fb23e2855a4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-18 13:01:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"570b969cc5ba2ced185ff22667cc2b7320513581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 14:42:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b2409de58bf4492b132e341e7b3fb91 : SUCCESS in 1h 28m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a418aae34f614d74b46cf918d526d3aa : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b511f63ab344b19aa711c8d1e7fe2c7 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dfcbd97bc594b3c90d176406663425f : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27baa33ccdb94c47b8ec8e9ab72cede4 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d6b541a503c4834a0dfb027943ce3ac : FAILURE in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a990656b1745658b3d44b1c47fc3c9 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b2432ed73fe4011b3cb6df0c9f0ad5d : SUCCESS in 4m 30s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/2df9e53d0b174f91b667b3f4989cc8c9 : SUCCESS in 5m 09s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/938306d1c74e469bad21138e1baf5ae8 : FAILURE in 6m 25s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/44be5bf53f194407b6701d5ac795abc8 : SUCCESS in 5m 36s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/0cbb161dc35d43b4b416e6d15ff3497a : FAILURE in 6m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bf70fa2463c24155a5ca1e96970dde27 : SUCCESS in 1h 24m 45s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e1fb517352e14feabe5b4a052776a9d6 : SUCCESS in 1h 30m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8a8c0c8b132b42ee9c840b26b267ed89 : SUCCESS in 1h 26m 52s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/587a57931bfa43a3ac21fc224edb7457 : SUCCESS in 1h 03m 31s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/84ea17f28d0b4f39b70e17e47e0ef32c : SUCCESS in 1h 09m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1746a33c1508e8c997809e722352d66cebf523b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 14:43:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e6655f0822fd7edcfd69bc2e2be0d60d79e06f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:20:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1fd6c275ef7f446ebb536855bed4243f : SUCCESS in 1h 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86935f6a203b46d4ae3e9faa8ea53290 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddbc302d4c4249eabf6364955da1e592 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/401464204bf94eb392ee4b6d79e0c7a3 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25034c9eea53407b85b598e378c677aa : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e19a20261f1644f7b1be5cc629e949c5 : SUCCESS in 4m 32s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/11c9b3b66baf4c7e905ed756771693f6 : SUCCESS in 5m 44s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/42b6139336924ac09046f165d7f2b161 : SUCCESS in 1h 27m 15s","accounts_in_message":[],"_revision_number":1},{"id":"18bb812968de09371d1fdf66ec26b3b18ffff7fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:20:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"beace4dfcebd065266449131ebeee705e36ff2b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:28:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b9a0d826f1544069ed9dff2afe70a1a : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/530e1e92a11c46b996a3ef91ac421428 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"672c00a67dd9a878dfefda148a9fbbaf911e4416","revisions":{"672c00a67dd9a878dfefda148a9fbbaf911e4416":{"kind":"REWORK","_number":1,"created":"2020-11-04 09:02:53.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/07/761307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/07/761307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/07/761307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/07/761307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/07/761307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/07/761307/1"}}},"commit":{"parents":[{"commit":"0c4203bbdc463010d756a71bbd06720867122bcb","subject":"Python 3.9: isAlive is removed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0c4203bbdc463010d756a71bbd06720867122bcb"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-04 09:02:47.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-04 09:02:47.000000000","tz":60},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update this at\neach new cycle.\n\nWallaby support officially the following runtimes [1]:\n- Python 3.6\n- Python 3.8\n\nDuring Victoria Python 3.7 was used as the default runtime [2] however this\nversion isn\u0027t longer officially supported.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/wallaby.html#python-runtimes-for-wallaby\n[2] https://governance.openstack.org/tc/reference/runtimes/victoria.html#python-runtimes-for-victoria\n\nChange-Id: I2c517401aa1ae464e6563800ecb0a459076655b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/672c00a67dd9a878dfefda148a9fbbaf911e4416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/672c00a67dd9a878dfefda148a9fbbaf911e4416"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
