)]}'
{"id":"openstack%2Foslo.messaging~762993","triplet_id":"openstack%2Foslo.messaging~stable%2Ftrain~I23844887722f77e827f97390fc8986f322fc7a78","project":"openstack/oslo.messaging","branch":"stable/train","topic":"non_voting_oslo.messaging-tox-py27-func-rabbit","hashtags":[],"change_id":"I23844887722f77e827f97390fc8986f322fc7a78","subject":"Add oslo.messaging-tox-py27-func-rabbit non-voting.","status":"MERGED","created":"2020-11-17 11:15:53.000000000","updated":"2020-11-19 15:29:27.000000000","submitted":"2020-11-19 15:23:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762993-1605799404451-4f1731f4","meta_rev_id":"69edbad30bff7aeb8f201147c6c02408b8539522","_number":762993,"virtual_id_number":762993,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 15:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2020-11-19 13:26:07.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2020-11-17 15:36:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":1,"date":"2020-11-19 13:26:07.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-17 15:36:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-19 13:26:07.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-11-19 15:23:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9dcf35ff9d23142f76e06277e39abf0b44a9fb9e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-11-17 11:15:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87f9580fcc3e9b7521c1158b30faa54fb48bbc2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 13:11:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3171e021c97c4b3ebe5d889b8c91ad8a : SUCCESS in 1h 36m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e98f367f3874ab49897a4e74adb0197 : SUCCESS in 1h 31m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35d549f6a94249c282c925d5e799f327 : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87f60e76542e4cb58d04e10a13fb53f6 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0d1c76e11f94b148a84e09f12a480aa : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa03d3ec4e9f4f40895cdb2396d6ac2b : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd2d1c80f4534d12b4c9ab28aaa5eff0 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef1867f17b2b4785a97c1fb0b0d30bb1 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569bfb2100a640c9a92c54407b2f02db : SUCCESS in 4m 32s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/6d8d2ee76df34cb4aa22e54a95fb55d5 : SUCCESS in 4m 41s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/31f3702112f44f12a1a03f1f41a08ce6 : FAILURE in 12m 38s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/5c8a6c71a2e341648d42f8e3e5579b65 : SUCCESS in 5m 05s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/b291159b57d34263b579d67d4fe76f39 : SUCCESS in 5m 12s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/f7a4c8a2659348208b45793c56be0f05 : FAILURE in 12m 26s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/47eed84534614f1488c9e4f393cf9033 : SUCCESS in 6m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e2faf4147cbc4cbba5f3c1a52f4835c0 : SUCCESS in 1h 36m 43s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/aa5cb8bcc44a46faa91175527136cb86 : SUCCESS in 1h 28m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/1a5fefce04854c5e8a31563486ab8ac6 : FAILURE in 17m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c9437c2e5b124802b4a92aa1e5fb3b58 : SUCCESS in 1h 46m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/78a28a17fd8645c1afbab948cbc0c723 : FAILURE in 17m 38s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/efce595ae9b549ea98a60e9e912d74b8 : SUCCESS in 1h 06m 09s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/14b47a09589148458c1b6436e0a5a84f : FAILURE in 26m 39s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/6347eb315f184e04a3b67e714accc7d4 : FAILURE in 21m 35s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/f83c18c09ff149088969b01e98444288 : FAILURE in 49m 36s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/64bcbfb0f8f84b7d8499fdc0a8271198 : FAILURE in 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e58a5afd6220edf64e90e29ef0daab0ba209be2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-17 15:36:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"85291f4b637a9852517c4e745d0338392cc464af","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-17 15:38:02.000000000","message":"Patch Set 1:\n\nAs discussed during the previous meeting these changes move functional tests to non voting on old branches http://eavesdrop.openstack.org/meetings/oslo/2020/oslo.2020-11-16-15.00.log.txt","accounts_in_message":[],"_revision_number":1},{"id":"81eca3eef63f189236b4e13b90e8a0056a60f7b8","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-11-19 13:26:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8849cb5bfc42eadd94d85cc821efd077ae3a4ea3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 13:26:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba5316676b971ca0558d6cfb6a85d35ad667ef46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 15:23:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b7d5e8a1a81529c11a5d94b3288f13419255c8b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 15:23:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4cad1d91d01a4e1e87bee53a8957ca0f : SUCCESS in 1h 45m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6437bc2c49e64fd6a445796c57bc716c : SUCCESS in 1h 49m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d9ba30251f34fea9d4c53b24128bf77 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea2d8c3a138f4a26b08df3ee856b3105 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a9e17f8204e489a94cae92e3be4174c : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b511b32dfe76425bbdaaebf7f4def1c2 : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0bcdcb1294d249af9f65f8993e73c190 : SUCCESS in 14m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5226931ca2cd497a84a578f3701c0f7b : SUCCESS in 6m 27s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/5973f3a2831b4e8c976adb7af8c7b962 : SUCCESS in 5m 45s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7a80f83cc71a479b92eb12b0adc0084e : SUCCESS in 1h 24m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a03f98c6ac7136de098a953e3f2f64ecf746a40c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 15:29:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36ba1a4e2df4bd6837c1171448f2f4b : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d98245cae84b06e3595034f1d0c3407c8231a976","revisions":{"d98245cae84b06e3595034f1d0c3407c8231a976":{"kind":"REWORK","_number":1,"created":"2020-11-17 11:15:53.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/93/762993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/93/762993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/762993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/762993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/762993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/93/762993/1"}}},"commit":{"parents":[{"commit":"1613b7a968f33ef7dc683d05fde05be0c458c8fc","subject":"Print warning message when connection running out","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1613b7a968f33ef7dc683d05fde05be0c458c8fc"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-11-17 11:21:54.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-11-17 11:21:54.000000000","tz":60},"subject":"Add oslo.messaging-tox-py27-func-rabbit non-voting.","message":"Add oslo.messaging-tox-py27-func-rabbit non-voting.\n\nAt the moment train is block by this job. We have a random failure. Move\nthe job to non-voting.\n\nChange-Id: I23844887722f77e827f97390fc8986f322fc7a78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d98245cae84b06e3595034f1d0c3407c8231a976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d98245cae84b06e3595034f1d0c3407c8231a976"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
