)]}'
{"id":"openstack%2Foslo.messaging~763343","triplet_id":"openstack%2Foslo.messaging~stable%2Fstein~I23844887722f77e827f97390fc8986f322fc7a78","project":"openstack/oslo.messaging","branch":"stable/stein","topic":"non_voting_oslo.messaging-tox-py27-func-rabbit_stein","hashtags":[],"change_id":"I23844887722f77e827f97390fc8986f322fc7a78","subject":"Add oslo.messaging-tox-py27-func-rabbit non-voting.","status":"MERGED","created":"2020-11-19 10:45:51.000000000","updated":"2020-11-25 12:11:44.000000000","submitted":"2020-11-25 12:06:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763343-non_voting_oslo.messaging-tox-py27-func-rabbit_stein","meta_rev_id":"6cbefcbadf86f1712e665681eb3532a2be6ef1ca","_number":763343,"virtual_id_number":763343,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-25 12:06:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-25 08:36:54.000000000","_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2020-11-19 13:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2020-11-19 10:46:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":1,"date":"2020-11-19 13:28:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 10:46:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-19 13:28:23.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-11-20 12:26:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56d080f61a630035461cce8b91067b81ba977a47","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-11-19 10:45:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08f910e7ba5c73405351cc1490b8a5c2094854dc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-19 10:46:39.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"d2636b092fe229ee68f0f8a343d203fb24fb423a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 12:29:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/05100c9ac8384165b947c161eb3cb460 : SUCCESS in 1h 32m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc608058094e4fe39c26877d61c364d7 : SUCCESS in 1h 12m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3efcf23d7aef4f859813bc5799c8e1c1 : SUCCESS in 9m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/769805717e2d412d9939d58c9a09bf83 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d6a7304ee84fba997b1abc25a2b571 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a4b483deaf34b0f800469b13b41b7f6 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/897e82646ee047ba9336956d4255ca83 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af6436af5e68458ca20efce9f72d986b : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c545551c38084660ae83f3cb805fa45c : SUCCESS in 3m 50s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/b5579d77c5394fd2a9f20d0673e58244 : FAILURE in 4m 50s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/33936cabd48d43689bfafc578a4eca17 : SUCCESS in 5m 34s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/8dd28a95093b43719dc13054dd253fac : FAILURE in 4m 34s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/d9818b3eb28e431facc9b6fddb6d36bb : SUCCESS in 5m 23s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/33848523a2334b8c81ae21984bc27d86 : SUCCESS in 25m 36s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/3f7832440b30413b858c3c741a030415 : FAILURE in 8m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1080335f714547f5989e825f475f7dba : SUCCESS in 1h 34m 55s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/cae6ace0a5c344328fd4e8327f8f387f : SUCCESS in 1h 36m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/41f86f8dc3fd44a7bd479929fb7efc94 : FAILURE in 13m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/88076cf775fe4fb2992a16545eb93ec5 : SUCCESS in 1h 26m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/909cd94a1f5148cfaaa163454018315c : FAILURE in 18m 29s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/3dc93d5b8b344fc0a90c5f5d3a7459a2 : SUCCESS in 1h 19m 13s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/7427e87e22a04f0d90bcd87669ba54a6 : FAILURE in 33m 55s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/fca58712be6f4fc5a93fada4b7d581d2 : FAILURE in 33m 30s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/93d3185c9b1a49b48c590f437e535494 : FAILURE in 30m 54s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/de87814a74a84523aeafc8a0fd04298d : FAILURE in 35m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a2acf2fe3f6ef92f09efe36c34135fd56fcd85c","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-11-19 13:28:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a796ae88838a8985464cccbebe5369ec0a1024d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 13:28:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a5e328e7b9af7b92b15ad6599c781cfe0f370a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 15:23:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f36cf8e556fc4fca8f7b576fabd2e450 : SUCCESS in 1h 30m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/029684e59d684593823f2b9f3b5f4d1c : SUCCESS in 1h 21m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c01830102761408eb55288fda111c224 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d2ee9c9a6024a4dbe4788ce8a841f4f : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/900afc85a117486cbf67e81a0c0ae19d : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a7becf54a154c6d964710780ee1f472 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d03ea0446784fbe9a1d9baf3c878f1f : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa744b22c3041c2b80bf82104906e4b : SUCCESS in 5m 04s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/d43aa9c6031647fb9bddcf9d87193b72 : FAILURE in 6m 08s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5a2aa0f835cb488dbb95f2146176a4f9 : SUCCESS in 1h 21m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c95144d61b241e97af0ff4df7f200885988884be","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-11-19 15:59:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0bebe2455c2b9c2bcac74d5e45964d6fbbbfe156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:56:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/52cc1503d78643b68acfdd5ba05b1ecc : SUCCESS in 1h 48m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1d33831fff6b44ac9b23bdbab8da8f72 : SUCCESS in 1h 51m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5402b6b859ae43d2a1487a27fd8bec5c : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/285f9519578c42fc8879a018d9cb290f : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/257c329857c347cdb71497cb204e3ddc : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5267a9e5caf94a2eaf1d2a3f5ff7b37c : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0655284dc554a4da44d393270bfac6f : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/535a43f1cc4a4e8885836eb98e1313d5 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8367df03e1684f67846db32f6539e984 : SUCCESS in 6m 27s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/82b9e9c4f6544ca6a27926ae359f3c97 : SUCCESS in 6m 46s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/f9dc2c62143046f094d6d890ba04fc94 : SUCCESS in 4m 35s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/7b727da8a9db47ab83873c3d3506aa86 : FAILURE in 4m 15s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/dacea36084e64e6fb3846cbd8866354a : FAILURE in 5m 26s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/61035fa80843499498eacd3afa66c241 : SUCCESS in 5m 21s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/80c9373cbbec46be9272f84190b2b935 : FAILURE in 5m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b47c35c1f71c465a97870c904573d564 : SUCCESS in 1h 41m 09s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d0f470dcdfc941239ab70cacf616d6a1 : SUCCESS in 1h 28m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/3ae1df43286e43589d9a26c214e93e18 : FAILURE in 19m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/fa5f5affbe754939a769d649f73fdaf8 : SUCCESS in 1h 32m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/0966afe080a840ec908caff15af3140c : FAILURE in 20m 22s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/9591bc8f3c2048a984500e69f4782f97 : SUCCESS in 50m 52s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/4e6311c55e6d442ca382682bceaaa90a : FAILURE in 1h 20m 35s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/ff4edd5d16554241a08a6c94cf55a671 : FAILURE in 29m 55s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/932bf63bd96845adb31488e1dff52ebf : FAILURE in 49m 25s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/1da8cb17f1cd49608fdf3ad91f7b64a3 : FAILURE in 31m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c6abdc9ca3eecebeec4e722c4bbffe7198140b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:56:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc29d5e1ecf78610da12ed85d418787abfc17285","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 19:39:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d29a040766cb407aac4aea4d792f866c : SUCCESS in 1h 40m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e1d1d8743294098acf3a412a6b094db : SUCCESS in 1h 26m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d249dc62256436388b7b859016a1fa4 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bb7900ffd964ae4acfcb77333e915e3 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09305792ca404421bdd3359ff91b7ab0 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de623b9e0c1e4078aa6f68159715c1ac : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee0eeb4c07994f4e98df7a403e0a6ce0 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2daf6fd00c497e8986b2fd2a7b1ae1 : SUCCESS in 5m 03s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/840aaf63c934494fb97861a2cb01ff0b : FAILURE in 6m 24s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5435b55965aa47c583adba829db8dcb0 : FAILURE in 14m 45s","accounts_in_message":[],"_revision_number":1},{"id":"bdb14529a95c8b7584c384ecab46f6948e547d91","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-11-20 10:47:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"023c587119187503f691daa76d63b241bd4c7513","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 12:26:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2cf061a09bb842fe8f7ec8da14502ebc : FAILURE in 1h 25m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c22a64bfc9ae44f385b4c4d1848dc4b1 : SUCCESS in 1h 17m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd4b1579dafd4c459300f6271e85ca20 : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b9b92ef1ed943b380b142143c997d8a : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8769df63f9104bc8a2fb458067e9ba85 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/03362356e86f47de8c7ad74597315b99 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae6f709884cb425a998994db82161387 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d694274355ed47d7ab896a9d66c6b844 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e4b419f5e1c4890ae634bd1d08019da : SUCCESS in 5m 15s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/b14448784e7b4587bc0ae18f87856625 : SUCCESS in 4m 19s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/27c1525941364b03bfe82e542d9fc819 : FAILURE in 3m 51s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/d4a42a1b3fe24a5c8791ef866ebd0cf4 : FAILURE in 4m 22s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/f0177ba6e7054be2806586c9fcec3b98 : SUCCESS in 5m 42s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/2727c7da3899436597fa475cd42c9bb0 : SUCCESS in 4m 57s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/f09b5b76b8414f38a641b2a79ba628e9 : FAILURE in 4m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3aaf9250210f40bfb877df62dfeb08fc : FAILURE in 1h 15m 58s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/10a1fc80dc2c4ec298d5f5e966b9e9cf : FAILURE in 54m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/66e75248611f4a2b8987d3c83d5cea09 : FAILURE in 17m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9f5b3f9afab945698fa026f488ec5004 : FAILURE in 1h 24m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/1b9c7af6b9e34c118c1c2d13a61171f1 : FAILURE in 10m 03s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/31ab026f33e14b0bab44912c73677eae : SUCCESS in 58m 27s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/1209496d65bc4a1b837772b54ccd5041 : FAILURE in 1h 03m 06s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/82f45cd215d04ffbb595061a98a7badf : FAILURE in 25m 07s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/0f2f04b0418d4eca90e0bdd6e6ef0f51 : FAILURE in 31m 34s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/56112681a53643d7aeaa9f4012e492a6 : FAILURE in 36m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af427de9fabf1b9a27d6b7740deed4c256f7c63a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-23 09:14:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8ecc7e25eaf56c1cc1bfc7a573be04fdfc34f1e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 10:50:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c812132b03194591997eef2bf7d6363e : FAILURE in 1h 08m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f17eacbce58442aa20dd62b38142016 : SUCCESS in 1h 24m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ed1507e9dae433aa530ee3c7c07b8de : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7c9925bf21b4f63bdeb70ca080ffc7c : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5d1e7753db34db0ba2a60e6baf43cc6 : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2de223cb6f714dff929affa9a201fb89 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a1836fcad7647d6984b6a3e6e5af1e5 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/603a469dea38496ab7f0777bfba6ad70 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37203f8dfda74f9da5bfc9fae6fdd36f : SUCCESS in 4m 17s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/f4f3d7e884384430b2a0af5c81b82eb8 : SUCCESS in 4m 21s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/281a0e46278846b58a31c91695526e45 : SUCCESS in 4m 29s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/fbafb4f0fdd44aef9e39618feef81457 : FAILURE in 3m 59s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/bcbd5677c53943639d5842d0ccf74dbf : SUCCESS in 5m 56s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/9e51f2d7a4da48eca139527ef2be987f : SUCCESS in 5m 11s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/e456befba697443781e470c4036fa62b : FAILURE in 4m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/61f1a21666ff4ddab3e9ad49bda95be8 : FAILURE in 1h 12m 45s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/60f039a280574434a5f398f1a2404076 : FAILURE in 1h 27m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/ef2f005d09134a69872d1a1bf1ca9f59 : FAILURE in 16m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2502fc1d4d60470e86f6c05a6303e284 : FAILURE in 1h 08m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/1bd8e6b71ce142219d5cdeab70b78429 : FAILURE in 10m 46s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/3556f8c2f4934b2c86a5e79e22749740 : SUCCESS in 53m 20s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/a09b93ef644749e990bfc0b5f31e6a8c : FAILURE in 59m 41s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/3ea80562e4fd461db0033fc2f6e01496 : FAILURE in 30m 49s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/1afc6128633e43149803cd560fa07810 : FAILURE in 31m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/edd839a10be541f7a280683775e825a7 : FAILURE in 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0651649dde17a58abe04fc4f903c262a06a318dd","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-11-23 14:25:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"751a62cefa50fd5099e5c5810ca5b0eb6b07189f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 15:58:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8de2d52140ca430aace53733d9c054de : FAILURE in 1h 25m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97d77ad67e2d44ebaef019d2b103c579 : FAILURE in 1h 03m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3db518e644b7452d86a3afafdd7574f6 : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6981b8a17da04a588ea6f1cc4e7559c4 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b6f88d5d56746c1b79f160767071852 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ec51a510d3b4c2fa99057564b33af2c : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9da06dccd184037a239b7bef2f30a9e : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a162bb3f18e4c309bc5a55f9ece4965 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dff62f756fa3492db9f7a7c29a1fa84f : SUCCESS in 4m 11s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/0c8cd459a8f0432298dc432d63127cd1 : SUCCESS in 4m 57s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/6a67c451fead4768bf31f0b84e0198ec : SUCCESS in 4m 20s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/8685f2ce5b8648e187dd36bacafedc0d : FAILURE in 4m 26s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/f85f085353ff4028bbd23e72fc0c3eb0 : SUCCESS in 4m 57s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/e2c883d71b0341f69a3144b1251a3972 : SUCCESS in 4m 42s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/9d004e2a33d640e380c391d92754474c : FAILURE in 4m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7a9cc284407f4f70b2da1ffd8f657f35 : FAILURE in 1h 23m 41s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/06ca713dd7d242eaa82c5aabe671b7e9 : FAILURE in 1h 06m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/e6e73d3c676140f8a7dff3e6b8076cd1 : FAILURE in 9m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1f0b8c48204b4b1abb7af510aa1b626c : FAILURE in 1h 18m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/48a3d8bcff224a48858ec3cda5ddc39a : FAILURE in 18m 19s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/6b189b19da2e42e28ac11fc9bee25bf5 : SUCCESS in 56m 54s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/84034cc12fcc4461a3fd0012298f3bc0 : FAILURE in 49m 40s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/e937605df3874cd0a0f4e4d82d02e872 : FAILURE in 33m 14s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/60af360faaf4494c8c77cb1ec9c85bea : FAILURE in 30m 53s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/3bf565f5ca1b4dd98e579a599c829505 : FAILURE in 25m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a379cf3cc829e0658a6cabaa9aa4b2b0e8cf0b0","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-11-24 13:50:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"32cb05ddefd7674eaf7d6d011e44381e64bccd0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 15:33:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b73372d0a09241f8b63c9f5bde43998e : FAILURE in 1h 36m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20d0d5a96f46482faf46cd35ee9cc495 : SUCCESS in 1h 33m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d27dc28c3fb4f56b427e6da9f79d535 : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/998fea192fd649f1a633873b6fa36082 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a9c14d64764438285008e24f9ff11e9 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/142f62166a63412fad201dd7af320204 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f94deb336a884528b6e8496c2c5b4d2e : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70265c6d322444d4b68675576af05bd2 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1881841c3df43d7ba20fdadb46f3d0a : SUCCESS in 3m 48s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/e2816681523e4d0fa51017654cf36101 : SUCCESS in 4m 32s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/1421e5d35a2f44a98cdca905793ea203 : SUCCESS in 4m 13s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/47a19e297e30405da332c464bea02e98 : FAILURE in 4m 30s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/1fec09541bdd45f9a0de8140be07e835 : SUCCESS in 5m 01s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/22ff6748da6a4c8f9a5b686de07545a1 : SUCCESS in 7m 26s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/72d462335ebf47a5bf935c329a676d23 : FAILURE in 8m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/18f53b39386549afa35780e395437cf7 : FAILURE in 1h 01m 41s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/0ceece2ad9474d9d9f3ae157d9cb1e72 : FAILURE in 1h 04m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/dfaa379b88d344d8b0f57003e3d8645c : FAILURE in 18m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/cb6923407d2c4713b07a51992778e559 : FAILURE in 1h 12m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/04d57842ff6b48e4acf3ea4e5fdb337f : FAILURE in 11m 02s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/b7313485315b4a41a61e106b5c286d24 : SUCCESS in 1h 07m 04s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/026709e80adc4c27ae9e250d1dea03ea : FAILURE in 1h 09m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/5b70bb1e057548d090c80196f953df56 : FAILURE in 17m 40s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/30aea0f4e090444aabcaef236c455037 : FAILURE in 17m 37s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/6c48c3f45cd24eeeb5eb50cdbd0b3b49 : FAILURE in 29m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77c74c6673e7f7e1ef6d1f1596d3ecaec133746c","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-11-25 08:36:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"794e6508f386cf807027ab30b9ecc69bebb19c67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 10:26:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/527a66bb614c43f6ab1bd4a117940e32 : SUCCESS in 1h 29m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0913da25daf448f185114c9c609ca41b : SUCCESS in 1h 32m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08a3ba081bfa4101801c0bbb06548c40 : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f28d6a0a983d41a093ca187c2e7a91d2 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03b163ce64154fd9abc0907bc9edc76a : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1133a800f9174c2a9ded1f3431a569cb : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/beb3139f63de4fa8a71a4097bb541297 : SUCCESS in 7m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab375478c6f349c2811a888edeaec7e0 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b20a9e4f2f4bd88dcb36bd4c83c474 : SUCCESS in 5m 02s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/ecf72bde15f84e608450f6742fac6751 : SUCCESS in 5m 13s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/bd4add7409684b90a40fb9868aaf6bee : SUCCESS in 5m 35s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/259fdb17cd8e4a3dbc96aacf0ee770ea : FAILURE in 4m 18s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/682ab1f403014b9da14aa83f9c874153 : SUCCESS in 4m 40s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/19686ba4f3dc414591a4e6c008893f4b : SUCCESS in 5m 13s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/17624d2518ec470b8e52c535e6eeeb5b : FAILURE in 5m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0ac718045f87449199d94f87b69d9518 : SUCCESS in 1h 24m 39s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ebc3a4f9d83c45ca972d3e2b167f6d18 : SUCCESS in 1h 26m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/08f2356c307f45bb9f43866b31f6e003 : FAILURE in 18m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/fe1eef34b32941d4a6a3de1b657f1340 : SUCCESS in 1h 38m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/01cf527530024bc1864d25ede7c80462 : FAILURE in 13m 13s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/0f106799ecee46f4b6c937ba6e0979ac : SUCCESS in 1h 00m 05s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/03475fc4911e4655be7e85511103c06d : FAILURE in 53m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/ebd95794520640c49333ff99d19db01d : FAILURE in 32m 08s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/9a6b2b7c9c4145afafa1dda027932002 : FAILURE in 34m 49s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/6ccf710f1b834e51a12b9a8899dc139c : FAILURE in 33m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dcb47a3ca65a54222514876b8fc1f91fd126c419","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 10:26:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f2560f95d498b7cd0ccaf861d9bd3b4686dcf7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 12:06:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1e8873b1598845219de809001c6f6a1c : SUCCESS in 1h 34m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d95e857a13014e92bea90fdbd3c1561e : SUCCESS in 1h 19m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a84414e4a184370bcc0c16ad577b2ad : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89afe4c46bc744eba47b2a613a016a7c : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f241c45a17954939b40acca6f6dac27c : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98c3bc06b14442ecb7f4808336489216 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70f731d975f046c584e642de05176148 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1a67f32e6704ba99ef7d3098dfb4686 : SUCCESS in 3m 47s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/4ae779798e3f4c27be5ca05df1c052ae : SUCCESS in 4m 59s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0156750651fe4c1293de13b2f460eae1 : SUCCESS in 1h 27m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2e7508003da664d61ea177d09287fcb5e194e85f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 12:06:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6cbefcbadf86f1712e665681eb3532a2be6ef1ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 12:11:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d5735ba720140e5b7a1e806bdd10d4b : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9939f46946a88fd882b6eabb13c51d11269d059","revisions":{"a9939f46946a88fd882b6eabb13c51d11269d059":{"kind":"REWORK","_number":1,"created":"2020-11-19 10:45:51.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/43/763343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/43/763343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/43/763343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/43/763343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/43/763343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/43/763343/1"}}},"commit":{"parents":[{"commit":"fd795bbb8f5aaa740fa810f2248646f6ec3a211e","subject":"Print warning message when connection running out","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fd795bbb8f5aaa740fa810f2248646f6ec3a211e"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-11-17 11:21:54.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-11-19 10:52:17.000000000","tz":60},"subject":"Add oslo.messaging-tox-py27-func-rabbit non-voting.","message":"Add oslo.messaging-tox-py27-func-rabbit non-voting.\n\nAt the moment train is block by this job. We have a random failure. Move\nthe job to non-voting.\n\nChange-Id: I23844887722f77e827f97390fc8986f322fc7a78\n(cherry picked from commit d98245cae84b06e3595034f1d0c3407c8231a976)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9939f46946a88fd882b6eabb13c51d11269d059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9939f46946a88fd882b6eabb13c51d11269d059"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
