)]}'
{"id":"openstack%2Foslo.messaging~763975","triplet_id":"openstack%2Foslo.messaging~master~I6039d09533c86d651b5c63b500058aac3c9f0a7f","project":"openstack/oslo.messaging","branch":"master","topic":"fix-var-name","hashtags":[],"change_id":"I6039d09533c86d651b5c63b500058aac3c9f0a7f","subject":"fix variable name","status":"MERGED","created":"2020-11-24 13:49:52.000000000","updated":"2021-01-08 14:17:10.000000000","submitted":"2021-01-08 14:15:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763975-fix-var-name","meta_rev_id":"727dbe8783f1ad63520b2b56689a7d4335cfbae1","_number":763975,"virtual_id_number":763975,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-08 14:15:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-12-11 15:31:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-12-11 15:31:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 16:04:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 15:31:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"dd13613e37e5cc3f2cbcfa6b8e83c78601279997","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-24 13:49:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fde5f5af11001116a4d16aaad21364234fc381fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 16:04:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78c949ebe5454a9f9539886440d99b7a : SUCCESS in 1h 35m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae582fb0107e49e8a0c8e6700546a38a : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/616523d6b9e842b8b66a8660890f0416 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/649e477f0b85453f940f420de3d914df : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3137f99b39df41e48f99e07bd6b32da8 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c40d257fa56d410180aa5bc234b265a7 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23b6bd480dec4f919e65a2114b399f9a : FAILURE in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df99d88d99047b89de2790fb1e21952 : SUCCESS in 3m 50s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/0cecb773745f49d29c96deed2e56b0ec : SUCCESS in 5m 34s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/aaa12de03a6447319e4288bdf7162cb1 : FAILURE in 6m 13s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/55aec939e34248b9818df281d5eae4f8 : SUCCESS in 5m 42s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/33d32261afc04b80a167b60a3319433e : FAILURE in 9m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1276e9be4ff54c57b36175dcc67fb274 : FAILURE in 1h 16m 35s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/10dcb9644d104dcfa6a5d45a88fd744d : FAILURE in 1h 14m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/025ba44e019542f2be1b9a49166c09c2 : FAILURE in 2h 07m 11s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/89e78e6b0ddf4195856522a285b22e43 : SUCCESS in 1h 07m 27s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc9379e36fe042d2b9f0e8bf8bfc1cfd : SUCCESS in 1h 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eba952252fa04523c8c070438f698a2240ac0fd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-11 15:31:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e53d9266ae32f4be5a7600e4e0a3dad7b3751fad","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-11 15:31:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6c6f388e6f646675151f879470f4d0df23c216c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 17:32:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd70c65684d941f7a4b643364d6ec5d2 : SUCCESS in 1h 39m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a594223ad0e0433bad693319ffe7452e : SUCCESS in 7m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/088288d7f6f54380a406b8db27cfeead : FAILURE in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/144c01811ed045eca0c5b0f3ecd74da8 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b07039b4388245498b579a702fe86476 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6f8647ba7af447695ed8d630c70acee : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/886602c368894d279dda0bea42c9c41c : FAILURE in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60e9826d103a4572b11d9c529e6dfa2f : SUCCESS in 4m 46s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/de4ffa4bb7924202aa86e80d57015ac4 : SUCCESS in 5m 51s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/e40bcac5573f46d39ce360b0707dabb7 : FAILURE in 6m 17s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/24b33dee26074af697f52b522d298f9b : SUCCESS in 6m 33s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/84aebf710cd24373b6b8f13c39be3584 : FAILURE in 6m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c29c140e4ab74320b09381729480f1ad : FAILURE in 1h 24m 13s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a375f7b702da4009b63f9703d3592980 : SUCCESS in 1h 50m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d8d942aa82914f19aedf8e96163ca8bf : SUCCESS in 1h 32m 31s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b8fa766c7136493bbdae75d0d7355894 : SUCCESS in 1h 06m 28s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/81055393fa0645a1a49ec33e351557a5 : SUCCESS in 1h 16m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bcab109f2dadc719fe7e5bcd9dbfd0cf93b044d6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-08 10:24:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e91c3719d4c777db7828b6507f78a286e6b27dc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 12:32:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7067730b4e964e14b77f5695efa67aa4 : SUCCESS in 1h 28m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/564fb1ef379c4693a208e589d9bcf886 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f627db2a9a74887bbd7b92c5dfc68e9 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f87a0c69bea8428aa174ab90c54a42d2 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cc7ee0336334720af4d195c531b41b7 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f19c9cedd74847bf93e202ff88bc853f : FAILURE in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0214a3743f2642ce944484cb8c7d537b : SUCCESS in 5m 48s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/a2821e067877406daf5bc04ea0278dfc : SUCCESS in 5m 19s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/dff2940c307444929b396a73a80d465d : FAILURE in 6m 17s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/675b7f268e444b22b70ed8a3648bf97c : SUCCESS in 5m 08s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/9e8584de2e104bcd89c6de8349325cd5 : FAILURE in 6m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b25d6249d2484960b7f871a9b8eb29f6 : SUCCESS in 1h 33m 57s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ec48f8f7ea96473daedf937a55fdba1d : SUCCESS in 1h 38m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f2177f2b6baa4a7a88540d1cb28eed66 : SUCCESS in 1h 26m 25s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/38aafde0fad94b3581bd9980d8397e81 : SUCCESS in 59m 24s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/216a563a6d9b4327b3745889d3ca01dd : SUCCESS in 1h 20m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"236c1b0a5996963fc897a5c146d15833cbefe0b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 12:32:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3864c0a0339ad6cadb41f3e08aea0e6e506f8bf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 14:15:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9a3b9302e9c4cd18d49c48493d8188f : SUCCESS in 1h 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/728a05199591403ea016483f9704a96d : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/851c9f166dc4490ca9c26c393bb09b44 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/370eb2e53b3d4f07b5000decebc790f8 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de7c7a35be8644e7b04c31e2ce258807 : SUCCESS in 4m 33s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/a6d22f37c87a49b89a24f6b794d1bfd5 : SUCCESS in 5m 15s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a83396f9e67544c5b24e4a96d0b3fb0a : SUCCESS in 1h 36m 56s","accounts_in_message":[],"_revision_number":2},{"id":"343646f26986bcf03d249dcb8c301ee88f30db20","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 14:15:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"727dbe8783f1ad63520b2b56689a7d4335cfbae1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 14:17:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75d0920ac8f74f6b8f866b7f0a573dc5 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0e4ec379c0bb44bd2b1dc25a2602ae9a944b8c4d","revisions":{"9221136293960b54aa9de2323fe8d3bd7c88b0b0":{"kind":"REWORK","_number":1,"created":"2020-11-24 13:49:52.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/75/763975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/75/763975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/75/763975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/75/763975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/75/763975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/75/763975/1"}}},"commit":{"parents":[{"commit":"0c4203bbdc463010d756a71bbd06720867122bcb","subject":"Python 3.9: isAlive is removed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0c4203bbdc463010d756a71bbd06720867122bcb"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-24 13:49:39.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-24 13:49:39.000000000","tz":60},"subject":"fix variable name","message":"fix variable name\n\nChange-Id: I6039d09533c86d651b5c63b500058aac3c9f0a7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9221136293960b54aa9de2323fe8d3bd7c88b0b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9221136293960b54aa9de2323fe8d3bd7c88b0b0"}]},"branch":"refs/heads/master"},"0e4ec379c0bb44bd2b1dc25a2602ae9a944b8c4d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-11 15:31:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/75/763975/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/75/763975/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/75/763975/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/75/763975/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/75/763975/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/75/763975/2"}}},"commit":{"parents":[{"commit":"54f8724be281b1acfae175d7ec1c14e439269117","subject":"Fix doc title rendering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/54f8724be281b1acfae175d7ec1c14e439269117"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-24 13:49:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-12-11 15:31:42.000000000","tz":0},"subject":"fix variable name","message":"fix variable name\n\nChange-Id: I6039d09533c86d651b5c63b500058aac3c9f0a7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0e4ec379c0bb44bd2b1dc25a2602ae9a944b8c4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0e4ec379c0bb44bd2b1dc25a2602ae9a944b8c4d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
