)]}'
{"id":"openstack%2Foslo.messaging~764776","triplet_id":"openstack%2Foslo.messaging~master~I1cbafff03349f7da9224de46285707fbf2a81a68","project":"openstack/oslo.messaging","branch":"master","topic":"flag","hashtags":[],"change_id":"I1cbafff03349f7da9224de46285707fbf2a81a68","subject":"Deprecate the mandatory flag","status":"MERGED","created":"2020-11-30 15:41:51.000000000","updated":"2021-02-01 16:49:12.000000000","submitted":"2021-02-01 16:47:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"764776-flag","meta_rev_id":"63e1ececb34970e60768a678ee388dedfce3fc56","_number":764776,"virtual_id_number":764776,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-01 16:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2021-02-01 14:14:57.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2021-02-01 15:09:13.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2021-02-01 15:09:13.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 15:56:08.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-11-30 17:26:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-02 16:20:24.000000000","updated_by":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"reviewer":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"state":"REVIEWER"},{"updated":"2020-12-02 16:39:20.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2021-01-27 16:25:11.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"CC"},{"updated":"2021-02-01 15:09:13.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"}],"messages":[{"id":"7a0151b625cfbde4a4bd890a6dc9497be0507ef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-30 15:41:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1d6edb430b8305ad241168935a9ade02e29e093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 17:26:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3348bff5ea0540a092074b939bf988d5 : SUCCESS in 1h 27m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a9527a2b316421b83054a220d239042 : SUCCESS in 7m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20eb3076f81e487cb5699a3700210a7a : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/456a007f7aca42fd964aef5a7f52a8d2 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27aae28c0a5d490b85eed40bb861fc22 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dfb5b8cd717480fb0f1b2f3ae26167a : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c4b6b1fbb004533acf4801f2045023d : FAILURE in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62547775eac846e2b1c2ddf1d3a0cd82 : SUCCESS in 5m 01s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/0593432c19594975abdc2d24711f12cb : SUCCESS in 6m 28s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/06408cee4b8d4734a53fd3d51b7b4fa8 : FAILURE in 7m 52s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/313ecece0d854ca38360911f16f2b103 : SUCCESS in 19m 59s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/eabc8e14d78c4683a2ed21e4d163ed11 : FAILURE in 22m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/495f97f01cce47548f544a6e83561d30 : SUCCESS in 1h 25m 57s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/55ee8cace1064cd79a88614c44c7ff62 : SUCCESS in 1h 29m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d99b09c107a242128f55a2328c04d221 : FAILURE in 1h 22m 53s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/26c005be3e48464f95b103ea30dea751 : FAILURE in 16m 01s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c84e4231f1148a892a32bd1d7ab2119 : FAILURE in 22m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"134b80a3d7596a039c8868e7620499473ce81532","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-11-30 21:14:49.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"624db0d5bf741df1f392f9a79524f449836e5efe","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2020-12-02 16:20:24.000000000","message":"Patch Set 1: Code-Review-1\n\nI commented on the lp bug here:\n\nhttps://bugs.launchpad.net/oslo.messaging/+bug/1905965/comments/1\n\nBut in summary, I don\u0027t think this would fix the problem and would actually make things worse.","accounts_in_message":[],"_revision_number":1},{"id":"87f374a73d7e067a0029d5c3bd9c34c12ba1a163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-11 09:43:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e6053bfe4fee5ce1d84b91e7e92b1a3848976cc0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-11 09:50:12.000000000","message":"Patch Set 2:\n\nI wonder if we can drop the mandatory flag option and only keep the `at_least_once` option.\n\nThe mandatory flag is useful and could be used everytime and I don\u0027t see reason to deactivate, I think it\u0027s preferable to only allow user to decide to loop over or simply re-raise it.\n\nNow that we can define a more granular failing strategy by using at_least_once, I think we can abandon the `direct_mandatory_flag` and assume to use the mandatory flag everytime.\n\nThoughts?","accounts_in_message":[],"_revision_number":2},{"id":"a6ef891a3f76e4a5ec0e976c7c4db0d567a7f5a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 11:41:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0a8e7b4af804ce49b922f370361346a : SUCCESS in 1h 20m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8354a1118804d42b60ba4a354eda448 : TIMED_OUT in 51m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e700f0ea8bf74db3a24d0d510cfb7b00 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5554ec3f2ce54d14b99a8bcffde29815 : TIMED_OUT in 40m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bd59e6ad5014dfca7c1bfa3942bf91e : TIMED_OUT in 41m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/526bc0d9c00e4175acffecc7d22641af : FAILURE in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9820ddaf576c4d0bb7f0aed5514097c0 : SUCCESS in 4m 10s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/dfea0dd436be4176b90ef7b177f83c98 : SUCCESS in 5m 44s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/c1f563f1874a462986ccf7eb3b7dd9d1 : FAILURE in 8m 14s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/97c94bc7823f403398689d0c09f43126 : FAILURE in 6m 33s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/ba263a4b26cf47f6a5b24d9e0e88e75f : FAILURE in 7m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/27a6373e55744a358bd7423c09897885 : SUCCESS in 1h 29m 49s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/0031a4adf6dd43bfa0086cdf36a441ab : FAILURE in 29m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2b06d25c9958488b98c0f75b52a8f113 : SUCCESS in 1h 45m 28s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/73604be10b0d440cabbe9cf1cc4162f0 : SUCCESS in 1h 01m 30s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d444ed58dc034d44a1d598a8e8b22f54 : SUCCESS in 1h 53m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dafd4151fcaa556d52acc34c5f19256d5bd9e930","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-11 14:14:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5b8d017e4e84bf48680e9880771a3460b9049289","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-11 14:16:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3dcf2f7e32b8cf52f553c374b8c7eff5b993df17","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-11 14:33:02.000000000","message":"Patch Set 3:\n\nI tried to simulate the scenario with oslo.messaging\u0027s simulator, unfortunatelly I couldn\u0027t reproduce the MessageUndeliverable exception.\n\nFind bellow the followed process, maybe you will have an idea oon how to simulate that.\n\n1. Run a rabbitmq server and allow to access to the dashboard to monitor things:\n\n```\n$ sudo podman run -d -e RABBITMQ_NODENAME\u003drabbitmq \\\n    -p 5672:5672 \\\n    -p 15672:15672 \\\n    --name rabbitmq rabbitmq:3-management\n```\n\n2. Prepare some configuration to use with our simulator:\n\n```\n# tuto.conf\n[DEFAULT]\ntransport_url \u003d rabbit://localhost/\nat_least_once \u003d False\n```\n\n3. Run the RPC server\n\n```\n$ tox -e venv --  python tools/simulator.py --config-file ./tuto.conf -d rpc-server -w 40\n```\n\nNotice that I asked to the server to wait before answer to allow me to manually delete the reply queue on the rabbitmq server side.\n\n4. Run the RPC client\n\n```\n$ tox -e venv --  python tools/simulator.py --config-file ./tuto.conf -d rpc-client\n```\n\n5. Delete the corresponding reply queue\n\n```\n$ sudo podman exec -it rabbitmq rabbitmqctl delete_queue reply_\u003cid\u003e\n```\n\nUnfortunatelly I can\u0027t fall into the expected situation by following this process... \n\nI suppose that some other mechanismes catch that something went wrong with this scenario and I didn\u0027t reach my expected scenario.","accounts_in_message":[],"_revision_number":3},{"id":"87b6506723f21cd1216548f947f0f693e4e810d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 16:23:26.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/703e2a2e34ce4e09a22b857c9dbc2e21 : SUCCESS in 1h 51m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3388dc4b943e46799cfd6b4168fb897c : TIMED_OUT in 51m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/945b33b592ed4e07915f31c828d2c8b7 : FAILURE in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/661992505594456b818ef50c5eb008db : TIMED_OUT in 41m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4587b47423234630af2530536e734334 : TIMED_OUT in 40m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51e2fbaf7eb6417bbff58b74d3875a18 : FAILURE in 11m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/676e4b5829cd49c585e9a2200294ecef : SUCCESS in 3m 53s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/e3b0e317fe1e461896fb95e9a527420a : SUCCESS in 5m 17s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/a3c7dc77f5ce4aba8f7d4bda7a5a0db0 : FAILURE in 6m 45s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/68d18ae4b226497a9d6c544bea4d9ebb : FAILURE in 5m 43s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/938a01f5286f49649bc742cf33e2e4af : FAILURE in 5m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/922fd06be5f14c16abd2277cf64d2425 : SUCCESS in 1h 16m 56s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/50230f868920428b98814a5919dcf371 : FAILURE in 29m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/fb9a0241722b4139b749ac6603396566 : SUCCESS in 1h 35m 04s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0b3a3443d74448f499a2554802eeafae : SUCCESS in 1h 07m 56s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebdeed284491444fb8d3e9b5bc0ddfa9 : SUCCESS in 1h 12m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0210e39262c5e7b8d3ddc78b3238d661c9e1f714","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2021-01-11 18:44:16.000000000","message":"Patch Set 3:\n\nStepping back a bit, I\u0027m wondering if we\u0027re doing the right thing using the \u0027mandatory\u0027 flag at all...\n\nAFAIKT the original bug report that sent us down the path of using the mandatory flag is [0].  Re-reading that bug report leads me to believe the original desire was to have a way for the RPC _client_ to fail fast when the RPC server is not present.\n\nThis makes me question why we\u0027re using \u0027mandatory\u0027 on the RPC server side at all.  In fact, I\u0027d argue the right solution to that original bug - RPC clients issuing calls before the RPC server is present - would be better solved outside of oslo.messaging by implementing a bespoke \"ping\" call with a short timeout.\n\nTL;DR - should be reconsider using the mandatory flag at all?\n\n\n[0] https://bugs.launchpad.net/oslo.messaging/+bug/1661510","accounts_in_message":[],"_revision_number":3},{"id":"03c8640f3d338884dbaefef616f88275c2f556a9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-13 13:36:45.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e Stepping back a bit, I\u0027m wondering if we\u0027re doing the right thing using the \u0027mandatory\u0027 flag at all...\n\u003e \n\u003e AFAIKT the original bug report that sent us down the path of using the mandatory flag is [0].  Re-reading that bug report leads me to believe the original desire was to have a way for the RPC _client_ to fail fast when the RPC server is not present.\n\u003e \n\u003e This makes me question why we\u0027re using \u0027mandatory\u0027 on the RPC server side at all.  In fact, I\u0027d argue the right solution to that original bug - RPC clients issuing calls before the RPC server is present - would be better solved outside of oslo.messaging by implementing a bespoke \"ping\" call with a short timeout.\n\u003e \n\u003e TL;DR - should be reconsider using the mandatory flag at all?\n\u003e \n\u003e \n\u003e [0] https://bugs.launchpad.net/oslo.messaging/+bug/1661510\n\nI agree with you.\n\nWe could follow these steps:\n- first, simply deactivate the mandatory flag by default and communicate loud about this to avoid back and forth with users that could started to support that flag on their side.\n- Second, remove the mandatory flag usage from the RPC server side\n- Third, focusing us on the RPC client side to try to keep the mandatory flag around to continue to fail early.\n\noslo.config\u0027s set_defaults could be used here to force the deactivation https://opendev.org/openstack/oslo.policy/src/commit/a626ad12fe5a3abd49d70e3e5b95589d279ab578/oslo_policy/opts.py#L121\n\nThese steps could let\u0027s us a chance for the future to improve O.M with the mandatory feature, and they could allow us to fix our present issue.\n\nLet\u0027s wait a bit for John\u0027s feedback.","accounts_in_message":[],"_revision_number":3},{"id":"7099d7bd6674662c763a3127da90ffae7bf1500c","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-01-14 22:26:13.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 3:\n\u003e \u003e \n\u003e \u003e Stepping back a bit, I\u0027m wondering if we\u0027re doing the right thing using the \u0027mandatory\u0027 flag at all...\n\u003e \u003e \n\u003e \u003e AFAIKT the original bug report that sent us down the path of using the mandatory flag is [0].  Re-reading that bug report leads me to believe the original desire was to have a way for the RPC _client_ to fail fast when the RPC server is not present.\n\u003e \u003e \n\u003e \u003e This makes me question why we\u0027re using \u0027mandatory\u0027 on the RPC server side at all.  In fact, I\u0027d argue the right solution to that original bug - RPC clients issuing calls before the RPC server is present - would be better solved outside of oslo.messaging by implementing a bespoke \"ping\" call with a short timeout.\n\u003e \u003e \n\u003e \u003e TL;DR - should be reconsider using the mandatory flag at all?\n\u003e \u003e \n\u003e \u003e \n\u003e \u003e [0] https://bugs.launchpad.net/oslo.messaging/+bug/1661510\n\u003e \n\u003e I agree with you.\n\u003e \n\u003e We could follow these steps:\n\u003e - first, simply deactivate the mandatory flag by default and communicate loud about this to avoid back and forth with users that could started to support that flag on their side.\n\u003e - Second, remove the mandatory flag usage from the RPC server side\n\u003e - Third, focusing us on the RPC client side to try to keep the mandatory flag around to continue to fail early.\n\u003e \n\u003e oslo.config\u0027s set_defaults could be used here to force the deactivation https://opendev.org/openstack/oslo.policy/src/commit/a626ad12fe5a3abd49d70e3e5b95589d279ab578/oslo_policy/opts.py#L121\n\u003e \n\u003e These steps could let\u0027s us a chance for the future to improve O.M with the mandatory feature, and they could allow us to fix our present issue.\n\u003e \n\u003e Let\u0027s wait a bit for John\u0027s feedback.\n\nYeah this sounds good to me.  Definitely remove the mandatory flag usage on server side for now so we can unbreak that.  I think longer-term there is potential to benefit from the mandatory flag as it lets us fail fast in situations where we otherwise would wait for an rpc timeout.  Plus I think we could completely get rid of the exchange-probing that happens with the AMQPDestinationNotFound stuff and replace it with mandatory; that\u0027s a much more straightforward usage to achieve the same end.","accounts_in_message":[],"_revision_number":3},{"id":"67c9ce700e1a1815dd7a86432af87d24d8d3bbf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-18 15:12:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b0923f781e18166d43f21b5fc09a4cc95365edf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:54:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4793af0cccd542ee889007cc277241f2 : SUCCESS in 1h 26m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/277e89ae71a9455598d1bf9397cb2519 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07f062bb9e23444c8748e6c16779d4ca : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac5cc425c92b4238ad238c28cf327341 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64232e7e3f8149f6a85ec6139432aae4 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/010e20904e2846bb98ec93b44bca1c56 : FAILURE in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cd3245cff21466fa36b5e5bb868bdab : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94e653ebf49141aa923913fd597a4546 : SUCCESS in 3m 36s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/11bbec8db9e443a1bf367115471f1a2b : SUCCESS in 6m 01s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/8dab56a558e545acbff1eee5066ab69a : FAILURE in 6m 24s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/4adc62db22684806b7b5f0cad07d5b7f : SUCCESS in 5m 25s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/fe370007d96f4299872b97965c3c5abc : FAILURE in 5m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/517b4ca291e74ee5a65fb05b25cecd4a : FAILURE in 1h 19m 35s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/dc141d7d953d450a8063a3d0ae7876a2 : SUCCESS in 1h 33m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/4144c820cb8a4c78b5d737ab39c9b87c : SUCCESS in 1h 31m 41s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/eccf286023e54e59b2bdb1342eff8421 : SUCCESS in 1h 03m 17s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d242b0b4e754d51b5e32378851d4d7a : SUCCESS in 1h 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c5a7fbb042f7a1c45e7f6cd73371fca6e378c540","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-19 08:41:49.000000000","message":"Patch Set 4:\n\nFYI I kept the previous PS active on a follow up patch to allow us to refine our use of the mandatory flag in a client mode.\n\nhttps://review.opendev.org/c/openstack/oslo.messaging/+/771232/","accounts_in_message":[],"_revision_number":4},{"id":"6eca39f508c3932ef53c3d194a03b87a03810c28","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-19 08:42:31.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0f170f6c7a73771bb64b694a764c3d9d8ceb3e18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 10:26:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b71f08a57e347f6be3fbe99342afaec : SUCCESS in 1h 27m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d7832d7029347b09cbe9331214debdb : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e6bfb0abbf4c31be4d6b0f20fb2909 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfa4ccc1ab854aedb23f3ef31e304fdd : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5aeaeeed0f3340ada8f9a249c92ca9c3 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74ea19de2546487080c0cb4217141fc9 : FAILURE in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee88ee7e6f324788be264fe99b28306b : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e54fe3ecba2457dbfe9c4939eb2a0bd : SUCCESS in 3m 51s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/ea903713834b4f51ab21e9138091ada4 : SUCCESS in 5m 29s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/2cd0c5ee66b74838a6fb3d255f08b324 : FAILURE in 6m 22s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/68fb79443f3349cba9b30590acf720f3 : SUCCESS in 7m 13s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/0ff90e3a5a1c4b33857acb9770763e03 : FAILURE in 6m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/31efdecf388f4c46b35781697fe2b7a4 : SUCCESS in 1h 40m 29s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1003db28a40c46c79b236ffc3f86c9ce : SUCCESS in 1h 33m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7cd83b87f8c047afa75e8fe81c9a2ddd : SUCCESS in 1h 35m 21s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f8422288acf447b5ae838dc87b7acc48 : SUCCESS in 1h 07m 48s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/8fa7c56cdaf0487f9121861178a6b420 : SUCCESS in 1h 05m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d30416365e1c9824cc022adaac65c553b5ea5330","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-22 11:07:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ced69e744a27a24b2e82da7957bff82ddf00e032","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 12:50:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/609c7d01f6ae4f4a9f4ee152f03d04c2 : SUCCESS in 1h 31m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1cbb59f94434397a517d0c20a683f56 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d83c955ebc449da3dd2c64d39d4065 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea03cd44fc20443bad39369389b5bced : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed6e4762b61a4ba4a5934c6fbd676a53 : SUCCESS in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e6c900662b34cb0bfc825586767bb71 : FAILURE in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92047d108d0d41c6acdae9c0236f803f : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/658c610112b54312a2467daee31a5c3f : SUCCESS in 4m 36s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/39913845573948b3829d7d347adcb1af : SUCCESS in 5m 28s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/5a4607b7d7bb4e669120d6f1d8a0341c : FAILURE in 6m 58s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/122568883ee3483fb0d11f33ced7db75 : SUCCESS in 5m 59s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/832221ce735945ed82f0b53709ae57d1 : FAILURE in 7m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c866107bb48f47b4a3208b70f1a33392 : SUCCESS in 1h 23m 17s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/49adaf61d751436f91962057cb3c1207 : SUCCESS in 1h 38m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/875dbd650e62424c9308aa0932733125 : SUCCESS in 1h 32m 34s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/127d7a6b7fba4638abf4dd5bcbb2e96d : SUCCESS in 1h 04m 28s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd4713aa6b3d43698180546eb81ddb51 : SUCCESS in 1h 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8de9978fe9fc372c112cc246695d287c40c00257","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-22 13:40:31.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"af85cd1e48dbeff32a502e865b2e07291e0e3085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 16:30:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e25aba49d21448b8991f067032ab34d3 : SUCCESS in 1h 15m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/317aa60eeef74df38e5517fa48651921 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7e82776e8e94591bbd42d9073f9437b : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba903c8d4f474806927c6e52a4214df8 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57ca55988fb84784a704af29048ab329 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adbe82ebb504444aa9320c4294abbc82 : FAILURE in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b36e2c82c046ea90c3b344f4d86ec5 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96f7598044c54ba09c8ce6e4b62663f5 : SUCCESS in 4m 27s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/30965b7d9b084e4d8c75487d2307c202 : SUCCESS in 6m 21s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/b9f1ec195b0d4ff48d68ad0275779e0b : FAILURE in 7m 39s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/dd382ca92053476190242ae7562ad11c : SUCCESS in 7m 30s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/59dd7a1439274c7db83f061ba54c8ea9 : FAILURE in 7m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/81d74f4f23b2475fb186c6f636dc3159 : SUCCESS in 1h 31m 42s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d66f4394e9d0494ab5fbff17220ad17a : SUCCESS in 1h 28m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/de7063034db6472691074f63e48f4c21 : FAILURE in 2h 21m 51s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/98d76fd46ac74aa28f1d1157f6428433 : SUCCESS in 1h 06m 47s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/591211090a714e6dbc06c7ff17c21db2 : SUCCESS in 1h 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"665c795aa97aa05c11f1444d925909cf948fe8aa","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2021-01-25 16:31:20.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nAside from the in-line suggestion this patch LGTM.","accounts_in_message":[],"_revision_number":6},{"id":"a7f317cadf2a2e2b1cfd0f021b381388ddec7cfc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-25 19:49:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0022bbf338ec5f79ed5691637ea3c68bab78c02f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-25 20:04:38.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a54b901e333b72ba8fe4c20b2d02eaac2e9916ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 21:50:35.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/497d3e66f2554aa483a46cb44f90aecd : SUCCESS in 1h 23m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdd1924af67d49ceb14a53fe5a89d32d : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f675ccd4567043a7b00d4c865d559f75 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bbee0ebf8a74c40801e6746940b2fe1 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b29db63edac24ff397f24d14b2c90f88 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e1db92461894f50bc5fe2bc5931d210 : FAILURE in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58781b35856c407cb1d4c47037d344c4 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7000fb017ee84461ae3c3a384b1b962d : SUCCESS in 4m 13s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/97b74181734d43f9a25ce97386786b4d : SUCCESS in 6m 15s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/83461df8c173455989036a1b61e22490 : FAILURE in 7m 35s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/3cc4c28b71ce4481b95463fe7700d61b : SUCCESS in 6m 38s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/7cb69b6de2f44db881bac77e650a8d26 : FAILURE in 7m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d2abe0aa6468479bbecd22d25ee75885 : FAILURE in 1h 40m 00s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/cf0e7d787ebb45ad9ff00871d2fcae83 : SUCCESS in 1h 31m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c3d3baaa8c194ee48d85b7f257777f03 : SUCCESS in 1h 34m 52s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/791620aa17c4430e865138f85c70fcc1 : SUCCESS in 1h 16m 24s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5937d9ffbdf443b93d402377625c2e3 : SUCCESS in 1h 21m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"aeb2d311092b4cf9061f04549a9b4f4e1b96cf9f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-27 08:59:09.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"4fff15fc0dd63c75042c893d70a55cc4fb2fc782","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 10:42:44.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b61b616d7c845c6a529aaa7df213f2b : SUCCESS in 1h 33m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1650166638a479780a27db93ea43a30 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad43aa46b6b5481c98314716f76192d2 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0c0aa9b66bc4251a5d64b4ad0208338 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7048c786309243a48920f8902ff723d5 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46f85b9a224a4267ae81ab39285d7332 : FAILURE in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e3c40bf335842518b0a3d89eefbe90a : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83368f8cf6c74036b412a8f7db98a38c : SUCCESS in 4m 39s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/6d2cc48fa9ae482880cdfbf39fd0ce77 : SUCCESS in 6m 23s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/09bad052ae3c4f2c9da5d26d73640b7d : FAILURE in 6m 08s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/2d8a4b0f04e24540960a0a33d67ab5e4 : SUCCESS in 6m 51s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/d1d01eb033424c6dad83c790c8117b1c : FAILURE in 5m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2a452325ad1d4a5ca86be91926fdfd9b : SUCCESS in 1h 37m 29s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/fbcaab6955eb4bc09c1c4479a39875f4 : SUCCESS in 1h 25m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a3e182f125ea473e8f764a989765dbc7 : SUCCESS in 1h 39m 02s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/26699808519346b49d8201062257501b : SUCCESS in 54m 04s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/a14005427fa848edaf5d3f4772ad1091 : SUCCESS in 1h 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e9b174913f5429401715a9aa9b5550049acefcdf","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-01-27 16:25:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bc08b00e6797ec5f03c0d88628157628d56e1aa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-01 09:31:45.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"3bd0fad7872ead33a9f4cc1139fbe79a6e03cdb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 11:13:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42480d0c507a44a5b4a2eede91c053d0 : SUCCESS in 1h 36m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12ed4d7e9e49468aa501e433fb173b21 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40b8b03e233e414da7098926d414be89 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd2523dd101c48fcaffbda58bdd6dbf2 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fe01991ecae4f32bef84f0371415741 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d007f82c75254e58aa80e6226f182ed5 : FAILURE in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c13c1c674cf494a903d428b9b8f105b : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1586d147cd54b69b0cb3d32928630d3 : SUCCESS in 4m 19s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/7b559a2972f7423d904036d0722d0250 : SUCCESS in 5m 36s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/870daebfc18f4e9588980e5ad41ef787 : FAILURE in 6m 30s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/7faa546be90b4fbd94b8c2fe9f0658d8 : SUCCESS in 6m 23s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/c782b91530fc45ddb63da9729b0fe71a : FAILURE in 6m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/da6e16881b8143d89d015a04fee16dcb : SUCCESS in 1h 26m 09s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/9b3e2dcedb7d42bcb70ca03d62423219 : SUCCESS in 1h 27m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9ec98be3ada8459cb219c886efd63750 : SUCCESS in 1h 31m 35s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/75070f3865d14938aea39bb196a8401b : SUCCESS in 1h 07m 17s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc16d1ca82c94a6789f8021a25060c8b : SUCCESS in 1h 22m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1092be46cd1606313ac503ce56534e9be098f958","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2021-02-01 14:14:57.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"3d3a0df94cd7bfbf4b863ba18319a28cc2ce0977","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-02-01 15:09:13.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"de7ba58558f34186c6cfb80665794d3563acad67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 15:09:47.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6f7c09375aa2bb742d6632d1ba508ca40644ec67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 16:47:18.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe16679371f34ab2aa26998e8ba5e485 : SUCCESS in 1h 32m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9bd5683390f4c84ac7409f3133ee1ae : SUCCESS in 10m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/047d5f87adcc4a14b511ff113a0774ef : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04f981ef9ccc46b5b0f01855e9ea93c7 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5540d5b0c7d40919ffe0037aee52826 : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/995df24e5f2246ad9cb2ddc046462df8 : SUCCESS in 4m 36s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/2157e496095d41d99bb0a71f8767000a : SUCCESS in 5m 17s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8218b6a33b6a40de9656e4a2486cae11 : SUCCESS in 1h 18m 53s","accounts_in_message":[],"_revision_number":8},{"id":"1eb57dd4bd294619422be545636f8f844f67270b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 16:47:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"63e1ececb34970e60768a678ee388dedfce3fc56","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 16:49:12.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73da8c380a2e45818265144f3c180d03 : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b397c82b562046ca85c3b69bca3de394 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"2b89d97888c9b6cd8b4f23e5ae7ce3103d5f9e5b","revisions":{"b61a4306934ed0bc0f0ea2b8af07f95e8635638d":{"kind":"REWORK","_number":1,"created":"2020-11-30 15:41:51.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/764776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/76/764776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/1"}}},"commit":{"parents":[{"commit":"0c4203bbdc463010d756a71bbd06720867122bcb","subject":"Python 3.9: isAlive is removed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0c4203bbdc463010d756a71bbd06720867122bcb"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-30 15:25:47.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-30 15:25:47.000000000","tz":60},"subject":"Don\u0027t pass the mandatory flag with the direct RPC","message":"Don\u0027t pass the mandatory flag with the direct RPC\n\nThe direct reply-to feature allows RPC clients to receive replies\ndirectly from their RPC server, without going through a reply queue.\n\"Directly\" here still means going through the same connection and a\nRabbitMQ node; there is no direct network connection between RPC client\nand RPC server processes.\n\nThe RPC server will then see a reply-to property with a generated name.\nIt should publish to the default exchange (\"\") with the routing key set\nto this value (i.e. just as if it were sending to a reply queue as\nusual). The message will then be sent straight to the client consumer.\n\nHowever this feature have some caveats and limitations [5]...\n\nEspecially the fact that the name `amq.rabbitmq.reply-to` is used in\n`basic.consume` and the `reply-to` property as if it were a queue;\nhowever it is not. It cannot be deleted, and does not appear in the\nmanagement plugin or `rabbitmqctl list_queues`.\n\nIf the RPC server publishes with the mandatory flag set then\n`amq.rabbitmq.reply-to.*` is treated as not a queue [5]; i.e. if the\nserver only publishes to this name then the message will be considered\n\"not routed\" [5][6]; a `basic.return` will be sent if the mandatory flag\nwas set.\n\nExample with nova. nova-compute\u0027s RPC server by replying by sending a direct\n`reply-to` to the client saw these message non routed.\nAfter awhile the client reached the timeout as the\nmessage was never delivered to him, and then this had for side effect to\nleft a block_device_mapping record in the database, leaving it listed in\nthe os-volume_attachments API and `nova volume-attachments` output. This\nalso causes the subsequent attempts to attach the volume to fail as we\nalready have an active bdm record for it.\n\nHowever even without this patch this bug could be easily solved by\ndisabling the `direct_mandatory_flag` option, it could be a simple\nworkaround to unlock those who face a similar issue.\n\nIf the customers/operators seeing this repeatedly in their environment\nthen they could try to disable the\n`[oslo_messaging_rabbit]/direct_mandatory_flag` option in the computes\n`nova.conf`.\n\nChange-Id: I1cbafff03349f7da9224de46285707fbf2a81a68\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b61a4306934ed0bc0f0ea2b8af07f95e8635638d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b61a4306934ed0bc0f0ea2b8af07f95e8635638d"}]},"branch":"refs/heads/master"},"30c47d030a28b5dab366da40abd576279e470892":{"kind":"REWORK","_number":2,"created":"2021-01-11 09:43:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/764776/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/76/764776/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/2"}}},"commit":{"parents":[{"commit":"7f823b406ea289101694e920c724b63f35b4fe64","subject":"Merge \"Fix type of direct_mandatory_flag opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f823b406ea289101694e920c724b63f35b4fe64"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-11 09:43:23.000000000","tz":60},"subject":"[WIP] Allow to loop for timeout when a MessageUndeliverable is raised","message":"[WIP] Allow to loop for timeout when a MessageUndeliverable is raised\n\nMessageUnderliverable exception have been introduced by recently added\nmandatory flag feature, however clients aren\u0027t prepared to handle\nrecover on their side and so it doesn\u0027t lets a chance to the requester to\nrestablishes itself.\n\nThese changes leave the mandatory flag feature as it was implemented\nhowever they add the possibility to also define `at_least_once` in\nconfiguration to allow to loop for a timeout while a message remains\nunderliverable, as did with AMQPDestinationNotFound exception.\n\nIf the client explicitly set at_least_once\u003dTrue, then when\nMessageUndeliverable is raised, we re-raise it back to the\nclient, with the assumption that it knows how to handle it.\n\nIf the client did not explicitly ask for it (the current case where nova\nand such do not use this API), then we can still use the\nfunctionality (the mandatory flag) within oslo.messaging, but we catch\nit and not re-raise it, the same way AMQPDestinationNotFound works currently.\n\nChange-Id: I1cbafff03349f7da9224de46285707fbf2a81a68\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/30c47d030a28b5dab366da40abd576279e470892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/30c47d030a28b5dab366da40abd576279e470892"}]},"branch":"refs/heads/master"},"146f5ba5e931e03e9d1d5d2d9dea45847a56eb7e":{"kind":"REWORK","_number":3,"created":"2021-01-11 14:16:54.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/764776/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/76/764776/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/3"}}},"commit":{"parents":[{"commit":"7f823b406ea289101694e920c724b63f35b4fe64","subject":"Merge \"Fix type of direct_mandatory_flag opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f823b406ea289101694e920c724b63f35b4fe64"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-11 14:16:11.000000000","tz":60},"subject":"[WIP] Allow to loop for timeout when a MessageUndeliverable is caught","message":"[WIP] Allow to loop for timeout when a MessageUndeliverable is caught\n\nMessageUnderliverable exception have been introduced by recently added\nmandatory flag feature, however clients aren\u0027t prepared to handle\nrecover on their side and so it doesn\u0027t lets a chance to the requester to\nrestablishes itself.\n\nThese changes leave the mandatory flag feature as it was implemented\nhowever they add the possibility to also define `at_least_once` in\nconfiguration to allow to loop for a timeout while a message remains\nunderliverable, as did with AMQPDestinationNotFound exception.\n\nIf the client explicitly set at_least_once\u003dTrue, then when\nMessageUndeliverable is raised, we re-raise it back to the\nclient, with the assumption that it knows how to handle it.\n\nIf the client did not explicitly ask for it (the current case where nova\nand such do not use this API), then we can still use the\nfunctionality (the mandatory flag) within oslo.messaging, but we catch\nit and not re-raise it, the same way AMQPDestinationNotFound works currently.\n\nChange-Id: I1cbafff03349f7da9224de46285707fbf2a81a68\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/146f5ba5e931e03e9d1d5d2d9dea45847a56eb7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/146f5ba5e931e03e9d1d5d2d9dea45847a56eb7e"}]},"branch":"refs/heads/master"},"46a971bda31444faa8ec2dfffda61200bc1c1eb7":{"kind":"REWORK","_number":4,"created":"2021-01-18 15:12:15.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/764776/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/76/764776/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/4"}}},"commit":{"parents":[{"commit":"7f823b406ea289101694e920c724b63f35b4fe64","subject":"Merge \"Fix type of direct_mandatory_flag opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f823b406ea289101694e920c724b63f35b4fe64"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-18 13:14:21.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-18 15:11:53.000000000","tz":60},"subject":"Disabling the mandatory flag by default","message":"Disabling the mandatory flag by default\n\nThis feature was mostly introduced to have a way for the RPC _client_ to\nfail fast when the RPC server is not present.\n\nWe shouldn\u0027t use the mandatory flag with the RPC server.\n\nAlso we didn\u0027t make enough noise related to this feature and to the\nfact that we want to fail early and so services aren\u0027t ready to\nhandle this case and on oslo.messaging we didn\u0027t bring a way to\nlets the time to redeclar exchanges properly. That lead us to a lot\nof inconsistent state and unexpected behaviors.\n\nThe goal is to disabling this feature for a while to let\u0027s us time to\ncatch/handle this case more properly with follow up patches.\n\nChange-Id: I1cbafff03349f7da9224de46285707fbf2a81a68\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/46a971bda31444faa8ec2dfffda61200bc1c1eb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/46a971bda31444faa8ec2dfffda61200bc1c1eb7"}]},"branch":"refs/heads/master"},"af47f18e82aba43b256074a7d49d482b5e295b80":{"kind":"REWORK","_number":5,"created":"2021-01-22 11:07:28.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/764776/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/76/764776/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/5"}}},"commit":{"parents":[{"commit":"7f823b406ea289101694e920c724b63f35b4fe64","subject":"Merge \"Fix type of direct_mandatory_flag opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f823b406ea289101694e920c724b63f35b4fe64"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-18 13:14:21.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-22 11:05:37.000000000","tz":60},"subject":"Deprecate the mandatory flag","message":"Deprecate the mandatory flag\n\nIt will not be possible to deactivate this fonctionality anymore.\n\nChange-Id: I1cbafff03349f7da9224de46285707fbf2a81a68\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/af47f18e82aba43b256074a7d49d482b5e295b80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/af47f18e82aba43b256074a7d49d482b5e295b80"}]},"branch":"refs/heads/master"},"2a3b3f8975526ecd6b141b9367f3ee7fc29db90b":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-01-22 13:40:31.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/764776/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/76/764776/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/6"}}},"commit":{"parents":[{"commit":"7f823b406ea289101694e920c724b63f35b4fe64","subject":"Merge \"Fix type of direct_mandatory_flag opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f823b406ea289101694e920c724b63f35b4fe64"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-18 13:14:21.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-22 13:39:56.000000000","tz":60},"subject":"Deprecate the mandatory flag","message":"Deprecate the mandatory flag\n\nIt will not be possible to deactivate this fonctionality anymore.\n\nChange-Id: I1cbafff03349f7da9224de46285707fbf2a81a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2a3b3f8975526ecd6b141b9367f3ee7fc29db90b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2a3b3f8975526ecd6b141b9367f3ee7fc29db90b"}]},"branch":"refs/heads/master"},"5d3387181e0f11d205ffec9b39046be12ccebc18":{"kind":"REWORK","_number":7,"created":"2021-01-25 20:04:38.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/764776/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/76/764776/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/7"}}},"commit":{"parents":[{"commit":"7f823b406ea289101694e920c724b63f35b4fe64","subject":"Merge \"Fix type of direct_mandatory_flag opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f823b406ea289101694e920c724b63f35b4fe64"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-18 13:14:21.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-25 20:02:41.000000000","tz":60},"subject":"Deprecate the mandatory flag","message":"Deprecate the mandatory flag\n\nIt will not be possible to deactivate this fonctionality anymore.\n\nChange-Id: I1cbafff03349f7da9224de46285707fbf2a81a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5d3387181e0f11d205ffec9b39046be12ccebc18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5d3387181e0f11d205ffec9b39046be12ccebc18"}]},"branch":"refs/heads/master"},"2b89d97888c9b6cd8b4f23e5ae7ce3103d5f9e5b":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-02-01 09:31:45.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/764776/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/76/764776/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/76/764776/8"}}},"commit":{"parents":[{"commit":"7f823b406ea289101694e920c724b63f35b4fe64","subject":"Merge \"Fix type of direct_mandatory_flag opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f823b406ea289101694e920c724b63f35b4fe64"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-18 13:14:21.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-01 09:31:16.000000000","tz":60},"subject":"Deprecate the mandatory flag","message":"Deprecate the mandatory flag\n\nIt will not be possible to deactivate this functionality anymore.\n\nChange-Id: I1cbafff03349f7da9224de46285707fbf2a81a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2b89d97888c9b6cd8b4f23e5ae7ce3103d5f9e5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2b89d97888c9b6cd8b4f23e5ae7ce3103d5f9e5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
